)]}'
{"id":"openstack%2Fpuppet-swift~942046","triplet_id":"openstack%2Fpuppet-swift~master~Ibab11aff6c0fede46d53fc36c93238c6138d45e0","project":"openstack/puppet-swift","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-02-20 16:57:46.000000000","reason":"Change was submitted"},"14985":{"account":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"last_update":"2025-02-20 15:15:58.000000000","reason":"removed on reply"},"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2025-02-20 07:52:30.000000000","reason":"\u003cGERRIT_ACCOUNT_16137\u003e replied on the change","reason_account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}},"hashtags":[],"change_id":"Ibab11aff6c0fede46d53fc36c93238c6138d45e0","subject":"Deprecate support for [oslo_messaging_rabbit] heartbeat_in_pthread","status":"MERGED","created":"2025-02-18 02:47:30.000000000","updated":"2025-02-20 16:58:41.000000000","submitted":"2025-02-20 16:57:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"942046","meta_rev_id":"7b3b8ef8e7c5761e446f65880c4bfdc214fde18c","_number":942046,"virtual_id_number":942046,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-20 16:57:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2025-02-20 07:52:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2025-02-20 15:15:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2025-02-20 15:15:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-18 04:00:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-18 15:08:47.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2025-02-19 14:18:04.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"15989ffb91090f0fec6224098095e7b33fa06558","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-02-18 02:47:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2edf90083f5576201fe0995fdd46c0c7cb817ab4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-02-18 02:58:28.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"0a9769bf70725eba9494aad04725f8859ba5dca2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-18 04:00:56.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ca779a29dd814b82a8311b204f7eafff\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d98d81feceb241708134ff0dbec65b67 : FAILURE in 3m 31s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/390887c53b174df89d3cb0d9a21957fe : FAILURE in 2m 47s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/1c1e42f49993406d9985120e99d0337d : FAILURE in 6m 19s (non-voting)\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/fdd90ac612714aa687dc064c592ea61e : FAILURE in 11m 22s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d665aad1216343088e01f60b57c3830e : FAILURE in 5m 02s\n- puppet-openstack-unit-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/a343a141f54c4dcb8febf0720f484380 : FAILURE in 11m 45s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6dfc959d74dc4d5386a5ce5de19d8867 : FAILURE in 10m 31s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/455d05e9e3be43aba01f1f14d34a1e78 : FAILURE in 5m 44s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/091617056da447f4ad2ce2c3a78f10a1 : FAILURE in 9m 56s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/685cc183499f4a259937a1649880b5b5 : FAILURE in 10m 38s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/ead63882956f40c8932ccdf5b6315750 : SUCCESS in 41m 47s\n- puppet-openstack-litmus-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ca1b69ea93264ba0ba3655b22a3d6ed6 : SUCCESS in 16m 28s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/3888f4e47c5b462a9d15b706f9ad7416 : SUCCESS in 45m 55s\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6702addd0dc74c3e8fea08e127a21682 : SUCCESS in 14m 11s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4517b98b329da1dde24690f177fca6142eea6985","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-02-18 10:22:28.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"13e6787149c486553ddccd700015c694ca3bbb22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-18 12:20:01.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b4e50ca24cfc4800bea43517ad8ec10c\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4ee2939624a4410e8155d2585dc03a7d : SUCCESS in 2m 11s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9ea9372c51b84a1db5c45ce81bae892d : SUCCESS in 1m 59s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/3e04a4386bd943358fe72eaf5268f4d7 : SUCCESS in 3m 39s (non-voting)\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/ae2f01e98cef43b1877843176d16f6eb : SUCCESS in 7m 16s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e9ee00d717cb47909209a25dc8da8642 : SUCCESS in 5m 29s\n- puppet-openstack-unit-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/322b5afa882d4908921354c8368c0df1 : SUCCESS in 7m 12s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a36bd715b1eb40c2abfc49e492578bb7 : SUCCESS in 4m 47s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6f48079b6f04482cb3e0b0a1f9c2dfef : SUCCESS in 57m 02s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/749483444185413fa998d963295e320c : SUCCESS in 1h 49m 59s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ee6627354ffb49dabc11c47d230b8c41 : SUCCESS in 1h 20m 56s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/51daa2b3434a47a684c8484604994f7b : SUCCESS in 33m 50s\n- puppet-openstack-litmus-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5604bc2803eb41c18a3cea9a782934db : SUCCESS in 15m 01s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/92df7472f74a40fe983bb78e5b584678 : SUCCESS in 38m 45s\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/323ee7c5558943ac930fc38722faf758 : SUCCESS in 18m 58s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"71f70cf5224b72945872a94acb8b12e4bc7f8a08","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2025-02-18 15:08:47.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"cd38978d491ad198a0383d710b1ab8843fcb66e0","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-02-19 14:18:04.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"25cd43f6546307852673bf1d6174560cebefc685","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-02-19 15:00:38.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"12d9194f0e88154066a0b760be4067f8f52eca9f","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2025-02-19 15:05:13.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"62911b420c728a854ac26be13341fdd1f1697ebd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-02-19 15:06:20.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":5},{"id":"83154ec15fd7917d882bb317eb9c4a8580cdf53a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-02-19 15:06:29.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"12d0a2ed19de88f874f1c867f48e7a4c477e21d5","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-02-19 15:06:42.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"9e2d605090b4b777195899d9859ca3176f6a5675","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-19 16:55:13.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b019677a00674352b63a7e412a26e7e3\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/544f1d96e0c746d984ec476e85886fe4 : SUCCESS in 2m 26s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b778299967bc425da30846e0507f5509 : SUCCESS in 2m 46s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/8c693ea688514930b2017eddbb8ac660 : SUCCESS in 5m 59s (non-voting)\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/7bac72990981439f9779bd11b23d0dc3 : SUCCESS in 8m 24s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4c88ef66e5e246f0a3f33afa6b7af594 : SUCCESS in 8m 08s\n- puppet-openstack-unit-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/9dab097fa024459086dbb1bef1e2caea : SUCCESS in 11m 42s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/52ca2bc1314947b293be9cc7da691b27 : SUCCESS in 8m 52s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/265d686ee5e24a488c929da73689229e : FAILURE in 1h 09m 54s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/50438e68055140d2a8c3733e647dcaec : SUCCESS in 1h 15m 19s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/162c14914c6544fd95028ee59fcb8f8e : SUCCESS in 1h 45m 44s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/f685f6ae91044794858530979214b496 : SUCCESS in 22m 54s\n- puppet-openstack-litmus-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/52208ea0fc61486fb850f1c61c9d7780 : SUCCESS in 16m 33s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/96610563360744a6b81839a5843f53d6 : SUCCESS in 29m 20s\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0450610f440f47eeae4e6d3bef20e1da : SUCCESS in 10m 09s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6370431b92914dc286427681c6a55ad2 : SUCCESS in 2m 46s","accounts_in_message":[],"_revision_number":5},{"id":"25f248d0a673c973310af40d169439259eebe2fb","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-02-19 22:06:06.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"0836494daa7b6f8ae0d603c452968f4e56595607","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-02-20 00:55:35.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"2a3c0bdcf7b5f07bbbf0653b45daae28c4f5f173","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 02:33:13.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd172a4159bb4b1ebd47915d8b6eca1c\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/87be3358d71d45109beac5a3abd34ccf : SUCCESS in 2m 46s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6bf42a21d6864eb1abcf8df79bad9081 : SUCCESS in 2m 45s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/c2a4a5da4ac44e549af374fc054c0d8c : SUCCESS in 4m 48s (non-voting)\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/55489c4c1f5247b997aa7e2f7f10d2a6 : SUCCESS in 10m 12s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5ad2d37280914781ab32ca7bffab2dda : SUCCESS in 8m 16s\n- puppet-openstack-unit-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/ac326fefcaf44fd5a73fd75a8de5d0bd : SUCCESS in 6m 33s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4ebd7e12873a4adb8ec5fa7664ff19f0 : SUCCESS in 9m 11s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4491a3173d1741d7a326c50e2066619b : SUCCESS in 58m 39s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d8ac1fd5e8e944feaaef9494df8e631b : SUCCESS in 1h 29m 29s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a234f1c1aa894b5ab5327c230c349f0b : SUCCESS in 1h 15m 40s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/c70fb917e717491d9cb7f31f8e6f81a4 : SUCCESS in 30m 30s\n- puppet-openstack-litmus-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/85430070264842e5acf7875f3f78e214 : SUCCESS in 8m 19s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/77a82acf14ce4983a0f3e2cfaedecda5 : SUCCESS in 38m 32s\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a90f5929eb1a43a7a47d6997db3b89c0 : SUCCESS in 7m 44s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9df58179492040978eb8069d2649df3f : SUCCESS in 2m 26s","accounts_in_message":[],"_revision_number":6},{"id":"163c23925d05adbcb2e6cc3996af487c1a787efb","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-02-20 07:52:30.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"9a0fed74f196e48739174d8fa5317be03ad02362","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2025-02-20 15:15:58.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"be87a36d140c8a9b22fcb569d61df6deb9d419d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 15:17:25.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"1381aa11ffb9087705a4a863101b7b0f2d369aa1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 16:57:46.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/93b4c60df0f84057b74dc4db47fd3e38\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3bf7e438cd404630b022553820208b36 : SUCCESS in 2m 39s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/11a0e1c495744e838edd9cad5e1176c1 : SUCCESS in 5m 41s\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/56aaf0029fba41328a9bf6008711963c : SUCCESS in 17m 32s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ef4a506a346a424c888c5e324590df14 : SUCCESS in 8m 09s\n- puppet-openstack-unit-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/333bc078249349b5ac52bcdf8e18d415 : SUCCESS in 11m 39s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/30f0837aa822417b93a216d49c15a583 : SUCCESS in 1h 39m 47s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/f6286e5073a14edf868b02347d2cee1a : SUCCESS in 36m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b4c3a7a06a014845baf485b2045c877f : SUCCESS in 2m 03s","accounts_in_message":[],"_revision_number":6},{"id":"325d98145074663c9cef26a5753cccc1268241d5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 16:57:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"7b3b8ef8e7c5761e446f65880c4bfdc214fde18c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 16:58:41.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e00f8ec01d124b769b0762f8f5846f98\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6379060828c14586a9150a3b76abe43c : SUCCESS in 36s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"88d364c8850ab796cbac2fa7ecfcdbabe217888a","revisions":{"7f52af4553f550c403b60768a6fb62d9e4084f85":{"kind":"REWORK","_number":1,"created":"2025-02-18 02:47:30.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/46/942046/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/46/942046/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/46/942046/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/46/942046/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/46/942046/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/46/942046/1"}}},"commit":{"parents":[{"commit":"b7fbe8740c001ed30eebfef9685e733f5ff260ac","subject":"Merge \"Mark puppet 8 supported\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/b7fbe8740c001ed30eebfef9685e733f5ff260ac"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-02-18 02:47:08.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-02-18 02:47:08.000000000","tz":540},"subject":"Deprecate support for rabbit_heartbeat_in_pthread","message":"Deprecate support for rabbit_heartbeat_in_pthread\n\n... because the option has been deprecated.\n\nNote that the deprecation warning may be added in puppet-oslo so is not\nimplemented in individual module.\n\nDepends-on: https://review.opendev.org/925778\nChange-Id: Ibab11aff6c0fede46d53fc36c93238c6138d45e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/7f52af4553f550c403b60768a6fb62d9e4084f85"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/7f52af4553f550c403b60768a6fb62d9e4084f85"}]},"branch":"refs/heads/master"},"a553888c2a2261916ddf8426c4647a52717dd983":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-02-18 02:58:28.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/46/942046/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/46/942046/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/46/942046/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/46/942046/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/46/942046/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/46/942046/2"}}},"commit":{"parents":[{"commit":"b7fbe8740c001ed30eebfef9685e733f5ff260ac","subject":"Merge \"Mark puppet 8 supported\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/b7fbe8740c001ed30eebfef9685e733f5ff260ac"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-02-18 02:47:08.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-02-18 02:58:16.000000000","tz":540},"subject":"Deprecate support for [oslo_messaging_rabbit] heartbeat_in_pthread","message":"Deprecate support for [oslo_messaging_rabbit] heartbeat_in_pthread\n\n... because the option has been deprecated.\n\nNote that the deprecation warning may be added in puppet-oslo so is not\nimplemented in individual module.\n\nDepends-on: https://review.opendev.org/925778\nChange-Id: Ibab11aff6c0fede46d53fc36c93238c6138d45e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/a553888c2a2261916ddf8426c4647a52717dd983"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/a553888c2a2261916ddf8426c4647a52717dd983"}]},"branch":"refs/heads/master"},"3a070246166a3844aa0fa96fc148cb11385a1605":{"kind":"REWORK","_number":3,"created":"2025-02-18 10:22:28.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/46/942046/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/46/942046/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/46/942046/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/46/942046/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/46/942046/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/46/942046/3"}}},"commit":{"parents":[{"commit":"b7fbe8740c001ed30eebfef9685e733f5ff260ac","subject":"Merge \"Mark puppet 8 supported\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/b7fbe8740c001ed30eebfef9685e733f5ff260ac"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-02-18 02:47:08.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-02-18 10:22:28.000000000","tz":0},"subject":"Deprecate support for [oslo_messaging_rabbit] heartbeat_in_pthread","message":"Deprecate support for [oslo_messaging_rabbit] heartbeat_in_pthread\n\n... because the option has been deprecated.\n\nNote that the deprecation warning may be added in puppet-oslo so is not\nimplemented in individual module.\n\nDepends-on: https://review.opendev.org/925778\nChange-Id: Ibab11aff6c0fede46d53fc36c93238c6138d45e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/3a070246166a3844aa0fa96fc148cb11385a1605"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/3a070246166a3844aa0fa96fc148cb11385a1605"}]},"branch":"refs/heads/master"},"40b8341be78a1c58ddefb94bde6972ead847af06":{"kind":"REWORK","_number":4,"created":"2025-02-19 15:00:38.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/46/942046/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/46/942046/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/46/942046/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/46/942046/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/46/942046/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/46/942046/4"}}},"commit":{"parents":[{"commit":"b7fbe8740c001ed30eebfef9685e733f5ff260ac","subject":"Merge \"Mark puppet 8 supported\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/b7fbe8740c001ed30eebfef9685e733f5ff260ac"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-02-18 02:47:08.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-02-19 15:00:29.000000000","tz":540},"subject":"Deprecate support for [oslo_messaging_rabbit] heartbeat_in_pthread","message":"Deprecate support for [oslo_messaging_rabbit] heartbeat_in_pthread\n\n... because the option has been deprecated.\n\nNote that the deprecation warning may be added in puppet-oslo so is not\nimplemented in individual module.\n\nDepends-on: https://review.opendev.org/925778\nChange-Id: Ibab11aff6c0fede46d53fc36c93238c6138d45e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/40b8341be78a1c58ddefb94bde6972ead847af06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/40b8341be78a1c58ddefb94bde6972ead847af06"}]},"branch":"refs/heads/master"},"af51a7fcd3f3534547f72661d1bc931cc74d59b6":{"kind":"REWORK","_number":5,"created":"2025-02-19 15:06:20.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/46/942046/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/46/942046/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/46/942046/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/46/942046/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/46/942046/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/46/942046/5"}}},"commit":{"parents":[{"commit":"b7fbe8740c001ed30eebfef9685e733f5ff260ac","subject":"Merge \"Mark puppet 8 supported\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/b7fbe8740c001ed30eebfef9685e733f5ff260ac"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-02-18 02:47:08.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-02-19 15:06:11.000000000","tz":540},"subject":"Deprecate support for [oslo_messaging_rabbit] heartbeat_in_pthread","message":"Deprecate support for [oslo_messaging_rabbit] heartbeat_in_pthread\n\n... because the option has been deprecated.\n\nNote that the deprecation warning may be added in puppet-oslo so is not\nimplemented in individual module.\n\nDepends-on: https://review.opendev.org/925778\nChange-Id: Ibab11aff6c0fede46d53fc36c93238c6138d45e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/af51a7fcd3f3534547f72661d1bc931cc74d59b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/af51a7fcd3f3534547f72661d1bc931cc74d59b6"}]},"branch":"refs/heads/master"},"88d364c8850ab796cbac2fa7ecfcdbabe217888a":{"kind":"REWORK","_number":6,"created":"2025-02-20 00:55:35.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/46/942046/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/46/942046/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/46/942046/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/46/942046/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/46/942046/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/46/942046/6"}}},"commit":{"parents":[{"commit":"b7fbe8740c001ed30eebfef9685e733f5ff260ac","subject":"Merge \"Mark puppet 8 supported\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/b7fbe8740c001ed30eebfef9685e733f5ff260ac"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-02-18 02:47:08.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-02-20 00:55:26.000000000","tz":540},"subject":"Deprecate support for [oslo_messaging_rabbit] heartbeat_in_pthread","message":"Deprecate support for [oslo_messaging_rabbit] heartbeat_in_pthread\n\n... because the option has been deprecated.\n\nNote that the deprecation warning may be added in puppet-oslo so is not\nimplemented in individual module.\n\nDepends-on: https://review.opendev.org/925778\nChange-Id: Ibab11aff6c0fede46d53fc36c93238c6138d45e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/88d364c8850ab796cbac2fa7ecfcdbabe217888a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/88d364c8850ab796cbac2fa7ecfcdbabe217888a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
