)]}'
{"id":"openstack%2Fpuppet-tacker~727004","triplet_id":"openstack%2Fpuppet-tacker~master~Ia329cfcbd6c4a1b92e52b9b5bd0b68eb99c2649c","project":"openstack/puppet-tacker","branch":"master","topic":"bump-reno","hashtags":[],"change_id":"Ia329cfcbd6c4a1b92e52b9b5bd0b68eb99c2649c","subject":"Bump reno to 2.5.0","status":"MERGED","created":"2020-05-12 02:51:07.000000000","updated":"2020-05-12 03:21:58.000000000","submitted":"2020-05-12 03:20:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"727004-1589253617582-2515f420","meta_rev_id":"28a77bdab85c4cd83929ed9836bb8efcaccb9a48","_number":727004,"virtual_id_number":727004,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-12 03:20:17.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2020-05-12 03:14:17.000000000","_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":1,"date":"2020-05-12 03:14:17.000000000","_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-12 03:20:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6c8b1f0f165dda4326471e3b0408edabee760b83","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-05-12 02:51:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d43dcb66357cef9a9f8ef0f25e30ca5dedd29586","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 02:55:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9ab49403dfa84d9bbbcef8981df2c7b4 : SUCCESS in 3m 18s","accounts_in_message":[],"_revision_number":1},{"id":"e0f6e1ac138344bd1aefc02431d8f4e31ed7f7f8","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-05-12 03:14:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d2561a3a952a8d5ac76b9dcce1dbdb4e353662e6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 03:14:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3303d0a18700d410b83736869e3e5e3db10d2309","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 03:20:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f7a4298ff3b7463da92fdbb5550e9b0b : SUCCESS in 4m 19s","accounts_in_message":[],"_revision_number":1},{"id":"bead374873cf324a6a1e2ed7d7daf698bb4fe967","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 03:20:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c745d36e77efab55d2ff4481d633d0617ad3b38c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 03:21:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4f5dd55383244cc5aa79c0ec49f9f880 : SUCCESS in 1m 23s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7578e709cb7c1579bf39e90990a8e053d15e1c98","revisions":{"7578e709cb7c1579bf39e90990a8e053d15e1c98":{"kind":"REWORK","_number":1,"created":"2020-05-12 02:51:07.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/04/727004/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tacker","ref":"refs/changes/04/727004/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tacker refs/changes/04/727004/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tacker refs/changes/04/727004/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tacker refs/changes/04/727004/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tacker refs/changes/04/727004/1"}}},"commit":{"parents":[{"commit":"3c00785d03688df9bfa7f0f02e5be674c650eece","subject":"Merge \"Add CentOS8 to nodesets\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tacker/commit/3c00785d03688df9bfa7f0f02e5be674c650eece"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-05-12 02:50:07.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-05-12 02:50:07.000000000","tz":480},"subject":"Bump reno to 2.5.0","message":"Bump reno to 2.5.0\n\nChange-Id: Ia329cfcbd6c4a1b92e52b9b5bd0b68eb99c2649c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tacker/commit/7578e709cb7c1579bf39e90990a8e053d15e1c98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tacker/commit/7578e709cb7c1579bf39e90990a8e053d15e1c98"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
