)]}'
{"id":"openstack%2Fpuppet-tacker~763691","triplet_id":"openstack%2Fpuppet-tacker~master~I3e075edf9e97995ff9d014f41bea72ad758c72e2","project":"openstack/puppet-tacker","branch":"master","topic":"bug/1904962","hashtags":[],"change_id":"I3e075edf9e97995ff9d014f41bea72ad758c72e2","subject":"Allow db sync timeouts to be configurable","status":"MERGED","created":"2020-11-23 01:39:40.000000000","updated":"2020-11-25 11:34:15.000000000","submitted":"2020-11-25 11:32:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"763691-bug/1904962","meta_rev_id":"6cc463fe1a69f409bde5c1eb5a35732ebebb31d6","_number":763691,"virtual_id_number":763691,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-25 11:32:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2020-11-25 11:06:26.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2020-11-24 16:02:44.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2020-11-25 11:06:26.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-23 02:01:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-24 16:02:44.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-11-25 10:57:26.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"4f8864f8089d816f4ec5aa8f5296a97c433d4bea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-11-23 01:39:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"03a3fa8fed5db36b524f8eb2504353f5533923b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-23 02:01:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/81a815fbe8b34836b596a4ea7296ee6a : SUCCESS in 4m 00s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/29182c2f5c2e47a38fae2ea10f47ce85 : SUCCESS in 3m 48s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/e05d6a95de1540ef8fd48d5f701a90de : SUCCESS in 11m 23s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0b52f1d5c81947619888814f10ce406d : SUCCESS in 5m 32s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/129bcbefb6bb4a4d85f95bc0058c3f0a : SUCCESS in 6m 03s (non-voting)\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/440ca04ebaed43b4a98cfc1776be61a7 : SUCCESS in 14m 51s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/197e90c60a9f44239295d4c736dce93d : FAILURE in 13m 03s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/23d171c1a4464e4ba7a70f3c2cabd9cd : SUCCESS in 3m 05s","accounts_in_message":[],"_revision_number":1},{"id":"09ad1c38de8fa2b899348d72677e2049d03fe5f5","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-11-24 16:02:44.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bdd944efd76514b7d2bc4c7ee8fbb572cdb7538f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-11-25 10:57:26.000000000","message":"Patch Set 1: Code-Review-1\n\nAdding -1 since my comment for another patch[1] is applied to this patch as well.\n\n[1] https://review.opendev.org/c/openstack/puppet-openstack-cookiecutter/+/763698","accounts_in_message":[],"_revision_number":1},{"id":"e1fb8b5c8f68232cf1b0fb8f2eaaee27e88b5707","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-11-25 11:06:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nI noticed that we already have the db_sync_timeout in puppet-nova and neutron.\nThen we can keep the current name to be consistent with these existing parameters...","accounts_in_message":[],"_revision_number":1},{"id":"2333c6bcf42b23f19f5ed126e7a4f953c2e2b3ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-25 11:06:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"def5598ec848e639597248d62de5574ad3e27235","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-25 11:32:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a83b9d7121fc4529b42e6ff960011101 : SUCCESS in 2m 56s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ce670f53d2874d619fc1f27363891ebe : SUCCESS in 3m 08s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/25f0ec2e31f7485ea840f46d57ee9657 : SUCCESS in 11m 31s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a02f44a40c854eaab605c53624830ec1 : SUCCESS in 8m 59s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/9c71de40dc304223af44e824cc9f84be : SUCCESS in 20m 29s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f97d9aa7f1e74a58b9d2aeb41f61c81e : SUCCESS in 11m 49s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b783cdb486ad40649333dd1f2a6fa26f : SUCCESS in 3m 23s","accounts_in_message":[],"_revision_number":1},{"id":"6b0aa10b7f17225295fcc76b45fc37835f3f3248","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-25 11:32:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"6cc463fe1a69f409bde5c1eb5a35732ebebb31d6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-25 11:34:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/47b7891a58d543ea8bfc2db394212274 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a90ce490158196726f21cf69d65e2a9a6eeb0089","revisions":{"a90ce490158196726f21cf69d65e2a9a6eeb0089":{"kind":"REWORK","_number":1,"created":"2020-11-23 01:39:40.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/91/763691/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tacker","ref":"refs/changes/91/763691/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tacker refs/changes/91/763691/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tacker refs/changes/91/763691/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tacker refs/changes/91/763691/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tacker refs/changes/91/763691/1"}}},"commit":{"parents":[{"commit":"bd04fe0a99b0ce34f7c7bae26e76e1c77fc4f89e","subject":"Merge \"Add support for the keystone_authtoken/service_type parameter\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tacker/commit/bd04fe0a99b0ce34f7c7bae26e76e1c77fc4f89e"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-11-23 01:27:52.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-11-23 01:27:52.000000000","tz":480},"subject":"Allow db sync timeouts to be configurable","message":"Allow db sync timeouts to be configurable\n\nAs Openstack projects continue to have longer database migration\nchains, the Puppet default timeout of 300 seconds for an execution\nis becoming too short a duration on some hardware, leading to timeouts.\nAs projects continue to add more migration scripts without pruning\nthe base, timeouts will continue to become more frequent unless\nthis time can be expanded.\n\nChange-Id: I3e075edf9e97995ff9d014f41bea72ad758c72e2\nCloses-Bug: #1904962\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tacker/commit/a90ce490158196726f21cf69d65e2a9a6eeb0089"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tacker/commit/a90ce490158196726f21cf69d65e2a9a6eeb0089"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
