)]}'
{"id":"openstack%2Fpuppet-tempest~250428","triplet_id":"openstack%2Fpuppet-tempest~master~I7c5671695787fb1660d26fde689ff891895980fa","project":"openstack/puppet-tempest","branch":"master","hashtags":[],"change_id":"I7c5671695787fb1660d26fde689ff891895980fa","subject":"Allow to activate Murano tests","status":"MERGED","created":"2015-11-26 15:04:07.000000000","updated":"2015-11-26 15:58:09.000000000","submitted":"2015-11-26 15:58:08.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"fc1aaa15f85882b1169409e126a3b35c0cbbb4ba","_number":250428,"virtual_id_number":250428,"owner":{"_account_id":13752,"name":"Alexey Deryugin","email":"aderiugin@mirantis.com","username":"velovec"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":7745,"name":"Denis Egorenko","email":"degorenko@mirantis.com","username":"degorenko"},{"value":2,"date":"2015-11-26 15:58:08.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2015-11-26 15:04:49.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2015-11-26 15:15:52.000000000","_account_id":7745,"name":"Denis Egorenko","email":"degorenko@mirantis.com","username":"degorenko"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7745,"name":"Denis Egorenko","email":"degorenko@mirantis.com","username":"degorenko"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2015-11-26 15:45:09.000000000","_account_id":7745,"name":"Denis Egorenko","email":"degorenko@mirantis.com","username":"degorenko"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":7745,"name":"Denis Egorenko","email":"degorenko@mirantis.com","username":"degorenko"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2015-11-26 15:04:49.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2015-11-26 15:45:09.000000000","updated_by":{"_account_id":7745,"name":"Denis Egorenko","email":"degorenko@mirantis.com","username":"degorenko"},"reviewer":{"_account_id":7745,"name":"Denis Egorenko","email":"degorenko@mirantis.com","username":"degorenko"},"state":"REVIEWER"},{"updated":"2015-11-26 15:58:08.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"9c8c9d4c42770cb0c7ece22a271542a0e2c32e45","author":{"_account_id":13752,"name":"Alexey Deryugin","email":"aderiugin@mirantis.com","username":"velovec"},"date":"2015-11-26 15:04:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9800e98908339ca9b4bf2c8f822e26dc6bab8977","author":{"_account_id":13752,"name":"Alexey Deryugin","email":"aderiugin@mirantis.com","username":"velovec"},"date":"2015-11-26 15:04:21.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":1},{"id":"ed58294a36f0308def137e9ca7c9e7d49a5d0453","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2015-11-26 15:04:49.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"40601f568714763b75ef498660876ee48222d87f","author":{"_account_id":7745,"name":"Denis Egorenko","email":"degorenko@mirantis.com","username":"degorenko"},"date":"2015-11-26 15:15:52.000000000","message":"Patch Set 2: Code-Review+2\n\nNice! Waiting for Ci.","accounts_in_message":[],"_revision_number":2},{"id":"9a0500f7fad7438f666116c9137969b11e0f1457","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-11-26 15:44:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-puppet-tempest-puppet-lint http://logs.openstack.org/28/250428/2/check/gate-puppet-tempest-puppet-lint/717d8dc/ : SUCCESS in 32s\n- gate-puppet-tempest-puppet-syntax-3-dsvm-centos7 http://logs.openstack.org/28/250428/2/check/gate-puppet-tempest-puppet-syntax-3-dsvm-centos7/991cbc2/ : SUCCESS in 4m 22s\n- gate-puppet-tempest-puppet-syntax-4-dsvm-centos7 http://logs.openstack.org/28/250428/2/check/gate-puppet-tempest-puppet-syntax-4-dsvm-centos7/178bf2a/ : SUCCESS in 3m 12s\n- gate-puppet-tempest-puppet-unit-3.3-dsvm-centos7 http://logs.openstack.org/28/250428/2/check/gate-puppet-tempest-puppet-unit-3.3-dsvm-centos7/f10117b/ : SUCCESS in 3m 51s\n- gate-puppet-tempest-puppet-unit-3.4-dsvm-centos7 http://logs.openstack.org/28/250428/2/check/gate-puppet-tempest-puppet-unit-3.4-dsvm-centos7/5a00a0f/ : SUCCESS in 5m 29s\n- gate-puppet-tempest-puppet-unit-3.6-dsvm-centos7 http://logs.openstack.org/28/250428/2/check/gate-puppet-tempest-puppet-unit-3.6-dsvm-centos7/c26c73f/ : SUCCESS in 5m 30s\n- gate-puppet-tempest-puppet-unit-3.7-dsvm-centos7 http://logs.openstack.org/28/250428/2/check/gate-puppet-tempest-puppet-unit-3.7-dsvm-centos7/9c27f1e/ : SUCCESS in 5m 31s\n- gate-puppet-tempest-puppet-unit-3.8-dsvm-centos7 http://logs.openstack.org/28/250428/2/check/gate-puppet-tempest-puppet-unit-3.8-dsvm-centos7/aa7e995/ : SUCCESS in 5m 40s\n- gate-puppet-tempest-puppet-unit-4.0-dsvm-centos7 http://logs.openstack.org/28/250428/2/check/gate-puppet-tempest-puppet-unit-4.0-dsvm-centos7/07a162e/ : SUCCESS in 4m 07s\n- gate-puppet-tempest-puppet-unit-latest-dsvm-centos7 http://logs.openstack.org/28/250428/2/check/gate-puppet-tempest-puppet-unit-latest-dsvm-centos7/6235866/ : SUCCESS in 6m 23s (non-voting)\n- gate-puppet-tempest-puppet-beaker-rspec-dsvm-trusty http://logs.openstack.org/28/250428/2/check/gate-puppet-tempest-puppet-beaker-rspec-dsvm-trusty/2347406/ : SUCCESS in 8m 29s\n- gate-puppet-tempest-puppet-beaker-rspec-dsvm-centos7 http://logs.openstack.org/28/250428/2/check/gate-puppet-tempest-puppet-beaker-rspec-dsvm-centos7/64505be/ : SUCCESS in 7m 21s\n- gate-puppet-openstack-integration-scenario001-dsvm-trusty http://logs.openstack.org/28/250428/2/check/gate-puppet-openstack-integration-scenario001-dsvm-trusty/6b10c8b/ : SUCCESS in 28m 54s (non-voting)\n- gate-puppet-openstack-integration-scenario001-dsvm-centos7 http://logs.openstack.org/28/250428/2/check/gate-puppet-openstack-integration-scenario001-dsvm-centos7/12dff7e/ : SUCCESS in 19m 14s (non-voting)\n- gate-puppet-openstack-integration-scenario002-dsvm-trusty http://logs.openstack.org/28/250428/2/check/gate-puppet-openstack-integration-scenario002-dsvm-trusty/4cbd0fa/ : SUCCESS in 29m 47s (non-voting)\n- gate-puppet-openstack-integration-scenario002-dsvm-centos7 http://logs.openstack.org/28/250428/2/check/gate-puppet-openstack-integration-scenario002-dsvm-centos7/302d821/ : SUCCESS in 22m 04s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7eb1fd1f45b05ce3e321fa3855fdf8e617aa662f","author":{"_account_id":7745,"name":"Denis Egorenko","email":"degorenko@mirantis.com","username":"degorenko"},"date":"2015-11-26 15:45:09.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f6d05fdb45a8f3e87e4ad37f0fde308cb626b9d2","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-11-26 15:45:15.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":2},{"id":"cac35373180fd9b23596e00b7c5847eead1e576e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-11-26 15:58:08.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-puppet-tempest-puppet-lint http://logs.openstack.org/28/250428/2/gate/gate-puppet-tempest-puppet-lint/ed0fd0b/ : SUCCESS in 32s\n- gate-puppet-tempest-puppet-syntax-3-dsvm-centos7 http://logs.openstack.org/28/250428/2/gate/gate-puppet-tempest-puppet-syntax-3-dsvm-centos7/535ba2a/ : SUCCESS in 2m 44s\n- gate-puppet-tempest-puppet-syntax-4-dsvm-centos7 http://logs.openstack.org/28/250428/2/gate/gate-puppet-tempest-puppet-syntax-4-dsvm-centos7/22df09e/ : SUCCESS in 2m 52s\n- gate-puppet-tempest-puppet-unit-3.3-dsvm-centos7 http://logs.openstack.org/28/250428/2/gate/gate-puppet-tempest-puppet-unit-3.3-dsvm-centos7/cd5f7e6/ : SUCCESS in 3m 33s\n- gate-puppet-tempest-puppet-unit-3.4-dsvm-centos7 http://logs.openstack.org/28/250428/2/gate/gate-puppet-tempest-puppet-unit-3.4-dsvm-centos7/8766921/ : SUCCESS in 3m 44s\n- gate-puppet-tempest-puppet-unit-3.6-dsvm-centos7 http://logs.openstack.org/28/250428/2/gate/gate-puppet-tempest-puppet-unit-3.6-dsvm-centos7/f0b3455/ : SUCCESS in 4m 14s\n- gate-puppet-tempest-puppet-unit-3.7-dsvm-centos7 http://logs.openstack.org/28/250428/2/gate/gate-puppet-tempest-puppet-unit-3.7-dsvm-centos7/97f03e8/ : SUCCESS in 3m 39s\n- gate-puppet-tempest-puppet-unit-3.8-dsvm-centos7 http://logs.openstack.org/28/250428/2/gate/gate-puppet-tempest-puppet-unit-3.8-dsvm-centos7/f48f762/ : SUCCESS in 3m 39s\n- gate-puppet-tempest-puppet-unit-4.0-dsvm-centos7 http://logs.openstack.org/28/250428/2/gate/gate-puppet-tempest-puppet-unit-4.0-dsvm-centos7/ee3e03b/ : SUCCESS in 3m 20s\n- gate-puppet-tempest-puppet-beaker-rspec-dsvm-trusty http://logs.openstack.org/28/250428/2/gate/gate-puppet-tempest-puppet-beaker-rspec-dsvm-trusty/23c55e3/ : SUCCESS in 7m 09s\n- gate-puppet-tempest-puppet-beaker-rspec-dsvm-centos7 http://logs.openstack.org/28/250428/2/gate/gate-puppet-tempest-puppet-beaker-rspec-dsvm-centos7/84ce6e1/ : SUCCESS in 6m 06s","accounts_in_message":[],"_revision_number":2},{"id":"fc1aaa15f85882b1169409e126a3b35c0cbbb4ba","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-11-26 15:58:09.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a7c0208d813d2f31a3921de6b52faefe4a4f4e8d","revisions":{"70f9286d607a1843a5529129a8ae1c0a91afc147":{"kind":"REWORK","_number":1,"created":"2015-11-26 15:04:07.000000000","uploader":{"_account_id":13752,"name":"Alexey Deryugin","email":"aderiugin@mirantis.com","username":"velovec"},"ref":"refs/changes/28/250428/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tempest","ref":"refs/changes/28/250428/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tempest refs/changes/28/250428/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tempest refs/changes/28/250428/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tempest refs/changes/28/250428/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tempest refs/changes/28/250428/1"}}},"commit":{"parents":[{"commit":"f03b104975f7ca1500f839c83b5a2e4c19d538fa","subject":"Prepare 6.1.0 release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tempest/commit/f03b104975f7ca1500f839c83b5a2e4c19d538fa"}]}],"author":{"name":"Alexey Deryugin","email":"aderyugin@mirantis.com","date":"2015-11-13 14:18:06.000000000","tz":180},"committer":{"name":"Alexey Deryugin","email":"aderyugin@mirantis.com","date":"2015-11-26 15:04:01.000000000","tz":180},"subject":"Allow to activate Murano tests","message":"Allow to activate Murano tests\n\nEnable murano parameter to run Application Catalog tests in Tempest.\n\nChange-Id: I7c5671695787fb1660d26fde689ff891895980fa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tempest/commit/70f9286d607a1843a5529129a8ae1c0a91afc147"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tempest/commit/70f9286d607a1843a5529129a8ae1c0a91afc147"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"f03b104975f7ca1500f839c83b5a2e4c19d538fa","is_merged_in_target_branch":true,"change_id":"I63260479da87738fe3d918836c25d5c98088e7c4","change_number":237164,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"a7c0208d813d2f31a3921de6b52faefe4a4f4e8d":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2015-11-26 15:04:21.000000000","uploader":{"_account_id":13752,"name":"Alexey Deryugin","email":"aderiugin@mirantis.com","username":"velovec"},"ref":"refs/changes/28/250428/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tempest","ref":"refs/changes/28/250428/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tempest refs/changes/28/250428/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tempest refs/changes/28/250428/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tempest refs/changes/28/250428/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tempest refs/changes/28/250428/2"}}},"commit":{"parents":[{"commit":"499c06ccfa39bc697c40ec88cfa7f0c9262d5a70","subject":"Checkout openstack-integration from master","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tempest/commit/499c06ccfa39bc697c40ec88cfa7f0c9262d5a70"}]}],"author":{"name":"Alexey Deryugin","email":"aderyugin@mirantis.com","date":"2015-11-13 14:18:06.000000000","tz":180},"committer":{"name":"Alexey Deryugin","email":"aderyugin@mirantis.com","date":"2015-11-26 15:04:21.000000000","tz":0},"subject":"Allow to activate Murano tests","message":"Allow to activate Murano tests\n\nEnable murano parameter to run Application Catalog tests in Tempest.\n\nChange-Id: I7c5671695787fb1660d26fde689ff891895980fa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tempest/commit/a7c0208d813d2f31a3921de6b52faefe4a4f4e8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tempest/commit/a7c0208d813d2f31a3921de6b52faefe4a4f4e8d"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"499c06ccfa39bc697c40ec88cfa7f0c9262d5a70","is_merged_in_target_branch":true,"change_id":"I04cf42657d020b2dd3019ada320fe13e08061cbd","change_number":250300,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
