)]}'
{"id":"openstack%2Fpuppet-tempest~411187","triplet_id":"openstack%2Fpuppet-tempest~master~Ida5c8c541fc45e200e146ae740c36c3462da6e72","project":"openstack/puppet-tempest","branch":"master","hashtags":[],"change_id":"Ida5c8c541fc45e200e146ae740c36c3462da6e72","subject":"Allow for enabling/disabling image api versions","status":"MERGED","created":"2016-12-15 09:30:10.000000000","updated":"2016-12-29 05:51:06.000000000","submitted":"2016-12-29 05:51:06.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"278dfe63dd816cea44ad8251441356ff0f972ccd","_number":411187,"virtual_id_number":411187,"owner":{"_account_id":6159,"name":"Flavio Percoco Premoli","display_name":"flaper87","email":"flavio.percoco@flyrlabs.com","username":"flaper87"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2016-12-29 05:51:06.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2016-12-23 18:30:55.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2016-12-29 05:09:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2016-12-29 05:09:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2016-12-23 18:30:55.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2016-12-29 05:09:18.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2016-12-29 05:51:06.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"5dd0e5a867fa20a54d00096003c9cafb682e7272","author":{"_account_id":6159,"name":"Flavio Percoco Premoli","display_name":"flaper87","email":"flavio.percoco@flyrlabs.com","username":"flaper87"},"date":"2016-12-15 09:30:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"349c4a385e69b672a57c5f8c4c6cabb4b9b431ae","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-12-15 10:11:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-puppet-tempest-puppet-lint http://logs.openstack.org/87/411187/1/check/gate-puppet-tempest-puppet-lint/b4a9c20/ : FAILURE in 1m 32s\n- gate-puppet-tempest-puppet-syntax-3-centos-7 http://logs.openstack.org/87/411187/1/check/gate-puppet-tempest-puppet-syntax-3-centos-7/f204485/ : SUCCESS in 3m 14s\n- gate-puppet-tempest-puppet-syntax-4-centos-7 http://logs.openstack.org/87/411187/1/check/gate-puppet-tempest-puppet-syntax-4-centos-7/71fcf1c/ : SUCCESS in 1m 54s\n- gate-puppet-tempest-puppet-unit-3.6-centos-7 http://logs.openstack.org/87/411187/1/check/gate-puppet-tempest-puppet-unit-3.6-centos-7/a0dbe36/ : SUCCESS in 3m 05s\n- gate-puppet-tempest-puppet-unit-3.8-centos-7 http://logs.openstack.org/87/411187/1/check/gate-puppet-tempest-puppet-unit-3.8-centos-7/5f4e460/ : SUCCESS in 3m 13s\n- gate-puppet-tempest-puppet-unit-4.5-centos-7 http://logs.openstack.org/87/411187/1/check/gate-puppet-tempest-puppet-unit-4.5-centos-7/4da046d/ : SUCCESS in 3m 52s\n- gate-puppet-tempest-puppet-unit-latest-centos-7 http://logs.openstack.org/87/411187/1/check/gate-puppet-tempest-puppet-unit-latest-centos-7/328780e/ : SUCCESS in 3m 28s (non-voting)\n- gate-puppet-tempest-puppet-beaker-rspec-centos-7 http://logs.openstack.org/87/411187/1/check/gate-puppet-tempest-puppet-beaker-rspec-centos-7/d5e2ce6/ : FAILURE in 4m 58s\n- gate-puppet-tempest-puppet-beaker-rspec-ubuntu-xenial http://logs.openstack.org/87/411187/1/check/gate-puppet-tempest-puppet-beaker-rspec-ubuntu-xenial/e6087ed/ : SUCCESS in 3m 24s\n- gate-puppet-openstack-integration-4-scenario001-tempest-centos-7 http://logs.openstack.org/87/411187/1/check/gate-puppet-openstack-integration-4-scenario001-tempest-centos-7/bd0cf81/ : SUCCESS in 33m 47s\n- gate-puppet-openstack-integration-4-scenario002-tempest-centos-7 http://logs.openstack.org/87/411187/1/check/gate-puppet-openstack-integration-4-scenario002-tempest-centos-7/b6c3929/ : SUCCESS in 38m 36s\n- gate-puppet-openstack-integration-4-scenario003-tempest-centos-7 http://logs.openstack.org/87/411187/1/check/gate-puppet-openstack-integration-4-scenario003-tempest-centos-7/ce3ec51/ : SUCCESS in 29m 53s\n- gate-puppet-openstack-integration-4-scenario004-tempest-centos-7 http://logs.openstack.org/87/411187/1/check/gate-puppet-openstack-integration-4-scenario004-tempest-centos-7/2079f34/ : SUCCESS in 23m 05s\n- gate-puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial http://logs.openstack.org/87/411187/1/check/gate-puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial/f29d382/ : SUCCESS in 39m 18s\n- gate-puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial http://logs.openstack.org/87/411187/1/check/gate-puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial/90f5090/ : SUCCESS in 36m 17s\n- gate-puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial http://logs.openstack.org/87/411187/1/check/gate-puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial/5f556cc/ : SUCCESS in 33m 14s\n- gate-puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial http://logs.openstack.org/87/411187/1/check/gate-puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial/4cb09e5/ : SUCCESS in 23m 33s","accounts_in_message":[],"_revision_number":1},{"id":"36045e9c38477da7d0888abb713d5fa1893df107","author":{"_account_id":6159,"name":"Flavio Percoco Premoli","display_name":"flaper87","email":"flavio.percoco@flyrlabs.com","username":"flaper87"},"date":"2016-12-15 10:35:19.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d5036c7e2532cec4a185fc3f9436497e98209be9","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-12-15 11:25:19.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-puppet-tempest-puppet-lint http://logs.openstack.org/87/411187/2/check/gate-puppet-tempest-puppet-lint/de42f1b/ : SUCCESS in 1m 35s\n- gate-puppet-tempest-puppet-syntax-3-centos-7 http://logs.openstack.org/87/411187/2/check/gate-puppet-tempest-puppet-syntax-3-centos-7/d804b76/ : SUCCESS in 2m 32s\n- gate-puppet-tempest-puppet-syntax-4-centos-7 http://logs.openstack.org/87/411187/2/check/gate-puppet-tempest-puppet-syntax-4-centos-7/d628668/ : SUCCESS in 2m 02s\n- gate-puppet-tempest-puppet-unit-3.6-centos-7 http://logs.openstack.org/87/411187/2/check/gate-puppet-tempest-puppet-unit-3.6-centos-7/1d09bcd/ : SUCCESS in 4m 07s\n- gate-puppet-tempest-puppet-unit-3.8-centos-7 http://logs.openstack.org/87/411187/2/check/gate-puppet-tempest-puppet-unit-3.8-centos-7/10e1104/ : SUCCESS in 3m 51s\n- gate-puppet-tempest-puppet-unit-4.5-centos-7 http://logs.openstack.org/87/411187/2/check/gate-puppet-tempest-puppet-unit-4.5-centos-7/062c1bb/ : SUCCESS in 4m 06s\n- gate-puppet-tempest-puppet-unit-latest-centos-7 http://logs.openstack.org/87/411187/2/check/gate-puppet-tempest-puppet-unit-latest-centos-7/e6f20fb/ : SUCCESS in 3m 13s (non-voting)\n- gate-puppet-tempest-puppet-beaker-rspec-centos-7 http://logs.openstack.org/87/411187/2/check/gate-puppet-tempest-puppet-beaker-rspec-centos-7/2907b48/ : FAILURE in 5m 24s\n- gate-puppet-tempest-puppet-beaker-rspec-ubuntu-xenial http://logs.openstack.org/87/411187/2/check/gate-puppet-tempest-puppet-beaker-rspec-ubuntu-xenial/d1867ca/ : SUCCESS in 5m 26s\n- gate-puppet-openstack-integration-4-scenario001-tempest-centos-7 http://logs.openstack.org/87/411187/2/check/gate-puppet-openstack-integration-4-scenario001-tempest-centos-7/c6de502/ : SUCCESS in 48m 09s\n- gate-puppet-openstack-integration-4-scenario002-tempest-centos-7 http://logs.openstack.org/87/411187/2/check/gate-puppet-openstack-integration-4-scenario002-tempest-centos-7/e881768/ : SUCCESS in 36m 58s\n- gate-puppet-openstack-integration-4-scenario003-tempest-centos-7 http://logs.openstack.org/87/411187/2/check/gate-puppet-openstack-integration-4-scenario003-tempest-centos-7/f0836af/ : SUCCESS in 30m 43s\n- gate-puppet-openstack-integration-4-scenario004-tempest-centos-7 http://logs.openstack.org/87/411187/2/check/gate-puppet-openstack-integration-4-scenario004-tempest-centos-7/d0547c9/ : SUCCESS in 25m 29s\n- gate-puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial http://logs.openstack.org/87/411187/2/check/gate-puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial/5644811/ : SUCCESS in 30m 12s\n- gate-puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial http://logs.openstack.org/87/411187/2/check/gate-puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial/33d7a20/ : SUCCESS in 47m 01s\n- gate-puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial http://logs.openstack.org/87/411187/2/check/gate-puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial/9b32c6d/ : SUCCESS in 30m 36s\n- gate-puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial http://logs.openstack.org/87/411187/2/check/gate-puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial/3487692/ : SUCCESS in 25m 19s","accounts_in_message":[],"_revision_number":2},{"id":"4d7a9bd0c3f6e5a80b3430d11269e38166f1f242","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2016-12-15 16:11:16.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"0d59170f613d419343b25df019e4fd40b7a06727","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-12-15 17:07:44.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-puppet-tempest-puppet-lint http://logs.openstack.org/87/411187/2/check/gate-puppet-tempest-puppet-lint/c195ccf/ : SUCCESS in 1m 37s\n- gate-puppet-tempest-puppet-syntax-3-centos-7 http://logs.openstack.org/87/411187/2/check/gate-puppet-tempest-puppet-syntax-3-centos-7/172f78f/ : SUCCESS in 2m 00s\n- gate-puppet-tempest-puppet-syntax-4-centos-7 http://logs.openstack.org/87/411187/2/check/gate-puppet-tempest-puppet-syntax-4-centos-7/74ece2a/ : SUCCESS in 2m 40s\n- gate-puppet-tempest-puppet-unit-3.6-centos-7 http://logs.openstack.org/87/411187/2/check/gate-puppet-tempest-puppet-unit-3.6-centos-7/9c42e10/ : SUCCESS in 3m 27s\n- gate-puppet-tempest-puppet-unit-3.8-centos-7 http://logs.openstack.org/87/411187/2/check/gate-puppet-tempest-puppet-unit-3.8-centos-7/584a59b/ : SUCCESS in 4m 49s\n- gate-puppet-tempest-puppet-unit-4.5-centos-7 http://logs.openstack.org/87/411187/2/check/gate-puppet-tempest-puppet-unit-4.5-centos-7/9e3c9b3/ : SUCCESS in 5m 51s\n- gate-puppet-tempest-puppet-unit-latest-centos-7 http://logs.openstack.org/87/411187/2/check/gate-puppet-tempest-puppet-unit-latest-centos-7/c0026c8/ : SUCCESS in 4m 53s (non-voting)\n- gate-puppet-tempest-puppet-beaker-rspec-centos-7 http://logs.openstack.org/87/411187/2/check/gate-puppet-tempest-puppet-beaker-rspec-centos-7/569d4bb/ : FAILURE in 5m 53s\n- gate-puppet-tempest-puppet-beaker-rspec-ubuntu-xenial http://logs.openstack.org/87/411187/2/check/gate-puppet-tempest-puppet-beaker-rspec-ubuntu-xenial/7eb1861/ : SUCCESS in 3m 55s\n- gate-puppet-openstack-integration-4-scenario001-tempest-centos-7 http://logs.openstack.org/87/411187/2/check/gate-puppet-openstack-integration-4-scenario001-tempest-centos-7/0e3b5fc/ : SUCCESS in 32m 28s\n- gate-puppet-openstack-integration-4-scenario002-tempest-centos-7 http://logs.openstack.org/87/411187/2/check/gate-puppet-openstack-integration-4-scenario002-tempest-centos-7/be09c70/ : SUCCESS in 53m 18s\n- gate-puppet-openstack-integration-4-scenario003-tempest-centos-7 http://logs.openstack.org/87/411187/2/check/gate-puppet-openstack-integration-4-scenario003-tempest-centos-7/c96ab17/ : SUCCESS in 43m 28s\n- gate-puppet-openstack-integration-4-scenario004-tempest-centos-7 http://logs.openstack.org/87/411187/2/check/gate-puppet-openstack-integration-4-scenario004-tempest-centos-7/3c82a65/ : SUCCESS in 25m 02s\n- gate-puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial http://logs.openstack.org/87/411187/2/check/gate-puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial/3c500e4/ : SUCCESS in 47m 55s\n- gate-puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial http://logs.openstack.org/87/411187/2/check/gate-puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial/fb1930c/ : SUCCESS in 37m 22s\n- gate-puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial http://logs.openstack.org/87/411187/2/check/gate-puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial/4c70d56/ : SUCCESS in 29m 57s\n- gate-puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial http://logs.openstack.org/87/411187/2/check/gate-puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial/479d224/ : SUCCESS in 23m 50s","accounts_in_message":[],"_revision_number":2},{"id":"5d6157a36d739beaad84c2e63677f7badb793a45","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2016-12-15 23:37:26.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"5e5b8de24e78a7aa21dd76ad3e2abbcf096a41ca","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-12-16 00:33:05.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-puppet-tempest-puppet-lint http://logs.openstack.org/87/411187/2/check/gate-puppet-tempest-puppet-lint/30eca9b/ : SUCCESS in 1m 45s\n- gate-puppet-tempest-puppet-syntax-3-centos-7 http://logs.openstack.org/87/411187/2/check/gate-puppet-tempest-puppet-syntax-3-centos-7/e056039/ : SUCCESS in 1m 47s\n- gate-puppet-tempest-puppet-syntax-4-centos-7 http://logs.openstack.org/87/411187/2/check/gate-puppet-tempest-puppet-syntax-4-centos-7/550b07b/ : SUCCESS in 2m 26s\n- gate-puppet-tempest-puppet-unit-3.6-centos-7 http://logs.openstack.org/87/411187/2/check/gate-puppet-tempest-puppet-unit-3.6-centos-7/e5be84f/ : SUCCESS in 3m 36s\n- gate-puppet-tempest-puppet-unit-3.8-centos-7 http://logs.openstack.org/87/411187/2/check/gate-puppet-tempest-puppet-unit-3.8-centos-7/4edbf8d/ : SUCCESS in 3m 09s\n- gate-puppet-tempest-puppet-unit-4.5-centos-7 http://logs.openstack.org/87/411187/2/check/gate-puppet-tempest-puppet-unit-4.5-centos-7/17c5d61/ : SUCCESS in 5m 11s\n- gate-puppet-tempest-puppet-unit-latest-centos-7 http://logs.openstack.org/87/411187/2/check/gate-puppet-tempest-puppet-unit-latest-centos-7/9447126/ : SUCCESS in 7m 04s (non-voting)\n- gate-puppet-tempest-puppet-beaker-rspec-centos-7 http://logs.openstack.org/87/411187/2/check/gate-puppet-tempest-puppet-beaker-rspec-centos-7/23aa2e9/ : FAILURE in 5m 57s\n- gate-puppet-tempest-puppet-beaker-rspec-ubuntu-xenial http://logs.openstack.org/87/411187/2/check/gate-puppet-tempest-puppet-beaker-rspec-ubuntu-xenial/b125ffe/ : SUCCESS in 4m 00s\n- gate-puppet-openstack-integration-4-scenario001-tempest-centos-7 http://logs.openstack.org/87/411187/2/check/gate-puppet-openstack-integration-4-scenario001-tempest-centos-7/4c40fab/ : SUCCESS in 34m 29s\n- gate-puppet-openstack-integration-4-scenario002-tempest-centos-7 http://logs.openstack.org/87/411187/2/check/gate-puppet-openstack-integration-4-scenario002-tempest-centos-7/f327094/ : SUCCESS in 48m 07s\n- gate-puppet-openstack-integration-4-scenario003-tempest-centos-7 http://logs.openstack.org/87/411187/2/check/gate-puppet-openstack-integration-4-scenario003-tempest-centos-7/14b565a/ : SUCCESS in 45m 21s\n- gate-puppet-openstack-integration-4-scenario004-tempest-centos-7 http://logs.openstack.org/87/411187/2/check/gate-puppet-openstack-integration-4-scenario004-tempest-centos-7/ba96a67/ : SUCCESS in 32m 38s\n- gate-puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial http://logs.openstack.org/87/411187/2/check/gate-puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial/9d7db2a/ : SUCCESS in 40m 14s\n- gate-puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial http://logs.openstack.org/87/411187/2/check/gate-puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial/4dcd8b5/ : SUCCESS in 48m 39s\n- gate-puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial http://logs.openstack.org/87/411187/2/check/gate-puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial/28a3101/ : SUCCESS in 37m 38s\n- gate-puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial http://logs.openstack.org/87/411187/2/check/gate-puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial/527e022/ : SUCCESS in 35m 47s","accounts_in_message":[],"_revision_number":2},{"id":"72d18aa9ccbce91fa66d2e4d75f973dc63c5057d","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2016-12-23 18:30:50.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"3c04ab5b53837b2f563f837dd796d69fe32681f3","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2016-12-23 18:30:55.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"347b2a1f70e86c6068bda6565a967cdeea621093","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-12-23 19:09:44.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-puppet-tempest-puppet-lint http://logs.openstack.org/87/411187/3/check/gate-puppet-tempest-puppet-lint/60e4ffb/ : SUCCESS in 1m 31s\n- gate-puppet-tempest-puppet-syntax-3-centos-7 http://logs.openstack.org/87/411187/3/check/gate-puppet-tempest-puppet-syntax-3-centos-7/05f943b/ : SUCCESS in 2m 43s\n- gate-puppet-tempest-puppet-syntax-4-centos-7 http://logs.openstack.org/87/411187/3/check/gate-puppet-tempest-puppet-syntax-4-centos-7/0b3616d/ : SUCCESS in 1m 43s\n- gate-puppet-tempest-puppet-unit-3.6-centos-7 http://logs.openstack.org/87/411187/3/check/gate-puppet-tempest-puppet-unit-3.6-centos-7/30dd63c/ : SUCCESS in 2m 54s\n- gate-puppet-tempest-puppet-unit-3.8-centos-7 http://logs.openstack.org/87/411187/3/check/gate-puppet-tempest-puppet-unit-3.8-centos-7/4b59d1b/ : SUCCESS in 2m 56s\n- gate-puppet-tempest-puppet-unit-4.5-centos-7 http://logs.openstack.org/87/411187/3/check/gate-puppet-tempest-puppet-unit-4.5-centos-7/afd5347/ : SUCCESS in 3m 20s\n- gate-puppet-tempest-puppet-unit-latest-centos-7 http://logs.openstack.org/87/411187/3/check/gate-puppet-tempest-puppet-unit-latest-centos-7/a086d34/ : SUCCESS in 3m 35s (non-voting)\n- gate-puppet-tempest-puppet-beaker-rspec-centos-7 http://logs.openstack.org/87/411187/3/check/gate-puppet-tempest-puppet-beaker-rspec-centos-7/cccca59/ : SUCCESS in 5m 19s\n- gate-puppet-tempest-puppet-beaker-rspec-ubuntu-xenial http://logs.openstack.org/87/411187/3/check/gate-puppet-tempest-puppet-beaker-rspec-ubuntu-xenial/46f295c/ : SUCCESS in 4m 03s\n- gate-puppet-openstack-integration-4-scenario001-tempest-centos-7 http://logs.openstack.org/87/411187/3/check/gate-puppet-openstack-integration-4-scenario001-tempest-centos-7/d1dc244/ : SUCCESS in 31m 18s\n- gate-puppet-openstack-integration-4-scenario002-tempest-centos-7 http://logs.openstack.org/87/411187/3/check/gate-puppet-openstack-integration-4-scenario002-tempest-centos-7/17617c6/ : SUCCESS in 35m 33s\n- gate-puppet-openstack-integration-4-scenario003-tempest-centos-7 http://logs.openstack.org/87/411187/3/check/gate-puppet-openstack-integration-4-scenario003-tempest-centos-7/972992b/ : SUCCESS in 35m 27s\n- gate-puppet-openstack-integration-4-scenario004-tempest-centos-7 http://logs.openstack.org/87/411187/3/check/gate-puppet-openstack-integration-4-scenario004-tempest-centos-7/783ad10/ : SUCCESS in 23m 31s\n- gate-puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial http://logs.openstack.org/87/411187/3/check/gate-puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial/3624da4/ : SUCCESS in 31m 42s\n- gate-puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial http://logs.openstack.org/87/411187/3/check/gate-puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial/2a82e49/ : SUCCESS in 38m 40s\n- gate-puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial http://logs.openstack.org/87/411187/3/check/gate-puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial/47eb6a9/ : SUCCESS in 30m 52s\n- gate-puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial http://logs.openstack.org/87/411187/3/check/gate-puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial/407c891/ : SUCCESS in 23m 49s","accounts_in_message":[],"_revision_number":3},{"id":"596202b4e1b50397706fba631f8f545218a5ad64","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2016-12-29 05:09:18.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"c24f22c2b65704b43739d0495ae782ca0ec6f1e9","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-12-29 05:09:29.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":3},{"id":"1570a000fa21558e1a015e32bbe695344cde5d23","date":"2016-12-29 05:51:06.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":3},{"id":"278dfe63dd816cea44ad8251441356ff0f972ccd","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-12-29 05:51:06.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-puppet-tempest-puppet-lint http://logs.openstack.org/87/411187/3/gate/gate-puppet-tempest-puppet-lint/1f603f0/ : SUCCESS in 1m 22s\n- gate-puppet-tempest-puppet-syntax-3-centos-7 http://logs.openstack.org/87/411187/3/gate/gate-puppet-tempest-puppet-syntax-3-centos-7/82c7709/ : SUCCESS in 2m 31s\n- gate-puppet-tempest-puppet-syntax-4-centos-7 http://logs.openstack.org/87/411187/3/gate/gate-puppet-tempest-puppet-syntax-4-centos-7/c723910/ : SUCCESS in 2m 03s\n- gate-puppet-tempest-puppet-unit-3.6-centos-7 http://logs.openstack.org/87/411187/3/gate/gate-puppet-tempest-puppet-unit-3.6-centos-7/84bfad4/ : SUCCESS in 4m 24s\n- gate-puppet-tempest-puppet-unit-3.8-centos-7 http://logs.openstack.org/87/411187/3/gate/gate-puppet-tempest-puppet-unit-3.8-centos-7/9313d9b/ : SUCCESS in 3m 00s\n- gate-puppet-tempest-puppet-unit-4.5-centos-7 http://logs.openstack.org/87/411187/3/gate/gate-puppet-tempest-puppet-unit-4.5-centos-7/4ddaa9d/ : SUCCESS in 2m 53s\n- gate-puppet-tempest-puppet-beaker-rspec-centos-7 http://logs.openstack.org/87/411187/3/gate/gate-puppet-tempest-puppet-beaker-rspec-centos-7/14f7612/ : SUCCESS in 5m 11s\n- gate-puppet-tempest-puppet-beaker-rspec-ubuntu-xenial http://logs.openstack.org/87/411187/3/gate/gate-puppet-tempest-puppet-beaker-rspec-ubuntu-xenial/4d08cc2/ : SUCCESS in 4m 56s\n- gate-puppet-openstack-integration-4-scenario001-tempest-centos-7 http://logs.openstack.org/87/411187/3/gate/gate-puppet-openstack-integration-4-scenario001-tempest-centos-7/9667563/ : SUCCESS in 34m 20s\n- gate-puppet-openstack-integration-4-scenario002-tempest-centos-7 http://logs.openstack.org/87/411187/3/gate/gate-puppet-openstack-integration-4-scenario002-tempest-centos-7/d99966f/ : SUCCESS in 32m 41s\n- gate-puppet-openstack-integration-4-scenario003-tempest-centos-7 http://logs.openstack.org/87/411187/3/gate/gate-puppet-openstack-integration-4-scenario003-tempest-centos-7/bf34fe8/ : SUCCESS in 26m 25s\n- gate-puppet-openstack-integration-4-scenario004-tempest-centos-7 http://logs.openstack.org/87/411187/3/gate/gate-puppet-openstack-integration-4-scenario004-tempest-centos-7/513fec5/ : SUCCESS in 21m 36s\n- gate-puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial http://logs.openstack.org/87/411187/3/gate/gate-puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial/43de244/ : SUCCESS in 31m 54s\n- gate-puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial http://logs.openstack.org/87/411187/3/gate/gate-puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial/555a736/ : SUCCESS in 35m 35s\n- gate-puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial http://logs.openstack.org/87/411187/3/gate/gate-puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial/d50bf1b/ : SUCCESS in 41m 35s\n- gate-puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial http://logs.openstack.org/87/411187/3/gate/gate-puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial/1eefc6a/ : SUCCESS in 33m 00s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f6b3ff4a747b1aab2b86a9f2f0319a7953d7288c","revisions":{"e0a165719fa00d9e6e45b672b42cded25b7c8a46":{"kind":"REWORK","_number":1,"created":"2016-12-15 09:30:10.000000000","uploader":{"_account_id":6159,"name":"Flavio Percoco Premoli","display_name":"flaper87","email":"flavio.percoco@flyrlabs.com","username":"flaper87"},"ref":"refs/changes/87/411187/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tempest","ref":"refs/changes/87/411187/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tempest refs/changes/87/411187/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tempest refs/changes/87/411187/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tempest refs/changes/87/411187/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tempest refs/changes/87/411187/1"}}},"commit":{"parents":[{"commit":"69239a9a208d3a582eb6d74fd1c158c4c17cfd7f","subject":"Prepare 10.1.0 release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tempest/commit/69239a9a208d3a582eb6d74fd1c158c4c17cfd7f"}]}],"author":{"name":"Flavio Percoco","email":"flaper87@gmail.com","date":"2016-12-15 09:25:26.000000000","tz":60},"committer":{"name":"Flavio Percoco","email":"flaper87@gmail.com","date":"2016-12-15 09:27:03.000000000","tz":60},"subject":"Allow for enabling/disabling image api versions","message":"Allow for enabling/disabling image api versions\n\nAdd options for enabling and/or disabling both, v1 and v2, versions of\nglance\u0027s API.\n\nChange-Id: Ida5c8c541fc45e200e146ae740c36c3462da6e72\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tempest/commit/e0a165719fa00d9e6e45b672b42cded25b7c8a46"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tempest/commit/e0a165719fa00d9e6e45b672b42cded25b7c8a46"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"69239a9a208d3a582eb6d74fd1c158c4c17cfd7f","is_merged_in_target_branch":true,"change_id":"I1b58e937c37e90a97ddcea840284389532094d6d","change_number":409244,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"a4e4132e79b7efbb0bf5993ed5645222bf4ac8b0":{"kind":"REWORK","_number":2,"created":"2016-12-15 10:35:19.000000000","uploader":{"_account_id":6159,"name":"Flavio Percoco Premoli","display_name":"flaper87","email":"flavio.percoco@flyrlabs.com","username":"flaper87"},"ref":"refs/changes/87/411187/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tempest","ref":"refs/changes/87/411187/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tempest refs/changes/87/411187/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tempest refs/changes/87/411187/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tempest refs/changes/87/411187/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tempest refs/changes/87/411187/2"}}},"commit":{"parents":[{"commit":"69239a9a208d3a582eb6d74fd1c158c4c17cfd7f","subject":"Prepare 10.1.0 release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tempest/commit/69239a9a208d3a582eb6d74fd1c158c4c17cfd7f"}]}],"author":{"name":"Flavio Percoco","email":"flaper87@gmail.com","date":"2016-12-15 09:25:26.000000000","tz":60},"committer":{"name":"Flavio Percoco","email":"flaper87@gmail.com","date":"2016-12-15 10:32:04.000000000","tz":60},"subject":"Allow for enabling/disabling image api versions","message":"Allow for enabling/disabling image api versions\n\nAdd options for enabling and/or disabling both, v1 and v2, versions of\nglance\u0027s API.\n\nChange-Id: Ida5c8c541fc45e200e146ae740c36c3462da6e72\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tempest/commit/a4e4132e79b7efbb0bf5993ed5645222bf4ac8b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tempest/commit/a4e4132e79b7efbb0bf5993ed5645222bf4ac8b0"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"69239a9a208d3a582eb6d74fd1c158c4c17cfd7f","is_merged_in_target_branch":true,"change_id":"I1b58e937c37e90a97ddcea840284389532094d6d","change_number":409244,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"f6b3ff4a747b1aab2b86a9f2f0319a7953d7288c":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2016-12-23 18:30:50.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/87/411187/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tempest","ref":"refs/changes/87/411187/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tempest refs/changes/87/411187/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tempest refs/changes/87/411187/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tempest refs/changes/87/411187/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tempest refs/changes/87/411187/3"}}},"commit":{"parents":[{"commit":"55f1dd10d1d20691979bd5e6bfdee299ba9132c7","subject":"Pin reno version to reno\u003e\u003d0.1.1,!\u003d2.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tempest/commit/55f1dd10d1d20691979bd5e6bfdee299ba9132c7"}]}],"author":{"name":"Flavio Percoco","email":"flaper87@gmail.com","date":"2016-12-15 09:25:26.000000000","tz":60},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2016-12-23 18:30:50.000000000","tz":0},"subject":"Allow for enabling/disabling image api versions","message":"Allow for enabling/disabling image api versions\n\nAdd options for enabling and/or disabling both, v1 and v2, versions of\nglance\u0027s API.\n\nChange-Id: Ida5c8c541fc45e200e146ae740c36c3462da6e72\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tempest/commit/f6b3ff4a747b1aab2b86a9f2f0319a7953d7288c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tempest/commit/f6b3ff4a747b1aab2b86a9f2f0319a7953d7288c"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"55f1dd10d1d20691979bd5e6bfdee299ba9132c7","is_merged_in_target_branch":true,"change_id":"I75c110a0f0fef6622d48dac7dd14fd801cd58de3","change_number":414055,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
