)]}'
{"id":"openstack%2Fpuppet-tempest~448959","triplet_id":"openstack%2Fpuppet-tempest~master~I93329bfa84d06aeb1cb9e1f83cf3ae28e2dc33d0","project":"openstack/puppet-tempest","branch":"master","topic":"update_test_requirements","hashtags":[],"change_id":"I93329bfa84d06aeb1cb9e1f83cf3ae28e2dc33d0","subject":"Update test-requirements.txt","status":"MERGED","created":"2017-03-23 06:55:16.000000000","updated":"2017-03-29 03:24:20.000000000","submitted":"2017-03-29 03:24:20.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"af9aac6a2c0fea27507135234d30b1d5cf96f078","_number":448959,"virtual_id_number":448959,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2017-03-29 03:24:20.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2017-03-23 16:42:16.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2017-03-23 16:42:16.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-03-23 16:42:16.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2017-03-29 03:24:20.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"97192285d4944fb7a9c1e58e1e91c40f312af96a","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2017-03-23 06:55:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7c3cb066c8691d5c5660a5d5120b2a776bb763c8","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-03-23 07:48:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-puppet-tempest-puppet-lint http://logs.openstack.org/59/448959/1/check/gate-puppet-tempest-puppet-lint/948e1f4/ : SUCCESS in 2m 14s\n- gate-puppet-tempest-puppet-syntax-4-centos-7 http://logs.openstack.org/59/448959/1/check/gate-puppet-tempest-puppet-syntax-4-centos-7/a801ebc/ : SUCCESS in 2m 31s\n- gate-puppet-tempest-puppet-unit-3.6-legacy-centos-7 http://logs.openstack.org/59/448959/1/check/gate-puppet-tempest-puppet-unit-3.6-legacy-centos-7/24d10e3/ : SUCCESS in 4m 43s\n- gate-puppet-tempest-puppet-unit-3.8-legacy-centos-7 http://logs.openstack.org/59/448959/1/check/gate-puppet-tempest-puppet-unit-3.8-legacy-centos-7/bf1806c/ : SUCCESS in 4m 17s\n- gate-puppet-tempest-puppet-unit-4.5-centos-7 http://logs.openstack.org/59/448959/1/check/gate-puppet-tempest-puppet-unit-4.5-centos-7/47114b4/ : SUCCESS in 5m 08s\n- gate-puppet-tempest-puppet-unit-latest-centos-7 http://logs.openstack.org/59/448959/1/check/gate-puppet-tempest-puppet-unit-latest-centos-7/178d79c/ : SUCCESS in 6m 04s (non-voting)\n- gate-puppet-tempest-puppet-beaker-rspec-centos-7 http://logs.openstack.org/59/448959/1/check/gate-puppet-tempest-puppet-beaker-rspec-centos-7/462305c/ : SUCCESS in 8m 01s\n- gate-puppet-tempest-puppet-beaker-rspec-ubuntu-xenial http://logs.openstack.org/59/448959/1/check/gate-puppet-tempest-puppet-beaker-rspec-ubuntu-xenial/892e080/ : SUCCESS in 4m 03s\n- gate-puppet-openstack-integration-4-scenario001-tempest-centos-7 http://logs.openstack.org/59/448959/1/check/gate-puppet-openstack-integration-4-scenario001-tempest-centos-7/c2dcca1/ : SUCCESS in 34m 34s\n- gate-puppet-openstack-integration-4-scenario002-tempest-centos-7 http://logs.openstack.org/59/448959/1/check/gate-puppet-openstack-integration-4-scenario002-tempest-centos-7/f367a5f/ : SUCCESS in 49m 57s\n- gate-puppet-openstack-integration-4-scenario003-tempest-centos-7 http://logs.openstack.org/59/448959/1/check/gate-puppet-openstack-integration-4-scenario003-tempest-centos-7/d9dca94/ : SUCCESS in 42m 42s\n- gate-puppet-openstack-integration-4-scenario004-tempest-centos-7 http://logs.openstack.org/59/448959/1/check/gate-puppet-openstack-integration-4-scenario004-tempest-centos-7/f1c4d3e/ : SUCCESS in 31m 25s\n- gate-puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial http://logs.openstack.org/59/448959/1/check/gate-puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial/3d1f49f/ : SUCCESS in 32m 12s\n- gate-puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial http://logs.openstack.org/59/448959/1/check/gate-puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial/63eeb62/ : SUCCESS in 32m 12s\n- gate-puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial http://logs.openstack.org/59/448959/1/check/gate-puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial/5280842/ : SUCCESS in 27m 11s\n- gate-puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial http://logs.openstack.org/59/448959/1/check/gate-puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial/09ec5bf/ : SUCCESS in 30m 20s\n- gate-puppet-tempest-releasenotes http://docs-draft.openstack.org/59/448959/1/check/gate-puppet-tempest-releasenotes/75abcb7//releasenotes/build/html/ : SUCCESS in 1m 26s","accounts_in_message":[],"_revision_number":1},{"id":"51c365060889ce6a45dac40c6e68010a559b953f","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2017-03-23 16:42:16.000000000","message":"Patch Set 1: Workflow+1 Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"01c7d62ab62ed01c9ae1f34da23f8736c29ec188","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-03-23 16:42:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"3297609446d95c8f0aeb1eec7c5c81c30e48f964","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-03-24 05:26:28.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-puppet-tempest-puppet-lint http://logs.openstack.org/59/448959/1/gate/gate-puppet-tempest-puppet-lint/b21004a/ : SUCCESS in 1m 58s\n- gate-puppet-tempest-puppet-syntax-4-centos-7 http://logs.openstack.org/59/448959/1/gate/gate-puppet-tempest-puppet-syntax-4-centos-7/c797c9b/ : SUCCESS in 2m 45s\n- gate-puppet-tempest-puppet-unit-4.5-centos-7 http://logs.openstack.org/59/448959/1/gate/gate-puppet-tempest-puppet-unit-4.5-centos-7/5cb9921/ : SUCCESS in 4m 53s\n- gate-puppet-tempest-puppet-beaker-rspec-centos-7 http://logs.openstack.org/59/448959/1/gate/gate-puppet-tempest-puppet-beaker-rspec-centos-7/07f82a9/ : SUCCESS in 8m 55s\n- gate-puppet-tempest-puppet-beaker-rspec-ubuntu-xenial http://logs.openstack.org/59/448959/1/gate/gate-puppet-tempest-puppet-beaker-rspec-ubuntu-xenial/b4fe2db/ : FAILURE in 3m 36s\n- gate-puppet-openstack-integration-4-scenario001-tempest-centos-7 http://logs.openstack.org/59/448959/1/gate/gate-puppet-openstack-integration-4-scenario001-tempest-centos-7/d41686c/ : SUCCESS in 45m 27s\n- gate-puppet-openstack-integration-4-scenario002-tempest-centos-7 http://logs.openstack.org/59/448959/1/gate/gate-puppet-openstack-integration-4-scenario002-tempest-centos-7/4f4b9be/ : SUCCESS in 34m 07s\n- gate-puppet-openstack-integration-4-scenario003-tempest-centos-7 http://logs.openstack.org/59/448959/1/gate/gate-puppet-openstack-integration-4-scenario003-tempest-centos-7/e3f33ea/ : SUCCESS in 41m 36s\n- gate-puppet-openstack-integration-4-scenario004-tempest-centos-7 http://logs.openstack.org/59/448959/1/gate/gate-puppet-openstack-integration-4-scenario004-tempest-centos-7/8456295/ : SUCCESS in 29m 15s\n- gate-puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial http://logs.openstack.org/59/448959/1/gate/gate-puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial/d1e7706/ : SUCCESS in 35m 59s\n- gate-puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial http://logs.openstack.org/59/448959/1/gate/gate-puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial/cf8b302/ : SUCCESS in 44m 54s\n- gate-puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial http://logs.openstack.org/59/448959/1/gate/gate-puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial/4b05eb4/ : SUCCESS in 25m 17s\n- gate-puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial http://logs.openstack.org/59/448959/1/gate/gate-puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial/77d8008/ : SUCCESS in 27m 15s\n- gate-puppet-tempest-releasenotes http://docs-draft.openstack.org/59/448959/1/gate/gate-puppet-tempest-releasenotes/b773fbb//releasenotes/build/html/ : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1},{"id":"c00d22f3356623d247f918b90580367fcb1b3df4","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2017-03-24 06:01:53.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"15e7106c7dd08c99ba68b4f5831f4abc98ef302e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-03-24 06:51:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-puppet-tempest-puppet-lint http://logs.openstack.org/59/448959/1/check/gate-puppet-tempest-puppet-lint/67d2b54/ : SUCCESS in 2m 53s\n- gate-puppet-tempest-puppet-syntax-4-centos-7 http://logs.openstack.org/59/448959/1/check/gate-puppet-tempest-puppet-syntax-4-centos-7/2efc374/ : SUCCESS in 1m 46s\n- gate-puppet-tempest-puppet-unit-4.5-centos-7 http://logs.openstack.org/59/448959/1/check/gate-puppet-tempest-puppet-unit-4.5-centos-7/06fd4ed/ : SUCCESS in 3m 42s\n- gate-puppet-tempest-puppet-unit-latest-centos-7 http://logs.openstack.org/59/448959/1/check/gate-puppet-tempest-puppet-unit-latest-centos-7/960d9bc/ : SUCCESS in 6m 18s (non-voting)\n- gate-puppet-tempest-puppet-beaker-rspec-centos-7 http://logs.openstack.org/59/448959/1/check/gate-puppet-tempest-puppet-beaker-rspec-centos-7/1ba15db/ : SUCCESS in 9m 30s\n- gate-puppet-tempest-puppet-beaker-rspec-ubuntu-xenial http://logs.openstack.org/59/448959/1/check/gate-puppet-tempest-puppet-beaker-rspec-ubuntu-xenial/888bdae/ : FAILURE in 3m 17s\n- gate-puppet-openstack-integration-4-scenario001-tempest-centos-7 http://logs.openstack.org/59/448959/1/check/gate-puppet-openstack-integration-4-scenario001-tempest-centos-7/259f433/ : SUCCESS in 44m 58s\n- gate-puppet-openstack-integration-4-scenario002-tempest-centos-7 http://logs.openstack.org/59/448959/1/check/gate-puppet-openstack-integration-4-scenario002-tempest-centos-7/2b813a4/ : SUCCESS in 43m 07s\n- gate-puppet-openstack-integration-4-scenario003-tempest-centos-7 http://logs.openstack.org/59/448959/1/check/gate-puppet-openstack-integration-4-scenario003-tempest-centos-7/c2f0ed0/ : SUCCESS in 43m 19s\n- gate-puppet-openstack-integration-4-scenario004-tempest-centos-7 http://logs.openstack.org/59/448959/1/check/gate-puppet-openstack-integration-4-scenario004-tempest-centos-7/41a6c62/ : SUCCESS in 27m 41s\n- gate-puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial http://logs.openstack.org/59/448959/1/check/gate-puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial/21763ea/ : SUCCESS in 27m 26s\n- gate-puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial http://logs.openstack.org/59/448959/1/check/gate-puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial/5e756f4/ : SUCCESS in 48m 52s\n- gate-puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial http://logs.openstack.org/59/448959/1/check/gate-puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial/e8d4df6/ : SUCCESS in 33m 07s\n- gate-puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial http://logs.openstack.org/59/448959/1/check/gate-puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial/b1a5355/ : SUCCESS in 27m 40s\n- gate-puppet-tempest-releasenotes http://docs-draft.openstack.org/59/448959/1/check/gate-puppet-tempest-releasenotes/43a8348//releasenotes/build/html/ : SUCCESS in 1m 22s","accounts_in_message":[],"_revision_number":1},{"id":"89b297d10a8027017fbfe7a4ad6789f9c956aa9d","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2017-03-24 09:09:30.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"71418fcaaf3a753b095d077c8537d040c9686939","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-03-24 10:02:36.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-puppet-tempest-puppet-lint http://logs.openstack.org/59/448959/1/check/gate-puppet-tempest-puppet-lint/e6ffe80/ : SUCCESS in 1m 55s\n- gate-puppet-tempest-puppet-syntax-4-centos-7 http://logs.openstack.org/59/448959/1/check/gate-puppet-tempest-puppet-syntax-4-centos-7/afb9596/ : SUCCESS in 2m 35s\n- gate-puppet-tempest-puppet-unit-4.5-centos-7 http://logs.openstack.org/59/448959/1/check/gate-puppet-tempest-puppet-unit-4.5-centos-7/5a61b85/ : SUCCESS in 4m 49s\n- gate-puppet-tempest-puppet-unit-latest-centos-7 http://logs.openstack.org/59/448959/1/check/gate-puppet-tempest-puppet-unit-latest-centos-7/8a2078b/ : SUCCESS in 4m 47s (non-voting)\n- gate-puppet-tempest-puppet-beaker-rspec-centos-7 http://logs.openstack.org/59/448959/1/check/gate-puppet-tempest-puppet-beaker-rspec-centos-7/684199d/ : SUCCESS in 9m 16s\n- gate-puppet-tempest-puppet-beaker-rspec-ubuntu-xenial http://logs.openstack.org/59/448959/1/check/gate-puppet-tempest-puppet-beaker-rspec-ubuntu-xenial/acba977/ : FAILURE in 2m 41s\n- gate-puppet-openstack-integration-4-scenario001-tempest-centos-7 http://logs.openstack.org/59/448959/1/check/gate-puppet-openstack-integration-4-scenario001-tempest-centos-7/82deb42/ : SUCCESS in 29m 08s\n- gate-puppet-openstack-integration-4-scenario002-tempest-centos-7 http://logs.openstack.org/59/448959/1/check/gate-puppet-openstack-integration-4-scenario002-tempest-centos-7/ee18e7d/ : SUCCESS in 49m 50s\n- gate-puppet-openstack-integration-4-scenario003-tempest-centos-7 http://logs.openstack.org/59/448959/1/check/gate-puppet-openstack-integration-4-scenario003-tempest-centos-7/6e4baa7/ : SUCCESS in 49m 59s\n- gate-puppet-openstack-integration-4-scenario004-tempest-centos-7 http://logs.openstack.org/59/448959/1/check/gate-puppet-openstack-integration-4-scenario004-tempest-centos-7/99c7e92/ : SUCCESS in 31m 29s\n- gate-puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial http://logs.openstack.org/59/448959/1/check/gate-puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial/c2209af/ : SUCCESS in 25m 16s\n- gate-puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial http://logs.openstack.org/59/448959/1/check/gate-puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial/6eee131/ : SUCCESS in 29m 58s\n- gate-puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial http://logs.openstack.org/59/448959/1/check/gate-puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial/5ed0f8c/ : SUCCESS in 27m 41s\n- gate-puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial http://logs.openstack.org/59/448959/1/check/gate-puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial/a709694/ : SUCCESS in 20m 18s\n- gate-puppet-tempest-releasenotes http://docs-draft.openstack.org/59/448959/1/check/gate-puppet-tempest-releasenotes/e704165//releasenotes/build/html/ : SUCCESS in 35s","accounts_in_message":[],"_revision_number":1},{"id":"94a2d58d4ec82d5ee2d6ac8a1fbb7596451cbe0e","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2017-03-25 04:38:37.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d480bfbfa23f7dcabdfb8b65d4a699736b0f5684","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-03-25 05:31:31.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-puppet-tempest-puppet-lint http://logs.openstack.org/59/448959/1/check/gate-puppet-tempest-puppet-lint/a110da9/ : FAILURE in 1m 09s\n- gate-puppet-tempest-puppet-syntax-4-centos-7 http://logs.openstack.org/59/448959/1/check/gate-puppet-tempest-puppet-syntax-4-centos-7/f80d89d/ : SUCCESS in 6m 38s\n- gate-puppet-tempest-puppet-unit-4.5-centos-7 http://logs.openstack.org/59/448959/1/check/gate-puppet-tempest-puppet-unit-4.5-centos-7/4354ca2/ : FAILURE in 10m 01s\n- gate-puppet-tempest-puppet-unit-latest-centos-7 http://logs.openstack.org/59/448959/1/check/gate-puppet-tempest-puppet-unit-latest-centos-7/4ca7a4c/ : SUCCESS in 12m 40s (non-voting)\n- gate-puppet-tempest-puppet-beaker-rspec-centos-7 http://logs.openstack.org/59/448959/1/check/gate-puppet-tempest-puppet-beaker-rspec-centos-7/62c9e35/ : SUCCESS in 21m 06s\n- gate-puppet-tempest-puppet-beaker-rspec-ubuntu-xenial http://logs.openstack.org/59/448959/1/check/gate-puppet-tempest-puppet-beaker-rspec-ubuntu-xenial/7464293/ : SUCCESS in 4m 48s\n- gate-puppet-openstack-integration-4-scenario001-tempest-centos-7 http://logs.openstack.org/59/448959/1/check/gate-puppet-openstack-integration-4-scenario001-tempest-centos-7/4ed4917/ : SUCCESS in 45m 46s\n- gate-puppet-openstack-integration-4-scenario002-tempest-centos-7 http://logs.openstack.org/59/448959/1/check/gate-puppet-openstack-integration-4-scenario002-tempest-centos-7/dfa2d68/ : SUCCESS in 47m 25s\n- gate-puppet-openstack-integration-4-scenario003-tempest-centos-7 http://logs.openstack.org/59/448959/1/check/gate-puppet-openstack-integration-4-scenario003-tempest-centos-7/8de919a/ : SUCCESS in 51m 08s\n- gate-puppet-openstack-integration-4-scenario004-tempest-centos-7 http://logs.openstack.org/59/448959/1/check/gate-puppet-openstack-integration-4-scenario004-tempest-centos-7/b8ea941/ : FAILURE in 21m 35s\n- gate-puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial http://logs.openstack.org/59/448959/1/check/gate-puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial/f0ccea1/ : SUCCESS in 31m 20s\n- gate-puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial http://logs.openstack.org/59/448959/1/check/gate-puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial/64249dd/ : SUCCESS in 35m 30s\n- gate-puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial http://logs.openstack.org/59/448959/1/check/gate-puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial/67d8c2a/ : SUCCESS in 32m 21s\n- gate-puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial http://logs.openstack.org/59/448959/1/check/gate-puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial/7ce71b7/ : SUCCESS in 26m 46s\n- gate-puppet-tempest-releasenotes http://docs-draft.openstack.org/59/448959/1/check/gate-puppet-tempest-releasenotes/358bb22//releasenotes/build/html/ : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1},{"id":"648bdc15f3c3abf426e5e30c019ec835521b2900","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2017-03-25 15:00:47.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"dcfdd14cb80a056d0aaf8f13c85c252f9db71a3b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-03-25 15:52:52.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-puppet-tempest-puppet-lint http://logs.openstack.org/59/448959/1/check/gate-puppet-tempest-puppet-lint/b6f695d/ : SUCCESS in 1m 29s\n- gate-puppet-tempest-puppet-syntax-4-centos-7 http://logs.openstack.org/59/448959/1/check/gate-puppet-tempest-puppet-syntax-4-centos-7/7f9d0a5/ : SUCCESS in 3m 55s\n- gate-puppet-tempest-puppet-unit-4.5-centos-7 http://logs.openstack.org/59/448959/1/check/gate-puppet-tempest-puppet-unit-4.5-centos-7/289c88b/ : SUCCESS in 4m 44s\n- gate-puppet-tempest-puppet-unit-latest-centos-7 http://logs.openstack.org/59/448959/1/check/gate-puppet-tempest-puppet-unit-latest-centos-7/a348119/ : SUCCESS in 5m 37s (non-voting)\n- gate-puppet-tempest-puppet-beaker-rspec-centos-7 http://logs.openstack.org/59/448959/1/check/gate-puppet-tempest-puppet-beaker-rspec-centos-7/d194376/ : SUCCESS in 11m 24s\n- gate-puppet-tempest-puppet-beaker-rspec-ubuntu-xenial http://logs.openstack.org/59/448959/1/check/gate-puppet-tempest-puppet-beaker-rspec-ubuntu-xenial/454056e/ : SUCCESS in 7m 49s\n- gate-puppet-openstack-integration-4-scenario001-tempest-centos-7 http://logs.openstack.org/59/448959/1/check/gate-puppet-openstack-integration-4-scenario001-tempest-centos-7/174f49b/ : SUCCESS in 48m 47s\n- gate-puppet-openstack-integration-4-scenario002-tempest-centos-7 http://logs.openstack.org/59/448959/1/check/gate-puppet-openstack-integration-4-scenario002-tempest-centos-7/d27ca67/ : SUCCESS in 48m 11s\n- gate-puppet-openstack-integration-4-scenario003-tempest-centos-7 http://logs.openstack.org/59/448959/1/check/gate-puppet-openstack-integration-4-scenario003-tempest-centos-7/a27dda4/ : SUCCESS in 47m 23s\n- gate-puppet-openstack-integration-4-scenario004-tempest-centos-7 http://logs.openstack.org/59/448959/1/check/gate-puppet-openstack-integration-4-scenario004-tempest-centos-7/2f51a19/ : FAILURE in 24m 39s\n- gate-puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial http://logs.openstack.org/59/448959/1/check/gate-puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial/346f46a/ : SUCCESS in 24m 36s\n- gate-puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial http://logs.openstack.org/59/448959/1/check/gate-puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial/4789377/ : SUCCESS in 29m 34s\n- gate-puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial http://logs.openstack.org/59/448959/1/check/gate-puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial/f370cfc/ : SUCCESS in 26m 24s\n- gate-puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial http://logs.openstack.org/59/448959/1/check/gate-puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial/a508109/ : SUCCESS in 21m 20s\n- gate-puppet-tempest-releasenotes http://docs-draft.openstack.org/59/448959/1/check/gate-puppet-tempest-releasenotes/f5148a6//releasenotes/build/html/ : SUCCESS in 1m 23s","accounts_in_message":[],"_revision_number":1},{"id":"23bc53c9e8c2d5b074756bbd8e8ba88abf236536","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2017-03-26 12:22:53.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c911402e5eeb699f6604d3d5af4ddb782dc90972","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-03-26 13:15:28.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-puppet-tempest-puppet-lint http://logs.openstack.org/59/448959/1/check/gate-puppet-tempest-puppet-lint/5bbbb8d/ : SUCCESS in 1m 37s\n- gate-puppet-tempest-puppet-syntax-4-centos-7 http://logs.openstack.org/59/448959/1/check/gate-puppet-tempest-puppet-syntax-4-centos-7/236f75e/ : SUCCESS in 2m 44s\n- gate-puppet-tempest-puppet-unit-4.5-centos-7 http://logs.openstack.org/59/448959/1/check/gate-puppet-tempest-puppet-unit-4.5-centos-7/9cb1dbd/ : SUCCESS in 4m 40s\n- gate-puppet-tempest-puppet-unit-latest-centos-7 http://logs.openstack.org/59/448959/1/check/gate-puppet-tempest-puppet-unit-latest-centos-7/1cb65f3/ : SUCCESS in 4m 33s (non-voting)\n- gate-puppet-tempest-puppet-beaker-rspec-centos-7 http://logs.openstack.org/59/448959/1/check/gate-puppet-tempest-puppet-beaker-rspec-centos-7/7545e9c/ : SUCCESS in 10m 00s\n- gate-puppet-tempest-puppet-beaker-rspec-ubuntu-xenial http://logs.openstack.org/59/448959/1/check/gate-puppet-tempest-puppet-beaker-rspec-ubuntu-xenial/861516e/ : SUCCESS in 4m 10s\n- gate-puppet-openstack-integration-4-scenario001-tempest-centos-7 http://logs.openstack.org/59/448959/1/check/gate-puppet-openstack-integration-4-scenario001-tempest-centos-7/a445603/ : SUCCESS in 46m 12s\n- gate-puppet-openstack-integration-4-scenario002-tempest-centos-7 http://logs.openstack.org/59/448959/1/check/gate-puppet-openstack-integration-4-scenario002-tempest-centos-7/78b4933/ : SUCCESS in 39m 26s\n- gate-puppet-openstack-integration-4-scenario003-tempest-centos-7 http://logs.openstack.org/59/448959/1/check/gate-puppet-openstack-integration-4-scenario003-tempest-centos-7/e534777/ : SUCCESS in 48m 49s\n- gate-puppet-openstack-integration-4-scenario004-tempest-centos-7 http://logs.openstack.org/59/448959/1/check/gate-puppet-openstack-integration-4-scenario004-tempest-centos-7/0b6dfee/ : FAILURE in 21m 59s\n- gate-puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial http://logs.openstack.org/59/448959/1/check/gate-puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial/6d397a2/ : SUCCESS in 25m 02s\n- gate-puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial http://logs.openstack.org/59/448959/1/check/gate-puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial/7158706/ : SUCCESS in 44m 57s\n- gate-puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial http://logs.openstack.org/59/448959/1/check/gate-puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial/934dd5b/ : SUCCESS in 25m 47s\n- gate-puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial http://logs.openstack.org/59/448959/1/check/gate-puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial/1f3330a/ : SUCCESS in 21m 02s\n- gate-puppet-tempest-releasenotes http://docs-draft.openstack.org/59/448959/1/check/gate-puppet-tempest-releasenotes/71ebcc0//releasenotes/build/html/ : SUCCESS in 24s","accounts_in_message":[],"_revision_number":1},{"id":"37d48f0c82d7019b98ccb5403ede2b53e4431c73","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2017-03-27 01:12:56.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3d1bf880eaf22858317c1a539cd2533352e21716","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2017-03-27 01:13:07.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"3589eba64f6127866789b95311c6302198764bbf","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-03-27 02:05:38.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-puppet-tempest-puppet-lint http://logs.openstack.org/59/448959/1/check/gate-puppet-tempest-puppet-lint/5512716/ : SUCCESS in 2m 37s\n- gate-puppet-tempest-puppet-syntax-4-centos-7 http://logs.openstack.org/59/448959/1/check/gate-puppet-tempest-puppet-syntax-4-centos-7/3fc6f04/ : SUCCESS in 3m 08s\n- gate-puppet-tempest-puppet-unit-4.5-centos-7 http://logs.openstack.org/59/448959/1/check/gate-puppet-tempest-puppet-unit-4.5-centos-7/54e2286/ : SUCCESS in 13m 11s\n- gate-puppet-tempest-puppet-unit-latest-centos-7 http://logs.openstack.org/59/448959/1/check/gate-puppet-tempest-puppet-unit-latest-centos-7/983a06e/ : SUCCESS in 6m 32s (non-voting)\n- gate-puppet-tempest-puppet-beaker-rspec-centos-7 http://logs.openstack.org/59/448959/1/check/gate-puppet-tempest-puppet-beaker-rspec-centos-7/4e36126/ : SUCCESS in 10m 11s\n- gate-puppet-tempest-puppet-beaker-rspec-ubuntu-xenial http://logs.openstack.org/59/448959/1/check/gate-puppet-tempest-puppet-beaker-rspec-ubuntu-xenial/ba3d307/ : SUCCESS in 4m 23s\n- gate-puppet-openstack-integration-4-scenario001-tempest-centos-7 http://logs.openstack.org/59/448959/1/check/gate-puppet-openstack-integration-4-scenario001-tempest-centos-7/f6fb29e/ : SUCCESS in 48m 17s\n- gate-puppet-openstack-integration-4-scenario002-tempest-centos-7 http://logs.openstack.org/59/448959/1/check/gate-puppet-openstack-integration-4-scenario002-tempest-centos-7/f1874d7/ : SUCCESS in 49m 28s\n- gate-puppet-openstack-integration-4-scenario003-tempest-centos-7 http://logs.openstack.org/59/448959/1/check/gate-puppet-openstack-integration-4-scenario003-tempest-centos-7/94849df/ : SUCCESS in 49m 25s\n- gate-puppet-openstack-integration-4-scenario004-tempest-centos-7 http://logs.openstack.org/59/448959/1/check/gate-puppet-openstack-integration-4-scenario004-tempest-centos-7/62c49dc/ : FAILURE in 23m 03s\n- gate-puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial http://logs.openstack.org/59/448959/1/check/gate-puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial/508c24e/ : SUCCESS in 30m 11s\n- gate-puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial http://logs.openstack.org/59/448959/1/check/gate-puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial/d949259/ : SUCCESS in 41m 37s\n- gate-puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial http://logs.openstack.org/59/448959/1/check/gate-puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial/29f297e/ : SUCCESS in 33m 56s\n- gate-puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial http://logs.openstack.org/59/448959/1/check/gate-puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial/15bc1b0/ : SUCCESS in 26m 46s\n- gate-puppet-tempest-releasenotes http://docs-draft.openstack.org/59/448959/1/check/gate-puppet-tempest-releasenotes/88392c7//releasenotes/build/html/ : SUCCESS in 51s","accounts_in_message":[],"_revision_number":1},{"id":"5e763bcb028448fb20f77e7f7e67cbd9a7f136dc","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2017-03-27 14:49:40.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"917e9c47c826bbc82c24999572ca94c9a6c3a2dc","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-03-27 15:42:33.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-puppet-tempest-puppet-lint http://logs.openstack.org/59/448959/1/check/gate-puppet-tempest-puppet-lint/280531c/ : SUCCESS in 3m 05s\n- gate-puppet-tempest-puppet-syntax-4-centos-7 http://logs.openstack.org/59/448959/1/check/gate-puppet-tempest-puppet-syntax-4-centos-7/29c3e39/ : SUCCESS in 2m 55s\n- gate-puppet-tempest-puppet-unit-4.5-centos-7 http://logs.openstack.org/59/448959/1/check/gate-puppet-tempest-puppet-unit-4.5-centos-7/589f084/ : SUCCESS in 5m 50s\n- gate-puppet-tempest-puppet-unit-latest-centos-7 http://logs.openstack.org/59/448959/1/check/gate-puppet-tempest-puppet-unit-latest-centos-7/757d9e6/ : SUCCESS in 5m 18s (non-voting)\n- gate-puppet-tempest-puppet-beaker-rspec-centos-7 http://logs.openstack.org/59/448959/1/check/gate-puppet-tempest-puppet-beaker-rspec-centos-7/d6476f5/ : SUCCESS in 14m 02s\n- gate-puppet-tempest-puppet-beaker-rspec-ubuntu-xenial http://logs.openstack.org/59/448959/1/check/gate-puppet-tempest-puppet-beaker-rspec-ubuntu-xenial/bff0650/ : SUCCESS in 4m 49s\n- gate-puppet-openstack-integration-4-scenario001-tempest-centos-7 http://logs.openstack.org/59/448959/1/check/gate-puppet-openstack-integration-4-scenario001-tempest-centos-7/40e0304/ : SUCCESS in 45m 42s\n- gate-puppet-openstack-integration-4-scenario002-tempest-centos-7 http://logs.openstack.org/59/448959/1/check/gate-puppet-openstack-integration-4-scenario002-tempest-centos-7/a1e2280/ : SUCCESS in 49m 07s\n- gate-puppet-openstack-integration-4-scenario003-tempest-centos-7 http://logs.openstack.org/59/448959/1/check/gate-puppet-openstack-integration-4-scenario003-tempest-centos-7/5580e59/ : SUCCESS in 47m 38s\n- gate-puppet-openstack-integration-4-scenario004-tempest-centos-7 http://logs.openstack.org/59/448959/1/check/gate-puppet-openstack-integration-4-scenario004-tempest-centos-7/9966333/ : FAILURE in 22m 09s\n- gate-puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial http://logs.openstack.org/59/448959/1/check/gate-puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial/9fec50e/ : SUCCESS in 29m 37s\n- gate-puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial http://logs.openstack.org/59/448959/1/check/gate-puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial/3425be7/ : SUCCESS in 32m 43s\n- gate-puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial http://logs.openstack.org/59/448959/1/check/gate-puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial/5d95df8/ : SUCCESS in 30m 54s\n- gate-puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial http://logs.openstack.org/59/448959/1/check/gate-puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial/4f1f43e/ : SUCCESS in 21m 53s\n- gate-puppet-tempest-releasenotes http://docs-draft.openstack.org/59/448959/1/check/gate-puppet-tempest-releasenotes/0a41399//releasenotes/build/html/ : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":1},{"id":"a6ff46481f70f6d9333899e23a9934cd9c98c18d","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2017-03-28 00:58:09.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8c5502586366e9915f1de7a05cb56b3592d8e0d3","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-03-28 01:48:01.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-puppet-tempest-puppet-lint http://logs.openstack.org/59/448959/1/check/gate-puppet-tempest-puppet-lint/a1d202b/ : SUCCESS in 1m 39s\n- gate-puppet-tempest-puppet-syntax-4-centos-7 http://logs.openstack.org/59/448959/1/check/gate-puppet-tempest-puppet-syntax-4-centos-7/2c93911/ : SUCCESS in 3m 41s\n- gate-puppet-tempest-puppet-unit-4.5-centos-7 http://logs.openstack.org/59/448959/1/check/gate-puppet-tempest-puppet-unit-4.5-centos-7/eea49c9/ : SUCCESS in 5m 41s\n- gate-puppet-tempest-puppet-unit-latest-centos-7 http://logs.openstack.org/59/448959/1/check/gate-puppet-tempest-puppet-unit-latest-centos-7/78d6414/ : SUCCESS in 5m 32s (non-voting)\n- gate-puppet-tempest-puppet-beaker-rspec-centos-7 http://logs.openstack.org/59/448959/1/check/gate-puppet-tempest-puppet-beaker-rspec-centos-7/7c69a9d/ : SUCCESS in 10m 47s\n- gate-puppet-tempest-puppet-beaker-rspec-ubuntu-xenial http://logs.openstack.org/59/448959/1/check/gate-puppet-tempest-puppet-beaker-rspec-ubuntu-xenial/ae069e2/ : SUCCESS in 5m 52s\n- gate-puppet-openstack-integration-4-scenario001-tempest-centos-7 http://logs.openstack.org/59/448959/1/check/gate-puppet-openstack-integration-4-scenario001-tempest-centos-7/bd758e6/ : SUCCESS in 42m 56s\n- gate-puppet-openstack-integration-4-scenario002-tempest-centos-7 http://logs.openstack.org/59/448959/1/check/gate-puppet-openstack-integration-4-scenario002-tempest-centos-7/14fa42a/ : SUCCESS in 43m 15s\n- gate-puppet-openstack-integration-4-scenario003-tempest-centos-7 http://logs.openstack.org/59/448959/1/check/gate-puppet-openstack-integration-4-scenario003-tempest-centos-7/cd3597d/ : SUCCESS in 46m 49s\n- gate-puppet-openstack-integration-4-scenario004-tempest-centos-7 http://logs.openstack.org/59/448959/1/check/gate-puppet-openstack-integration-4-scenario004-tempest-centos-7/fe6227b/ : FAILURE in 18m 37s\n- gate-puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial http://logs.openstack.org/59/448959/1/check/gate-puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial/ccdd217/ : SUCCESS in 31m 59s\n- gate-puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial http://logs.openstack.org/59/448959/1/check/gate-puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial/a07b0fd/ : SUCCESS in 41m 35s\n- gate-puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial http://logs.openstack.org/59/448959/1/check/gate-puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial/21508a1/ : SUCCESS in 34m 44s\n- gate-puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial http://logs.openstack.org/59/448959/1/check/gate-puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial/ae706c4/ : SUCCESS in 27m 31s\n- gate-puppet-tempest-releasenotes http://docs-draft.openstack.org/59/448959/1/check/gate-puppet-tempest-releasenotes/f1e5aee//releasenotes/build/html/ : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1},{"id":"8eef0e47db99cc81d480a33930ac97f1537d9206","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2017-03-29 01:22:49.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"bc825870f957ae38f3f66f75a90d939cfca7ed70","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-03-29 02:12:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-puppet-tempest-puppet-lint http://logs.openstack.org/59/448959/1/check/gate-puppet-tempest-puppet-lint/54c61fa/ : SUCCESS in 1m 31s\n- gate-puppet-tempest-puppet-syntax-4-centos-7 http://logs.openstack.org/59/448959/1/check/gate-puppet-tempest-puppet-syntax-4-centos-7/56cd397/ : SUCCESS in 2m 49s\n- gate-puppet-tempest-puppet-unit-4.5-centos-7 http://logs.openstack.org/59/448959/1/check/gate-puppet-tempest-puppet-unit-4.5-centos-7/30fb28f/ : SUCCESS in 5m 18s\n- gate-puppet-tempest-puppet-unit-latest-centos-7 http://logs.openstack.org/59/448959/1/check/gate-puppet-tempest-puppet-unit-latest-centos-7/0519ef1/ : SUCCESS in 5m 13s (non-voting)\n- gate-puppet-tempest-puppet-beaker-rspec-centos-7 http://logs.openstack.org/59/448959/1/check/gate-puppet-tempest-puppet-beaker-rspec-centos-7/b9a226f/ : SUCCESS in 11m 05s\n- gate-puppet-tempest-puppet-beaker-rspec-ubuntu-xenial http://logs.openstack.org/59/448959/1/check/gate-puppet-tempest-puppet-beaker-rspec-ubuntu-xenial/d371470/ : SUCCESS in 3m 33s\n- gate-puppet-openstack-integration-4-scenario001-tempest-centos-7 http://logs.openstack.org/59/448959/1/check/gate-puppet-openstack-integration-4-scenario001-tempest-centos-7/f4cf971/ : SUCCESS in 45m 56s\n- gate-puppet-openstack-integration-4-scenario002-tempest-centos-7 http://logs.openstack.org/59/448959/1/check/gate-puppet-openstack-integration-4-scenario002-tempest-centos-7/0544d6d/ : SUCCESS in 45m 28s\n- gate-puppet-openstack-integration-4-scenario003-tempest-centos-7 http://logs.openstack.org/59/448959/1/check/gate-puppet-openstack-integration-4-scenario003-tempest-centos-7/6288247/ : SUCCESS in 41m 32s\n- gate-puppet-openstack-integration-4-scenario004-tempest-centos-7 http://logs.openstack.org/59/448959/1/check/gate-puppet-openstack-integration-4-scenario004-tempest-centos-7/30c70c2/ : SUCCESS in 28m 49s\n- gate-puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial http://logs.openstack.org/59/448959/1/check/gate-puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial/847f347/ : SUCCESS in 24m 54s\n- gate-puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial http://logs.openstack.org/59/448959/1/check/gate-puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial/bb5fd44/ : SUCCESS in 40m 34s\n- gate-puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial http://logs.openstack.org/59/448959/1/check/gate-puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial/605b438/ : SUCCESS in 25m 39s\n- gate-puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial http://logs.openstack.org/59/448959/1/check/gate-puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial/6e64b25/ : SUCCESS in 26m 52s\n- gate-puppet-tempest-releasenotes http://docs-draft.openstack.org/59/448959/1/check/gate-puppet-tempest-releasenotes/79425e8//releasenotes/build/html/ : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1},{"id":"f9281956ec95f62cba6e0209ce061ab170c56954","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-03-29 02:13:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"62ea3f53d82174975491acf3e84065601b519b6c","date":"2017-03-29 03:24:20.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":1},{"id":"13416440d6d0e521f745252d910f52edb27760f4","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-03-29 03:24:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-puppet-tempest-puppet-lint http://logs.openstack.org/59/448959/1/gate/gate-puppet-tempest-puppet-lint/b045780/ : SUCCESS in 1m 30s\n- gate-puppet-tempest-puppet-syntax-4-centos-7 http://logs.openstack.org/59/448959/1/gate/gate-puppet-tempest-puppet-syntax-4-centos-7/ae2b21f/ : SUCCESS in 2m 26s\n- gate-puppet-tempest-puppet-unit-4.5-centos-7 http://logs.openstack.org/59/448959/1/gate/gate-puppet-tempest-puppet-unit-4.5-centos-7/892ff1a/ : SUCCESS in 4m 07s\n- gate-puppet-tempest-puppet-beaker-rspec-centos-7 http://logs.openstack.org/59/448959/1/gate/gate-puppet-tempest-puppet-beaker-rspec-centos-7/0b5fbb7/ : SUCCESS in 7m 41s\n- gate-puppet-tempest-puppet-beaker-rspec-ubuntu-xenial http://logs.openstack.org/59/448959/1/gate/gate-puppet-tempest-puppet-beaker-rspec-ubuntu-xenial/7e790c7/ : SUCCESS in 3m 28s\n- gate-puppet-openstack-integration-4-scenario001-tempest-centos-7 http://logs.openstack.org/59/448959/1/gate/gate-puppet-openstack-integration-4-scenario001-tempest-centos-7/8d60daf/ : SUCCESS in 34m 28s\n- gate-puppet-openstack-integration-4-scenario002-tempest-centos-7 http://logs.openstack.org/59/448959/1/gate/gate-puppet-openstack-integration-4-scenario002-tempest-centos-7/1f31ce2/ : SUCCESS in 44m 45s\n- gate-puppet-openstack-integration-4-scenario003-tempest-centos-7 http://logs.openstack.org/59/448959/1/gate/gate-puppet-openstack-integration-4-scenario003-tempest-centos-7/37c4e8b/ : SUCCESS in 35m 33s\n- gate-puppet-openstack-integration-4-scenario004-tempest-centos-7 http://logs.openstack.org/59/448959/1/gate/gate-puppet-openstack-integration-4-scenario004-tempest-centos-7/956b201/ : SUCCESS in 23m 35s\n- gate-puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial http://logs.openstack.org/59/448959/1/gate/gate-puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial/75cd3a5/ : SUCCESS in 23m 42s\n- gate-puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial http://logs.openstack.org/59/448959/1/gate/gate-puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial/d42e8e8/ : SUCCESS in 28m 39s\n- gate-puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial http://logs.openstack.org/59/448959/1/gate/gate-puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial/e19d833/ : SUCCESS in 36m 04s\n- gate-puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial http://logs.openstack.org/59/448959/1/gate/gate-puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial/7464f64/ : SUCCESS in 28m 14s\n- gate-puppet-tempest-releasenotes http://docs-draft.openstack.org/59/448959/1/gate/gate-puppet-tempest-releasenotes/1785c1f//releasenotes/build/html/ : SUCCESS in 27s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cb8d37b0b420e66e5a4435598c8d573a05c13502","revisions":{"cb8d37b0b420e66e5a4435598c8d573a05c13502":{"kind":"REWORK","_number":1,"created":"2017-03-23 06:55:16.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/59/448959/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tempest","ref":"refs/changes/59/448959/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tempest refs/changes/59/448959/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tempest refs/changes/59/448959/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tempest refs/changes/59/448959/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tempest refs/changes/59/448959/1"}}},"commit":{"parents":[{"commit":"a93b9331509238fb16556a4b3688ac54ccd388d5","subject":"Allow to set api_extensions to test for neutron","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tempest/commit/a93b9331509238fb16556a4b3688ac54ccd388d5"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2017-03-23 06:55:09.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2017-03-23 06:55:09.000000000","tz":480},"subject":"Update test-requirements.txt","message":"Update test-requirements.txt\n\nThis patch brings it in line with the latest global requirements.\nBy being in line with global requirements, this is a more accurate\nlist with respect to how things are actually tested in the gate.\nThe following links[0][1][2] show the latest reno, oslosphinx, sphinx\npackages.\n\n[0]https://github.com/openstack/requirements/blob/master/global-requirements.txt#L263\n[1]https://github.com/openstack/requirements/blob/master/global-requirements.txt#L372\n[2]https://github.com/openstack/requirements/blob/master/global-requirements.txt#L395\n\nChange-Id: I93329bfa84d06aeb1cb9e1f83cf3ae28e2dc33d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tempest/commit/cb8d37b0b420e66e5a4435598c8d573a05c13502"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tempest/commit/cb8d37b0b420e66e5a4435598c8d573a05c13502"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"a93b9331509238fb16556a4b3688ac54ccd388d5","is_merged_in_target_branch":true,"change_id":"I31105600dfd5632ec4341b35ca3e8a13af3bb945","change_number":446109,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
