)]}'
{"id":"openstack%2Fpuppet-tempest~484438","triplet_id":"openstack%2Fpuppet-tempest~master~I5671813a52e56da9ee858d0f1cac177b7d20b1d7","project":"openstack/puppet-tempest","branch":"master","topic":"fix-tempest","hashtags":[],"change_id":"I5671813a52e56da9ee858d0f1cac177b7d20b1d7","subject":"Fixed the tempest package name for neutron l2gw plugin","status":"MERGED","created":"2017-07-17 17:54:34.000000000","updated":"2017-07-18 05:56:15.000000000","submitted":"2017-07-18 05:56:15.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"25ba8b61ca5e99a913653b043eda2ac838c458e1","_number":484438,"virtual_id_number":484438,"owner":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2017-07-18 05:56:15.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":21626,"name":"Ricardo","email":"rnoriega@redhat.com","username":"rnoriega"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2017-07-17 17:56:51.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2017-07-17 18:21:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2017-07-17 20:54:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21626,"name":"Ricardo","email":"rnoriega@redhat.com","username":"rnoriega"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2017-07-17 18:21:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":21626,"name":"Ricardo","email":"rnoriega@redhat.com","username":"rnoriega"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":21626,"name":"Ricardo","email":"rnoriega@redhat.com","username":"rnoriega"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-07-17 17:56:51.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2017-07-17 18:21:50.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2017-07-17 20:54:14.000000000","updated_by":{"_account_id":21626,"name":"Ricardo","email":"rnoriega@redhat.com","username":"rnoriega"},"reviewer":{"_account_id":21626,"name":"Ricardo","email":"rnoriega@redhat.com","username":"rnoriega"},"state":"REVIEWER"},{"updated":"2017-07-18 05:56:15.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"6281555871b15cc6500cd28d91dd00a9f2c796ae","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2017-07-17 17:54:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"07c9db0c1363ae8093d02646d267c4301085c7d3","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2017-07-17 17:56:51.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4441a3e299d75a6d0ceb18b8fea3fa481e57894a","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2017-07-17 18:21:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d64ba8f4160ff7ba3fad18f5642b2417a95e6e86","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-07-17 18:56:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-puppet-tempest-puppet-lint http://logs.openstack.org/38/484438/1/check/gate-puppet-tempest-puppet-lint/be5e044/ : SUCCESS in 1m 41s\n- gate-puppet-tempest-puppet-syntax-4-centos-7 http://logs.openstack.org/38/484438/1/check/gate-puppet-tempest-puppet-syntax-4-centos-7/7b0bd89/ : SUCCESS in 2m 39s\n- gate-puppet-tempest-puppet-unit-4.5-centos-7 http://logs.openstack.org/38/484438/1/check/gate-puppet-tempest-puppet-unit-4.5-centos-7/bb2b8df/ : SUCCESS in 8m 07s\n- gate-puppet-tempest-puppet-unit-latest-centos-7 http://logs.openstack.org/38/484438/1/check/gate-puppet-tempest-puppet-unit-latest-centos-7/f935ed5/ : SUCCESS in 4m 26s (non-voting)\n- gate-puppet-tempest-puppet-beaker-rspec-centos-7 http://logs.openstack.org/38/484438/1/check/gate-puppet-tempest-puppet-beaker-rspec-centos-7/e03d475/ : SUCCESS in 8m 43s\n- gate-puppet-tempest-puppet-beaker-rspec-ubuntu-xenial http://logs.openstack.org/38/484438/1/check/gate-puppet-tempest-puppet-beaker-rspec-ubuntu-xenial/4ec5afc/ : SUCCESS in 5m 41s\n- gate-puppet-openstack-integration-4-scenario001-tempest-centos-7 http://logs.openstack.org/38/484438/1/check/gate-puppet-openstack-integration-4-scenario001-tempest-centos-7/8305432/ : SUCCESS in 37m 40s\n- gate-puppet-openstack-integration-4-scenario002-tempest-centos-7 http://logs.openstack.org/38/484438/1/check/gate-puppet-openstack-integration-4-scenario002-tempest-centos-7/2218289/ : SUCCESS in 50m 22s\n- gate-puppet-openstack-integration-4-scenario003-tempest-centos-7 http://logs.openstack.org/38/484438/1/check/gate-puppet-openstack-integration-4-scenario003-tempest-centos-7/bf74f58/ : SUCCESS in 1h 00m 16s\n- gate-puppet-openstack-integration-4-scenario004-tempest-centos-7 http://logs.openstack.org/38/484438/1/check/gate-puppet-openstack-integration-4-scenario004-tempest-centos-7/19ff97a/ : SUCCESS in 28m 33s\n- gate-puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial http://logs.openstack.org/38/484438/1/check/gate-puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial/db1b98a/ : SUCCESS in 27m 03s\n- gate-puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial http://logs.openstack.org/38/484438/1/check/gate-puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial/4fc4de4/ : SUCCESS in 42m 48s\n- gate-puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial http://logs.openstack.org/38/484438/1/check/gate-puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial/b7de599/ : SUCCESS in 26m 24s\n- gate-puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial http://logs.openstack.org/38/484438/1/check/gate-puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial/70e7928/ : SUCCESS in 31m 46s","accounts_in_message":[],"_revision_number":1},{"id":"3f9447b53ad3d4d5fa156977d8926c03623a1e94","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-07-17 18:56:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"91e2ead978152e409a65aafaf32857c0cadf95c4","author":{"_account_id":21626,"name":"Ricardo","email":"rnoriega@redhat.com","username":"rnoriega"},"date":"2017-07-17 20:54:05.000000000","message":"Patch Set 1:\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"202d5c25f384ec9fd2b28ec104f59f27a3714da8","author":{"_account_id":21626,"name":"Ricardo","email":"rnoriega@redhat.com","username":"rnoriega"},"date":"2017-07-17 20:54:14.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"44b6c9dbbcd5647d5d78492cfeedddc06f5dd4b9","date":"2017-07-18 05:56:15.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":1},{"id":"7b25ba7f207d64d41a7eeaf456a7c496f30176d4","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-07-18 05:56:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-puppet-tempest-puppet-lint http://logs.openstack.org/38/484438/1/gate/gate-puppet-tempest-puppet-lint/be49ac3/ : SUCCESS in 2m 51s\n- gate-puppet-tempest-puppet-syntax-4-centos-7 http://logs.openstack.org/38/484438/1/gate/gate-puppet-tempest-puppet-syntax-4-centos-7/b684658/ : SUCCESS in 2m 25s\n- gate-puppet-tempest-puppet-unit-4.5-centos-7 http://logs.openstack.org/38/484438/1/gate/gate-puppet-tempest-puppet-unit-4.5-centos-7/2d520bb/ : SUCCESS in 7m 37s\n- gate-puppet-tempest-puppet-beaker-rspec-centos-7 http://logs.openstack.org/38/484438/1/gate/gate-puppet-tempest-puppet-beaker-rspec-centos-7/1eb2b3c/ : SUCCESS in 9m 13s\n- gate-puppet-tempest-puppet-beaker-rspec-ubuntu-xenial http://logs.openstack.org/38/484438/1/gate/gate-puppet-tempest-puppet-beaker-rspec-ubuntu-xenial/efba66e/ : SUCCESS in 4m 50s\n- gate-puppet-openstack-integration-4-scenario001-tempest-centos-7 http://logs.openstack.org/38/484438/1/gate/gate-puppet-openstack-integration-4-scenario001-tempest-centos-7/08dd1d9/ : SUCCESS in 37m 24s\n- gate-puppet-openstack-integration-4-scenario002-tempest-centos-7 http://logs.openstack.org/38/484438/1/gate/gate-puppet-openstack-integration-4-scenario002-tempest-centos-7/734b1aa/ : SUCCESS in 41m 02s\n- gate-puppet-openstack-integration-4-scenario003-tempest-centos-7 http://logs.openstack.org/38/484438/1/gate/gate-puppet-openstack-integration-4-scenario003-tempest-centos-7/ba73927/ : SUCCESS in 53m 53s\n- gate-puppet-openstack-integration-4-scenario004-tempest-centos-7 http://logs.openstack.org/38/484438/1/gate/gate-puppet-openstack-integration-4-scenario004-tempest-centos-7/ff48b2f/ : SUCCESS in 35m 13s\n- gate-puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial http://logs.openstack.org/38/484438/1/gate/gate-puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial/b6a06f3/ : SUCCESS in 25m 22s\n- gate-puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial http://logs.openstack.org/38/484438/1/gate/gate-puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial/b7be7f5/ : SUCCESS in 30m 36s\n- gate-puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial http://logs.openstack.org/38/484438/1/gate/gate-puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial/6d1dd92/ : SUCCESS in 25m 34s\n- gate-puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial http://logs.openstack.org/38/484438/1/gate/gate-puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial/b1b855a/ : SUCCESS in 22m 32s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"dc78cae90ce8fe7fd6f4f28dceed4b13143adf33","revisions":{"dc78cae90ce8fe7fd6f4f28dceed4b13143adf33":{"kind":"REWORK","_number":1,"created":"2017-07-17 17:54:34.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/38/484438/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tempest","ref":"refs/changes/38/484438/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tempest refs/changes/38/484438/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tempest refs/changes/38/484438/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tempest refs/changes/38/484438/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tempest refs/changes/38/484438/1"}}},"commit":{"parents":[{"commit":"58a49e092c63176ba6021c175dd61cdead3f248b","subject":"Merge \"Change author in setup.cfg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tempest/commit/58a49e092c63176ba6021c175dd61cdead3f248b"}]}],"author":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2017-07-17 17:53:13.000000000","tz":330},"committer":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2017-07-17 17:54:30.000000000","tz":330},"subject":"Fixed the tempest package name for neutron l2gw plugin","message":"Fixed the tempest package name for neutron l2gw plugin\n\nThe correct package name for neutron l2gw plugin is\npython-networking-l2gw-tests-tempest.\n\nChange-Id: I5671813a52e56da9ee858d0f1cac177b7d20b1d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tempest/commit/dc78cae90ce8fe7fd6f4f28dceed4b13143adf33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tempest/commit/dc78cae90ce8fe7fd6f4f28dceed4b13143adf33"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"58a49e092c63176ba6021c175dd61cdead3f248b","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
