)]}'
{"id":"openstack%2Fpuppet-tempest~489068","triplet_id":"openstack%2Fpuppet-tempest~master~I707ee3cfa2cb89f555069e0083564af07a7fd617","project":"openstack/puppet-tempest","branch":"master","topic":"update_requirements","hashtags":[],"change_id":"I707ee3cfa2cb89f555069e0083564af07a7fd617","subject":"Update openstackdocstheme\u003e\u003d1.16.0","status":"MERGED","created":"2017-07-31 03:03:43.000000000","updated":"2017-08-05 07:18:50.000000000","submitted":"2017-08-05 07:18:50.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ddf717b067cfd0e5da894f8f08dc8bc9e0f2fd4d","_number":489068,"virtual_id_number":489068,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":2,"date":"2017-08-05 07:18:50.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2017-08-04 22:40:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2017-08-04 22:40:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-08-04 22:40:12.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2017-08-05 07:18:50.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"dfb9bfc1b229f66bd80d5ff67ddad05421b9ebfb","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2017-07-31 03:03:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ac6be5a3f4953d05d8ff8988d76f83253aa9b290","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-07-31 03:06:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-puppet-tempest-puppet-lint http://logs.openstack.org/68/489068/1/check/gate-puppet-tempest-puppet-lint/c22105c/ : SUCCESS in 1m 45s\n- gate-puppet-tempest-releasenotes http://docs-draft.openstack.org/68/489068/1/check/gate-puppet-tempest-releasenotes/7139c67//releasenotes/build/html/ : SUCCESS in 26s","accounts_in_message":[],"_revision_number":1},{"id":"dc8f952d02928266e958c79e500de28a2d380232","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2017-08-04 22:40:12.000000000","message":"Patch Set 1: Workflow+1 Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"24d64687fb40866e9aea9e6f6783698036277a92","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-08-04 22:40:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"e9c4ac9623398aed40746f69890a9e819ad15916","date":"2017-08-05 07:18:50.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":1},{"id":"7f723f211eff5c1da75bb5b90e6cf6885f5fafc6","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-08-05 07:18:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-puppet-tempest-puppet-lint http://logs.openstack.org/68/489068/1/gate/gate-puppet-tempest-puppet-lint/5498904/ : SUCCESS in 2m 59s\n- gate-puppet-tempest-releasenotes http://docs-draft.openstack.org/68/489068/1/gate/gate-puppet-tempest-releasenotes/5fa74b6//releasenotes/build/html/ : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"abbde5981fe4682e19aed2b1fb671ffcdb8df649","revisions":{"abbde5981fe4682e19aed2b1fb671ffcdb8df649":{"kind":"REWORK","_number":1,"created":"2017-07-31 03:03:43.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/68/489068/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tempest","ref":"refs/changes/68/489068/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tempest refs/changes/68/489068/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tempest refs/changes/68/489068/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tempest refs/changes/68/489068/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tempest refs/changes/68/489068/1"}}},"commit":{"parents":[{"commit":"a5c3f4f0902e1d98a57cb9934562a43718f8c612","subject":"Drop -dev version for Pike M3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tempest/commit/a5c3f4f0902e1d98a57cb9934562a43718f8c612"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2017-07-31 03:04:10.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2017-07-31 03:04:10.000000000","tz":480},"subject":"Update openstackdocstheme\u003e\u003d1.16.0","message":"Update openstackdocstheme\u003e\u003d1.16.0\n\nChange-Id: I707ee3cfa2cb89f555069e0083564af07a7fd617\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tempest/commit/abbde5981fe4682e19aed2b1fb671ffcdb8df649"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tempest/commit/abbde5981fe4682e19aed2b1fb671ffcdb8df649"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"a5c3f4f0902e1d98a57cb9934562a43718f8c612","is_merged_in_target_branch":true,"change_id":"I32d75526d49f815cb3b52272b3f15dfeb42ec23c","change_number":487096,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
