)]}'
{"id":"openstack%2Fpuppet-tempest~512846","triplet_id":"openstack%2Fpuppet-tempest~master~Icbded3288de998f6e970e4a0597aaa1648eddb90","project":"openstack/puppet-tempest","branch":"master","topic":"prep-release-queens-1","hashtags":[],"change_id":"Icbded3288de998f6e970e4a0597aaa1648eddb90","subject":"Prepare for Queens Milestone 1","status":"MERGED","created":"2017-10-17 20:29:06.000000000","updated":"2017-10-28 01:00:13.000000000","submitted":"2017-10-28 01:00:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"512846-1509152413978-ca78a3df","meta_rev_id":"c33a180288849db032779fbbfff50f5b234136d6","_number":512846,"virtual_id_number":512846,"owner":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2017-10-19 01:55:56.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2017-10-28 01:00:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2017-10-18 02:21:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2017-10-18 02:01:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2017-10-18 02:21:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"date":"2017-10-19 03:48:17.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-10-18 02:01:57.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2017-10-18 02:21:08.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"},{"updated":"2017-10-19 01:55:56.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2017-10-28 01:00:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d3bb627cb0b73652ded084b564a07b207f8c35bb","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2017-10-17 20:29:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"674479e8108caa88d75253c07076df1e91dd5b6b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-17 20:59:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed.\n\n- puppet-openstack-lint http://logs.openstack.org/46/512846/1/check/puppet-openstack-lint/04ce442/ : SUCCESS in 5m 19s\n- puppet-openstack-syntax-4 http://logs.openstack.org/46/512846/1/check/puppet-openstack-syntax-4/2206667/ : SUCCESS in 3m 59s\n- puppet-openstack-module-build http://logs.openstack.org/46/512846/1/check/puppet-openstack-module-build/01b4512/ : FAILURE in 2m 51s\n- legacy-puppet-unit-4.8-centos-7 http://logs.openstack.org/46/512846/1/check/legacy-puppet-unit-4.8-centos-7/514c768/ : SUCCESS in 7m 02s\n- legacy-puppet-unit-latest-centos-7 http://logs.openstack.org/46/512846/1/check/legacy-puppet-unit-latest-centos-7/2643bc5/ : SUCCESS in 6m 31s (non-voting)\n- legacy-puppet-beaker-rspec-centos-7 http://logs.openstack.org/46/512846/1/check/legacy-puppet-beaker-rspec-centos-7/d47f92d/ : SUCCESS in 15m 27s\n- legacy-puppet-beaker-rspec http://logs.openstack.org/46/512846/1/check/legacy-puppet-beaker-rspec/444cd10/ : SUCCESS in 6m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"081b8882990093e99139ca6b2a9fa1cd48a96123","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2017-10-18 01:53:54.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"cb01024937a6dfac7cac6675065ce2222f220df9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-18 02:00:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded.\n\n- puppet-openstack-module-build http://logs.openstack.org/46/512846/2/check/puppet-openstack-module-build/18f4383/ : SUCCESS in 1m 21s","accounts_in_message":[],"_revision_number":2},{"id":"f65a9cdc1b79dcf72a02947157bdd3d58d2d2bba","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2017-10-18 02:01:57.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"89137d8bbf8da22fd8422f5f2e588aef55310b82","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2017-10-18 02:21:08.000000000","message":"Patch Set 2: Workflow+1 Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"8a0c39a5cc25f76af3500a907d5e5b2f52c86a82","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2017-10-18 13:18:42.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"ae26d1e2e81abd8982103d72e670d94244781a7a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-18 13:26:43.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed.\n\n- puppet-openstack-lint http://logs.openstack.org/46/512846/2/check/puppet-openstack-lint/fbaf587/ : SUCCESS in 2m 31s\n- puppet-openstack-syntax-4 http://logs.openstack.org/46/512846/2/check/puppet-openstack-syntax-4/b935a32/ : SUCCESS in 2m 35s\n- puppet-openstack-module-build http://logs.openstack.org/46/512846/2/check/puppet-openstack-module-build/262c804/ : FAILURE in 2m 14s\n- puppet-openstack-integration-4-scenario001-tempest-centos-7 puppet-openstack-integration-4-scenario001-tempest-centos-7 : ERROR No valid playbook found\n- puppet-openstack-integration-4-scenario002-tempest-centos-7 puppet-openstack-integration-4-scenario002-tempest-centos-7 : ERROR No valid playbook found\n- puppet-openstack-integration-4-scenario003-tempest-centos-7 puppet-openstack-integration-4-scenario003-tempest-centos-7 : ERROR No valid playbook found\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 puppet-openstack-integration-4-scenario004-tempest-centos-7 : ERROR No valid playbook found\n- puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial : ERROR No valid playbook found (non-voting)\n- puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial : ERROR No valid playbook found (non-voting)\n- puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial : ERROR No valid playbook found (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial : ERROR No valid playbook found (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"18969ab3245ff610c657334069379bf102f1b1dd","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2017-10-19 01:55:56.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"cf1c2b01c7f33d5a7b60fb0c5c60a8f2d8b2cac4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-19 03:11:26.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- puppet-openstack-module-build http://logs.openstack.org/46/512846/2/check/puppet-openstack-module-build/022e211/ : SUCCESS in 1m 37s\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 puppet-openstack-integration-4-scenario004-tempest-centos-7 : ERROR No valid playbook found\n- puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial : ERROR No valid playbook found","accounts_in_message":[],"_revision_number":2},{"id":"ceb2086560b9e2bb63ff98e2d166c63c6d387d1c","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2017-10-19 03:38:59.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"5c9860e94c6e0aa4c41e13f089f3782e93fe62e3","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2017-10-19 03:39:37.000000000","message":"Removed Workflow+1 by Alex Schultz \u003caschultz@redhat.com\u003e\n","accounts_in_message":[],"_revision_number":2},{"id":"0b8d1bc1bbb7108a1e67241f947defcdfff020a3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-19 03:45:27.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- puppet-openstack-module-build http://logs.openstack.org/46/512846/2/check/puppet-openstack-module-build/d6e8e6f/ : SUCCESS in 2m 15s\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 puppet-openstack-integration-4-scenario004-tempest-centos-7 : ERROR No valid playbook found\n- puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial : ERROR No valid playbook found","accounts_in_message":[],"_revision_number":2},{"id":"68f8be7ab44d9130a2bd64866206c8eb3aec3505","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2017-10-19 03:48:17.000000000","message":"Patch Set 2:\n\nhmm the job layout looks broken or something","accounts_in_message":[],"_revision_number":2},{"id":"9dfd4443b631af5cfe34e67c547aa01fe53a67fd","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2017-10-28 00:34:06.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"8faa643d5ccadf8fd54d771e12f0f1746a3538a9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-28 00:45:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-module-build http://logs.openstack.org/46/512846/2/check/puppet-openstack-module-build/0fdf664/ : SUCCESS in 2m 10s","accounts_in_message":[],"_revision_number":2},{"id":"3784882a167dbe3361a56bb23f36311140f47981","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-28 00:45:54.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d626a19b61056a8b195e36f5c0f939906a05c887","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-28 01:00:13.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-module-build http://logs.openstack.org/46/512846/2/gate/puppet-openstack-module-build/82c4daa/ : SUCCESS in 2m 40s","accounts_in_message":[],"_revision_number":2},{"id":"513b8dd52638795576cd3637085c565d04e51e15","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-28 01:00:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e031b12cd044e32131384cebf12ea5610068bc80","revisions":{"1928cc789b2e1c26dda7c8168b79cffa26b019b8":{"kind":"REWORK","_number":1,"created":"2017-10-17 20:29:06.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/46/512846/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tempest","ref":"refs/changes/46/512846/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tempest refs/changes/46/512846/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tempest refs/changes/46/512846/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tempest refs/changes/46/512846/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tempest refs/changes/46/512846/1"}}},"commit":{"parents":[{"commit":"da6fe3964bcb0592957ca37cd299ee1abcc32e6b","subject":"Set Keystone v2 API to disabled by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tempest/commit/da6fe3964bcb0592957ca37cd299ee1abcc32e6b"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2017-10-17 20:26:32.000000000","tz":-360},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2017-10-17 20:29:05.000000000","tz":-360},"subject":"Prepare for Queens Milestone 1","message":"Prepare for Queens Milestone 1\n\nPrepare the metadata configuration for the first Queens milestone.\n\nChange-Id: Icbded3288de998f6e970e4a0597aaa1648eddb90\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tempest/commit/1928cc789b2e1c26dda7c8168b79cffa26b019b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tempest/commit/1928cc789b2e1c26dda7c8168b79cffa26b019b8"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"da6fe3964bcb0592957ca37cd299ee1abcc32e6b","is_merged_in_target_branch":true,"change_id":"I558e91cddb0e191de4874174e7774c2ec29d26ae","change_number":507570,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/master"},"e031b12cd044e32131384cebf12ea5610068bc80":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2017-10-18 01:53:54.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/46/512846/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tempest","ref":"refs/changes/46/512846/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tempest refs/changes/46/512846/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tempest refs/changes/46/512846/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tempest refs/changes/46/512846/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tempest refs/changes/46/512846/2"}}},"commit":{"parents":[{"commit":"da6fe3964bcb0592957ca37cd299ee1abcc32e6b","subject":"Set Keystone v2 API to disabled by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tempest/commit/da6fe3964bcb0592957ca37cd299ee1abcc32e6b"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2017-10-17 20:26:32.000000000","tz":-360},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2017-10-18 01:53:07.000000000","tz":0},"subject":"Prepare for Queens Milestone 1","message":"Prepare for Queens Milestone 1\n\nPrepare the metadata configuration for the first Queens milestone.\n\nDepends-On: Id68ee1b443a4172d0c1d6d58a04908c52a566623\nChange-Id: Icbded3288de998f6e970e4a0597aaa1648eddb90\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tempest/commit/e031b12cd044e32131384cebf12ea5610068bc80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tempest/commit/e031b12cd044e32131384cebf12ea5610068bc80"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"da6fe3964bcb0592957ca37cd299ee1abcc32e6b","is_merged_in_target_branch":true,"change_id":"I558e91cddb0e191de4874174e7774c2ec29d26ae","change_number":507570,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
