)]}'
{"id":"openstack%2Fpuppet-tempest~515876","triplet_id":"openstack%2Fpuppet-tempest~stable%2Fnewton~I771b38e835a2c4826b7f77556dbb4d7b199cf0a4","project":"openstack/puppet-tempest","branch":"stable/newton","topic":"bindep/puppet","hashtags":[],"change_id":"I771b38e835a2c4826b7f77556dbb4d7b199cf0a4","subject":"[newton] Add Puppet package to bindep, for module build","status":"MERGED","created":"2017-10-27 20:30:07.000000000","updated":"2017-10-28 01:00:13.000000000","submitted":"2017-10-28 01:00:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"515876-1509152413158-b211ec01","meta_rev_id":"e3541aa94b470c2592654c151c0bd644787c0237","_number":515876,"virtual_id_number":515876,"owner":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2017-10-28 01:00:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":2,"date":"2017-10-27 23:02:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":1,"date":"2017-10-27 23:02:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-10-27 23:02:09.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2017-10-28 01:00:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5efd5225592ae085631233513fd07f2f382e064b","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2017-10-27 20:30:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"07106af0ff7da38cbe3e2e0c2b7a18658b860d9b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-27 22:04:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint http://logs.openstack.org/76/515876/1/check/puppet-openstack-lint/e10f60d/ : SUCCESS in 3m 27s\n- puppet-openstack-syntax-3-legacy http://logs.openstack.org/76/515876/1/check/puppet-openstack-syntax-3-legacy/214677d/ : SUCCESS in 9m 27s\n- puppet-openstack-syntax-4 http://logs.openstack.org/76/515876/1/check/puppet-openstack-syntax-4/7f51394/ : SUCCESS in 3m 24s\n- legacy-puppet-unit-3.6-legacy-centos-7 http://logs.openstack.org/76/515876/1/check/legacy-puppet-unit-3.6-legacy-centos-7/5fbd12c/ : SUCCESS in 8m 45s\n- legacy-puppet-unit-3.8-legacy-centos-7 http://logs.openstack.org/76/515876/1/check/legacy-puppet-unit-3.8-legacy-centos-7/592a075/ : SUCCESS in 7m 56s\n- legacy-puppet-unit-4.8-centos-7 http://logs.openstack.org/76/515876/1/check/legacy-puppet-unit-4.8-centos-7/ee2e01c/ : SUCCESS in 11m 15s\n- legacy-puppet-beaker-rspec-centos-7 http://logs.openstack.org/76/515876/1/check/legacy-puppet-beaker-rspec-centos-7/1f4c867/ : SUCCESS in 20m 02s\n- legacy-puppet-beaker-rspec http://logs.openstack.org/76/515876/1/check/legacy-puppet-beaker-rspec/5668053/ : SUCCESS in 11m 41s (non-voting)\n- puppet-openstack-integration-4-scenario001-tempest-centos-7 http://logs.openstack.org/76/515876/1/check/puppet-openstack-integration-4-scenario001-tempest-centos-7/36ebfd9/ : SUCCESS in 59m 31s\n- puppet-openstack-integration-4-scenario002-tempest-centos-7 http://logs.openstack.org/76/515876/1/check/puppet-openstack-integration-4-scenario002-tempest-centos-7/928c21a/ : SUCCESS in 51m 07s\n- puppet-openstack-integration-4-scenario003-tempest-centos-7 http://logs.openstack.org/76/515876/1/check/puppet-openstack-integration-4-scenario003-tempest-centos-7/77e2863/ : SUCCESS in 52m 27s\n- puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial http://logs.openstack.org/76/515876/1/check/puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial/447f044/ : SUCCESS in 27m 51s (non-voting)\n- puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial http://logs.openstack.org/76/515876/1/check/puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial/28dbcbe/ : SUCCESS in 44m 31s (non-voting)\n- puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial http://logs.openstack.org/76/515876/1/check/puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial/26c470b/ : SUCCESS in 36m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a75eb5de83695ddfad73cf174ffa33991013a1df","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2017-10-27 23:02:09.000000000","message":"Patch Set 1: Workflow+1 Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3142d8e4638b6fbf4ae72bc63970443abaf9eeb7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-27 23:02:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9f06f88d17163889152f2e89db40cd667bdc7bb5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-27 23:49:22.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ddca114f7312319b384e154f199361e4467001de","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-28 01:00:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint http://logs.openstack.org/76/515876/1/gate/puppet-openstack-lint/a451cb4/ : SUCCESS in 4m 43s\n- puppet-openstack-syntax-3-legacy http://logs.openstack.org/76/515876/1/gate/puppet-openstack-syntax-3-legacy/11ba640/ : SUCCESS in 3m 45s\n- puppet-openstack-syntax-4 http://logs.openstack.org/76/515876/1/gate/puppet-openstack-syntax-4/273c0ae/ : SUCCESS in 6m 51s\n- legacy-puppet-unit-3.6-legacy-centos-7 http://logs.openstack.org/76/515876/1/gate/legacy-puppet-unit-3.6-legacy-centos-7/6ff0b6e/ : SUCCESS in 11m 57s\n- legacy-puppet-unit-3.8-legacy-centos-7 http://logs.openstack.org/76/515876/1/gate/legacy-puppet-unit-3.8-legacy-centos-7/4bc6101/ : SUCCESS in 9m 08s\n- legacy-puppet-unit-4.8-centos-7 http://logs.openstack.org/76/515876/1/gate/legacy-puppet-unit-4.8-centos-7/657cd34/ : SUCCESS in 11m 16s\n- legacy-puppet-beaker-rspec-centos-7 http://logs.openstack.org/76/515876/1/gate/legacy-puppet-beaker-rspec-centos-7/2a9cc16/ : SUCCESS in 17m 01s\n- legacy-puppet-beaker-rspec http://logs.openstack.org/76/515876/1/gate/legacy-puppet-beaker-rspec/156ff3c/ : SUCCESS in 13m 02s (non-voting)\n- puppet-openstack-integration-4-scenario001-tempest-centos-7 http://logs.openstack.org/76/515876/1/gate/puppet-openstack-integration-4-scenario001-tempest-centos-7/0335f73/ : SUCCESS in 1h 01m 17s\n- puppet-openstack-integration-4-scenario002-tempest-centos-7 http://logs.openstack.org/76/515876/1/gate/puppet-openstack-integration-4-scenario002-tempest-centos-7/e0de8e5/ : SUCCESS in 53m 01s\n- puppet-openstack-integration-4-scenario003-tempest-centos-7 http://logs.openstack.org/76/515876/1/gate/puppet-openstack-integration-4-scenario003-tempest-centos-7/acbaa2b/ : SUCCESS in 51m 07s\n- puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial http://logs.openstack.org/76/515876/1/gate/puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial/c3a6de7/ : SUCCESS in 42m 17s\n- puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial http://logs.openstack.org/76/515876/1/gate/puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial/695fa93/ : SUCCESS in 38m 42s\n- puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial http://logs.openstack.org/76/515876/1/gate/puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial/587c2ac/ : SUCCESS in 41m 28s","accounts_in_message":[],"_revision_number":1},{"id":"e8e024887b3ef6730523caf81afe6e1c7a4386b9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-28 01:00:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"355c6bd96da2696f9977d8a55bee25224328a21c","revisions":{"355c6bd96da2696f9977d8a55bee25224328a21c":{"kind":"REWORK","_number":1,"created":"2017-10-27 20:30:07.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/76/515876/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tempest","ref":"refs/changes/76/515876/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tempest refs/changes/76/515876/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tempest refs/changes/76/515876/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tempest refs/changes/76/515876/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tempest refs/changes/76/515876/1"}}},"commit":{"parents":[{"commit":"9bcf09c6efdacc9ca08044fe943af4ff9e253676","subject":"Enable Keystone v3 support for providers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tempest/commit/9bcf09c6efdacc9ca08044fe943af4ff9e253676"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2017-10-27 20:30:02.000000000","tz":-420},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2017-10-27 20:30:02.000000000","tz":-420},"subject":"[newton] Add Puppet package to bindep, for module build","message":"[newton] Add Puppet package to bindep, for module build\n\nWe need Puppet package deployed from bindep so we can\nrun puppet module build with the new zuul v3 job.\n\nChange-Id: I771b38e835a2c4826b7f77556dbb4d7b199cf0a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tempest/commit/355c6bd96da2696f9977d8a55bee25224328a21c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tempest/commit/355c6bd96da2696f9977d8a55bee25224328a21c"}]},"parents_data":[{"branch_name":"refs/heads/stable/newton","commit_id":"9bcf09c6efdacc9ca08044fe943af4ff9e253676","is_merged_in_target_branch":true,"change_id":"I95ea682064ff6365622fc31c562b42014fc52303","change_number":471315,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/stable/newton"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
