)]}'
{"id":"openstack%2Fpuppet-tempest~524191","triplet_id":"openstack%2Fpuppet-tempest~master~I168e666ce936e650c6137920dce75ae1a59a1a76","project":"openstack/puppet-tempest","branch":"master","topic":"neutron-tempest-plugin","hashtags":[],"change_id":"I168e666ce936e650c6137920dce75ae1a59a1a76","subject":"Rename python-neutron-tests to python-neutron-tests-tempest","status":"MERGED","created":"2017-11-30 14:30:42.000000000","updated":"2017-12-02 07:40:52.000000000","submitted":"2017-12-02 07:40:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"524191-1512200452038-4b468b43","meta_rev_id":"496df9d6ca26cf3cb15f8275ab1800a5d0b49e28","_number":524191,"virtual_id_number":524191,"owner":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"date":"2017-12-01 09:08:59.000000000","_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2017-12-02 07:40:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2017-12-01 18:41:55.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":2,"date":"2017-12-01 22:24:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":1,"date":"2017-12-01 22:24:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-12-01 18:41:55.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2017-12-01 22:24:52.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2017-12-02 07:40:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e478398ea641380fd64fe5f7c59070066395d748","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2017-11-30 14:30:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a96a8d3e2b74a4093f01c7589e5fc62615f7a66f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-11-30 15:01:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint http://logs.openstack.org/91/524191/1/check/puppet-openstack-lint/eb62ae1/ : RETRY_LIMIT in 49s\n- puppet-openstack-syntax-4 http://logs.openstack.org/91/524191/1/check/puppet-openstack-syntax-4/5f0a355/ : RETRY_LIMIT in 1m 30s\n- legacy-puppet-unit-4.8-centos-7 http://logs.openstack.org/91/524191/1/check/legacy-puppet-unit-4.8-centos-7/298d991/ : RETRY_LIMIT in 4m 50s\n- legacy-puppet-unit-latest-centos-7 http://logs.openstack.org/91/524191/1/check/legacy-puppet-unit-latest-centos-7/5380b91/ : RETRY_LIMIT in 4m 59s (non-voting)\n- legacy-puppet-beaker-rspec-centos-7 http://logs.openstack.org/91/524191/1/check/legacy-puppet-beaker-rspec-centos-7/32536d0/ : RETRY_LIMIT in 3m 34s\n- legacy-puppet-beaker-rspec http://logs.openstack.org/91/524191/1/check/legacy-puppet-beaker-rspec/445fbe6/ : RETRY_LIMIT in 4m 38s\n- puppet-openstack-integration-4-scenario001-tempest-centos-7 http://logs.openstack.org/91/524191/1/check/puppet-openstack-integration-4-scenario001-tempest-centos-7/71ba666/ : RETRY_LIMIT in 3m 00s\n- puppet-openstack-integration-4-scenario002-tempest-centos-7 http://logs.openstack.org/91/524191/1/check/puppet-openstack-integration-4-scenario002-tempest-centos-7/c9c98de/ : RETRY_LIMIT in 5m 13s\n- puppet-openstack-integration-4-scenario003-tempest-centos-7 http://logs.openstack.org/91/524191/1/check/puppet-openstack-integration-4-scenario003-tempest-centos-7/768f2c4/ : RETRY_LIMIT in 2m 50s\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 http://logs.openstack.org/91/524191/1/check/puppet-openstack-integration-4-scenario004-tempest-centos-7/839bc9d/ : RETRY_LIMIT in 5m 49s\n- puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial http://logs.openstack.org/91/524191/1/check/puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial/836c51b/ : RETRY_LIMIT in 4m 35s (non-voting)\n- puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial http://logs.openstack.org/91/524191/1/check/puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial/9ef6793/ : RETRY_LIMIT in 3m 00s (non-voting)\n- puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial http://logs.openstack.org/91/524191/1/check/puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial/a3fdde9/ : RETRY_LIMIT in 4m 38s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial http://logs.openstack.org/91/524191/1/check/puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial/812594f/ : RETRY_LIMIT in 5m 03s (non-voting)\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/91/524191/1/check/puppet-openstack-unit-4.8-centos-7/5d2cba6/ : RETRY_LIMIT in 2m 27s\n- puppet-openstack-unit-latest-centos-7 http://logs.openstack.org/91/524191/1/check/puppet-openstack-unit-latest-centos-7/11d3db9/ : RETRY_LIMIT in 5m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"049a7e24041d89de992ce98fdb7256af7d6a50d8","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2017-11-30 17:36:52.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"cc9fdfa624b876ac3691eb66613d9de98731521a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-11-30 18:26:43.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint http://logs.openstack.org/91/524191/1/check/puppet-openstack-lint/160dd1b/ : SUCCESS in 3m 00s\n- puppet-openstack-syntax-4 http://logs.openstack.org/91/524191/1/check/puppet-openstack-syntax-4/352737b/ : SUCCESS in 3m 15s\n- legacy-puppet-unit-4.8-centos-7 http://logs.openstack.org/91/524191/1/check/legacy-puppet-unit-4.8-centos-7/84e83f2/ : SUCCESS in 9m 51s\n- legacy-puppet-unit-latest-centos-7 http://logs.openstack.org/91/524191/1/check/legacy-puppet-unit-latest-centos-7/66600fe/ : SUCCESS in 9m 24s (non-voting)\n- legacy-puppet-beaker-rspec-centos-7 http://logs.openstack.org/91/524191/1/check/legacy-puppet-beaker-rspec-centos-7/e7c79ac/ : SUCCESS in 16m 17s\n- legacy-puppet-beaker-rspec http://logs.openstack.org/91/524191/1/check/legacy-puppet-beaker-rspec/76d417f/ : SUCCESS in 7m 39s\n- puppet-openstack-integration-4-scenario001-tempest-centos-7 http://logs.openstack.org/91/524191/1/check/puppet-openstack-integration-4-scenario001-tempest-centos-7/e2215eb/ : FAILURE in 26m 06s\n- puppet-openstack-integration-4-scenario002-tempest-centos-7 http://logs.openstack.org/91/524191/1/check/puppet-openstack-integration-4-scenario002-tempest-centos-7/3b0b29b/ : FAILURE in 18m 17s\n- puppet-openstack-integration-4-scenario003-tempest-centos-7 http://logs.openstack.org/91/524191/1/check/puppet-openstack-integration-4-scenario003-tempest-centos-7/816c050/ : FAILURE in 22m 38s\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 http://logs.openstack.org/91/524191/1/check/puppet-openstack-integration-4-scenario004-tempest-centos-7/615c40c/ : FAILURE in 24m 26s\n- puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial http://logs.openstack.org/91/524191/1/check/puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial/66d67e6/ : SUCCESS in 35m 07s (non-voting)\n- puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial http://logs.openstack.org/91/524191/1/check/puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial/2a16f56/ : SUCCESS in 38m 42s (non-voting)\n- puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial http://logs.openstack.org/91/524191/1/check/puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial/d5c9301/ : SUCCESS in 38m 43s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial http://logs.openstack.org/91/524191/1/check/puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial/6277030/ : SUCCESS in 29m 37s (non-voting)\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/91/524191/1/check/puppet-openstack-unit-4.8-centos-7/2e06aa2/ : SUCCESS in 9m 40s\n- puppet-openstack-unit-latest-centos-7 http://logs.openstack.org/91/524191/1/check/puppet-openstack-unit-latest-centos-7/c0bb878/ : SUCCESS in 8m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4fb6b37a1c900829c09e2aa2411c72d95cbc65b2","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2017-12-01 09:08:59.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"35e76132271385a91cc272efdd59eacbaa31463b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-01 10:16:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint http://logs.openstack.org/91/524191/1/check/puppet-openstack-lint/16b761f/ : SUCCESS in 4m 05s\n- puppet-openstack-syntax-4 http://logs.openstack.org/91/524191/1/check/puppet-openstack-syntax-4/0880149/ : SUCCESS in 3m 20s\n- legacy-puppet-unit-4.8-centos-7 http://logs.openstack.org/91/524191/1/check/legacy-puppet-unit-4.8-centos-7/483af99/ : SUCCESS in 10m 31s\n- legacy-puppet-unit-latest-centos-7 http://logs.openstack.org/91/524191/1/check/legacy-puppet-unit-latest-centos-7/4d36107/ : SUCCESS in 7m 41s (non-voting)\n- legacy-puppet-beaker-rspec-centos-7 http://logs.openstack.org/91/524191/1/check/legacy-puppet-beaker-rspec-centos-7/792181c/ : SUCCESS in 19m 07s\n- legacy-puppet-beaker-rspec http://logs.openstack.org/91/524191/1/check/legacy-puppet-beaker-rspec/ebee6e8/ : SUCCESS in 6m 40s\n- puppet-openstack-integration-4-scenario001-tempest-centos-7 http://logs.openstack.org/91/524191/1/check/puppet-openstack-integration-4-scenario001-tempest-centos-7/6efabc6/ : SUCCESS in 57m 09s\n- puppet-openstack-integration-4-scenario002-tempest-centos-7 http://logs.openstack.org/91/524191/1/check/puppet-openstack-integration-4-scenario002-tempest-centos-7/2dd16fe/ : SUCCESS in 54m 53s\n- puppet-openstack-integration-4-scenario003-tempest-centos-7 http://logs.openstack.org/91/524191/1/check/puppet-openstack-integration-4-scenario003-tempest-centos-7/0dc64a3/ : SUCCESS in 54m 30s\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 http://logs.openstack.org/91/524191/1/check/puppet-openstack-integration-4-scenario004-tempest-centos-7/0a0058f/ : SUCCESS in 46m 13s\n- puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial http://logs.openstack.org/91/524191/1/check/puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial/c31b36c/ : FAILURE in 37m 35s (non-voting)\n- puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial http://logs.openstack.org/91/524191/1/check/puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial/792d889/ : FAILURE in 40m 24s (non-voting)\n- puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial http://logs.openstack.org/91/524191/1/check/puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial/cf0441d/ : SUCCESS in 30m 08s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial http://logs.openstack.org/91/524191/1/check/puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial/a0a57bb/ : SUCCESS in 35m 58s (non-voting)\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/91/524191/1/check/puppet-openstack-unit-4.8-centos-7/4e26f83/ : SUCCESS in 12m 23s\n- puppet-openstack-unit-latest-centos-7 http://logs.openstack.org/91/524191/1/check/puppet-openstack-unit-latest-centos-7/c99e8f4/ : SUCCESS in 12m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cd17b2b61a177f14660dc4ac6d615bae64fb4eb7","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2017-12-01 18:41:55.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"09635dd0f5c8785aa7d6413f456da3d5d4f78dea","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2017-12-01 22:24:52.000000000","message":"Patch Set 1: Workflow+1 Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bee301453aa3d8be1171989fcd9d7de60e514c57","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-01 22:25:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"afc31ba764ed37d4374a223b0dc7c6a1315c5295","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-02 01:30:34.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint http://logs.openstack.org/91/524191/1/gate/puppet-openstack-lint/0112ba7/ : SUCCESS in 3m 29s\n- puppet-openstack-syntax-4 http://logs.openstack.org/91/524191/1/gate/puppet-openstack-syntax-4/91be243/ : SUCCESS in 3m 46s\n- legacy-puppet-unit-4.8-centos-7 http://logs.openstack.org/91/524191/1/gate/legacy-puppet-unit-4.8-centos-7/8435f1e/ : SUCCESS in 9m 29s\n- legacy-puppet-beaker-rspec-centos-7 http://logs.openstack.org/91/524191/1/gate/legacy-puppet-beaker-rspec-centos-7/d4dd788/ : SUCCESS in 13m 34s\n- legacy-puppet-beaker-rspec http://logs.openstack.org/91/524191/1/gate/legacy-puppet-beaker-rspec/571cfe0/ : TIMED_OUT in 3h 04m 00s\n- puppet-openstack-integration-4-scenario001-tempest-centos-7 http://logs.openstack.org/91/524191/1/gate/puppet-openstack-integration-4-scenario001-tempest-centos-7/5901e89/ : TIMED_OUT in 1h 38m 48s\n- puppet-openstack-integration-4-scenario002-tempest-centos-7 http://logs.openstack.org/91/524191/1/gate/puppet-openstack-integration-4-scenario002-tempest-centos-7/fd80121/ : SUCCESS in 1h 17m 31s\n- puppet-openstack-integration-4-scenario003-tempest-centos-7 http://logs.openstack.org/91/524191/1/gate/puppet-openstack-integration-4-scenario003-tempest-centos-7/6091b2e/ : SUCCESS in 54m 16s\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 http://logs.openstack.org/91/524191/1/gate/puppet-openstack-integration-4-scenario004-tempest-centos-7/cb846d6/ : SUCCESS in 42m 12s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/91/524191/1/gate/puppet-openstack-unit-4.8-centos-7/7eabdc3/ : SUCCESS in 11m 48s","accounts_in_message":[],"_revision_number":1},{"id":"b66f31c893d55cf8fb782abe3006d6b82072117c","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2017-12-02 05:34:51.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"129b1863c2f5e7040649f07abe5d07d2f42d4aa5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-02 06:38:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint http://logs.openstack.org/91/524191/1/check/puppet-openstack-lint/87301e7/ : SUCCESS in 2m 46s\n- puppet-openstack-syntax-4 http://logs.openstack.org/91/524191/1/check/puppet-openstack-syntax-4/1f42290/ : SUCCESS in 3m 57s\n- legacy-puppet-unit-4.8-centos-7 http://logs.openstack.org/91/524191/1/check/legacy-puppet-unit-4.8-centos-7/e4b4973/ : SUCCESS in 8m 59s\n- legacy-puppet-unit-latest-centos-7 http://logs.openstack.org/91/524191/1/check/legacy-puppet-unit-latest-centos-7/e38762a/ : SUCCESS in 11m 34s (non-voting)\n- legacy-puppet-beaker-rspec-centos-7 http://logs.openstack.org/91/524191/1/check/legacy-puppet-beaker-rspec-centos-7/539f14c/ : SUCCESS in 18m 26s\n- legacy-puppet-beaker-rspec http://logs.openstack.org/91/524191/1/check/legacy-puppet-beaker-rspec/64b5e35/ : SUCCESS in 6m 10s\n- puppet-openstack-integration-4-scenario001-tempest-centos-7 http://logs.openstack.org/91/524191/1/check/puppet-openstack-integration-4-scenario001-tempest-centos-7/c7a234f/ : SUCCESS in 1h 00m 50s\n- puppet-openstack-integration-4-scenario002-tempest-centos-7 http://logs.openstack.org/91/524191/1/check/puppet-openstack-integration-4-scenario002-tempest-centos-7/d58e944/ : SUCCESS in 1h 02m 05s\n- puppet-openstack-integration-4-scenario003-tempest-centos-7 http://logs.openstack.org/91/524191/1/check/puppet-openstack-integration-4-scenario003-tempest-centos-7/0d4d62c/ : SUCCESS in 59m 27s\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 http://logs.openstack.org/91/524191/1/check/puppet-openstack-integration-4-scenario004-tempest-centos-7/25850f1/ : SUCCESS in 43m 38s\n- puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial http://logs.openstack.org/91/524191/1/check/puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial/b132b8d/ : SUCCESS in 45m 44s (non-voting)\n- puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial http://logs.openstack.org/91/524191/1/check/puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial/1dfb6a3/ : SUCCESS in 39m 10s (non-voting)\n- puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial http://logs.openstack.org/91/524191/1/check/puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial/3b30e4c/ : SUCCESS in 32m 34s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial http://logs.openstack.org/91/524191/1/check/puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial/b200937/ : SUCCESS in 37m 27s (non-voting)\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/91/524191/1/check/puppet-openstack-unit-4.8-centos-7/97b011e/ : SUCCESS in 8m 54s\n- puppet-openstack-unit-latest-centos-7 http://logs.openstack.org/91/524191/1/check/puppet-openstack-unit-latest-centos-7/84554d0/ : SUCCESS in 11m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fef38e0b8ca0d8db162f5d9e1a17757ce5e96d23","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-02 06:39:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"86eca462867f933d069360bee5b0418d711cac03","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-02 07:40:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint http://logs.openstack.org/91/524191/1/gate/puppet-openstack-lint/970f11f/ : SUCCESS in 3m 17s\n- puppet-openstack-syntax-4 http://logs.openstack.org/91/524191/1/gate/puppet-openstack-syntax-4/cdfbed8/ : SUCCESS in 3m 29s\n- legacy-puppet-unit-4.8-centos-7 http://logs.openstack.org/91/524191/1/gate/legacy-puppet-unit-4.8-centos-7/97df99e/ : SUCCESS in 12m 02s\n- legacy-puppet-beaker-rspec-centos-7 http://logs.openstack.org/91/524191/1/gate/legacy-puppet-beaker-rspec-centos-7/7291adc/ : SUCCESS in 16m 09s\n- legacy-puppet-beaker-rspec http://logs.openstack.org/91/524191/1/gate/legacy-puppet-beaker-rspec/29c6298/ : SUCCESS in 10m 34s\n- puppet-openstack-integration-4-scenario001-tempest-centos-7 http://logs.openstack.org/91/524191/1/gate/puppet-openstack-integration-4-scenario001-tempest-centos-7/58bbd30/ : SUCCESS in 55m 23s\n- puppet-openstack-integration-4-scenario002-tempest-centos-7 http://logs.openstack.org/91/524191/1/gate/puppet-openstack-integration-4-scenario002-tempest-centos-7/eb98068/ : SUCCESS in 49m 02s\n- puppet-openstack-integration-4-scenario003-tempest-centos-7 http://logs.openstack.org/91/524191/1/gate/puppet-openstack-integration-4-scenario003-tempest-centos-7/b9726d2/ : SUCCESS in 57m 28s\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 http://logs.openstack.org/91/524191/1/gate/puppet-openstack-integration-4-scenario004-tempest-centos-7/4d2e896/ : SUCCESS in 43m 30s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/91/524191/1/gate/puppet-openstack-unit-4.8-centos-7/70b8299/ : SUCCESS in 10m 00s","accounts_in_message":[],"_revision_number":1},{"id":"ee627ab99a1cfc362ac51c081b63a66076fe3c4f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-02 07:40:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f65101fc22f6553a894df0a6f584264df8396b1a","revisions":{"f65101fc22f6553a894df0a6f584264df8396b1a":{"kind":"REWORK","_number":1,"created":"2017-11-30 14:30:42.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/91/524191/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tempest","ref":"refs/changes/91/524191/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tempest refs/changes/91/524191/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tempest refs/changes/91/524191/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tempest refs/changes/91/524191/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tempest refs/changes/91/524191/1"}}},"commit":{"parents":[{"commit":"797422451013f958c9509997c7097c889484a9e4","subject":"Prepare for Queens Milestone 2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tempest/commit/797422451013f958c9509997c7097c889484a9e4"}]}],"author":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2017-11-30 14:29:16.000000000","tz":330},"committer":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2017-11-30 14:30:38.000000000","tz":330},"subject":"Rename python-neutron-tests to python-neutron-tests-tempest","message":"Rename python-neutron-tests to python-neutron-tests-tempest\n\n* https://review.rdoproject.org/r/#/c/10549/ marks the addition of\n  neutron-tempest-plugin in RDO from Queens release. We need to\n  rename the package to avoid breakage.\n\nChange-Id: I168e666ce936e650c6137920dce75ae1a59a1a76\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tempest/commit/f65101fc22f6553a894df0a6f584264df8396b1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tempest/commit/f65101fc22f6553a894df0a6f584264df8396b1a"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"797422451013f958c9509997c7097c889484a9e4","is_merged_in_target_branch":true,"change_id":"I2ddc95966d40429fc67e74b3aa42ad448491a177","change_number":523596,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
