)]}'
{"id":"openstack%2Fpuppet-tempest~524296","triplet_id":"openstack%2Fpuppet-tempest~stable%2Fpike~I2271a2599f44599b1436540cbb623fd07e6ad1a5","project":"openstack/puppet-tempest","branch":"stable/pike","topic":"post-pike-release","hashtags":[],"change_id":"I2271a2599f44599b1436540cbb623fd07e6ad1a5","subject":"Prepare for post-Pike updates","status":"MERGED","created":"2017-11-30 17:59:27.000000000","updated":"2017-11-30 18:36:12.000000000","submitted":"2017-11-30 18:36:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"524296-1512066972826-27d8552d","meta_rev_id":"d905383c035b848037a6b3f69a39632c1d3a77eb","_number":524296,"virtual_id_number":524296,"owner":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2017-11-30 18:36:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2017-11-30 18:30:32.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2017-11-30 18:30:32.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-11-30 18:30:32.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2017-11-30 18:36:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8a6bc38cce2476b94df2917371818e9f70de9eba","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2017-11-30 17:59:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"04c1291cc5d307465bb70eb3f96f0b356f9b5fe9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-11-30 18:07:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-releasenotes http://logs.openstack.org/96/524296/1/check/build-openstack-releasenotes/2daa226/html/ : SUCCESS in 1m 51s\n- build-openstack-puppet-tarball http://logs.openstack.org/96/524296/1/check/build-openstack-puppet-tarball/00ebfb9/ : SUCCESS in 1m 31s","accounts_in_message":[],"_revision_number":1},{"id":"6e8537b2c89944d120da1b94960a2fb9a3a94ee5","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2017-11-30 18:30:32.000000000","message":"Patch Set 1: Workflow+1 Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9de4613faf166c11e14b09e340c80ef995a97b0b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-11-30 18:30:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e5c67fcfa7c833519893f2e13b174a015e2d49f2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-11-30 18:36:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a839644388879d629cd48da0d1cd7905cffe35fc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-11-30 18:36:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-releasenotes http://logs.openstack.org/96/524296/1/gate/build-openstack-releasenotes/4e6adb2/html/ : SUCCESS in 2m 53s\n- build-openstack-puppet-tarball http://logs.openstack.org/96/524296/1/gate/build-openstack-puppet-tarball/4777c92/ : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2a6ae29cc80fb9fbb879b78211c6b5a505b9d3e3","revisions":{"2a6ae29cc80fb9fbb879b78211c6b5a505b9d3e3":{"kind":"REWORK","_number":1,"created":"2017-11-30 17:59:27.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/96/524296/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tempest","ref":"refs/changes/96/524296/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tempest refs/changes/96/524296/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tempest refs/changes/96/524296/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tempest refs/changes/96/524296/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tempest refs/changes/96/524296/1"}}},"commit":{"parents":[{"commit":"dcfca6d682d55332697cc5534dfa42e4c8e16a0f","subject":"[pike] Add Puppet package to bindep, for module build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tempest/commit/dcfca6d682d55332697cc5534dfa42e4c8e16a0f"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2017-11-30 17:55:58.000000000","tz":-420},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2017-11-30 17:59:27.000000000","tz":-420},"subject":"Prepare for post-Pike updates","message":"Prepare for post-Pike updates\n\nPrepare the metadata for a post-Pike release.\n\nChange-Id: I2271a2599f44599b1436540cbb623fd07e6ad1a5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tempest/commit/2a6ae29cc80fb9fbb879b78211c6b5a505b9d3e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tempest/commit/2a6ae29cc80fb9fbb879b78211c6b5a505b9d3e3"}]},"parents_data":[{"branch_name":"refs/heads/stable/pike","commit_id":"dcfca6d682d55332697cc5534dfa42e4c8e16a0f","is_merged_in_target_branch":true,"change_id":"I19a2a6546b6d4137af044c388f34b4858c803384","change_number":515939,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/stable/pike"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
