)]}'
{"id":"openstack%2Fpuppet-tempest~535684","triplet_id":"openstack%2Fpuppet-tempest~master~I030f4a40e063c685026c6c4736d90cf75e986137","project":"openstack/puppet-tempest","branch":"master","topic":"tempest_plugins","hashtags":[],"change_id":"I030f4a40e063c685026c6c4736d90cf75e986137","subject":"Update the package name for heat-tempest-plugin","status":"MERGED","created":"2018-01-19 09:16:34.000000000","updated":"2018-01-20 02:02:48.000000000","submitted":"2018-01-20 02:02:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"535684-1516413768000-e6ec8469","meta_rev_id":"33fc6822be6c41793d892c9ab29a536bc2721a95","_number":535684,"virtual_id_number":535684,"owner":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2018-01-20 02:02:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2018-01-19 19:25:17.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2018-01-19 11:13:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":2,"date":"2018-01-19 22:35:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":1,"date":"2018-01-19 22:35:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-01-19 11:13:10.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"},{"updated":"2018-01-19 19:25:17.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2018-01-19 22:35:16.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2018-01-20 02:02:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7eab671db3b8d4152759cd6546b69afd265e141b","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2018-01-19 09:16:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1d206f82a89eb775f0f7e143b8422ceba52527a8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-19 10:45:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint http://logs.openstack.org/84/535684/1/check/puppet-openstack-lint/ba9181a/ : SUCCESS in 4m 14s\n- puppet-openstack-syntax-4 http://logs.openstack.org/84/535684/1/check/puppet-openstack-syntax-4/f5c54a0/ : SUCCESS in 2m 51s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/84/535684/1/check/puppet-openstack-unit-4.8-centos-7/d0e79d6/ : SUCCESS in 9m 18s\n- puppet-openstack-unit-latest-centos-7 http://logs.openstack.org/84/535684/1/check/puppet-openstack-unit-latest-centos-7/57452a7/ : SUCCESS in 8m 28s (non-voting)\n- puppet-openstack-integration-4-scenario001-tempest-centos-7 http://logs.openstack.org/84/535684/1/check/puppet-openstack-integration-4-scenario001-tempest-centos-7/2f7b8c5/ : SUCCESS in 1h 03m 58s\n- puppet-openstack-integration-4-scenario002-tempest-centos-7 http://logs.openstack.org/84/535684/1/check/puppet-openstack-integration-4-scenario002-tempest-centos-7/a4cb891/ : SUCCESS in 1h 26m 07s\n- puppet-openstack-integration-4-scenario003-tempest-centos-7 http://logs.openstack.org/84/535684/1/check/puppet-openstack-integration-4-scenario003-tempest-centos-7/814b7f1/ : SUCCESS in 1h 02m 40s\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 http://logs.openstack.org/84/535684/1/check/puppet-openstack-integration-4-scenario004-tempest-centos-7/46db763/ : SUCCESS in 54m 16s\n- puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial http://logs.openstack.org/84/535684/1/check/puppet-openstack-integration-4-scenario001-tempest-ubuntu-xenial/86091f5/ : FAILURE in 12m 48s (non-voting)\n- puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial http://logs.openstack.org/84/535684/1/check/puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial/c089979/ : SUCCESS in 54m 11s (non-voting)\n- puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial http://logs.openstack.org/84/535684/1/check/puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial/34447f4/ : SUCCESS in 34m 33s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial http://logs.openstack.org/84/535684/1/check/puppet-openstack-integration-4-scenario004-tempest-ubuntu-xenial/7756c44/ : SUCCESS in 31m 27s (non-voting)\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/84/535684/1/check/puppet-openstack-beaker-centos-7/26704aa/ : SUCCESS in 15m 50s\n- puppet-openstack-beaker-xenial http://logs.openstack.org/84/535684/1/check/puppet-openstack-beaker-xenial/a9c4c54/ : SUCCESS in 11m 05s","accounts_in_message":[],"_revision_number":1},{"id":"e90c19db55dfd5b6e16ab9cbf30e0ea5134b7147","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2018-01-19 11:13:10.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e368b09b21b4c5a8aed4b23ed72e23dab8e9cfed","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2018-01-19 19:25:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"cd007d76cfc901f38e0d05713f508736e2ec9daf","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2018-01-19 22:35:16.000000000","message":"Patch Set 1: Workflow+1 Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"48ebbd7d37329cd79f3657f1fb2bcdd97c23ee29","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-19 22:35:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ed61724bdf2db0c88ca0db055a51634d14350ed9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-20 00:30:42.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b24c48143dbb418aaaf4c278a194044beb56c270","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-20 02:02:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint http://logs.openstack.org/84/535684/1/gate/puppet-openstack-lint/02c0f92/ : SUCCESS in 4m 12s\n- puppet-openstack-syntax-4 http://logs.openstack.org/84/535684/1/gate/puppet-openstack-syntax-4/5f007e9/ : SUCCESS in 3m 54s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/84/535684/1/gate/puppet-openstack-unit-4.8-centos-7/f863857/ : SUCCESS in 9m 46s\n- puppet-openstack-integration-4-scenario001-tempest-centos-7 http://logs.openstack.org/84/535684/1/gate/puppet-openstack-integration-4-scenario001-tempest-centos-7/286473c/ : SUCCESS in 59m 41s\n- puppet-openstack-integration-4-scenario002-tempest-centos-7 http://logs.openstack.org/84/535684/1/gate/puppet-openstack-integration-4-scenario002-tempest-centos-7/5408053/ : SUCCESS in 1h 30m 00s\n- puppet-openstack-integration-4-scenario003-tempest-centos-7 http://logs.openstack.org/84/535684/1/gate/puppet-openstack-integration-4-scenario003-tempest-centos-7/0f9133c/ : SUCCESS in 1h 08m 05s\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 http://logs.openstack.org/84/535684/1/gate/puppet-openstack-integration-4-scenario004-tempest-centos-7/507bc3c/ : SUCCESS in 52m 09s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/84/535684/1/gate/puppet-openstack-beaker-centos-7/90b6a88/ : SUCCESS in 19m 16s\n- puppet-openstack-beaker-xenial http://logs.openstack.org/84/535684/1/gate/puppet-openstack-beaker-xenial/c24a041/ : SUCCESS in 9m 04s","accounts_in_message":[],"_revision_number":1},{"id":"4d207471eecb244745afbc49da7dd64b3a31c3af","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-20 02:02:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1fd82da5551b37c1fff6f37902aed31593c2d9a4","revisions":{"1fd82da5551b37c1fff6f37902aed31593c2d9a4":{"kind":"REWORK","_number":1,"created":"2018-01-19 09:16:34.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/84/535684/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tempest","ref":"refs/changes/84/535684/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tempest refs/changes/84/535684/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tempest refs/changes/84/535684/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tempest refs/changes/84/535684/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tempest refs/changes/84/535684/1"}}},"commit":{"parents":[{"commit":"683fce59a5df708ebc21d07857703708376b5192","subject":"Merge \"Use telemetry-tempest plugin for telemetry projects\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tempest/commit/683fce59a5df708ebc21d07857703708376b5192"}]}],"author":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2018-01-19 09:14:50.000000000","tz":330},"committer":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2018-01-19 09:14:50.000000000","tz":330},"subject":"Update the package name for heat-tempest-plugin","message":"Update the package name for heat-tempest-plugin\n\n* https://review.rdoproject.org/r/#/q/topic:add-heat-tempest-plugin\n  adds the heat tempest plugin to RDO.\n\nChange-Id: I030f4a40e063c685026c6c4736d90cf75e986137\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tempest/commit/1fd82da5551b37c1fff6f37902aed31593c2d9a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tempest/commit/1fd82da5551b37c1fff6f37902aed31593c2d9a4"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"683fce59a5df708ebc21d07857703708376b5192","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
