)]}'
{"id":"openstack%2Fpuppet-tempest~598933","triplet_id":"openstack%2Fpuppet-tempest~stable%2Frocky~Ia2f0addc2fb668d82424c144246b5c24647ef7cc","project":"openstack/puppet-tempest","branch":"stable/rocky","topic":"python3-first","hashtags":[],"change_id":"Ia2f0addc2fb668d82424c144246b5c24647ef7cc","subject":"import zuul job settings from project-config","status":"MERGED","created":"2018-08-31 13:02:15.000000000","updated":"2018-09-02 04:23:33.000000000","submitted":"2018-09-02 04:23:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"598933-1535862213950-5c4522c2","meta_rev_id":"20a9218426b5608c2d808c2d948660609b53f045","_number":598933,"virtual_id_number":598933,"owner":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"date":"2018-09-02 01:58:46.000000000","_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2018-09-02 04:23:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"recommended":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":1,"date":"2018-08-31 17:00:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2018-08-31 16:17:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2018-08-31 16:17:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-08-31 16:17:37.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2018-08-31 17:00:23.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2018-09-02 01:58:46.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2018-09-02 04:23:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"625f579cede275faeac2f3f768d8400a7addccfa","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2018-08-31 13:02:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f1e11a12de85014e7b2b26ba5e2731257eff9086","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2018-08-31 16:17:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"47d8e0f4b5b06cc0fa868dd9868411b0c60fc72a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-31 16:39:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint http://logs.openstack.org/33/598933/1/check/puppet-openstack-lint/21d8bf7/ : SUCCESS in 4m 04s\n- puppet-openstack-syntax-4 http://logs.openstack.org/33/598933/1/check/puppet-openstack-syntax-4/f89bfda/ : SUCCESS in 3m 40s\n- puppet-openstack-syntax-5 http://logs.openstack.org/33/598933/1/check/puppet-openstack-syntax-5/cd519cb/ : SUCCESS in 4m 11s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/33/598933/1/check/puppet-openstack-unit-4.8-centos-7/50b1d5b/ : SUCCESS in 11m 01s\n- puppet-openstack-integration-4-scenario004-tempest-centos-7-mimic http://logs.openstack.org/33/598933/1/check/puppet-openstack-integration-4-scenario004-tempest-centos-7-mimic/15a8ee4/ : SUCCESS in 57m 58s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous http://logs.openstack.org/33/598933/1/check/puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous/4286339/ : FAILURE in 1h 24m 57s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 http://logs.openstack.org/33/598933/1/check/puppet-openstack-integration-5-scenario002-tempest-centos-7/0ebf0f0/ : SUCCESS in 1h 05m 08s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 http://logs.openstack.org/33/598933/1/check/puppet-openstack-integration-5-scenario003-tempest-centos-7/7820f0d/ : SUCCESS in 1h 04m 11s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-mimic http://logs.openstack.org/33/598933/1/check/puppet-openstack-integration-5-scenario004-tempest-centos-7-mimic/4c3c2c8/ : SUCCESS in 53m 20s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic http://logs.openstack.org/33/598933/1/check/puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic/f6dace4/ : SUCCESS in 47m 23s\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic http://logs.openstack.org/33/598933/1/check/puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic/d97ed62/ : SUCCESS in 49m 29s\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic http://logs.openstack.org/33/598933/1/check/puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic/ecc0d0c/ : SUCCESS in 52m 04s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic http://logs.openstack.org/33/598933/1/check/puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic/9190504/ : SUCCESS in 38m 50s (non-voting)\n- puppet-openstack-integration-4-scenario001-tempest-debian-stable-luminous http://logs.openstack.org/33/598933/1/check/puppet-openstack-integration-4-scenario001-tempest-debian-stable-luminous/de7f0ae/ : FAILURE in 40m 36s (non-voting)\n- puppet-openstack-integration-4-scenario002-tempest-debian-stable http://logs.openstack.org/33/598933/1/check/puppet-openstack-integration-4-scenario002-tempest-debian-stable/0eaf26f/ : FAILURE in 45m 16s (non-voting)\n- puppet-openstack-integration-4-scenario003-tempest-debian-stable http://logs.openstack.org/33/598933/1/check/puppet-openstack-integration-4-scenario003-tempest-debian-stable/db641de/ : FAILURE in 36m 09s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-debian-stable-luminous http://logs.openstack.org/33/598933/1/check/puppet-openstack-integration-4-scenario004-tempest-debian-stable-luminous/5b6d427/ : FAILURE in 7m 45s (non-voting)\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/33/598933/1/check/puppet-openstack-beaker-centos-7/1a58c96/ : SUCCESS in 18m 42s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/33/598933/1/check/puppet-openstack-beaker-ubuntu-bionic/6630612/ : SUCCESS in 11m 34s","accounts_in_message":[],"_revision_number":1},{"id":"7ed87f967d29ee92b17acc9448a99535c7dd4178","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2018-08-31 17:00:23.000000000","message":"Patch Set 1: Code-Review+1\n\nreckeck","accounts_in_message":[],"_revision_number":1},{"id":"871d976c19ded11616fa3c57ae7b74c3e7eadd73","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2018-09-01 11:41:17.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"64162339535f38182146abc7379cd6a394a5524f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-01 12:59:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint http://logs.openstack.org/33/598933/1/check/puppet-openstack-lint/774433a/ : SUCCESS in 3m 36s\n- puppet-openstack-syntax-4 http://logs.openstack.org/33/598933/1/check/puppet-openstack-syntax-4/d13a9bb/ : SUCCESS in 3m 55s\n- puppet-openstack-syntax-5 http://logs.openstack.org/33/598933/1/check/puppet-openstack-syntax-5/1203e20/ : SUCCESS in 3m 46s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/33/598933/1/check/puppet-openstack-unit-4.8-centos-7/a2f9345/ : SUCCESS in 12m 05s\n- puppet-openstack-integration-4-scenario004-tempest-centos-7-mimic http://logs.openstack.org/33/598933/1/check/puppet-openstack-integration-4-scenario004-tempest-centos-7-mimic/c41dda8/ : SUCCESS in 1h 15m 58s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous http://logs.openstack.org/33/598933/1/check/puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous/b995ee4/ : SUCCESS in 1h 01m 15s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 http://logs.openstack.org/33/598933/1/check/puppet-openstack-integration-5-scenario002-tempest-centos-7/c333ae6/ : SUCCESS in 1h 03m 45s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 http://logs.openstack.org/33/598933/1/check/puppet-openstack-integration-5-scenario003-tempest-centos-7/3ffa024/ : SUCCESS in 1h 12m 51s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-mimic http://logs.openstack.org/33/598933/1/check/puppet-openstack-integration-5-scenario004-tempest-centos-7-mimic/125892d/ : SUCCESS in 50m 45s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic http://logs.openstack.org/33/598933/1/check/puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic/80d6ea8/ : SUCCESS in 44m 21s\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic http://logs.openstack.org/33/598933/1/check/puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic/ae0fc46/ : SUCCESS in 46m 03s\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic http://logs.openstack.org/33/598933/1/check/puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic/f279ec4/ : FAILURE in 51m 42s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic http://logs.openstack.org/33/598933/1/check/puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic/695c558/ : SUCCESS in 41m 28s (non-voting)\n- puppet-openstack-integration-4-scenario001-tempest-debian-stable-luminous http://logs.openstack.org/33/598933/1/check/puppet-openstack-integration-4-scenario001-tempest-debian-stable-luminous/3f48373/ : FAILURE in 46m 45s (non-voting)\n- puppet-openstack-integration-4-scenario002-tempest-debian-stable http://logs.openstack.org/33/598933/1/check/puppet-openstack-integration-4-scenario002-tempest-debian-stable/0507358/ : FAILURE in 53m 40s (non-voting)\n- puppet-openstack-integration-4-scenario003-tempest-debian-stable http://logs.openstack.org/33/598933/1/check/puppet-openstack-integration-4-scenario003-tempest-debian-stable/67d354e/ : FAILURE in 38m 04s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-debian-stable-luminous http://logs.openstack.org/33/598933/1/check/puppet-openstack-integration-4-scenario004-tempest-debian-stable-luminous/32d77a9/ : FAILURE in 5m 04s (non-voting)\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/33/598933/1/check/puppet-openstack-beaker-centos-7/0b2b55d/ : SUCCESS in 19m 28s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/33/598933/1/check/puppet-openstack-beaker-ubuntu-bionic/5a0890d/ : SUCCESS in 7m 23s","accounts_in_message":[],"_revision_number":1},{"id":"fb58c65ce89f277c84d0629712edb1ca93742c6c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-01 12:59:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f05b55861836b60b514a4c40978e07cbad0f1b41","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-01 14:31:30.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint http://logs.openstack.org/33/598933/1/gate/puppet-openstack-lint/65b73e0/ : SUCCESS in 3m 46s\n- puppet-openstack-syntax-4 http://logs.openstack.org/33/598933/1/gate/puppet-openstack-syntax-4/ce5c823/ : SUCCESS in 3m 20s\n- puppet-openstack-syntax-5 http://logs.openstack.org/33/598933/1/gate/puppet-openstack-syntax-5/5f9d0e8/ : SUCCESS in 3m 25s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/33/598933/1/gate/puppet-openstack-unit-4.8-centos-7/ddb5548/ : SUCCESS in 11m 22s\n- puppet-openstack-integration-4-scenario004-tempest-centos-7-mimic http://logs.openstack.org/33/598933/1/gate/puppet-openstack-integration-4-scenario004-tempest-centos-7-mimic/9db1bf2/ : SUCCESS in 52m 40s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous http://logs.openstack.org/33/598933/1/gate/puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous/bfede50/ : SUCCESS in 1h 03m 02s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 http://logs.openstack.org/33/598933/1/gate/puppet-openstack-integration-5-scenario002-tempest-centos-7/6b27601/ : SUCCESS in 1h 01m 16s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 http://logs.openstack.org/33/598933/1/gate/puppet-openstack-integration-5-scenario003-tempest-centos-7/37f9154/ : TIMED_OUT in 1h 30m 45s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-mimic http://logs.openstack.org/33/598933/1/gate/puppet-openstack-integration-5-scenario004-tempest-centos-7-mimic/fe5ecbc/ : SUCCESS in 51m 57s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic http://logs.openstack.org/33/598933/1/gate/puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic/45d1eff/ : SUCCESS in 49m 54s\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic http://logs.openstack.org/33/598933/1/gate/puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic/6ac394b/ : SUCCESS in 44m 44s\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic http://logs.openstack.org/33/598933/1/gate/puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic/c124a38/ : SUCCESS in 50m 51s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic http://logs.openstack.org/33/598933/1/gate/puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic/10f5163/ : SUCCESS in 40m 20s (non-voting)\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/33/598933/1/gate/puppet-openstack-beaker-centos-7/7c9be48/ : SUCCESS in 15m 48s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/33/598933/1/gate/puppet-openstack-beaker-ubuntu-bionic/4590b31/ : SUCCESS in 6m 36s","accounts_in_message":[],"_revision_number":1},{"id":"342590dac6c01e91841f53f4fbe34b228d9242ca","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2018-09-02 01:58:46.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6ba0a53514b972ff23c7eab83ff6c3d0b4623dcd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-02 03:15:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint http://logs.openstack.org/33/598933/1/check/puppet-openstack-lint/9a8a975/ : SUCCESS in 3m 53s\n- puppet-openstack-syntax-4 http://logs.openstack.org/33/598933/1/check/puppet-openstack-syntax-4/c86cf2c/ : SUCCESS in 3m 41s\n- puppet-openstack-syntax-5 http://logs.openstack.org/33/598933/1/check/puppet-openstack-syntax-5/1ebd0b0/ : SUCCESS in 6m 36s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/33/598933/1/check/puppet-openstack-unit-4.8-centos-7/98e289e/ : SUCCESS in 9m 03s\n- puppet-openstack-integration-4-scenario004-tempest-centos-7-mimic http://logs.openstack.org/33/598933/1/check/puppet-openstack-integration-4-scenario004-tempest-centos-7-mimic/904fd96/ : SUCCESS in 46m 14s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous http://logs.openstack.org/33/598933/1/check/puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous/e10b39b/ : SUCCESS in 1h 04m 01s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 http://logs.openstack.org/33/598933/1/check/puppet-openstack-integration-5-scenario002-tempest-centos-7/39ab133/ : SUCCESS in 1h 00m 16s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 http://logs.openstack.org/33/598933/1/check/puppet-openstack-integration-5-scenario003-tempest-centos-7/76ffb1a/ : SUCCESS in 1h 11m 28s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-mimic http://logs.openstack.org/33/598933/1/check/puppet-openstack-integration-5-scenario004-tempest-centos-7-mimic/8f02f97/ : SUCCESS in 52m 51s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic http://logs.openstack.org/33/598933/1/check/puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic/e2b44cc/ : SUCCESS in 1h 08m 43s\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic http://logs.openstack.org/33/598933/1/check/puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic/82f9399/ : SUCCESS in 45m 49s\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic http://logs.openstack.org/33/598933/1/check/puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic/f9ab0ac/ : SUCCESS in 50m 25s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic http://logs.openstack.org/33/598933/1/check/puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic/5c7cd84/ : SUCCESS in 42m 15s (non-voting)\n- puppet-openstack-integration-4-scenario001-tempest-debian-stable-luminous http://logs.openstack.org/33/598933/1/check/puppet-openstack-integration-4-scenario001-tempest-debian-stable-luminous/51e6185/ : FAILURE in 40m 19s (non-voting)\n- puppet-openstack-integration-4-scenario002-tempest-debian-stable http://logs.openstack.org/33/598933/1/check/puppet-openstack-integration-4-scenario002-tempest-debian-stable/16f7574/ : FAILURE in 44m 20s (non-voting)\n- puppet-openstack-integration-4-scenario003-tempest-debian-stable http://logs.openstack.org/33/598933/1/check/puppet-openstack-integration-4-scenario003-tempest-debian-stable/6b82827/ : FAILURE in 5m 26s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-debian-stable-luminous http://logs.openstack.org/33/598933/1/check/puppet-openstack-integration-4-scenario004-tempest-debian-stable-luminous/eb1049f/ : FAILURE in 34m 11s (non-voting)\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/33/598933/1/check/puppet-openstack-beaker-centos-7/978b6c8/ : SUCCESS in 16m 54s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/33/598933/1/check/puppet-openstack-beaker-ubuntu-bionic/f637514/ : SUCCESS in 9m 00s","accounts_in_message":[],"_revision_number":1},{"id":"b3e47c91cc0e9f66cfa2e3f6384b6c2372833aed","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-02 03:15:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0432e8234a44b1d03ba5c553831c73b9e3e22019","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-02 04:23:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"715de7052609f1bcde00d4b03e1f5d6512e7f9df","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-02 04:23:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint http://logs.openstack.org/33/598933/1/gate/puppet-openstack-lint/485e91a/ : SUCCESS in 4m 10s\n- puppet-openstack-syntax-4 http://logs.openstack.org/33/598933/1/gate/puppet-openstack-syntax-4/98afd83/ : SUCCESS in 3m 19s\n- puppet-openstack-syntax-5 http://logs.openstack.org/33/598933/1/gate/puppet-openstack-syntax-5/bb8c4af/ : SUCCESS in 4m 00s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/33/598933/1/gate/puppet-openstack-unit-4.8-centos-7/a85725f/ : SUCCESS in 12m 05s\n- puppet-openstack-integration-4-scenario004-tempest-centos-7-mimic http://logs.openstack.org/33/598933/1/gate/puppet-openstack-integration-4-scenario004-tempest-centos-7-mimic/f1b24b5/ : SUCCESS in 45m 36s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous http://logs.openstack.org/33/598933/1/gate/puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous/25ec4ec/ : SUCCESS in 1h 01m 39s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 http://logs.openstack.org/33/598933/1/gate/puppet-openstack-integration-5-scenario002-tempest-centos-7/490f7d8/ : SUCCESS in 1h 01m 48s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 http://logs.openstack.org/33/598933/1/gate/puppet-openstack-integration-5-scenario003-tempest-centos-7/750e024/ : SUCCESS in 1h 07m 10s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-mimic http://logs.openstack.org/33/598933/1/gate/puppet-openstack-integration-5-scenario004-tempest-centos-7-mimic/17918a4/ : SUCCESS in 51m 41s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic http://logs.openstack.org/33/598933/1/gate/puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic/c54eac1/ : SUCCESS in 49m 05s\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic http://logs.openstack.org/33/598933/1/gate/puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic/c3b8a29/ : SUCCESS in 48m 13s\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic http://logs.openstack.org/33/598933/1/gate/puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic/0ae4f32/ : FAILURE in 50m 29s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic http://logs.openstack.org/33/598933/1/gate/puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic/c12a5a7/ : SUCCESS in 34m 28s (non-voting)\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/33/598933/1/gate/puppet-openstack-beaker-centos-7/845cb2c/ : SUCCESS in 17m 43s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/33/598933/1/gate/puppet-openstack-beaker-ubuntu-bionic/a9538c9/ : SUCCESS in 7m 21s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"dd3dda580773fcce4adf065aa3bba9a9e8a9f842","revisions":{"dd3dda580773fcce4adf065aa3bba9a9e8a9f842":{"kind":"REWORK","_number":1,"created":"2018-08-31 13:02:15.000000000","uploader":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"ref":"refs/changes/33/598933/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tempest","ref":"refs/changes/33/598933/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tempest refs/changes/33/598933/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tempest refs/changes/33/598933/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tempest refs/changes/33/598933/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tempest refs/changes/33/598933/1"}}},"commit":{"parents":[{"commit":"d50dd8c93a086ec992df65b34681310b012ff751","subject":"Update Gemfile for stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tempest/commit/d50dd8c93a086ec992df65b34681310b012ff751"}]}],"author":{"name":"Doug Hellmann","email":"doug@doughellmann.com","date":"2018-08-31 12:48:19.000000000","tz":-240},"committer":{"name":"Doug Hellmann","email":"doug@doughellmann.com","date":"2018-08-31 12:48:19.000000000","tz":-240},"subject":"import zuul job settings from project-config","message":"import zuul job settings from project-config\n\nThis is a mechanically generated patch to complete step 1 of moving\nthe zuul job settings out of project-config and into each project\nrepository.\n\nBecause there will be a separate patch on each branch, the branch\nspecifiers for branch-specific jobs have been removed.\n\nBecause this patch is generated by a script, there may be some\ncosmetic changes to the layout of the YAML file(s) as the contents are\nnormalized.\n\nSee the python3-first goal document for details:\nhttps://governance.openstack.org/tc/goals/stein/python3-first.html\n\nChange-Id: Ia2f0addc2fb668d82424c144246b5c24647ef7cc\nStory: #2002586\nTask: #24325\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tempest/commit/dd3dda580773fcce4adf065aa3bba9a9e8a9f842"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tempest/commit/dd3dda580773fcce4adf065aa3bba9a9e8a9f842"}]},"parents_data":[{"branch_name":"refs/heads/stable/rocky","commit_id":"d50dd8c93a086ec992df65b34681310b012ff751","is_merged_in_target_branch":true,"change_id":"I3fda91d2636639064f8868ce8ca05735d2ef7774","change_number":597096,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
