)]}'
{"id":"openstack%2Fpuppet-tempest~676081","triplet_id":"openstack%2Fpuppet-tempest~master~If5525382b4b900af8a18eb37b9a8c3c1e4f9fb65","project":"openstack/puppet-tempest","branch":"master","topic":"fix-openstackdocstheme","hashtags":[],"change_id":"If5525382b4b900af8a18eb37b9a8c3c1e4f9fb65","subject":"Bump the openstackdocstheme extension to 1.20","status":"MERGED","created":"2019-08-13 03:30:14.000000000","updated":"2019-08-13 06:08:09.000000000","submitted":"2019-08-13 06:08:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"676081-1565676489466-d4deb22e","meta_rev_id":"766a588f2f553d40c60c5447a245fd7c8741fe56","_number":676081,"virtual_id_number":676081,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2019-08-13 06:08:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2019-08-13 06:00:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":1,"date":"2019-08-13 06:00:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-13 06:08:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"59f2a87227ccf188d0257e0715a931c35a0e9f8d","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-08-13 03:30:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ab23f12e35137382a1d3cdb4e5636c94abad619d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-13 03:33:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/958256160da4419d9c22c14c1de71ee6 : SUCCESS in 2m 34s","accounts_in_message":[],"_revision_number":1},{"id":"7de131b55c24c574b9b1188fe7ca21d049f1152a","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-08-13 06:00:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5e7c3c907dc3b7d316f7da637d17f4c3296912ff","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-13 06:00:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9c094f2ca35d2a8814183a151c819adfa8dd5a20","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-13 06:08:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d7118c36b4e8cf1a16c6f773d416286aee80646a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-13 06:08:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1809e0c1f11c401dbfbdb53625dd8cf8 : SUCCESS in 4m 26s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0c43f799f73a0286d0c3147c16444f675b5ea60b","revisions":{"0c43f799f73a0286d0c3147c16444f675b5ea60b":{"kind":"REWORK","_number":1,"created":"2019-08-13 03:30:14.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/81/676081/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tempest","ref":"refs/changes/81/676081/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tempest refs/changes/81/676081/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tempest refs/changes/81/676081/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tempest refs/changes/81/676081/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tempest refs/changes/81/676081/1"}}},"commit":{"parents":[{"commit":"483805f5e8c3d6b16010f143f16a77f90e65df7f","subject":"Sync Sphinx requirement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tempest/commit/483805f5e8c3d6b16010f143f16a77f90e65df7f"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-08-13 03:29:08.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-08-13 03:29:08.000000000","tz":480},"subject":"Bump the openstackdocstheme extension to 1.20","message":"Bump the openstackdocstheme extension to 1.20\n\nSome options are now automatically configured by the version 1.20[1]:\n- project\n- html_last_updated_fmt\n- latex_engine\n- latex_elements\n\n[1]https://docs.openstack.org/openstackdocstheme/latest/#using-the-theme\n\nChange-Id: If5525382b4b900af8a18eb37b9a8c3c1e4f9fb65\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tempest/commit/0c43f799f73a0286d0c3147c16444f675b5ea60b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tempest/commit/0c43f799f73a0286d0c3147c16444f675b5ea60b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
