)]}'
{"id":"openstack%2Fpuppet-tempest~722288","triplet_id":"openstack%2Fpuppet-tempest~stable%2Ftrain~Iac6bd17448b52d6a357b6e68c2b4d6fc28dc4cee","project":"openstack/puppet-tempest","branch":"stable/train","hashtags":[],"change_id":"Iac6bd17448b52d6a357b6e68c2b4d6fc28dc4cee","subject":"Use python3-swift-tests package on CentOS8","status":"MERGED","created":"2020-04-23 12:19:16.000000000","updated":"2020-04-26 15:50:45.000000000","submitted":"2020-04-26 15:50:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"722288-1587916245262-9a8ae8f1","meta_rev_id":"85d5500b65bdd9c2ddba933625cea104df55e80d","_number":722288,"virtual_id_number":722288,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-26 15:50:45.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2020-04-24 21:30:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2020-04-26 14:49:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2020-04-26 14:49:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-24 21:30:32.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-04-26 14:49:29.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-04-26 15:50:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3a3090346fe71cd7f4c3f24f24fdf40511f525ee","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-23 12:19:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7bbdc341aa48f72d15b176c45dd669d7176c6f5c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 13:30:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/91f5f7c690de4158aa95bd95945be9d7 : SUCCESS in 4m 36s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3b5412c8a3414d0ebd0eb93aa254f603 : SUCCESS in 5m 49s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b405de37bec743e59ec121e326c4e94a : SUCCESS in 5m 08s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/8ea7ec0c898a48c9807a6c21c4bed099 : SUCCESS in 10m 13s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/83702105734741a9a8044ba5c4195cac : SUCCESS in 13m 23s\n- puppet-openstack-unit-6.4-centos-8 https://zuul.opendev.org/t/openstack/build/72048b4630324599a8f407f0071e4d66 : SUCCESS in 13m 25s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f107350c4a5d4973a0b4abb6b159ad3f : SUCCESS in 9m 41s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d7bc084331ba41bf853e2d1ba894ab9d : SUCCESS in 8m 37s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/e04a8dde4fe34b39924eaadcaf3be51f : SUCCESS in 52m 51s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/8840c34f688c4ecea77e70d9343ccfb3 : SUCCESS in 1h 04m 59s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/ef45dc07f06c4b318efea4cf1eed2869 : SUCCESS in 1h 00m 59s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/6b81f229a1d24e648a44c74cc060dfa6 : SUCCESS in 46m 28s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/d5dc36127bc34b1b894b0e8ef08d6ad2 : SUCCESS in 47m 31s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/99aaa1aa9b80496dad40e162995957a5 : SUCCESS in 50m 34s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d537d6d5a52c472da59c05fdcf056838 : SUCCESS in 43m 25s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/4654c79fc62f4bf7879597725277cc6e : SUCCESS in 38m 01s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/e3495e9ae70548e79d828aa73e4b76b7 : SUCCESS in 54m 52s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/c37ea37058854aa2805df9a4860c5cd9 : SUCCESS in 1h 04m 35s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/d67f5a9f02484115a018ebe0a538b340 : SUCCESS in 59m 57s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/434ad3bf988e44278083a373904fc7a1 : SUCCESS in 51m 35s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/cf66b804b8c342c9943d88cb94bac4c8 : SUCCESS in 17m 01s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b8f046458f04456eb287d6ea6002f877 : SUCCESS in 11m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4fac346f63b4a75364af10d1b571e5c89fa14e12","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-04-24 21:30:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d9d2601c0339081b1e675751de8dcf2f1c122bdd","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-04-24 21:30:35.000000000","message":"Patch Set 1: -Workflow","accounts_in_message":[],"_revision_number":1},{"id":"80cbf1573d3e193803b1013366210258fe986345","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-04-26 14:49:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7eaeec17f9069c5c16219d67cc19a1d0ca4b7786","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-26 14:49:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c9be9f2388597e762c179ddd259db397d107c2d2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-26 15:50:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bf4a79ff0c7846c4953ea508829ee6ec : SUCCESS in 4m 48s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/76a4a7c33ab54078955d63b01352d9e1 : SUCCESS in 4m 26s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fca0e10674d14b9d9a50a804320416b6 : SUCCESS in 4m 39s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/98603b8c4f304912b783a87aada3fe2d : SUCCESS in 9m 31s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8141b1171c644d1493835fb16a1b99ab : SUCCESS in 8m 30s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/149657de14174c0b88079e3a6de06ac1 : SUCCESS in 53m 36s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/b5af916bb9a54fa39ab7ad8ca96807a7 : SUCCESS in 55m 54s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/423d1ac7915f40298a7d4af4bb2f59f0 : SUCCESS in 1h 00m 08s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/8e961c9bab6e4ea08eb4e97b7922a95d : SUCCESS in 42m 08s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/4a4484df647447d1bdea07e866055f13 : SUCCESS in 14m 38s","accounts_in_message":[],"_revision_number":1},{"id":"93601a696ee88983854141842040d77a37ff6768","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-26 15:50:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8ad7ca52d4620705982df85ae367b793bdbfe7f7","revisions":{"8ad7ca52d4620705982df85ae367b793bdbfe7f7":{"kind":"REWORK","_number":1,"created":"2020-04-23 12:19:16.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/88/722288/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tempest","ref":"refs/changes/88/722288/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tempest refs/changes/88/722288/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tempest refs/changes/88/722288/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tempest refs/changes/88/722288/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tempest refs/changes/88/722288/1"}}},"commit":{"parents":[{"commit":"f74827bbbfe50d886580505d5ad40821be09fc0b","subject":"Merge \"Switch to Train\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tempest/commit/f74827bbbfe50d886580505d5ad40821be09fc0b"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-04-22 02:00:56.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-04-23 12:19:09.000000000","tz":540},"subject":"Use python3-swift-tests package on CentOS8","message":"Use python3-swift-tests package on CentOS8\n\nChange-Id: Iac6bd17448b52d6a357b6e68c2b4d6fc28dc4cee\n(cherry picked from commit 2f4f7d70de08ef66abb0d80d2439a32f82539bd5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tempest/commit/8ad7ca52d4620705982df85ae367b793bdbfe7f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tempest/commit/8ad7ca52d4620705982df85ae367b793bdbfe7f7"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
