)]}'
{"id":"openstack%2Fpuppet-tempest~847915","triplet_id":"openstack%2Fpuppet-tempest~master~Iecbf8ca8bde85dd011352c1f51516ed6f87e77d5","project":"openstack/puppet-tempest","branch":"master","topic":"ovn-octavia","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-05-02 18:08:43.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Iecbf8ca8bde85dd011352c1f51516ed6f87e77d5","subject":"wip: Add more octavia-tempest-plugin parameters","status":"ABANDONED","created":"2022-06-28 08:50:45.000000000","updated":"2024-05-02 18:08:43.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"88be8e06274a3eb5e5a93cf766fc32a52d3aeb70","_number":847915,"virtual_id_number":847915,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-06-29 18:37:13.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-28 09:57:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1814cefebd9eb8fb3fe4adc414c72aee225e64dc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-06-28 08:50:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f928aba41af57aea15db1179f9bd9412809af7c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-28 09:57:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/675ab8e07a414454be7dd54813bca740 : FAILURE in 3m 21s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b3112364fc5c4f27ac24d118741df21b : SUCCESS in 3m 16s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c6577c3137c4492e851c84c418a73a45 : SUCCESS in 3m 01s\n- puppet-openstack-unit-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/3fb6aa0bf52b485187c805cd0cae6b70 : FAILURE in 11m 04s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5161c73ec0cf4941831252c01b209d38 : FAILURE in 7m 49s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/84a2ff5dc31545f79e876f920e744022 : FAILURE in 7m 50s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/217facbf93f14e038e41c5f85f586edf : FAILURE in 7m 44s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e9c4f02ed70246688fb3ad6def2d61e8 : FAILURE in 25m 33s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c04ae6677d764207b0fbb9ea4ad06797 : FAILURE in 42m 47s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8e71c155ccbb4feba6eb4240fd82e548 : FAILURE in 36m 57s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/477d6300a92c43d496e2e15a0a754541 : FAILURE in 23m 22s (non-voting)\n- puppet-openstack-integration-6-scenario005-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/22582830014c41c980b65c0170c7fda2 : FAILURE in 33m 01s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a336197dd7234c6da93848bc127b98da : SUCCESS in 20m 34s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c23e8a34d01942cbbdf275d4fed6cd44 : SUCCESS in 53m 16s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e4c70858b201408082d12a8e2c4efbd2 : SUCCESS in 1h 01m 21s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/baa9ca79a8d141ab8cba4f645e7b18f8 : SUCCESS in 58m 29s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6a0bb9f4f299462f89a39ce3063cb369 : SUCCESS in 49m 00s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f3febe12f2664c78b71d320f22f475ed : SUCCESS in 37m 55s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bc4e12e2a3f74559b4f9e5314ccf6cd9 : SUCCESS in 8m 53s (non-voting)\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/1efd86344fc84341bdd6e36957a59d32 : SUCCESS in 15m 02s\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b91849dc96724263a00d81e0598ed1be : SUCCESS in 9m 01s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d6b2ac2d8bda40d98ab30bb45e471380 : SUCCESS in 8m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"932061e237a24b21cfc5afe6759120e3d04dda41","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-06-28 10:29:51.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0517851ad0d80e6067791fc553eb786691137fc4","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-06-28 10:29:57.000000000","message":"Topic set to ovn-octavia","accounts_in_message":[],"_revision_number":2},{"id":"bc948cff1a06e49a0cc8d3120786711d746c554a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-28 11:41:59.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7fa85924de7e42629d1146f99f8f5b47 : SUCCESS in 3m 33s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d7c0de3c59044ba1a96b3f8620a8f87f : SUCCESS in 2m 42s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4bd3767cd8a44cd2a8292e73579d392e : SUCCESS in 3m 36s\n- puppet-openstack-unit-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/610dddecc5bc48a9bcf106eeec79daa1 : FAILURE in 11m 37s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/abdab45ff0bc469da410d2050dd79a54 : FAILURE in 8m 46s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/81692292084e493e9516bbca2b9455a0 : FAILURE in 8m 35s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/04438c87e40f42909fc89f7fac6bd1d4 : FAILURE in 8m 32s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b0239c19d62f4d288423b672399c2802 : FAILURE in 24m 45s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e9ba9104f8fa40a196aeefa5d7029be0 : FAILURE in 38m 14s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2e5ee9430ce54d1c82dba911efe35469 : FAILURE in 37m 29s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f815244a698a446a83f64744983289fb : FAILURE in 24m 05s (non-voting)\n- puppet-openstack-integration-6-scenario005-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/54e51e2a9c0e42749927540719ae5776 : FAILURE in 29m 21s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b748538efda04c01afebbb474736e6a6 : SUCCESS in 21m 23s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d0984978ae544aec8da9d600c7e14d18 : SUCCESS in 56m 48s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/862ff18816de4b619e0713f4b592ef88 : SUCCESS in 1h 09m 14s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/535b2e9afe23414ea1a65f783536efe3 : SUCCESS in 59m 54s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9078d84a8676471998489167965c0bed : SUCCESS in 48m 08s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/431c8c09f81c4f50998e6b81f1f71e27 : SUCCESS in 35m 39s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/954d2860a48c4954977c8e44ba2fea90 : SUCCESS in 10m 04s (non-voting)\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/faa04aa9fa6a429db6da09f4cbc7ea32 : SUCCESS in 14m 27s\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/22a925a732b84a74b01f98bfe22c0e91 : SUCCESS in 9m 17s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/02eabbad5dc14986ae82f2267905a009 : SUCCESS in 9m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ab99357c18e93302c389d5049a59453b187f656b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-06-29 16:28:39.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"207a69e80d60af1657354571dbfd79e964c064ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-29 18:37:13.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/81fff2323f16423ea402496f0b5932a7 : SUCCESS in 3m 26s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a15d2bc43b3341c39ab0670e29c031b1 : SUCCESS in 2m 51s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/40efe4bed99445e68f4ecfcfbcc0acf8 : SUCCESS in 2m 57s\n- puppet-openstack-unit-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/79e883f259004d7caab4e2d226a83245 : FAILURE in 11m 18s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b8be98914a2a4100a4bf50229b2109e0 : FAILURE in 7m 34s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/40526138f66b4cdea1c0d2a9f71e5df4 : FAILURE in 7m 47s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/786a3881f7b24418a4a04a3f65198f3f : FAILURE in 7m 43s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7a48170569864393a86bae06e8bcd55b : FAILURE in 22m 07s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6d3a7ba4fb224faca7bf9d64e4baffa8 : FAILURE in 33m 48s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a38c9adb0171428bb74d3726d1c7c96c : FAILURE in 34m 28s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b6dec42ba20f48a6ba3d29ff31d5fc4e : FAILURE in 21m 15s (non-voting)\n- puppet-openstack-integration-6-scenario005-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d3e6ec257ef443bcb48831e50de8981d : FAILURE in 25m 56s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2c9d66e86704457f94aae4eb0a15c20f : SUCCESS in 18m 00s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c71a1c37bd604c46a6e7c0ca03ae568f : TIMED_OUT in 2h 04m 37s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7250956bd7854f9cab2b0a36f94598dd : SUCCESS in 55m 19s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3a4b1482df3c43df9d45fe4e261fe180 : SUCCESS in 54m 41s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ec484785c4274994989d3b8c9ceaece6 : SUCCESS in 45m 23s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8af1dde205eb41d8b3c69783a34724ba : SUCCESS in 31m 44s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b7abd9f007244f70abbf1ee276341896 : SUCCESS in 8m 58s (non-voting)\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/b9cf8fb4ff8541188b628b2e7e1d3bbd : SUCCESS in 13m 50s\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9e09914cc5964b9e862e3ee5b04d5ece : SUCCESS in 9m 05s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/88c036d5cea642cabb541d54de749b9b : SUCCESS in 8m 21s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"88be8e06274a3eb5e5a93cf766fc32a52d3aeb70","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-02 18:08:43.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"b0886d6267860644dbfac2e44b9e26e9c4bcd938","revisions":{"06644df1bec6a7e45dae813b215383edadd3164e":{"kind":"REWORK","_number":1,"created":"2022-06-28 08:50:45.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/15/847915/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tempest","ref":"refs/changes/15/847915/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tempest refs/changes/15/847915/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tempest refs/changes/15/847915/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tempest refs/changes/15/847915/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tempest refs/changes/15/847915/1"}}},"commit":{"parents":[{"commit":"5928ffdd14b309d303dda9bf3ba153d6c192caf1","subject":"Merge \"Allow arbitrary section name for *_id_setter\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tempest/commit/5928ffdd14b309d303dda9bf3ba153d6c192caf1"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-06-28 08:50:20.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-06-28 08:50:20.000000000","tz":540},"subject":"wip: Add more octavia-tempest-plugin parameters","message":"wip: Add more octavia-tempest-plugin parameters\n\nChange-Id: Iecbf8ca8bde85dd011352c1f51516ed6f87e77d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tempest/commit/06644df1bec6a7e45dae813b215383edadd3164e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tempest/commit/06644df1bec6a7e45dae813b215383edadd3164e"}]},"branch":"refs/heads/master"},"5f7c9e71c8d3d2362d67e9d322a0e6eb9de88364":{"kind":"REWORK","_number":2,"created":"2022-06-28 10:29:51.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/15/847915/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tempest","ref":"refs/changes/15/847915/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tempest refs/changes/15/847915/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tempest refs/changes/15/847915/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tempest refs/changes/15/847915/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tempest refs/changes/15/847915/2"}}},"commit":{"parents":[{"commit":"5928ffdd14b309d303dda9bf3ba153d6c192caf1","subject":"Merge \"Allow arbitrary section name for *_id_setter\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tempest/commit/5928ffdd14b309d303dda9bf3ba153d6c192caf1"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-06-28 08:50:20.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-06-28 10:29:42.000000000","tz":540},"subject":"wip: Add more octavia-tempest-plugin parameters","message":"wip: Add more octavia-tempest-plugin parameters\n\nChange-Id: Iecbf8ca8bde85dd011352c1f51516ed6f87e77d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tempest/commit/5f7c9e71c8d3d2362d67e9d322a0e6eb9de88364"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tempest/commit/5f7c9e71c8d3d2362d67e9d322a0e6eb9de88364"}]},"branch":"refs/heads/master"},"b0886d6267860644dbfac2e44b9e26e9c4bcd938":{"kind":"REWORK","_number":3,"created":"2022-06-29 16:28:39.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/15/847915/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tempest","ref":"refs/changes/15/847915/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tempest refs/changes/15/847915/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tempest refs/changes/15/847915/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tempest refs/changes/15/847915/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tempest refs/changes/15/847915/3"}}},"commit":{"parents":[{"commit":"5928ffdd14b309d303dda9bf3ba153d6c192caf1","subject":"Merge \"Allow arbitrary section name for *_id_setter\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tempest/commit/5928ffdd14b309d303dda9bf3ba153d6c192caf1"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-06-28 08:50:20.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-06-29 16:28:31.000000000","tz":540},"subject":"wip: Add more octavia-tempest-plugin parameters","message":"wip: Add more octavia-tempest-plugin parameters\n\nChange-Id: Iecbf8ca8bde85dd011352c1f51516ed6f87e77d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tempest/commit/b0886d6267860644dbfac2e44b9e26e9c4bcd938"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tempest/commit/b0886d6267860644dbfac2e44b9e26e9c4bcd938"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
