)]}'
{"id":"openstack%2Fpuppet-tempest~864493","triplet_id":"openstack%2Fpuppet-tempest~master~Iacb4a192fbd1573d89f62890bc0ef80998e41cd3","project":"openstack/puppet-tempest","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-11-16 12:27:28.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iacb4a192fbd1573d89f62890bc0ef80998e41cd3","subject":"Remove deprecated parameter for whitebox tempest plugin","status":"MERGED","created":"2022-11-15 05:57:36.000000000","updated":"2022-11-16 12:28:26.000000000","submitted":"2022-11-16 12:27:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"864493","meta_rev_id":"a9de876bed6d5b462c5de85ddd9709efaac8ea3e","_number":864493,"virtual_id_number":864493,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-16 12:27:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2022-11-15 15:12:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2022-11-16 11:11:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2022-11-16 11:11:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-15 07:13:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-15 15:12:59.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2022-11-16 11:11:16.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"8155a97b7557bc46311bdae103ddb7aa850889ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-15 05:57:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2b7e10c5b1138ccacf08895475b5c2600fb684c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-15 07:13:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/84937877cdca472e94bab444a0b7343e\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/692a617c3e2b41b78630be454eebc8e0 : SUCCESS in 3m 19s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5aea4968787c4f688de8f27484183f57 : SUCCESS in 3m 18s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a724b3e099bc462abe74669ab1c5d3a2 : SUCCESS in 2m 52s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/a6a6f6b9b7f54c80897a936b5828f9ab : SUCCESS in 8m 38s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f3dd3ddec2534a0bb8736427fc94306d : SUCCESS in 5m 37s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c0be280dbd6543e2b0202ea8e53f5720 : SUCCESS in 5m 29s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2434d1860d2047068457be63f4cd7fe7 : SUCCESS in 5m 24s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/395fffd223304cd1aa15f58f9b26a567 : SUCCESS in 47m 44s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e848e7198a3b4a9e9b0d9d667193e04a : SUCCESS in 56m 03s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7ca26207444849619496722aeb0aed92 : SUCCESS in 43m 47s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3096992afa1e43c694a74998ac184498 : SUCCESS in 34m 13s (non-voting)\n- puppet-openstack-integration-6-scenario005-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3f9f6b6f449747bcb33e435ac157ae2c : SUCCESS in 31m 07s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/21c715f4c406456eaba95ec93f2eec93 : SUCCESS in 20m 25s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/60a3f11117c94e13bf7b8dfb1362d686 : SUCCESS in 49m 11s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/27a0482d3ed4404499a72d9522413738 : SUCCESS in 1h 14m 35s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6b6b8d6938864bbd81350b46411faecb : SUCCESS in 50m 06s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/11ebc97761a24d5e87ef553eba16f136 : SUCCESS in 54m 39s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6feb88fba69e4dfb9b04f5ac408e312f : SUCCESS in 51m 12s\n- puppet-openstack-litmus-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a5ad2c60fe8d499c95b00731a69c0552 : SUCCESS in 8m 03s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/9b5207d30579480991965ed666eb0cc5 : SUCCESS in 11m 17s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0e4478e0c9c849f6b0f4290518a5db16 : SUCCESS in 8m 41s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d8dbc1fbaef04848951db62415f5d1ff : SUCCESS in 6m 33s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f2417cf81fb34402ba49936294b402ea : SUCCESS in 2m 51s","accounts_in_message":[],"_revision_number":1},{"id":"43a3dc37beb4bb5dedd6da6b74b681480b6302d3","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-11-15 15:12:59.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"526f1997c29557e4bc43df589f41e34d6d0ebeb0","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-11-16 11:11:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0cd001d69cf67ab20a7cb49def56d408fa373ad9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-16 11:12:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"569e800bc7556cf077c94db768e2b2582eb92df3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-16 12:27:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4fe28b46fae44cc78e2711b5abb7541e\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4cf33f19863849dea22988de117f0038 : SUCCESS in 2m 45s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9e9f903f97294f3f85690645a77bd7f1 : SUCCESS in 2m 39s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/724e8c4a5a5041e3be2b7e2511744933 : SUCCESS in 2m 45s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/5a9a5ca8432e42459049118ccd847dd5 : SUCCESS in 9m 20s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/51e4306ace8e4988b1b8ecb18b29cf82 : SUCCESS in 6m 11s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/20237150d4684d029fc19dcaeb27324c : SUCCESS in 5m 43s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4372e403d9824cef96f3af8d3b05caa4 : SUCCESS in 58m 23s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/751b030c8f9d4a6bb43ef9975dbeedd6 : SUCCESS in 1h 10m 43s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a092c835d7ef46698b54f34cb6b3062c : SUCCESS in 1h 01m 31s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/94530b4d9af34ac88bd26fb4eb7f4c84 : SUCCESS in 51m 06s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1ce6b77b9b9d45fc99a9abc8969b233e : SUCCESS in 52m 22s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/96ea13b487c74b4db8e28e166767712a : SUCCESS in 14m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/422e93445f0c47d0a29e9fda099cc9d3 : SUCCESS in 3m 01s","accounts_in_message":[],"_revision_number":1},{"id":"4b3c08b777d5d1ce0373da127a1ad3fae4746eca","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-16 12:27:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a9de876bed6d5b462c5de85ddd9709efaac8ea3e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-16 12:28:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a0ac6e04fd9c4c6a9141753b34ffbe96\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/522399ed835e4b29babfbfbb1e7e164f : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"786ed8421df880ed923f4b804b5a9b528ce50b06","revisions":{"786ed8421df880ed923f4b804b5a9b528ce50b06":{"kind":"REWORK","_number":1,"created":"2022-11-15 05:57:36.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/93/864493/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tempest","ref":"refs/changes/93/864493/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tempest refs/changes/93/864493/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tempest refs/changes/93/864493/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tempest refs/changes/93/864493/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tempest refs/changes/93/864493/1"}}},"commit":{"parents":[{"commit":"d42c8a1e241bdd36309388b601955c6f144f9c90","subject":"Update master for stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tempest/commit/d42c8a1e241bdd36309388b601955c6f144f9c90"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-11-15 05:56:44.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-11-15 05:57:25.000000000","tz":540},"subject":"Remove deprecated parameter for whitebox tempest plugin","message":"Remove deprecated parameter for whitebox tempest plugin\n\n... because it was deprecated during the previous cycle by [1].\n\n[1] 9f2d6ea65016f65e9ed21df8353fac0fb81ae369\n\nChange-Id: Iacb4a192fbd1573d89f62890bc0ef80998e41cd3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tempest/commit/786ed8421df880ed923f4b804b5a9b528ce50b06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tempest/commit/786ed8421df880ed923f4b804b5a9b528ce50b06"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[]}
