)]}'
{"id":"openstack%2Fpuppet-tempest~931758","triplet_id":"openstack%2Fpuppet-tempest~stable%2F2024.1~If1fddd56828c93c19499ac8dadb2337b3f2afc46","project":"openstack/puppet-tempest","branch":"stable/2024.1","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-10-09 15:41:19.000000000","reason":"Change was submitted"},"14985":{"account":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"last_update":"2024-10-09 14:16:27.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"If1fddd56828c93c19499ac8dadb2337b3f2afc46","subject":"Support customizing catalog_type options","status":"MERGED","created":"2024-10-08 10:08:04.000000000","updated":"2024-10-09 15:42:22.000000000","submitted":"2024-10-09 15:41:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"931758","meta_rev_id":"d22da74f828354442e53a2ed6800a3c66d3d4c28","_number":931758,"virtual_id_number":931758,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-09 15:41:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2024-10-09 05:59:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2024-10-09 14:16:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2024-10-09 14:16:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-08 12:27:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-08 15:22:29.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2024-10-09 05:59:32.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"8756ba6085f8f61b2d624e029b87b64da2d27532","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-08 10:08:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"71d1cc2b015b8d307804f5107cb213ee87a99beb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-08 11:19:19.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"4b10d79bc54c0b91c4c238bfda62da57db8bbf41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-08 12:27:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f6d5b4f05ddb43ee8a34a683d50b757d\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dd1c517ae9304c9f95ed04eb19a52b8d : SUCCESS in 2m 33s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8770bdc96ab749c09c9b4b0f0282320c : SUCCESS in 2m 20s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/4ab4d30bb5ee48cea2045febcd826a8d : SUCCESS in 5m 16s (non-voting)\n- puppet-openstack-unit-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/d625f7feee3e4a8db28783dea9ebdb36 : SUCCESS in 7m 25s\n- puppet-openstack-unit-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cb02837f27494ff5912e0c9986f4d676 : SUCCESS in 4m 38s\n- puppet-openstack-unit-8.5-centos-9-stream https://zuul.opendev.org/t/openstack/build/e6ee5b9f29b54d93babdd317e21e3946 : SUCCESS in 7m 17s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8918c745acd84e47825515186656636c : SUCCESS in 5m 17s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2aadb83449014430914fe7f12e0f555e : FAILURE in 45m 40s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6d2ae105cf43440cae79345d5a1ecf6f : FAILURE in 50m 40s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c6f96ebac7f940aaad957cb6de70537b : SUCCESS in 44m 58s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bcae2c5123e54b09aaabc9e3765ce614 : FAILURE in 49m 17s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9b39af5cade7465d887f26a955a62130 : FAILURE in 23m 53s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/92d9d852529d43cfac25e66972e9df4a : SUCCESS in 57m 26s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/90818ea52d1a4961b53ae520b69797ed : SUCCESS in 43m 50s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a9cb0333b0ce42c6941cddc8509c27b3 : SUCCESS in 55m 27s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/fdde06863e7c43208aaf083cc50aaa50 : SUCCESS in 1h 02m 29s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1637cd334d314e4684f1be288dd0fcaf : SUCCESS in 52m 27s\n- puppet-openstack-litmus-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/299afe50ab074e3ea2abdc598046b70a : SUCCESS in 8m 22s\n- puppet-openstack-litmus-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1a00722a4865481d8673bef92cd4f524 : SUCCESS in 6m 13s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/45c66c3086df431c9237548abb34f4f9 : SUCCESS in 7m 58s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b75b29c509314ae39f5fdfe3f072121e : SUCCESS in 3m 20s","accounts_in_message":[],"_revision_number":2},{"id":"abfb967260f16964461cef22763825c427a94f4a","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2024-10-08 15:22:29.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"4a3d02bbc24379872bc5489c9e03ffb9a624e4f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-09 02:07:27.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"5bf44b30c65bfd0a3259fd922c951840c5e031cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-09 04:11:03.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb1ca589809e470294baf7a92a128a27\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c61d15fa45d44158a9c7473fb5b40f4c : SUCCESS in 2m 45s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/06ce353b7ac24ed68ae59b4f16009c6e : SUCCESS in 3m 07s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/63681d4458164fd587647a62dba00862 : SUCCESS in 5m 32s (non-voting)\n- puppet-openstack-unit-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/9e80e56048574aa1a3871e2c76ca7adf : SUCCESS in 8m 02s\n- puppet-openstack-unit-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a844454ea5394fdba067ed0fc03fe38e : SUCCESS in 4m 42s\n- puppet-openstack-unit-8.5-centos-9-stream https://zuul.opendev.org/t/openstack/build/5e2fe095018942ebb1c29191fb3aa27d : SUCCESS in 7m 40s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dc447f8c727d4f11bc6425bc35ec2ed3 : SUCCESS in 4m 46s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/15e6143ce5344b6fb591d4c6136f9afd : FAILURE in 42m 29s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2709f9e8b1604669a20c361af3cc162f : FAILURE in 54m 29s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/af416a11c4524007b0b6724c6a4c0018 : SUCCESS in 52m 23s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/49f4eda0963f48dfab29a886d8df5738 : SUCCESS in 51m 39s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/208ca451b64241279c72ce776bf67fa1 : FAILURE in 44m 41s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6f8147bade52433cbdf7598e3508c2ab : SUCCESS in 54m 55s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/592c64c3e05e4ed2aab6b2cdad3d9341 : SUCCESS in 1h 16m 19s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e7caa0b5d4514f14b3dd6e86b050b6dd : SUCCESS in 32m 42s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/69548b17881d4057bbf464896d849a0f : SUCCESS in 1h 00m 25s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/95f29996c2844c6b9fb94470394eb348 : SUCCESS in 56m 38s\n- puppet-openstack-litmus-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/192cdca850244bc9b49bdb9a6fe2cb1b : SUCCESS in 7m 55s\n- puppet-openstack-litmus-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a6b4f44cebba416686e30906885882a7 : SUCCESS in 6m 00s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1e0670e8c1314122a3e36a7d2d688784 : SUCCESS in 5m 50s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b9ab970f1ca44322bffacdfead9b0ec2 : SUCCESS in 2m 33s","accounts_in_message":[],"_revision_number":3},{"id":"d13200d99bf30a0f7cb57802bb6ff56ba796ab76","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-10-09 05:59:32.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"47d8fd2e861e765cd02c1c55b4dafa273bfdde6c","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2024-10-09 14:16:27.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"c01547b9cf61573bf61c3f532cdb45279ab187d2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-09 14:17:24.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"700df0d906a86818018e022602235b0753ba9007","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-09 15:41:19.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/779cc3da149e4e1996ecd3643861f860\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c4eb1704dfae4fde81c564ac07e91966 : SUCCESS in 2m 29s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c7161b7458964ffc8574a0582a9b1273 : SUCCESS in 2m 43s\n- puppet-openstack-unit-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/5290d8bbf7314b75a7c3c4e0ed24d2b0 : SUCCESS in 7m 06s\n- puppet-openstack-unit-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8491530ad4e34020b1926873b2b6764d : SUCCESS in 4m 40s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6996f8b26e3b40c49225f1e445c49a64 : SUCCESS in 55m 26s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/84050dc2c2114aff80b2e97eee578c19 : SUCCESS in 1h 16m 19s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/56a9fb68bbb94f569063c6dfacf3e62d : SUCCESS in 1h 01m 42s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/da10d27dafac4877a96b45b22ec59d07 : SUCCESS in 48m 55s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0befa3352c324304a60cf55705dbe567 : SUCCESS in 53m 17s\n- puppet-openstack-litmus-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/ac9b895cd7344afda8cc492dc6563a06 : SUCCESS in 12m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dd752543ffd74e0ba8b47a210614f91f : SUCCESS in 2m 26s","accounts_in_message":[],"_revision_number":3},{"id":"0ff245a4f7ce46563478b0afea4f2b494d222e28","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-09 15:41:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"d22da74f828354442e53a2ed6800a3c66d3d4c28","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-09 15:42:22.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b10d8d1b39b24752bd0f3bd648f0a4e9\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1177a78cd88b49dcb4f809fa1c707176 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"849acae47bf3ac974bdf8b171d92fd438452b7ce","revisions":{"20c37e168e8d0c91aedbe5e0a1cdaecfd792a753":{"kind":"REWORK","_number":1,"created":"2024-10-08 10:08:04.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/58/931758/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tempest","ref":"refs/changes/58/931758/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tempest refs/changes/58/931758/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tempest refs/changes/58/931758/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tempest refs/changes/58/931758/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tempest refs/changes/58/931758/1"}}},"commit":{"parents":[{"commit":"8e72fcdd7108699121c15d710c1198d73e09efc4","subject":"magnum: Add manage_tests_packages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tempest/commit/8e72fcdd7108699121c15d710c1198d73e09efc4"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-05 03:50:00.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-08 10:07:47.000000000","tz":540},"subject":"Support customizing catalog_type options","message":"Support customizing catalog_type options\n\nConflicts:\n\tmanifests/init.pp\n\tmanifests/magnum.pp\n\tspec/classes/tempest_magnum_spec.rb\n\nChange-Id: If1fddd56828c93c19499ac8dadb2337b3f2afc46\n(cherry picked from commit cd44577546e66ce4c1806e2e8e31e21e5a818904)\n(cherry picked from commit 1d056d66606dc7b75ad60623941b70832384d672)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tempest/commit/20c37e168e8d0c91aedbe5e0a1cdaecfd792a753"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tempest/commit/20c37e168e8d0c91aedbe5e0a1cdaecfd792a753"}]},"branch":"refs/heads/stable/2024.1"},"4e138bdd898c989301e472ca08516d4cc11e1a68":{"kind":"REWORK","_number":2,"created":"2024-10-08 11:19:19.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/58/931758/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tempest","ref":"refs/changes/58/931758/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tempest refs/changes/58/931758/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tempest refs/changes/58/931758/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tempest refs/changes/58/931758/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tempest refs/changes/58/931758/2"}}},"commit":{"parents":[{"commit":"8e72fcdd7108699121c15d710c1198d73e09efc4","subject":"magnum: Add manage_tests_packages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tempest/commit/8e72fcdd7108699121c15d710c1198d73e09efc4"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-05 03:50:00.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-08 11:19:19.000000000","tz":0},"subject":"Support customizing catalog_type options","message":"Support customizing catalog_type options\n\nConflicts:\n\tmanifests/init.pp\n\tmanifests/magnum.pp\n\tspec/classes/tempest_magnum_spec.rb\n\nChange-Id: If1fddd56828c93c19499ac8dadb2337b3f2afc46\n(cherry picked from commit cd44577546e66ce4c1806e2e8e31e21e5a818904)\n(cherry picked from commit 1d056d66606dc7b75ad60623941b70832384d672)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tempest/commit/4e138bdd898c989301e472ca08516d4cc11e1a68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tempest/commit/4e138bdd898c989301e472ca08516d4cc11e1a68"}]},"branch":"refs/heads/stable/2024.1"},"849acae47bf3ac974bdf8b171d92fd438452b7ce":{"kind":"REWORK","_number":3,"created":"2024-10-09 02:07:27.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/58/931758/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tempest","ref":"refs/changes/58/931758/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tempest refs/changes/58/931758/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tempest refs/changes/58/931758/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tempest refs/changes/58/931758/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tempest refs/changes/58/931758/3"}}},"commit":{"parents":[{"commit":"8e72fcdd7108699121c15d710c1198d73e09efc4","subject":"magnum: Add manage_tests_packages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tempest/commit/8e72fcdd7108699121c15d710c1198d73e09efc4"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-05 03:50:00.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-09 02:06:51.000000000","tz":540},"subject":"Support customizing catalog_type options","message":"Support customizing catalog_type options\n\nBackport note:\nThis backport includes 5f84cf371b19c6aeffd423aa2001be7acd5c9c96 which\nfixed the parameter description order.\n\nConflicts:\n\tmanifests/init.pp\n\tmanifests/magnum.pp\n\tspec/classes/tempest_magnum_spec.rb\n\nChange-Id: If1fddd56828c93c19499ac8dadb2337b3f2afc46\n(cherry picked from commit cd44577546e66ce4c1806e2e8e31e21e5a818904)\n(cherry picked from commit d6fa19fa2d6978d0f7ec6b22790c532fac2cabad)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tempest/commit/849acae47bf3ac974bdf8b171d92fd438452b7ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tempest/commit/849acae47bf3ac974bdf8b171d92fd438452b7ce"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
