)]}'
{"id":"openstack%2Fpuppet-tripleo~344513","triplet_id":"openstack%2Fpuppet-tripleo~master~I67a15dc83a754fb6f5fe25c64ae9e7d29c58fcec","project":"openstack/puppet-tripleo","branch":"master","topic":"bug/1604624","hashtags":[],"change_id":"I67a15dc83a754fb6f5fe25c64ae9e7d29c58fcec","subject":"Only sync the gnocchi db on the pacemaker master","status":"MERGED","created":"2016-07-20 00:09:16.000000000","updated":"2016-07-20 07:41:06.000000000","submitted":"2016-07-20 07:41:06.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2db9c6fd894c444fdcca58a8e4e96b938236fac0","_number":344513,"virtual_id_number":344513,"owner":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2016-07-20 07:41:06.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},{"value":0,"_account_id":4328,"name":"Steven Hardy","email":"steven.hardy@suse.com","username":"shardy"},{"value":0,"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4328,"name":"Steven Hardy","email":"steven.hardy@suse.com","username":"shardy"},"recommended":{"_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2016-07-20 07:29:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},{"value":2,"date":"2016-07-20 07:31:57.000000000","_account_id":4328,"name":"Steven Hardy","email":"steven.hardy@suse.com","username":"shardy"},{"value":2,"date":"2016-07-20 04:19:33.000000000","_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4328,"name":"Steven Hardy","email":"steven.hardy@suse.com","username":"shardy"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},{"value":1,"date":"2016-07-20 07:31:57.000000000","_account_id":4328,"name":"Steven Hardy","email":"steven.hardy@suse.com","username":"shardy"},{"value":0,"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":4328,"name":"Steven Hardy","email":"steven.hardy@suse.com","username":"shardy"},{"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},{"_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2016-07-20 01:37:07.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2016-07-20 04:19:33.000000000","updated_by":{"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},"reviewer":{"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},"state":"REVIEWER"},{"updated":"2016-07-20 07:29:11.000000000","updated_by":{"_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},"reviewer":{"_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},"state":"REVIEWER"},{"updated":"2016-07-20 07:31:57.000000000","updated_by":{"_account_id":4328,"name":"Steven Hardy","email":"steven.hardy@suse.com","username":"shardy"},"reviewer":{"_account_id":4328,"name":"Steven Hardy","email":"steven.hardy@suse.com","username":"shardy"},"state":"REVIEWER"},{"updated":"2016-07-20 07:41:06.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"faf8f4187fc2bb46634e2cd3f8342c416c990b32","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2016-07-20 00:09:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"95a567708d99421942986c1ee7b4f8b43b0b1402","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-07-20 00:19:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-puppet-tripleo-puppet-lint http://logs.openstack.org/13/344513/1/check/gate-puppet-tripleo-puppet-lint/0b7e7a6/ : SUCCESS in 3m 16s\n- gate-puppet-tripleo-puppet-syntax-3-centos-7 http://logs.openstack.org/13/344513/1/check/gate-puppet-tripleo-puppet-syntax-3-centos-7/20b7721/ : SUCCESS in 6m 12s\n- gate-puppet-tripleo-puppet-syntax-4-centos-7 http://logs.openstack.org/13/344513/1/check/gate-puppet-tripleo-puppet-syntax-4-centos-7/fadba10/ : SUCCESS in 3m 34s\n- gate-puppet-tripleo-puppet-unit-3.6-centos-7 http://logs.openstack.org/13/344513/1/check/gate-puppet-tripleo-puppet-unit-3.6-centos-7/5fb87ed/ : SUCCESS in 6m 12s\n- gate-puppet-tripleo-puppet-unit-3.8-centos-7 http://logs.openstack.org/13/344513/1/check/gate-puppet-tripleo-puppet-unit-3.8-centos-7/d8e0551/ : SUCCESS in 4m 54s\n- gate-puppet-tripleo-puppet-unit-4.5-centos-7 http://logs.openstack.org/13/344513/1/check/gate-puppet-tripleo-puppet-unit-4.5-centos-7/dd43747/ : SUCCESS in 7m 09s\n- gate-puppet-tripleo-puppet-unit-latest-centos-7 http://logs.openstack.org/13/344513/1/check/gate-puppet-tripleo-puppet-unit-latest-centos-7/5b26f0e/ : SUCCESS in 5m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"35ab922883c7c92387ef213e0df6855b971c9e47","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2016-07-20 01:37:07.000000000","message":"Patch Set 1:\n\n(2 comments)\n\nnice catch, I just have one comment inline.","accounts_in_message":[],"_revision_number":1},{"id":"0c223453a0064975911caa9e67ecd8cc681a283a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-07-20 01:51:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (check-tripleo pipeline).\n\n- gate-tripleo-ci-centos-7-ovb-ha http://logs.openstack.org/13/344513/1/check-tripleo/gate-tripleo-ci-centos-7-ovb-ha/0f7cb54/ : SUCCESS in 1h 41m 30s","accounts_in_message":[],"_revision_number":1},{"id":"f351d9d3a0e0fbd7d749741a0da54a4bdc89d018","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2016-07-20 02:13:37.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2cfc549c8fcd0ec8affd0dd515bcc30e5e440465","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-07-20 02:24:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-puppet-tripleo-puppet-lint http://logs.openstack.org/13/344513/2/check/gate-puppet-tripleo-puppet-lint/d5b8385/ : SUCCESS in 3m 26s\n- gate-puppet-tripleo-puppet-syntax-3-centos-7 http://logs.openstack.org/13/344513/2/check/gate-puppet-tripleo-puppet-syntax-3-centos-7/4e61ed9/ : SUCCESS in 3m 30s\n- gate-puppet-tripleo-puppet-syntax-4-centos-7 http://logs.openstack.org/13/344513/2/check/gate-puppet-tripleo-puppet-syntax-4-centos-7/2e320bb/ : SUCCESS in 4m 40s\n- gate-puppet-tripleo-puppet-unit-3.6-centos-7 http://logs.openstack.org/13/344513/2/check/gate-puppet-tripleo-puppet-unit-3.6-centos-7/7b67971/ : SUCCESS in 7m 00s\n- gate-puppet-tripleo-puppet-unit-3.8-centos-7 http://logs.openstack.org/13/344513/2/check/gate-puppet-tripleo-puppet-unit-3.8-centos-7/27e4412/ : SUCCESS in 4m 09s\n- gate-puppet-tripleo-puppet-unit-4.5-centos-7 http://logs.openstack.org/13/344513/2/check/gate-puppet-tripleo-puppet-unit-4.5-centos-7/7d9ab58/ : SUCCESS in 5m 58s\n- gate-puppet-tripleo-puppet-unit-latest-centos-7 http://logs.openstack.org/13/344513/2/check/gate-puppet-tripleo-puppet-unit-latest-centos-7/5cdac82/ : SUCCESS in 4m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bbcd00b8866280caec2263e19b448dc04cccfd17","author":{"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},"date":"2016-07-20 04:19:33.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"83abc0ec85bf9b76b43aaa581b6d3ddfb02c3688","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-07-20 04:21:28.000000000","message":"Patch Set 2:\n\nBuild failed (check-tripleo pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- gate-tripleo-ci-centos-7-ovb-ha http://logs.openstack.org/13/344513/2/check-tripleo/gate-tripleo-ci-centos-7-ovb-ha/adb0715/ : FAILURE in 2h 07m 36s","accounts_in_message":[],"_revision_number":2},{"id":"06cb7ef4a8251c1795254130483da619be4564d4","author":{"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},"date":"2016-07-20 05:35:22.000000000","message":"Patch Set 2:\n\nrecheck timed out","accounts_in_message":[],"_revision_number":2},{"id":"ba60d5a7fa61d28e796173cded6938bfb54d1467","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-07-20 05:48:46.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- gate-puppet-tripleo-puppet-lint http://logs.openstack.org/13/344513/2/check/gate-puppet-tripleo-puppet-lint/a64d368/ : SUCCESS in 5m 03s\n- gate-puppet-tripleo-puppet-syntax-3-centos-7 http://logs.openstack.org/13/344513/2/check/gate-puppet-tripleo-puppet-syntax-3-centos-7/7e56b7d/ : SUCCESS in 4m 16s\n- gate-puppet-tripleo-puppet-syntax-4-centos-7 http://logs.openstack.org/13/344513/2/check/gate-puppet-tripleo-puppet-syntax-4-centos-7/b14050b/ : SUCCESS in 11m 36s\n- gate-puppet-tripleo-puppet-unit-3.6-centos-7 http://logs.openstack.org/13/344513/2/check/gate-puppet-tripleo-puppet-unit-3.6-centos-7/908eb72/ : SUCCESS in 6m 09s\n- gate-puppet-tripleo-puppet-unit-3.8-centos-7 http://logs.openstack.org/13/344513/2/check/gate-puppet-tripleo-puppet-unit-3.8-centos-7/6a28b56/ : SUCCESS in 4m 44s\n- gate-puppet-tripleo-puppet-unit-4.5-centos-7 http://logs.openstack.org/13/344513/2/check/gate-puppet-tripleo-puppet-unit-4.5-centos-7/67cf76c/ : SUCCESS in 5m 48s\n- gate-puppet-tripleo-puppet-unit-latest-centos-7 http://logs.openstack.org/13/344513/2/check/gate-puppet-tripleo-puppet-unit-latest-centos-7/d0abd6c/ : SUCCESS in 4m 39s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"eab8c498b49f2b1d24cdf8d17688acc23033404f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-07-20 07:27:26.000000000","message":"Patch Set 2:\n\nBuild succeeded (check-tripleo pipeline).\n\n- gate-tripleo-ci-centos-7-ovb-ha http://logs.openstack.org/13/344513/2/check-tripleo/gate-tripleo-ci-centos-7-ovb-ha/cdd4a25/ : SUCCESS in 1h 51m 52s","accounts_in_message":[],"_revision_number":2},{"id":"3d147478a3fcb7fba90c0ebfe7057b61e1f523fb","author":{"_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},"date":"2016-07-20 07:29:11.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"679dacde42d0f6e79826d60fa1fe6b18e9f0bb20","author":{"_account_id":4328,"name":"Steven Hardy","email":"steven.hardy@suse.com","username":"shardy"},"date":"2016-07-20 07:31:57.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"858567288d75be12eb049badfb335f72ef7128a9","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-07-20 07:32:07.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":2},{"id":"e52bfac0d6f4b63775772dffb8f5305be88b3e1d","date":"2016-07-20 07:41:06.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":2},{"id":"a7a5c7512c5489de8d72e7a72878e0b86175c3f4","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-07-20 07:41:06.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-puppet-tripleo-puppet-lint http://logs.openstack.org/13/344513/2/gate/gate-puppet-tripleo-puppet-lint/1f27872/ : SUCCESS in 4m 04s\n- gate-puppet-tripleo-puppet-syntax-3-centos-7 http://logs.openstack.org/13/344513/2/gate/gate-puppet-tripleo-puppet-syntax-3-centos-7/051f64f/ : SUCCESS in 4m 36s\n- gate-puppet-tripleo-puppet-syntax-4-centos-7 http://logs.openstack.org/13/344513/2/gate/gate-puppet-tripleo-puppet-syntax-4-centos-7/5bb2988/ : SUCCESS in 4m 28s\n- gate-puppet-tripleo-puppet-unit-3.6-centos-7 http://logs.openstack.org/13/344513/2/gate/gate-puppet-tripleo-puppet-unit-3.6-centos-7/7b7a4e5/ : SUCCESS in 4m 21s\n- gate-puppet-tripleo-puppet-unit-3.8-centos-7 http://logs.openstack.org/13/344513/2/gate/gate-puppet-tripleo-puppet-unit-3.8-centos-7/e084156/ : SUCCESS in 3m 52s\n- gate-puppet-tripleo-puppet-unit-4.5-centos-7 http://logs.openstack.org/13/344513/2/gate/gate-puppet-tripleo-puppet-unit-4.5-centos-7/1ca4f66/ : SUCCESS in 4m 32s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"88ac50880e208cc0b6fca16f0b325b4a0c86a3dd","revisions":{"6bf804970b4f02803d67340ef0927ad0cd739033":{"kind":"REWORK","_number":1,"created":"2016-07-20 00:09:16.000000000","uploader":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"ref":"refs/changes/13/344513/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/13/344513/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/13/344513/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/13/344513/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/13/344513/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/13/344513/1"}}},"commit":{"parents":[{"commit":"b05e8debde9ae61ab1baadd4bf23f532b29a9a5f","subject":"Add principal to certmonger\u0027s haproxy helper","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/b05e8debde9ae61ab1baadd4bf23f532b29a9a5f"}]}],"author":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2016-07-20 00:08:08.000000000","tz":-240},"committer":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2016-07-20 00:08:08.000000000","tz":-240},"subject":"Only sync the gnocchi db on the pacemaker master","message":"Only sync the gnocchi db on the pacemaker master\n\nThe gnocchi db sync is being run from multiple controllers causing errors in\nCI. See the bug for more details.\n\nChange-Id: I67a15dc83a754fb6f5fe25c64ae9e7d29c58fcec\nCloses-Bug: #1604624\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/6bf804970b4f02803d67340ef0927ad0cd739033"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/6bf804970b4f02803d67340ef0927ad0cd739033"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"b05e8debde9ae61ab1baadd4bf23f532b29a9a5f","is_merged_in_target_branch":true,"change_id":"Ie27848f522d11135b061aef766de2b696c77fcb9","change_number":343710,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"88ac50880e208cc0b6fca16f0b325b4a0c86a3dd":{"kind":"REWORK","_number":2,"created":"2016-07-20 02:13:37.000000000","uploader":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"ref":"refs/changes/13/344513/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/13/344513/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/13/344513/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/13/344513/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/13/344513/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/13/344513/2"}}},"commit":{"parents":[{"commit":"b05e8debde9ae61ab1baadd4bf23f532b29a9a5f","subject":"Add principal to certmonger\u0027s haproxy helper","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/b05e8debde9ae61ab1baadd4bf23f532b29a9a5f"}]}],"author":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2016-07-20 00:08:08.000000000","tz":-240},"committer":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2016-07-20 02:13:32.000000000","tz":-240},"subject":"Only sync the gnocchi db on the pacemaker master","message":"Only sync the gnocchi db on the pacemaker master\n\nThe gnocchi db sync is being run from multiple controllers causing errors in\nCI. See the bug for more details.\n\nChange-Id: I67a15dc83a754fb6f5fe25c64ae9e7d29c58fcec\nCloses-Bug: #1604624\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/88ac50880e208cc0b6fca16f0b325b4a0c86a3dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/88ac50880e208cc0b6fca16f0b325b4a0c86a3dd"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"b05e8debde9ae61ab1baadd4bf23f532b29a9a5f","is_merged_in_target_branch":true,"change_id":"Ie27848f522d11135b061aef766de2b696c77fcb9","change_number":343710,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
