)]}'
{"id":"openstack%2Fpuppet-tripleo~373513","triplet_id":"openstack%2Fpuppet-tripleo~master~I4cf4a04bddb8f042e8e8f7e1d1b69f846c533e3b","project":"openstack/puppet-tripleo","branch":"master","topic":"tripleo/puppet4","hashtags":[],"change_id":"I4cf4a04bddb8f042e8e8f7e1d1b69f846c533e3b","subject":"certmonger: improve orchestration for puppet4","status":"MERGED","created":"2016-09-20 19:54:41.000000000","updated":"2016-09-23 16:32:25.000000000","submitted":"2016-09-23 16:32:25.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e2729dc0dc4775c51d25b4dba1a87b82b4e34ac1","_number":373513,"virtual_id_number":373513,"owner":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":2,"date":"2016-09-23 16:32:25.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2016-09-21 13:31:17.000000000","_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},{"value":1,"date":"2016-09-21 17:19:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2016-09-23 13:33:43.000000000","_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2016-09-23 13:33:43.000000000","_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2016-09-21 13:31:17.000000000","updated_by":{"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},"reviewer":{"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},"state":"REVIEWER"},{"updated":"2016-09-21 17:19:45.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2016-09-23 13:33:43.000000000","updated_by":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"reviewer":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"state":"REVIEWER"},{"updated":"2016-09-23 16:32:25.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"28f91164a5f73599c77d7a4cfb8ddbe0cbbeedc0","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2016-09-20 19:54:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c54adbc46689fdd64ce4920f0d0c9983adb17fab","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2016-09-20 20:17:50.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"594043e3b4d7db742339ec411f6477270fa6787a","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2016-09-20 20:59:57.000000000","message":"Topic changed from puppet4 to tripleo/puppet4","accounts_in_message":[],"_revision_number":2},{"id":"63f80bda8bd9df614115b99eb51c6682ec6e8e15","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-09-20 21:38:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-puppet-tripleo-puppet-lint http://logs.openstack.org/13/373513/2/check/gate-puppet-tripleo-puppet-lint/ec223fa/ : SUCCESS in 2m 16s\n- gate-puppet-tripleo-puppet-syntax-3-centos-7 http://logs.openstack.org/13/373513/2/check/gate-puppet-tripleo-puppet-syntax-3-centos-7/dbf4f9a/ : SUCCESS in 4m 43s\n- gate-puppet-tripleo-puppet-syntax-4-centos-7 http://logs.openstack.org/13/373513/2/check/gate-puppet-tripleo-puppet-syntax-4-centos-7/0db8dcf/ : SUCCESS in 4m 59s\n- gate-puppet-tripleo-puppet-unit-3.6-centos-7 http://logs.openstack.org/13/373513/2/check/gate-puppet-tripleo-puppet-unit-3.6-centos-7/83c5485/ : SUCCESS in 6m 10s\n- gate-puppet-tripleo-puppet-unit-3.8-centos-7 http://logs.openstack.org/13/373513/2/check/gate-puppet-tripleo-puppet-unit-3.8-centos-7/534bcf9/ : SUCCESS in 5m 09s\n- gate-puppet-tripleo-puppet-unit-4.5-centos-7 http://logs.openstack.org/13/373513/2/check/gate-puppet-tripleo-puppet-unit-4.5-centos-7/0cfd5c2/ : SUCCESS in 3m 44s\n- gate-puppet-tripleo-puppet-unit-latest-centos-7 http://logs.openstack.org/13/373513/2/check/gate-puppet-tripleo-puppet-unit-latest-centos-7/c972507/ : SUCCESS in 4m 47s (non-voting)\n- gate-tripleo-ci-centos-7-nonha-multinode http://logs.openstack.org/13/373513/2/check/gate-tripleo-ci-centos-7-nonha-multinode/f9e7613/ : SUCCESS in 1h 17m 25s\n- gate-tripleo-ci-centos-7-undercloud http://logs.openstack.org/13/373513/2/check/gate-tripleo-ci-centos-7-undercloud/956a95a/ : SUCCESS in 18m 53s\n- gate-tripleo-ci-centos-7-nonha-multinode-updates-nv http://logs.openstack.org/13/373513/2/check/gate-tripleo-ci-centos-7-nonha-multinode-updates-nv/6b44429/ : FAILURE in 57m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"fe494ca4f395b0e713fcbaa7b6af25629b784c76","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-09-20 22:11:57.000000000","message":"Patch Set 2:\n\nBuild failed (check-tripleo pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- gate-tripleo-ci-centos-7-ovb-ha http://logs.openstack.org/13/373513/2/check-tripleo/gate-tripleo-ci-centos-7-ovb-ha/38eda0f/ : FAILURE in 1h 45m 04s\n- gate-tripleo-ci-centos-7-ovb-nonha http://logs.openstack.org/13/373513/2/check-tripleo/gate-tripleo-ci-centos-7-ovb-nonha/76b5822/ : SUCCESS in 1h 46m 24s","accounts_in_message":[],"_revision_number":2},{"id":"91b5e9f032a7c543948cd76985ce8891a0cd0429","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2016-09-20 23:42:42.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"525d003a5f91f79dd2813e1125256d9085a79ec0","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-09-21 00:25:50.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- gate-puppet-tripleo-puppet-lint http://logs.openstack.org/13/373513/2/check/gate-puppet-tripleo-puppet-lint/5738dbc/ : SUCCESS in 2m 19s\n- gate-puppet-tripleo-puppet-syntax-3-centos-7 http://logs.openstack.org/13/373513/2/check/gate-puppet-tripleo-puppet-syntax-3-centos-7/9c0dc36/ : SUCCESS in 6m 37s\n- gate-puppet-tripleo-puppet-syntax-4-centos-7 http://logs.openstack.org/13/373513/2/check/gate-puppet-tripleo-puppet-syntax-4-centos-7/5e69588/ : SUCCESS in 3m 40s\n- gate-puppet-tripleo-puppet-unit-3.6-centos-7 http://logs.openstack.org/13/373513/2/check/gate-puppet-tripleo-puppet-unit-3.6-centos-7/74e6b37/ : SUCCESS in 3m 25s\n- gate-puppet-tripleo-puppet-unit-3.8-centos-7 http://logs.openstack.org/13/373513/2/check/gate-puppet-tripleo-puppet-unit-3.8-centos-7/9fcae21/ : SUCCESS in 3m 36s\n- gate-puppet-tripleo-puppet-unit-4.5-centos-7 http://logs.openstack.org/13/373513/2/check/gate-puppet-tripleo-puppet-unit-4.5-centos-7/6836f40/ : SUCCESS in 3m 31s\n- gate-puppet-tripleo-puppet-unit-latest-centos-7 http://logs.openstack.org/13/373513/2/check/gate-puppet-tripleo-puppet-unit-latest-centos-7/c2da18e/ : SUCCESS in 3m 33s (non-voting)\n- gate-tripleo-ci-centos-7-nonha-multinode http://logs.openstack.org/13/373513/2/check/gate-tripleo-ci-centos-7-nonha-multinode/81e8e3e/ : SUCCESS in 40m 13s\n- gate-tripleo-ci-centos-7-undercloud http://logs.openstack.org/13/373513/2/check/gate-tripleo-ci-centos-7-undercloud/e1b488a/ : SUCCESS in 31m 22s\n- gate-tripleo-ci-centos-7-nonha-multinode-updates-nv http://logs.openstack.org/13/373513/2/check/gate-tripleo-ci-centos-7-nonha-multinode-updates-nv/8fc109c/ : FAILURE in 40m 33s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8522d0d52600dea476ec71e6450511cbee78bc8a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-09-21 01:39:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (check-tripleo pipeline).\n\n- gate-tripleo-ci-centos-7-ovb-ha http://logs.openstack.org/13/373513/2/check-tripleo/gate-tripleo-ci-centos-7-ovb-ha/817bdaa/ : SUCCESS in 1h 54m 57s\n- gate-tripleo-ci-centos-7-ovb-nonha http://logs.openstack.org/13/373513/2/check-tripleo/gate-tripleo-ci-centos-7-ovb-nonha/dae6a84/ : SUCCESS in 1h 46m 44s","accounts_in_message":[],"_revision_number":2},{"id":"e6aea4294ba21bbfb2ea5b499cd1e07478f192c0","author":{"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},"date":"2016-09-21 13:31:17.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"59c5e9f655fa3d42318b7fdbaff89a47cb313a94","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2016-09-21 17:19:45.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"78ec29d6bff2e7b0c5f30b9290e557890b1c2fe2","author":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"date":"2016-09-23 13:33:43.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f2402aead93a189da661eedb9872bf6c2ec2eceb","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-09-23 13:33:54.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":2},{"id":"dc8d0443ef83cf0cbb1a2195883bf69060eb9bc7","date":"2016-09-23 16:32:25.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":2},{"id":"e2729dc0dc4775c51d25b4dba1a87b82b4e34ac1","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-09-23 16:32:25.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-puppet-tripleo-puppet-lint http://logs.openstack.org/13/373513/2/gate/gate-puppet-tripleo-puppet-lint/e9f80f5/ : SUCCESS in 2m 55s\n- gate-puppet-tripleo-puppet-syntax-3-centos-7 http://logs.openstack.org/13/373513/2/gate/gate-puppet-tripleo-puppet-syntax-3-centos-7/91ea086/ : SUCCESS in 4m 34s\n- gate-puppet-tripleo-puppet-syntax-4-centos-7 http://logs.openstack.org/13/373513/2/gate/gate-puppet-tripleo-puppet-syntax-4-centos-7/59b5a92/ : SUCCESS in 5m 00s\n- gate-puppet-tripleo-puppet-unit-3.6-centos-7 http://logs.openstack.org/13/373513/2/gate/gate-puppet-tripleo-puppet-unit-3.6-centos-7/56a087b/ : SUCCESS in 5m 01s\n- gate-puppet-tripleo-puppet-unit-3.8-centos-7 http://logs.openstack.org/13/373513/2/gate/gate-puppet-tripleo-puppet-unit-3.8-centos-7/7ceb05c/ : SUCCESS in 3m 52s\n- gate-puppet-tripleo-puppet-unit-4.5-centos-7 http://logs.openstack.org/13/373513/2/gate/gate-puppet-tripleo-puppet-unit-4.5-centos-7/f36a5db/ : SUCCESS in 4m 13s\n- gate-tripleo-ci-centos-7-nonha-multinode http://logs.openstack.org/13/373513/2/gate/gate-tripleo-ci-centos-7-nonha-multinode/1a5a9de/ : SUCCESS in 1h 13m 16s\n- gate-tripleo-ci-centos-7-undercloud http://logs.openstack.org/13/373513/2/gate/gate-tripleo-ci-centos-7-undercloud/4f4b40a/ : SUCCESS in 27m 24s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9b974df7a2a20741b64b72e8980bc47c13b6166c","revisions":{"01eba7ff5b004d3ff4f18ce98ac0c72695c1b9f3":{"kind":"REWORK","_number":1,"created":"2016-09-20 19:54:41.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/13/373513/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/13/373513/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/13/373513/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/13/373513/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/13/373513/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/13/373513/1"}}},"commit":{"parents":[{"commit":"6a9429eeda549f85fdb7e3748188046d702852fb","subject":"Merge \"Terminate Zaqar websocket endpoint in HAProxy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/6a9429eeda549f85fdb7e3748188046d702852fb"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2016-09-20 19:52:18.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2016-09-20 19:52:18.000000000","tz":-240},"subject":"certmonger: improve orchestration for puppet4","message":"certmonger: improve orchestration for puppet4\n\nThe extract-and-trust-ca actually needs /var/lib/certmonger/local/creds\nfile to be created, which is created when certmonger is started, not\nwhen package is installed.\nThis patch change the exec dependency to run it only when service is\nstarted.\nAlso, since the service create the file, let\u0027s relax the Exec a little\nbit by allowing to retry 5 times after 1s break in case the Exec fails,\nfor example if service takes more than 5 seconds to create this file.\nIt will avoid us some race condition in the deployment.\n\nChange-Id: I4cf4a04bddb8f042e8e8f7e1d1b69f846c533e3b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/01eba7ff5b004d3ff4f18ce98ac0c72695c1b9f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/01eba7ff5b004d3ff4f18ce98ac0c72695c1b9f3"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"6a9429eeda549f85fdb7e3748188046d702852fb","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"9b974df7a2a20741b64b72e8980bc47c13b6166c":{"kind":"REWORK","_number":2,"created":"2016-09-20 20:17:50.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/13/373513/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/13/373513/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/13/373513/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/13/373513/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/13/373513/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/13/373513/2"}}},"commit":{"parents":[{"commit":"6a9429eeda549f85fdb7e3748188046d702852fb","subject":"Merge \"Terminate Zaqar websocket endpoint in HAProxy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/6a9429eeda549f85fdb7e3748188046d702852fb"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2016-09-20 19:52:18.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2016-09-20 20:17:43.000000000","tz":-240},"subject":"certmonger: improve orchestration for puppet4","message":"certmonger: improve orchestration for puppet4\n\nThe extract-and-trust-ca actually needs /var/lib/certmonger/local/creds\nfile to be created, which is created when certmonger is started, not\nwhen package is installed.\nThis patch change the exec dependency to run it only when service is\nstarted.\nAlso, since the service create the file, let\u0027s relax the Exec a little\nbit by allowing to retry 5 times after 1s break in case the Exec fails,\nfor example if service takes more than 5 seconds to create this file.\nIt will avoid us some race condition in the deployment.\n\nChange-Id: I4cf4a04bddb8f042e8e8f7e1d1b69f846c533e3b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/9b974df7a2a20741b64b72e8980bc47c13b6166c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/9b974df7a2a20741b64b72e8980bc47c13b6166c"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"6a9429eeda549f85fdb7e3748188046d702852fb","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
