)]}'
{"id":"openstack%2Fpuppet-tripleo~483500","triplet_id":"openstack%2Fpuppet-tripleo~master~Iebc547aa92f9f40e4a633c57d79e6c9cddb5dd28","project":"openstack/puppet-tripleo","branch":"master","topic":"bug/1702964","hashtags":[],"change_id":"Iebc547aa92f9f40e4a633c57d79e6c9cddb5dd28","subject":"Update resource references for dependencies","status":"MERGED","created":"2017-07-13 18:27:48.000000000","updated":"2017-07-18 01:49:25.000000000","submitted":"2017-07-18 01:49:25.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8f9e8a071af11299228489701f432aec14aafaa4","_number":483500,"virtual_id_number":483500,"owner":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2017-07-18 01:49:25.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":8042,"name":"Jiří Stránský","email":"jistr@redhat.com","username":"jistr"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2017-07-17 13:58:57.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2017-07-17 14:32:46.000000000","_account_id":8042,"name":"Jiří Stránský","email":"jistr@redhat.com","username":"jistr"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8042,"name":"Jiří Stránský","email":"jistr@redhat.com","username":"jistr"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2017-07-17 14:32:46.000000000","_account_id":8042,"name":"Jiří Stránský","email":"jistr@redhat.com","username":"jistr"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":8042,"name":"Jiří Stránský","email":"jistr@redhat.com","username":"jistr"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-07-17 13:58:57.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2017-07-17 14:32:46.000000000","updated_by":{"_account_id":8042,"name":"Jiří Stránský","email":"jistr@redhat.com","username":"jistr"},"reviewer":{"_account_id":8042,"name":"Jiří Stránský","email":"jistr@redhat.com","username":"jistr"},"state":"REVIEWER"},{"updated":"2017-07-18 01:49:25.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"fec68c65d2fbf6d2e932682f7c05aa2488fdde11","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2017-07-13 18:27:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"65ae4e68dfb90bcc1419b5852e0b36d2c6b41eaf","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-07-13 21:14:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-puppet-tripleo-puppet-lint http://logs.openstack.org/00/483500/1/check/gate-puppet-tripleo-puppet-lint/ec483be/ : SUCCESS in 1m 58s\n- gate-puppet-tripleo-puppet-syntax-4-centos-7 http://logs.openstack.org/00/483500/1/check/gate-puppet-tripleo-puppet-syntax-4-centos-7/1de498a/ : SUCCESS in 3m 29s\n- gate-tripleo-ci-centos-7-nonha-multinode-oooq http://logs.openstack.org/00/483500/1/check/gate-tripleo-ci-centos-7-nonha-multinode-oooq/6c66b3e/ : SUCCESS in 1h 41m 09s\n- gate-tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/00/483500/1/check/gate-tripleo-ci-centos-7-containers-multinode/22211e4/ : SUCCESS in 1h 53m 03s\n- gate-tripleo-ci-centos-7-nonha-multinode-updates-nv http://logs.openstack.org/00/483500/1/check/gate-tripleo-ci-centos-7-nonha-multinode-updates-nv/6a57dfb/ : SUCCESS in 1h 25m 58s (non-voting)\n- gate-tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/00/483500/1/check/gate-tripleo-ci-centos-7-undercloud-containers/b5553c1/ : SUCCESS in 29m 52s\n- gate-tripleo-ci-centos-7-undercloud-oooq http://logs.openstack.org/00/483500/1/check/gate-tripleo-ci-centos-7-undercloud-oooq/8908295/ : SUCCESS in 40m 27s\n- gate-tripleo-ci-centos-7-undercloud-upgrades-nv http://logs.openstack.org/00/483500/1/check/gate-tripleo-ci-centos-7-undercloud-upgrades-nv/b6d9314/ : SUCCESS in 46m 38s (non-voting)\n- gate-tripleo-ci-centos-7-multinode-upgrades-nv http://logs.openstack.org/00/483500/1/check/gate-tripleo-ci-centos-7-multinode-upgrades-nv/83e592e/ : FAILURE in 2h 39m 19s (non-voting)\n- gate-tripleo-ci-centos-7-containers-multinode-upgrades-nv http://logs.openstack.org/00/483500/1/check/gate-tripleo-ci-centos-7-containers-multinode-upgrades-nv/8cb3c94/ : SUCCESS in 1h 45m 30s (non-voting)\n- gate-puppet-tripleo-puppet-unit-4.5-centos-7 http://logs.openstack.org/00/483500/1/check/gate-puppet-tripleo-puppet-unit-4.5-centos-7/580dd8a/ : FAILURE in 6m 43s\n- gate-puppet-tripleo-puppet-unit-latest-centos-7 http://logs.openstack.org/00/483500/1/check/gate-puppet-tripleo-puppet-unit-latest-centos-7/b29e785/ : FAILURE in 6m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7428725dbb0e6b8dc69cfe54f7d20ba985687a99","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-07-13 23:07:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (check-tripleo pipeline).\n\n- gate-tripleo-ci-centos-7-ovb-ha-oooq http://logs.openstack.org/00/483500/1/check-tripleo/gate-tripleo-ci-centos-7-ovb-ha-oooq/b6a4ad9/ : SUCCESS in 2h 33m 52s\n- gate-tripleo-ci-centos-7-ovb-containers-oooq http://logs.openstack.org/00/483500/1/check-tripleo/gate-tripleo-ci-centos-7-ovb-containers-oooq/2689c10/ : SUCCESS in 2h 26m 00s\n- gate-tripleo-ci-centos-7-ovb-1ctlr_1comp_1ceph-featureset024 http://logs.openstack.org/00/483500/1/check-tripleo/gate-tripleo-ci-centos-7-ovb-1ctlr_1comp_1ceph-featureset024/ee5b96f/ : SUCCESS in 2h 33m 27s","accounts_in_message":[],"_revision_number":1},{"id":"70652e02c50bed340ce6aa7d09702f861dee481c","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2017-07-14 17:11:41.000000000","message":"Patch Set 1:\n\nrecheck github timeout","accounts_in_message":[],"_revision_number":1},{"id":"36283d7f826ffb6bf56b318150339bbdb22ce2b6","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-07-14 19:15:30.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-puppet-tripleo-puppet-lint http://logs.openstack.org/00/483500/1/check/gate-puppet-tripleo-puppet-lint/dd35943/ : SUCCESS in 3m 20s\n- gate-puppet-tripleo-puppet-syntax-4-centos-7 http://logs.openstack.org/00/483500/1/check/gate-puppet-tripleo-puppet-syntax-4-centos-7/9190184/ : SUCCESS in 3m 01s\n- gate-tripleo-ci-centos-7-nonha-multinode-oooq http://logs.openstack.org/00/483500/1/check/gate-tripleo-ci-centos-7-nonha-multinode-oooq/74dfd58/ : SUCCESS in 1h 13m 34s\n- gate-tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/00/483500/1/check/gate-tripleo-ci-centos-7-containers-multinode/6fe1086/ : SUCCESS in 1h 18m 33s\n- gate-tripleo-ci-centos-7-nonha-multinode-updates-nv http://logs.openstack.org/00/483500/1/check/gate-tripleo-ci-centos-7-nonha-multinode-updates-nv/83e2416/ : SUCCESS in 1h 35m 32s (non-voting)\n- gate-tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/00/483500/1/check/gate-tripleo-ci-centos-7-undercloud-containers/460f9da/ : SUCCESS in 44m 39s\n- gate-tripleo-ci-centos-7-undercloud-oooq http://logs.openstack.org/00/483500/1/check/gate-tripleo-ci-centos-7-undercloud-oooq/14ee1ed/ : SUCCESS in 1h 01m 42s\n- gate-tripleo-ci-centos-7-undercloud-upgrades-nv http://logs.openstack.org/00/483500/1/check/gate-tripleo-ci-centos-7-undercloud-upgrades-nv/9d13d02/ : SUCCESS in 45m 18s (non-voting)\n- gate-tripleo-ci-centos-7-multinode-upgrades-nv http://logs.openstack.org/00/483500/1/check/gate-tripleo-ci-centos-7-multinode-upgrades-nv/5963049/ : FAILURE in 2h 00m 23s (non-voting)\n- gate-tripleo-ci-centos-7-containers-multinode-upgrades-nv http://logs.openstack.org/00/483500/1/check/gate-tripleo-ci-centos-7-containers-multinode-upgrades-nv/d149ae2/ : SUCCESS in 2h 01m 47s (non-voting)\n- gate-puppet-tripleo-puppet-unit-4.5-centos-7 http://logs.openstack.org/00/483500/1/check/gate-puppet-tripleo-puppet-unit-4.5-centos-7/286657f/ : FAILURE in 22m 55s\n- gate-puppet-tripleo-puppet-unit-latest-centos-7 http://logs.openstack.org/00/483500/1/check/gate-puppet-tripleo-puppet-unit-latest-centos-7/0ee6c5d/ : FAILURE in 21m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"65637a8b8430c970fbb9270372b83569a6b37039","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-07-14 20:49:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (check-tripleo pipeline).\n\n- gate-tripleo-ci-centos-7-ovb-ha-oooq http://logs.openstack.org/00/483500/1/check-tripleo/gate-tripleo-ci-centos-7-ovb-ha-oooq/7a544c4/ : SUCCESS in 2h 44m 05s\n- gate-tripleo-ci-centos-7-ovb-containers-oooq http://logs.openstack.org/00/483500/1/check-tripleo/gate-tripleo-ci-centos-7-ovb-containers-oooq/610c351/ : SUCCESS in 2h 25m 44s\n- gate-tripleo-ci-centos-7-ovb-1ctlr_1comp_1ceph-featureset024 http://logs.openstack.org/00/483500/1/check-tripleo/gate-tripleo-ci-centos-7-ovb-1ctlr_1comp_1ceph-featureset024/9412791/ : SUCCESS in 2h 29m 23s","accounts_in_message":[],"_revision_number":1},{"id":"891382c73122bb280c4b06b9fcd20b749eaea92b","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2017-07-15 16:26:12.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5861ab84cfed3dd762d595bef50b7df6944980e5","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-07-15 18:50:15.000000000","message":"Patch Set 2:\n\nBuild succeeded (check-tripleo pipeline).\n\n- gate-tripleo-ci-centos-7-ovb-ha-oooq http://logs.openstack.org/00/483500/2/check-tripleo/gate-tripleo-ci-centos-7-ovb-ha-oooq/9a52d1b/ : SUCCESS in 2h 23m 50s\n- gate-tripleo-ci-centos-7-ovb-containers-oooq http://logs.openstack.org/00/483500/2/check-tripleo/gate-tripleo-ci-centos-7-ovb-containers-oooq/d8b5fb2/ : SUCCESS in 2h 16m 42s\n- gate-tripleo-ci-centos-7-ovb-1ctlr_1comp_1ceph-featureset024 http://logs.openstack.org/00/483500/2/check-tripleo/gate-tripleo-ci-centos-7-ovb-1ctlr_1comp_1ceph-featureset024/1287d72/ : SUCCESS in 2h 18m 48s","accounts_in_message":[],"_revision_number":2},{"id":"dd9f48ede0a405bdef866eaf90bcf3c00d65d5c2","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-07-15 19:08:28.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-puppet-tripleo-puppet-lint http://logs.openstack.org/00/483500/2/check/gate-puppet-tripleo-puppet-lint/3ec353f/ : SUCCESS in 1m 47s\n- gate-puppet-tripleo-puppet-syntax-4-centos-7 http://logs.openstack.org/00/483500/2/check/gate-puppet-tripleo-puppet-syntax-4-centos-7/4bf02e0/ : SUCCESS in 2m 22s\n- gate-tripleo-ci-centos-7-nonha-multinode-oooq http://logs.openstack.org/00/483500/2/check/gate-tripleo-ci-centos-7-nonha-multinode-oooq/4842c2b/ : SUCCESS in 1h 10m 21s\n- gate-tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/00/483500/2/check/gate-tripleo-ci-centos-7-containers-multinode/2f92d21/ : SUCCESS in 1h 17m 56s\n- gate-tripleo-ci-centos-7-nonha-multinode-updates-nv http://logs.openstack.org/00/483500/2/check/gate-tripleo-ci-centos-7-nonha-multinode-updates-nv/71ed372/ : SUCCESS in 2h 03m 41s (non-voting)\n- gate-tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/00/483500/2/check/gate-tripleo-ci-centos-7-undercloud-containers/1ca0e31/ : SUCCESS in 27m 59s\n- gate-tripleo-ci-centos-7-undercloud-oooq http://logs.openstack.org/00/483500/2/check/gate-tripleo-ci-centos-7-undercloud-oooq/6b6b4cf/ : SUCCESS in 35m 59s\n- gate-tripleo-ci-centos-7-undercloud-upgrades-nv http://logs.openstack.org/00/483500/2/check/gate-tripleo-ci-centos-7-undercloud-upgrades-nv/e7e0daf/ : SUCCESS in 43m 49s (non-voting)\n- gate-tripleo-ci-centos-7-multinode-upgrades-nv http://logs.openstack.org/00/483500/2/check/gate-tripleo-ci-centos-7-multinode-upgrades-nv/6c2d3da/ : FAILURE in 2h 39m 58s (non-voting)\n- gate-tripleo-ci-centos-7-containers-multinode-upgrades-nv http://logs.openstack.org/00/483500/2/check/gate-tripleo-ci-centos-7-containers-multinode-upgrades-nv/6442747/ : SUCCESS in 2h 10m 51s (non-voting)\n- gate-puppet-tripleo-puppet-unit-4.5-centos-7 http://logs.openstack.org/00/483500/2/check/gate-puppet-tripleo-puppet-unit-4.5-centos-7/d684774/ : FAILURE in 22m 26s\n- gate-puppet-tripleo-puppet-unit-latest-centos-7 http://logs.openstack.org/00/483500/2/check/gate-puppet-tripleo-puppet-unit-latest-centos-7/d00865c/ : FAILURE in 22m 57s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8d3613a7c482530c2b7b570104d78e78fff862f8","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2017-07-16 05:45:34.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"23af14904e164bcff8caab1c12c86167ca8a298d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-07-16 07:52:36.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-puppet-tripleo-puppet-lint http://logs.openstack.org/00/483500/3/check/gate-puppet-tripleo-puppet-lint/8352a99/ : SUCCESS in 3m 08s\n- gate-puppet-tripleo-puppet-syntax-4-centos-7 http://logs.openstack.org/00/483500/3/check/gate-puppet-tripleo-puppet-syntax-4-centos-7/bf276f8/ : SUCCESS in 4m 02s\n- gate-tripleo-ci-centos-7-nonha-multinode-oooq http://logs.openstack.org/00/483500/3/check/gate-tripleo-ci-centos-7-nonha-multinode-oooq/ed26d1c/ : SUCCESS in 1h 41m 30s\n- gate-tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/00/483500/3/check/gate-tripleo-ci-centos-7-containers-multinode/0ad1e7c/ : SUCCESS in 1h 13m 22s\n- gate-tripleo-ci-centos-7-nonha-multinode-updates-nv http://logs.openstack.org/00/483500/3/check/gate-tripleo-ci-centos-7-nonha-multinode-updates-nv/24e6bf4/ : SUCCESS in 1h 23m 04s (non-voting)\n- gate-tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/00/483500/3/check/gate-tripleo-ci-centos-7-undercloud-containers/b286702/ : SUCCESS in 37m 09s\n- gate-tripleo-ci-centos-7-undercloud-oooq http://logs.openstack.org/00/483500/3/check/gate-tripleo-ci-centos-7-undercloud-oooq/d4c4075/ : SUCCESS in 54m 05s\n- gate-tripleo-ci-centos-7-undercloud-upgrades-nv http://logs.openstack.org/00/483500/3/check/gate-tripleo-ci-centos-7-undercloud-upgrades-nv/ac7d1af/ : SUCCESS in 39m 53s (non-voting)\n- gate-tripleo-ci-centos-7-multinode-upgrades-nv http://logs.openstack.org/00/483500/3/check/gate-tripleo-ci-centos-7-multinode-upgrades-nv/f838222/ : SUCCESS in 1h 59m 19s (non-voting)\n- gate-tripleo-ci-centos-7-containers-multinode-upgrades-nv http://logs.openstack.org/00/483500/3/check/gate-tripleo-ci-centos-7-containers-multinode-upgrades-nv/aad712b/ : SUCCESS in 2h 04m 59s (non-voting)\n- gate-puppet-tripleo-puppet-unit-4.5-centos-7 http://logs.openstack.org/00/483500/3/check/gate-puppet-tripleo-puppet-unit-4.5-centos-7/aedaf6a/ : SUCCESS in 21m 26s\n- gate-puppet-tripleo-puppet-unit-latest-centos-7 http://logs.openstack.org/00/483500/3/check/gate-puppet-tripleo-puppet-unit-latest-centos-7/61c0aa2/ : FAILURE in 16m 52s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"cb870a967bce4264d559b6ff1e6589994364e0c5","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-07-16 08:00:16.000000000","message":"Patch Set 3:\n\nBuild succeeded (check-tripleo pipeline).\n\n- gate-tripleo-ci-centos-7-ovb-ha-oooq http://logs.openstack.org/00/483500/3/check-tripleo/gate-tripleo-ci-centos-7-ovb-ha-oooq/d5e0229/ : SUCCESS in 2h 14m 28s\n- gate-tripleo-ci-centos-7-ovb-containers-oooq http://logs.openstack.org/00/483500/3/check-tripleo/gate-tripleo-ci-centos-7-ovb-containers-oooq/40f86e8/ : SUCCESS in 2h 11m 24s\n- gate-tripleo-ci-centos-7-ovb-1ctlr_1comp_1ceph-featureset024 http://logs.openstack.org/00/483500/3/check-tripleo/gate-tripleo-ci-centos-7-ovb-1ctlr_1comp_1ceph-featureset024/19036dd/ : SUCCESS in 2h 10m 09s","accounts_in_message":[],"_revision_number":3},{"id":"b773f6ef29107b0e82a90bb789caeafd204b4c85","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2017-07-17 13:58:57.000000000","message":"Patch Set 3: Code-Review+2\n\nLooks like the other failures are related to puppet-swift needing to be fixed","accounts_in_message":[],"_revision_number":3},{"id":"06e9fca7fe0b287755281e5c122e87e46eb98320","author":{"_account_id":8042,"name":"Jiří Stránský","email":"jistr@redhat.com","username":"jistr"},"date":"2017-07-17 14:32:46.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"6bce9a2205f18c42a333ce77fb03cd82dccb4dc2","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-07-17 14:32:56.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":3},{"id":"1fcf6bd52285cca2782f18eb7b9cd3a60f4cb86c","date":"2017-07-18 01:49:25.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":3},{"id":"eeadf261e1fc9ea70b18a3e6c1146a1a2517a245","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-07-18 01:49:25.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-puppet-tripleo-puppet-lint http://logs.openstack.org/00/483500/3/gate/gate-puppet-tripleo-puppet-lint/deba190/ : SUCCESS in 1m 58s\n- gate-puppet-tripleo-puppet-syntax-4-centos-7 http://logs.openstack.org/00/483500/3/gate/gate-puppet-tripleo-puppet-syntax-4-centos-7/48eb5f1/ : SUCCESS in 2m 59s\n- gate-tripleo-ci-centos-7-nonha-multinode-oooq http://logs.openstack.org/00/483500/3/gate/gate-tripleo-ci-centos-7-nonha-multinode-oooq/3a26ab8/ : SUCCESS in 1h 11m 47s\n- gate-tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/00/483500/3/gate/gate-tripleo-ci-centos-7-containers-multinode/04fab4a/ : SUCCESS in 1h 20m 45s\n- gate-tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/00/483500/3/gate/gate-tripleo-ci-centos-7-undercloud-containers/3de2129/ : SUCCESS in 28m 57s\n- gate-tripleo-ci-centos-7-undercloud-oooq http://logs.openstack.org/00/483500/3/gate/gate-tripleo-ci-centos-7-undercloud-oooq/31f7780/ : SUCCESS in 37m 08s\n- gate-puppet-tripleo-puppet-unit-4.5-centos-7 http://logs.openstack.org/00/483500/3/gate/gate-puppet-tripleo-puppet-unit-4.5-centos-7/6b8a6a0/ : SUCCESS in 23m 08s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"3ee568800d9295bc2a4f843bd3ce74b109613c6b","revisions":{"d3bbd4cb57ef5e5b4867980769e6e7f31762ec7b":{"kind":"REWORK","_number":1,"created":"2017-07-13 18:27:48.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/00/483500/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/00/483500/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/00/483500/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/00/483500/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/00/483500/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/00/483500/1"}}},"commit":{"parents":[{"commit":"dbc3c765e5747a526ef154f0497e136a8c8cdfb4","subject":"Merge \"Leverage kolla config_files to copy config into containers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/dbc3c765e5747a526ef154f0497e136a8c8cdfb4"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2017-07-13 18:26:54.000000000","tz":-420},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2017-07-13 18:27:11.000000000","tz":-420},"subject":"Update resource references for dependencies","message":"Update resource references for dependencies\n\nThe latest version of puppet now reports these as catalog failures so\nthis change removes the unnecessary references and the references\nshould be updated.\n\nCloses-Bug: #1702964\nChange-Id: Iebc547aa92f9f40e4a633c57d79e6c9cddb5dd28\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/d3bbd4cb57ef5e5b4867980769e6e7f31762ec7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/d3bbd4cb57ef5e5b4867980769e6e7f31762ec7b"}]},"branch":"refs/heads/master"},"498e6e30ddf281b88abd117ecf494cbe7b1e31be":{"kind":"REWORK","_number":2,"created":"2017-07-15 16:26:12.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/00/483500/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/00/483500/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/00/483500/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/00/483500/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/00/483500/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/00/483500/2"}}},"commit":{"parents":[{"commit":"152d224c73cdbd3243a560a27f6a18e5ffbb2e3c","subject":"Merge \"Contrail: Fix controlplane/dataplane network asignments \u0026 enable optional dpdk\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/152d224c73cdbd3243a560a27f6a18e5ffbb2e3c"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2017-07-13 18:26:54.000000000","tz":-420},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2017-07-15 16:26:03.000000000","tz":-420},"subject":"Update resource references for dependencies","message":"Update resource references for dependencies\n\nThe latest version of puppet now reports these as catalog failures so\nthis change removes the unnecessary references and the references\nshould be updated.\n\nCloses-Bug: #1702964\nChange-Id: Iebc547aa92f9f40e4a633c57d79e6c9cddb5dd28\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/498e6e30ddf281b88abd117ecf494cbe7b1e31be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/498e6e30ddf281b88abd117ecf494cbe7b1e31be"}]},"branch":"refs/heads/master"},"3ee568800d9295bc2a4f843bd3ce74b109613c6b":{"kind":"REWORK","_number":3,"created":"2017-07-16 05:45:34.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/00/483500/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/00/483500/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/00/483500/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/00/483500/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/00/483500/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/00/483500/3"}}},"commit":{"parents":[{"commit":"9c49a5c39a0705a5222985700131cdc19be0cfcc","subject":"Merge \"Add new profile for the Veritas HyperScale\u0027s cinder backend.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/9c49a5c39a0705a5222985700131cdc19be0cfcc"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2017-07-13 18:26:54.000000000","tz":-420},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2017-07-16 05:21:18.000000000","tz":-420},"subject":"Update resource references for dependencies","message":"Update resource references for dependencies\n\nThe latest version of puppet now reports these as catalog failures so\nthis change removes the unnecessary references and the references\nshould be updated.\n\nCloses-Bug: #1702964\nChange-Id: Iebc547aa92f9f40e4a633c57d79e6c9cddb5dd28\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/3ee568800d9295bc2a4f843bd3ce74b109613c6b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/3ee568800d9295bc2a4f843bd3ce74b109613c6b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
