)]}'
{"id":"openstack%2Fpuppet-tripleo~522608","triplet_id":"openstack%2Fpuppet-tripleo~stable%2Focata~Idc77e54d0a17825634d5450c97fe80d1890ca631","project":"openstack/puppet-tripleo","branch":"stable/ocata","topic":"test","hashtags":[],"change_id":"Idc77e54d0a17825634d5450c97fe80d1890ca631","subject":"TEST DNM","status":"ABANDONED","created":"2017-11-23 16:46:48.000000000","updated":"2017-11-23 19:39:23.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b29bddbc2cc8cfe291b8f22e2d2aec4107e01421","_number":522608,"virtual_id_number":522608,"owner":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":-1,"date":"2017-11-23 17:11:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-11-23 17:11:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"07ee7a5c35629296bbdc9820b8ada815d6bb118f","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2017-11-23 16:46:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"abb2a0b0c6cf2dc9a304ac50f727a2e313670f58","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-11-23 17:11:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint http://logs.openstack.org/08/522608/1/check/puppet-openstack-lint/ad68556/ : SUCCESS in 3m 23s\n- puppet-openstack-syntax-3-legacy http://logs.openstack.org/08/522608/1/check/puppet-openstack-syntax-3-legacy/6c6386e/ : SUCCESS in 2m 43s\n- puppet-openstack-syntax-4 http://logs.openstack.org/08/522608/1/check/puppet-openstack-syntax-4/a62c505/ : SUCCESS in 2m 44s\n- puppet-openstack-unit-3.6-centos-7 http://logs.openstack.org/08/522608/1/check/puppet-openstack-unit-3.6-centos-7/d4bd6c6/ : FAILURE in 21m 08s\n- puppet-openstack-unit-3.8-centos-7 http://logs.openstack.org/08/522608/1/check/puppet-openstack-unit-3.8-centos-7/7fae056/ : FAILURE in 22m 09s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/08/522608/1/check/puppet-openstack-unit-4.8-centos-7/a026e4c/ : SUCCESS in 21m 05s","accounts_in_message":[],"_revision_number":1},{"id":"1ef213533152dd26914aa1598e2068e344cd12cf","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2017-11-23 19:39:23.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"359f6d5b5805fea0b8806e13645a07cc7cc6f5c7","revisions":{"359f6d5b5805fea0b8806e13645a07cc7cc6f5c7":{"kind":"REWORK","_number":1,"created":"2017-11-23 16:46:48.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/08/522608/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/08/522608/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/08/522608/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/08/522608/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/08/522608/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/08/522608/1"}}},"commit":{"parents":[{"commit":"b6e8d2018093de9b650a371d6c4d0fde79d61d92","subject":"Puppetfile: pin puppetlabs-haproxy to 1.5 (ocata commit)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/b6e8d2018093de9b650a371d6c4d0fde79d61d92"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2017-11-23 16:46:30.000000000","tz":-480},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2017-11-23 16:46:35.000000000","tz":-480},"subject":"TEST DNM","message":"TEST DNM\n\nChange-Id: Idc77e54d0a17825634d5450c97fe80d1890ca631\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/359f6d5b5805fea0b8806e13645a07cc7cc6f5c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/359f6d5b5805fea0b8806e13645a07cc7cc6f5c7"}]},"parents_data":[{"branch_name":"refs/heads/stable/ocata","commit_id":"b6e8d2018093de9b650a371d6c4d0fde79d61d92","is_merged_in_target_branch":false,"change_id":"Ib7ea0eaf156121652ace7ee74a7b7003dd1376cd","change_number":521967,"patch_set_number":7,"change_status":"MERGED"}],"branch":"refs/heads/stable/ocata"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
