)]}'
{"id":"openstack%2Fpuppet-tripleo~655813","triplet_id":"openstack%2Fpuppet-tripleo~stable%2Fstein~If8eb685dd21791b65845afd44fb483592b8bcc0c","project":"openstack/puppet-tripleo","branch":"stable/stein","topic":"bug/1825146-stable/stein","hashtags":[],"change_id":"If8eb685dd21791b65845afd44fb483592b8bcc0c","subject":"Add support to set ovs_nb_connection","status":"MERGED","created":"2019-04-25 23:26:08.000000000","updated":"2019-04-26 18:31:44.000000000","submitted":"2019-04-26 18:31:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"655813-1556303504250-cdd0ea5d","meta_rev_id":"a80a4ce61e63cbdbce6e7601aac52ca3a5c811e2","_number":655813,"virtual_id_number":655813,"owner":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-04-26 18:31:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-04-26 02:31:59.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":2,"date":"2019-04-26 14:18:18.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"date":"2019-04-25 23:26:08.000000000","_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":1,"date":"2019-04-26 14:18:18.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-25 23:26:08.000000000","updated_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"reviewer":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"state":"REVIEWER"},{"updated":"2019-04-26 02:31:59.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-04-26 14:18:18.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-04-26 18:31:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3970d7dbefeb225a2c6c4c1b593ddf2f8a3bd80f","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-04-25 23:26:08.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"eda9e7c24054c4fc4d262bbeb1752603f1fdecf3","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-04-26 02:31:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/13/655813/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/6a9efc2/ : SUCCESS in 2h 45m 11s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/13/655813/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/12f0228/ : SUCCESS in 3h 02m 19s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/13/655813/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/f55da19/ : SUCCESS in 3h 04m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fa39f3d4244dc4f22cedcd28a2248ba965cac14a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 03:43:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/13/655813/1/check/puppet-openstack-lint-ubuntu-bionic/aa23c73/ : SUCCESS in 4m 53s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/13/655813/1/check/puppet-openstack-syntax-4-ubuntu-bionic/6fa8a43/ : SUCCESS in 4m 21s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/13/655813/1/check/puppet-openstack-syntax-5-ubuntu-bionic/a00788e/ : SUCCESS in 4m 36s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/13/655813/1/check/puppet-openstack-syntax-6-ubuntu-bionic/98874dc/ : SUCCESS in 4m 56s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/13/655813/1/check/puppet-openstack-unit-4.8-centos-7/cc7ad81/ : SUCCESS in 26m 47s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/13/655813/1/check/puppet-openstack-unit-5.5-centos-7/7ec950c/ : SUCCESS in 30m 03s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/13/655813/1/check/tripleo-ci-centos-7-undercloud-containers/36a355e/ : SUCCESS in 1h 34m 48s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/13/655813/1/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/32d3a72/ : FAILURE in 2h 10m 08s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/13/655813/1/check/tripleo-ci-centos-7-containers-multinode/1c23b07/ : SUCCESS in 1h 57m 11s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/13/655813/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/3881693/ : SUCCESS in 2h 16m 23s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/13/655813/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/a7f5b9a/ : SUCCESS in 2h 37m 37s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/13/655813/1/check/tripleo-ci-centos-7-standalone/67a7f1b/ : SUCCESS in 1h 55m 33s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/13/655813/1/check/tripleo-ci-fedora-28-standalone/0d45d43/ : SUCCESS in 1h 41m 42s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/13/655813/1/check/tripleo-ci-centos-7-standalone-upgrade/0f0fffa/ : FAILURE in 2h 11m 46s (non-voting)\n- build-openstack-releasenotes http://logs.openstack.org/13/655813/1/check/build-openstack-releasenotes/3275cd1/html/ : SUCCESS in 3m 54s\n- tripleo-ci-centos-7-scenario010-standalone http://logs.openstack.org/13/655813/1/check/tripleo-ci-centos-7-scenario010-standalone/04fa35e/ : SUCCESS in 1h 15m 17s (non-voting)\n- tripleo-ci-centos-7-scenario010-multinode-oooq-container http://logs.openstack.org/13/655813/1/check/tripleo-ci-centos-7-scenario010-multinode-oooq-container/983962a/ : SUCCESS in 2h 21m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7a51e6f9738a7e00374ab8a6c23086dbb0a646eb","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-04-26 14:18:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0dbe7bcc2b3915d0a3728ed2418f22113f7acb4f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 14:18:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0192931969e15cedd10ebfcea1d5d55e5eb1daf7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 16:04:38.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4891418aa218db84f4c8b86257317469baedc55c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 18:31:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/13/655813/1/gate/puppet-openstack-lint-ubuntu-bionic/3092658/ : SUCCESS in 5m 32s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/13/655813/1/gate/puppet-openstack-syntax-4-ubuntu-bionic/3b821cb/ : SUCCESS in 7m 30s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/13/655813/1/gate/puppet-openstack-syntax-5-ubuntu-bionic/cb756c8/ : SUCCESS in 6m 57s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/13/655813/1/gate/puppet-openstack-syntax-6-ubuntu-bionic/31242dd/ : SUCCESS in 7m 19s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/13/655813/1/gate/puppet-openstack-unit-4.8-centos-7/fc61a84/ : SUCCESS in 31m 15s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/13/655813/1/gate/puppet-openstack-unit-5.5-centos-7/89be356/ : SUCCESS in 32m 41s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/13/655813/1/gate/tripleo-ci-centos-7-undercloud-containers/9fced9d/ : SUCCESS in 1h 33m 34s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/13/655813/1/gate/tripleo-ci-centos-7-containers-multinode/b4c5d07/ : SUCCESS in 2h 02m 43s\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/13/655813/1/gate/tripleo-ci-centos-7-standalone/20c2f41/ : SUCCESS in 1h 49m 33s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/13/655813/1/gate/tripleo-ci-fedora-28-standalone/9702d0a/ : SUCCESS in 1h 43m 41s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/13/655813/1/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/614f7e8/ : SUCCESS in 2h 16m 02s\n- build-openstack-releasenotes http://logs.openstack.org/13/655813/1/gate/build-openstack-releasenotes/9c5c90b/html/ : SUCCESS in 3m 06s","accounts_in_message":[],"_revision_number":1},{"id":"e38d1a0f9dd913681dc6547fae7f7ca96e31ee9a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 18:31:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ad0e5975fb8a8fae50eb07b1a2b692f79d10ba8c","revisions":{"ad0e5975fb8a8fae50eb07b1a2b692f79d10ba8c":{"kind":"REWORK","_number":1,"created":"2019-04-25 23:26:08.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/13/655813/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/13/655813/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/13/655813/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/13/655813/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/13/655813/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/13/655813/1"}}},"commit":{"parents":[{"commit":"bc825d05baf1d800f46ea9d4480db471871f9de8","subject":"Merge \"Set octavia provider_drivers config option correctly\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/bc825d05baf1d800f46ea9d4480db471871f9de8"}]}],"author":{"name":"Rabi Mishra","email":"ramishra@redhat.com","date":"2019-04-18 03:55:47.000000000","tz":330},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-04-25 23:26:08.000000000","tz":0},"subject":"Add support to set ovs_nb_connection","message":"Add support to set ovs_nb_connection\n\nWhen using ovn provider unless set, it would try to\nconnect to 127.0.0.1.\n\nChange-Id: If8eb685dd21791b65845afd44fb483592b8bcc0c\nDepends-On: https://review.openstack.org/653673\nCloses-Bug: #1825146\n(cherry picked from commit d88db3ad716561e5526afc1a0cdb4de08e599741)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/ad0e5975fb8a8fae50eb07b1a2b692f79d10ba8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/ad0e5975fb8a8fae50eb07b1a2b692f79d10ba8c"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
