)]}'
{"id":"openstack%2Fpuppet-tripleo~658329","triplet_id":"openstack%2Fpuppet-tripleo~master~I5e5aae692482ee1de6de298eb18c49563a2d7d29","project":"openstack/puppet-tripleo","branch":"master","topic":"rsyslog","hashtags":[],"change_id":"I5e5aae692482ee1de6de298eb18c49563a2d7d29","subject":"Base profile for rsyslogd","status":"MERGED","created":"2019-05-10 10:57:04.000000000","updated":"2019-06-14 22:11:46.000000000","submitted":"2019-06-14 22:11:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"658329-1560550306336-91065db7","meta_rev_id":"469ccee7919f5ebdff6e80c4c66471e3ef46ea06","_number":658329,"virtual_id_number":658329,"owner":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":30402,"name":"gaoqiang","email":"gqgqgqdyx@126.com"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},{"value":0,"_account_id":22954,"name":"Juan Badia Payno","email":"jbadiapa@redhat.com","username":"jbadiapa"},{"value":2,"date":"2019-06-14 22:11:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-12 08:12:07.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2019-06-14 15:07:47.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":30402,"name":"gaoqiang","email":"gqgqgqdyx@126.com"},{"value":2,"date":"2019-06-12 16:20:38.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},{"value":0,"_account_id":22954,"name":"Juan Badia Payno","email":"jbadiapa@redhat.com","username":"jbadiapa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2019-06-14 15:07:47.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":30402,"name":"gaoqiang","email":"gqgqgqdyx@126.com"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},{"value":0,"_account_id":22954,"name":"Juan Badia Payno","email":"jbadiapa@redhat.com","username":"jbadiapa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22954,"name":"Juan Badia Payno","email":"jbadiapa@redhat.com","username":"jbadiapa"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":30402,"name":"gaoqiang","email":"gqgqgqdyx@126.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-10 11:46:12.000000000","updated_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"reviewer":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"state":"REVIEWER"},{"updated":"2019-05-14 10:33:19.000000000","updated_by":{"_account_id":30402,"name":"gaoqiang","email":"gqgqgqdyx@126.com"},"reviewer":{"_account_id":30402,"name":"gaoqiang","email":"gqgqgqdyx@126.com"},"state":"REVIEWER"},{"updated":"2019-06-10 09:36:31.000000000","updated_by":{"_account_id":22954,"name":"Juan Badia Payno","email":"jbadiapa@redhat.com","username":"jbadiapa"},"reviewer":{"_account_id":22954,"name":"Juan Badia Payno","email":"jbadiapa@redhat.com","username":"jbadiapa"},"state":"REVIEWER"},{"updated":"2019-06-12 08:12:07.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-12 16:20:38.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-06-14 15:07:47.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-06-14 22:11:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"526016839952384af297952d66e84e7175da59b1","author":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"date":"2019-05-10 10:57:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d03d568843be5e49c9a6150db3477a6a319d2b48","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2019-05-10 11:46:12.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"baceee9ef38dc4e5e75f790219284e4288f82261","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-10 12:37:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/29/658329/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/7c10af2/ : FAILURE in 1h 39m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b62d08953f3ea06bf32c32e228c3003f44d5167e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 13:36:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/29/658329/1/check/puppet-openstack-lint-ubuntu-bionic/8b98e3f/ : FAILURE in 6m 18s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/29/658329/1/check/puppet-openstack-syntax-4-ubuntu-bionic/119e143/ : SUCCESS in 6m 23s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/29/658329/1/check/puppet-openstack-syntax-5-ubuntu-bionic/39cd00c/ : SUCCESS in 6m 07s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/29/658329/1/check/puppet-openstack-syntax-6-ubuntu-bionic/8272da9/ : SUCCESS in 6m 35s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/29/658329/1/check/puppet-openstack-unit-5.5-centos-7/11177db/ : SUCCESS in 28m 38s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/29/658329/1/check/puppet-openstack-unit-6.0-ubuntu-bionic/67d2be2/ : SUCCESS in 13m 02s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/29/658329/1/check/puppet-openstack-unit-latest-ubuntu-bionic/764e342/ : SUCCESS in 16m 05s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/29/658329/1/check/tripleo-ci-centos-7-undercloud-containers/c15f493/ : SUCCESS in 1h 43m 14s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/29/658329/1/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/14e2313/ : FAILURE in 1h 39m 13s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/29/658329/1/check/tripleo-ci-centos-7-containers-multinode/e30d8c2/ : SUCCESS in 1h 56m 01s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/29/658329/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/4255cd1/ : SUCCESS in 2h 16m 02s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/29/658329/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/131622b/ : SUCCESS in 2h 26m 05s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/29/658329/1/check/tripleo-ci-centos-7-standalone/acbfccf/ : SUCCESS in 1h 44m 33s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/29/658329/1/check/tripleo-ci-fedora-28-standalone/f3c5337/ : SUCCESS in 1h 51m 00s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/29/658329/1/check/tripleo-ci-centos-7-standalone-upgrade/0c468ae/ : SUCCESS in 2h 03m 58s (non-voting)\n- build-openstack-releasenotes http://logs.openstack.org/29/658329/1/check/build-openstack-releasenotes/13575d1/html/ : SUCCESS in 3m 42s\n- tripleo-ci-centos-7-scenario001-standalone http://logs.openstack.org/29/658329/1/check/tripleo-ci-centos-7-scenario001-standalone/d5602f2/ : SUCCESS in 1h 36m 46s\n- tripleo-ci-centos-7-standalone-rocky http://logs.openstack.org/29/658329/1/check/tripleo-ci-centos-7-standalone-rocky/6abc907/ : SUCCESS in 1h 31m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d18469f42b50676b5f515e3013fefd5859854b26","author":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"date":"2019-05-10 20:07:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c19e5ddab4435043e414e80911c3be54430f63f0","author":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"date":"2019-05-13 14:28:34.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2dada8ae3f564523c7665d253f242b70e44cb12a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-13 16:30:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/29/658329/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/98e3d0c/ : FAILURE in 1h 30m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2d59355044ad71c5455466c92507b6f507e5b70c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-13 18:41:36.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/29/658329/2/check/puppet-openstack-lint-ubuntu-bionic/77836fd/ : FAILURE in 7m 09s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/29/658329/2/check/puppet-openstack-syntax-4-ubuntu-bionic/d257e0c/ : SUCCESS in 6m 35s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/29/658329/2/check/puppet-openstack-syntax-5-ubuntu-bionic/1554d9c/ : SUCCESS in 5m 54s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/29/658329/2/check/puppet-openstack-syntax-6-ubuntu-bionic/5de9ad4/ : SUCCESS in 5m 37s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/29/658329/2/check/puppet-openstack-unit-5.5-centos-7/59172cd/ : SUCCESS in 27m 10s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/29/658329/2/check/puppet-openstack-unit-6.0-ubuntu-bionic/eb0b867/ : SUCCESS in 16m 40s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/29/658329/2/check/puppet-openstack-unit-latest-ubuntu-bionic/2717205/ : SUCCESS in 17m 10s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/29/658329/2/check/tripleo-ci-centos-7-undercloud-containers/01c68b3/ : SUCCESS in 1h 25m 50s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/29/658329/2/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/c7394e1/ : FAILURE in 1h 39m 59s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/29/658329/2/check/tripleo-ci-centos-7-containers-multinode/235a110/ : SUCCESS in 2h 03m 43s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/29/658329/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/bbacefc/ : SUCCESS in 2h 07m 13s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/29/658329/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/99b1bba/ : SUCCESS in 2h 05m 24s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/29/658329/2/check/tripleo-ci-centos-7-standalone/6114a72/ : SUCCESS in 1h 45m 15s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/29/658329/2/check/tripleo-ci-fedora-28-standalone/c7383f0/ : SUCCESS in 1h 54m 22s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/29/658329/2/check/tripleo-ci-centos-7-standalone-upgrade/f76601b/ : SUCCESS in 2h 00m 58s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/29/658329/2/check/tripleo-ci-centos-7-standalone-upgrade-stein/0e08a23/ : SUCCESS in 1h 54m 48s\n- build-openstack-releasenotes http://logs.openstack.org/29/658329/2/check/build-openstack-releasenotes/c55c2d8/html/ : SUCCESS in 2m 54s\n- tripleo-ci-centos-7-scenario001-standalone http://logs.openstack.org/29/658329/2/check/tripleo-ci-centos-7-scenario001-standalone/95d9ae3/ : SUCCESS in 1h 48m 10s\n- tripleo-ci-centos-7-standalone-rocky http://logs.openstack.org/29/658329/2/check/tripleo-ci-centos-7-standalone-rocky/f827167/ : SUCCESS in 1h 38m 50s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"006ba87dd065110ab39f10921a1078bd99a23cc4","author":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"date":"2019-05-13 18:43:40.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"40015e9055e5bbce225e8eb97b948810dd31753a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-13 21:11:08.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/29/658329/3/check/puppet-openstack-lint-ubuntu-bionic/d9e261f/ : FAILURE in 5m 05s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/29/658329/3/check/puppet-openstack-syntax-4-ubuntu-bionic/0b27bc1/ : SUCCESS in 4m 03s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/29/658329/3/check/puppet-openstack-syntax-5-ubuntu-bionic/aec5b24/ : SUCCESS in 3m 44s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/29/658329/3/check/puppet-openstack-syntax-6-ubuntu-bionic/81a9e0d/ : SUCCESS in 3m 45s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/29/658329/3/check/puppet-openstack-unit-5.5-centos-7/be86e30/ : SUCCESS in 30m 15s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/29/658329/3/check/puppet-openstack-unit-6.0-ubuntu-bionic/67012a9/ : SUCCESS in 14m 08s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/29/658329/3/check/puppet-openstack-unit-latest-ubuntu-bionic/91f29e4/ : SUCCESS in 17m 11s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/29/658329/3/check/tripleo-ci-centos-7-undercloud-containers/f167f2d/ : SUCCESS in 1h 26m 30s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/29/658329/3/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/4439ffd/ : FAILURE in 1h 44m 25s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/29/658329/3/check/tripleo-ci-centos-7-containers-multinode/089571f/ : SUCCESS in 2h 02m 41s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/29/658329/3/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/51e2429/ : SUCCESS in 2h 09m 08s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/29/658329/3/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/cd52189/ : SUCCESS in 2h 16m 30s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/29/658329/3/check/tripleo-ci-centos-7-standalone/455fe37/ : SUCCESS in 1h 40m 40s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/29/658329/3/check/tripleo-ci-fedora-28-standalone/b99e2cd/ : SUCCESS in 1h 47m 26s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/29/658329/3/check/tripleo-ci-centos-7-standalone-upgrade/64a1ca0/ : FAILURE in 1h 34m 15s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/29/658329/3/check/tripleo-ci-centos-7-standalone-upgrade-stein/65cc34e/ : SUCCESS in 2h 21m 22s\n- build-openstack-releasenotes http://logs.openstack.org/29/658329/3/check/build-openstack-releasenotes/e6ec559/html/ : SUCCESS in 2m 28s\n- tripleo-ci-centos-7-scenario001-standalone http://logs.openstack.org/29/658329/3/check/tripleo-ci-centos-7-scenario001-standalone/cecc255/ : SUCCESS in 1h 35m 45s\n- tripleo-ci-centos-7-standalone-rocky http://logs.openstack.org/29/658329/3/check/tripleo-ci-centos-7-standalone-rocky/d0f033c/ : SUCCESS in 1h 35m 19s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"377117230960d9c5884891e68bf9e309bc246eba","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-13 21:49:51.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/29/658329/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/b225eba/ : SUCCESS in 3h 04m 38s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3506b84f51b7949d03d17fe1074258b5efac0a37","author":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"date":"2019-05-14 07:34:10.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"d058bc1269f69e895e619faa7c05bf9390c05652","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 10:23:23.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/29/658329/4/check/puppet-openstack-lint-ubuntu-bionic/a109702/ : SUCCESS in 4m 00s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/29/658329/4/check/puppet-openstack-syntax-4-ubuntu-bionic/0a3e767/ : SUCCESS in 5m 29s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/29/658329/4/check/puppet-openstack-syntax-5-ubuntu-bionic/a2cc526/ : SUCCESS in 4m 02s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/29/658329/4/check/puppet-openstack-syntax-6-ubuntu-bionic/232be6c/ : SUCCESS in 3m 55s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/29/658329/4/check/puppet-openstack-unit-5.5-centos-7/d153db3/ : SUCCESS in 29m 09s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/29/658329/4/check/puppet-openstack-unit-6.0-ubuntu-bionic/d198446/ : FAILURE in 7m 04s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/29/658329/4/check/puppet-openstack-unit-latest-ubuntu-bionic/fb998c6/ : SUCCESS in 13m 24s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/29/658329/4/check/tripleo-ci-centos-7-undercloud-containers/b0e5803/ : SUCCESS in 1h 25m 37s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/29/658329/4/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/c8d77bd/ : FAILURE in 1h 32m 16s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/29/658329/4/check/tripleo-ci-centos-7-containers-multinode/4745464/ : SUCCESS in 1h 52m 23s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/29/658329/4/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/4dc4ab3/ : SUCCESS in 1h 56m 37s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/29/658329/4/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/cbb61a3/ : SUCCESS in 2h 28m 20s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/29/658329/4/check/tripleo-ci-centos-7-standalone/d8a5b26/ : SUCCESS in 2h 00m 18s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/29/658329/4/check/tripleo-ci-fedora-28-standalone/c57a7f4/ : SUCCESS in 1h 53m 32s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/29/658329/4/check/tripleo-ci-centos-7-standalone-upgrade/8730962/ : FAILURE in 1h 40m 02s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/29/658329/4/check/tripleo-ci-centos-7-standalone-upgrade-stein/e889a92/ : SUCCESS in 2h 17m 59s\n- build-openstack-releasenotes http://logs.openstack.org/29/658329/4/check/build-openstack-releasenotes/29195c5/html/ : SUCCESS in 3m 37s\n- tripleo-ci-centos-7-scenario001-standalone http://logs.openstack.org/29/658329/4/check/tripleo-ci-centos-7-scenario001-standalone/23e04b9/ : SUCCESS in 1h 59m 47s\n- tripleo-ci-centos-7-standalone-rocky http://logs.openstack.org/29/658329/4/check/tripleo-ci-centos-7-standalone-rocky/6ca0cb3/ : SUCCESS in 1h 54m 15s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"5022827104dc8950bf4602b05b52fdda701d3dc8","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-14 10:28:55.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/29/658329/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/c48d9f6/ : SUCCESS in 2h 53m 50s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"792de1cb56d8b7b58a1ef4d6310a9c90595e8aff","author":{"_account_id":30402,"name":"gaoqiang","email":"gqgqgqdyx@126.com"},"date":"2019-05-14 10:33:19.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"0f870151d0370bc6636af786525f6ffea9ab2e00","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-05-30 23:54:35.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1f5cdf828ec65e02282609fad4f632b9fd4fdace","author":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"date":"2019-06-06 12:48:56.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"12c6d10ffd0ca2b24bf9cf258f8fd602da26f7e5","author":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"date":"2019-06-06 12:50:16.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"7542ed77ee0f50caba05389d14fc49dcb686585b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-06 15:36:37.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/29/658329/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/85fc3fe/ : SUCCESS in 2h 45m 30s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"c1e27983756e341eb78fb3606a031c5066d229b6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 19:18:55.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/29/658329/6/check/puppet-openstack-lint-ubuntu-bionic/a41b567/ : SUCCESS in 4m 37s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/29/658329/6/check/puppet-openstack-syntax-4-ubuntu-bionic/5082741/ : SUCCESS in 4m 41s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/29/658329/6/check/puppet-openstack-syntax-5-ubuntu-bionic/2437548/ : SUCCESS in 5m 11s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/29/658329/6/check/puppet-openstack-syntax-6-ubuntu-bionic/451aa8a/ : SUCCESS in 4m 08s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/29/658329/6/check/puppet-openstack-unit-5.5-centos-7/ff63744/ : FAILURE in 22m 52s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/29/658329/6/check/puppet-openstack-unit-6.0-ubuntu-bionic/bdf8f63/ : FAILURE in 10m 55s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/29/658329/6/check/puppet-openstack-unit-latest-ubuntu-bionic/fa8b87a/ : FAILURE in 8m 55s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/29/658329/6/check/tripleo-ci-centos-7-undercloud-containers/7f8f8df/ : SUCCESS in 1h 51m 09s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/29/658329/6/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/a6314a7/ : FAILURE in 1h 26m 03s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/29/658329/6/check/tripleo-ci-centos-7-containers-multinode/92570b5/ : SUCCESS in 1h 58m 55s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/29/658329/6/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/2a3b49e/ : SUCCESS in 2h 09m 13s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/29/658329/6/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/b5f16c5/ : SUCCESS in 2h 31m 33s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/29/658329/6/check/tripleo-ci-centos-7-standalone/04f9611/ : SUCCESS in 1h 51m 45s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/29/658329/6/check/tripleo-ci-fedora-28-standalone/efa0cb4/ : SUCCESS in 1h 58m 45s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/29/658329/6/check/tripleo-ci-centos-7-standalone-upgrade/7f01537/ : SUCCESS in 2h 13m 16s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/29/658329/6/check/tripleo-ci-centos-7-standalone-upgrade-stein/46140e1/ : SUCCESS in 1h 53m 35s\n- build-openstack-releasenotes http://logs.openstack.org/29/658329/6/check/build-openstack-releasenotes/e85a245/html/ : SUCCESS in 3m 08s\n- tripleo-ci-centos-7-scenario001-standalone http://logs.openstack.org/29/658329/6/check/tripleo-ci-centos-7-scenario001-standalone/c217563/ : SUCCESS in 1h 47m 19s","accounts_in_message":[],"_revision_number":6},{"id":"ec0030e77269603f70812648e428f6b663a29725","author":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"date":"2019-06-06 21:13:01.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"00b009d7cf1c75376efe818fa5b1dd76a1ac67bc","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-06 23:48:35.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/29/658329/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/97970fc/ : SUCCESS in 2h 34m 47s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"6ae2ac4af853abc9e4ce6e8bf2cce1d85cac289d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 04:12:53.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/29/658329/7/check/puppet-openstack-lint-ubuntu-bionic/eea2185/ : SUCCESS in 4m 27s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/29/658329/7/check/puppet-openstack-syntax-4-ubuntu-bionic/0a0ddaf/ : SUCCESS in 4m 13s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/29/658329/7/check/puppet-openstack-syntax-5-ubuntu-bionic/356b349/ : SUCCESS in 3m 56s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/29/658329/7/check/puppet-openstack-syntax-6-ubuntu-bionic/3bd4ba8/ : SUCCESS in 3m 47s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/29/658329/7/check/puppet-openstack-unit-5.5-centos-7/a1105d1/ : FAILURE in 13m 42s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/29/658329/7/check/puppet-openstack-unit-6.0-ubuntu-bionic/347b803/ : FAILURE in 8m 09s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/29/658329/7/check/puppet-openstack-unit-latest-ubuntu-bionic/23f0797/ : FAILURE in 8m 26s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/29/658329/7/check/tripleo-ci-centos-7-undercloud-containers/dd6132a/ : SUCCESS in 1h 26m 44s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/29/658329/7/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/25142ab/ : FAILURE in 1h 19m 34s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/29/658329/7/check/tripleo-ci-centos-7-containers-multinode/0255761/ : SUCCESS in 1h 56m 56s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/29/658329/7/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/7b2374e/ : SUCCESS in 2h 17m 47s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/29/658329/7/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/16d6238/ : SUCCESS in 2h 23m 52s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/29/658329/7/check/tripleo-ci-centos-7-standalone/afc3b23/ : SUCCESS in 1h 36m 59s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/29/658329/7/check/tripleo-ci-fedora-28-standalone/1ff5368/ : SUCCESS in 1h 37m 20s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/29/658329/7/check/tripleo-ci-centos-7-standalone-upgrade/a1df42b/ : SUCCESS in 2h 02m 06s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/29/658329/7/check/tripleo-ci-centos-7-standalone-upgrade-stein/7f9109f/ : SUCCESS in 2h 07m 31s\n- build-openstack-releasenotes http://logs.openstack.org/29/658329/7/check/build-openstack-releasenotes/305c1c4/html/ : SUCCESS in 2m 52s\n- tripleo-ci-centos-7-scenario001-standalone http://logs.openstack.org/29/658329/7/check/tripleo-ci-centos-7-scenario001-standalone/f48884d/ : SUCCESS in 1h 47m 14s","accounts_in_message":[],"_revision_number":7},{"id":"b3c81a40a86e71010daaebf81ced55f10d651168","author":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"date":"2019-06-07 08:13:52.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"fbf18c640c105a1270636bc46fb29c9e064e82d1","author":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"date":"2019-06-07 09:15:55.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"246aca1f2b44155eb311260dfdb292a70301a4fc","author":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"date":"2019-06-07 10:07:39.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"e99bb157300f0237bd1a62cf5e59975862d4d056","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-07 12:53:53.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/29/658329/10/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/158c567/ : SUCCESS in 2h 45m 24s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"eb80afbdb4f9020b286f572d507b2b0942b2dbfe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 14:16:51.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/29/658329/10/check/puppet-openstack-lint-ubuntu-bionic/7cfa7dc/ : SUCCESS in 6m 48s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/29/658329/10/check/puppet-openstack-syntax-4-ubuntu-bionic/822bdb4/ : SUCCESS in 3m 54s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/29/658329/10/check/puppet-openstack-syntax-5-ubuntu-bionic/fe9eb25/ : SUCCESS in 3m 57s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/29/658329/10/check/puppet-openstack-syntax-6-ubuntu-bionic/b2cc09d/ : SUCCESS in 3m 50s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/29/658329/10/check/puppet-openstack-unit-5.5-centos-7/ae49005/ : FAILURE in 31m 09s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/29/658329/10/check/puppet-openstack-unit-6.0-ubuntu-bionic/1f1a931/ : FAILURE in 15m 36s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/29/658329/10/check/puppet-openstack-unit-latest-ubuntu-bionic/f1b7b18/ : FAILURE in 14m 36s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/29/658329/10/check/tripleo-ci-centos-7-undercloud-containers/58e3008/ : SUCCESS in 1h 34m 15s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/29/658329/10/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/fcdd5aa/ : FAILURE in 1h 23m 16s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/29/658329/10/check/tripleo-ci-centos-7-containers-multinode/3c115e2/ : SUCCESS in 2h 05m 23s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/29/658329/10/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/db17b91/ : SUCCESS in 2h 15m 25s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/29/658329/10/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/b1791d7/ : SUCCESS in 2h 26m 43s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/29/658329/10/check/tripleo-ci-centos-7-standalone/018eae0/ : SUCCESS in 1h 49m 05s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/29/658329/10/check/tripleo-ci-fedora-28-standalone/396c652/ : SUCCESS in 2h 07m 47s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/29/658329/10/check/tripleo-ci-centos-7-standalone-upgrade/6592b78/ : SUCCESS in 2h 11m 25s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/29/658329/10/check/tripleo-ci-centos-7-standalone-upgrade-stein/7127e45/ : SUCCESS in 2h 19m 22s\n- build-openstack-releasenotes http://logs.openstack.org/29/658329/10/check/build-openstack-releasenotes/2214ec2/html/ : SUCCESS in 2m 52s\n- tripleo-ci-centos-7-scenario001-standalone http://logs.openstack.org/29/658329/10/check/tripleo-ci-centos-7-scenario001-standalone/589c806/ : SUCCESS in 1h 35m 15s","accounts_in_message":[],"_revision_number":10},{"id":"4135507a378c2e40b1cf932c8e61ca99df79c58b","author":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"date":"2019-06-11 21:20:50.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"8fa33cde1eb19c9d4daf4a8d37d8127ac21b24d6","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-12 00:12:18.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/29/658329/11/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/7684b57/ : SUCCESS in 2h 50m 38s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"9939bfbd36046616408745984057b6779d797591","author":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"date":"2019-06-12 06:13:08.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"ee82de5cb3cfb559313d07a98d44d8867c61fadc","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-12 08:12:07.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/29/658329/12/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/aff5410/ : FAILURE in 1h 58m 21s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"2ecd7e458cafc445a4595503b916e56042ccf9ac","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 16:16:57.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/29/658329/12/check/puppet-openstack-lint-ubuntu-bionic/018b3a2/ : SUCCESS in 4m 09s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/29/658329/12/check/puppet-openstack-syntax-4-ubuntu-bionic/a19655c/ : SUCCESS in 4m 32s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/29/658329/12/check/puppet-openstack-syntax-5-ubuntu-bionic/3adef73/ : SUCCESS in 4m 07s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/29/658329/12/check/puppet-openstack-syntax-6-ubuntu-bionic/ff8f1b3/ : SUCCESS in 4m 40s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/29/658329/12/check/puppet-openstack-unit-5.5-centos-7/e2324b1/ : SUCCESS in 28m 49s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/29/658329/12/check/puppet-openstack-unit-6.0-ubuntu-bionic/458340c/ : SUCCESS in 14m 19s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/29/658329/12/check/puppet-openstack-unit-latest-ubuntu-bionic/e1d219f/ : SUCCESS in 14m 48s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/29/658329/12/check/tripleo-ci-centos-7-undercloud-containers/6aed760/ : SUCCESS in 1h 30m 16s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/29/658329/12/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/e100b56/ : FAILURE in 1h 43m 19s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/29/658329/12/check/tripleo-ci-centos-7-containers-multinode/ba35bea/ : SUCCESS in 2h 22m 14s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/29/658329/12/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/a5efe3a/ : SUCCESS in 2h 28m 06s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/29/658329/12/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/26af9fd/ : SUCCESS in 2h 16m 27s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/29/658329/12/check/tripleo-ci-centos-7-standalone/294f699/ : SUCCESS in 2h 18m 36s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/29/658329/12/check/tripleo-ci-fedora-28-standalone/36b1c43/ : SUCCESS in 1h 42m 29s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/29/658329/12/check/tripleo-ci-centos-7-standalone-upgrade/41ea443/ : SUCCESS in 2h 10m 42s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/29/658329/12/check/tripleo-ci-centos-7-standalone-upgrade-stein/c6f9eb8/ : SUCCESS in 2h 09m 02s\n- build-openstack-releasenotes http://logs.openstack.org/29/658329/12/check/build-openstack-releasenotes/e9a0fd1/html/ : SUCCESS in 3m 31s\n- tripleo-ci-centos-7-scenario001-standalone http://logs.openstack.org/29/658329/12/check/tripleo-ci-centos-7-scenario001-standalone/782ae16/ : SUCCESS in 1h 44m 47s","accounts_in_message":[],"_revision_number":12},{"id":"631498f2ab68cb2c84f707ab7c32b1b4a25d2616","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-12 16:20:38.000000000","message":"Patch Set 12: Code-Review+2","accounts_in_message":[],"_revision_number":12},{"id":"ad659c8c7510ff80fa614b0c0f40391bbb1f2ba0","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-06-14 15:07:47.000000000","message":"Patch Set 12: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":12},{"id":"6b649071aeb8d32f339f96603806a11446b10ae1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 18:59:45.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"27910d66f934981f767cc5d47eee3c8273366b20","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 22:11:45.000000000","message":"Patch Set 12: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/29/658329/12/gate/puppet-openstack-lint-ubuntu-bionic/6ce8135/ : SUCCESS in 5m 14s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/29/658329/12/gate/puppet-openstack-syntax-5-ubuntu-bionic/7f4f4ee/ : SUCCESS in 5m 14s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/29/658329/12/gate/puppet-openstack-syntax-6-ubuntu-bionic/5a05d5b/ : SUCCESS in 4m 48s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/29/658329/12/gate/puppet-openstack-unit-5.5-centos-7/bdcdd2b/ : SUCCESS in 32m 34s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/29/658329/12/gate/puppet-openstack-unit-6.0-ubuntu-bionic/bfb3136/ : SUCCESS in 15m 46s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/29/658329/12/gate/tripleo-ci-centos-7-undercloud-containers/5f05a12/ : SUCCESS in 1h 38m 39s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/29/658329/12/gate/tripleo-ci-centos-7-containers-multinode/12c6d36/ : SUCCESS in 2h 11m 41s\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/29/658329/12/gate/tripleo-ci-centos-7-standalone/f535fdf/ : SUCCESS in 2h 10m 59s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/29/658329/12/gate/tripleo-ci-fedora-28-standalone/aca6181/ : SUCCESS in 2h 16m 21s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/29/658329/12/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/a5b5035/ : SUCCESS in 2h 34m 47s\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/29/658329/12/gate/tripleo-ci-centos-7-standalone-upgrade-stein/d3f1249/ : SUCCESS in 2h 05m 21s\n- build-openstack-releasenotes http://logs.openstack.org/29/658329/12/gate/build-openstack-releasenotes/80bedb0/html/ : SUCCESS in 4m 03s\n- tripleo-ci-centos-7-scenario001-standalone http://logs.openstack.org/29/658329/12/gate/tripleo-ci-centos-7-scenario001-standalone/945d052/ : SUCCESS in 1h 47m 51s","accounts_in_message":[],"_revision_number":12},{"id":"53499448e2eac40647bbb0dc6df5b40780bae583","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 22:11:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"ff13fa1304949292985a6719b0fe16bf1c701929","revisions":{"923d8e041d27ea357e92390690e0c44c69285b6b":{"kind":"REWORK","_number":1,"created":"2019-05-10 10:57:04.000000000","uploader":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"ref":"refs/changes/29/658329/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/29/658329/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/29/658329/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/29/658329/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/29/658329/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/29/658329/1"}}},"commit":{"parents":[{"commit":"5460291800995ffc4b0439a879a7dd41a9da1374","subject":"Update version in metadata.json","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/5460291800995ffc4b0439a879a7dd41a9da1374"}]}],"author":{"name":"Martin Magr","email":"mmagr@redhat.com","date":"2019-05-10 10:47:28.000000000","tz":120},"committer":{"name":"Martin Magr","email":"mmagr@redhat.com","date":"2019-05-10 10:47:28.000000000","tz":120},"subject":"Base profile for rsyslogd","message":"Base profile for rsyslogd\n\nAdds profile for rsyslogd composable service which aims to replace fluentd\nwith the same behaviour. This means that rsyslog will be tailing OpenStack\nlog files and forwarding it to central log collector (ELK)\n\nChange-Id: I5e5aae692482ee1de6de298eb18c49563a2d7d29\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/923d8e041d27ea357e92390690e0c44c69285b6b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/923d8e041d27ea357e92390690e0c44c69285b6b"}]},"branch":"refs/heads/master"},"fecb53f723e16a8e39c1a0a129f8b96e6647358b":{"kind":"REWORK","_number":2,"created":"2019-05-13 14:28:34.000000000","uploader":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"ref":"refs/changes/29/658329/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/29/658329/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/29/658329/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/29/658329/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/29/658329/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/29/658329/2"}}},"commit":{"parents":[{"commit":"5460291800995ffc4b0439a879a7dd41a9da1374","subject":"Update version in metadata.json","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/5460291800995ffc4b0439a879a7dd41a9da1374"}]}],"author":{"name":"Martin Magr","email":"mmagr@redhat.com","date":"2019-05-10 10:47:28.000000000","tz":120},"committer":{"name":"Martin Magr","email":"mmagr@redhat.com","date":"2019-05-13 14:28:11.000000000","tz":120},"subject":"Base profile for rsyslogd","message":"Base profile for rsyslogd\n\nAdds profile for rsyslogd composable service which aims to replace fluentd\nwith the same behaviour. This means that rsyslog will be tailing OpenStack\nlog files and forwarding it to central log collector (ELK)\n\nChange-Id: I5e5aae692482ee1de6de298eb18c49563a2d7d29\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/fecb53f723e16a8e39c1a0a129f8b96e6647358b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/fecb53f723e16a8e39c1a0a129f8b96e6647358b"}]},"branch":"refs/heads/master"},"9f3bd2b5f8993806e0f98293dcded12516e79b27":{"kind":"REWORK","_number":3,"created":"2019-05-13 18:43:40.000000000","uploader":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"ref":"refs/changes/29/658329/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/29/658329/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/29/658329/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/29/658329/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/29/658329/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/29/658329/3"}}},"commit":{"parents":[{"commit":"5460291800995ffc4b0439a879a7dd41a9da1374","subject":"Update version in metadata.json","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/5460291800995ffc4b0439a879a7dd41a9da1374"}]}],"author":{"name":"Martin Magr","email":"mmagr@redhat.com","date":"2019-05-10 10:47:28.000000000","tz":120},"committer":{"name":"Martin Magr","email":"mmagr@redhat.com","date":"2019-05-13 18:43:26.000000000","tz":120},"subject":"Base profile for rsyslogd","message":"Base profile for rsyslogd\n\nAdds profile for rsyslogd composable service which aims to replace fluentd\nwith the same behaviour. This means that rsyslog will be tailing OpenStack\nlog files and forwarding it to central log collector (ELK)\n\nChange-Id: I5e5aae692482ee1de6de298eb18c49563a2d7d29\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/9f3bd2b5f8993806e0f98293dcded12516e79b27"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/9f3bd2b5f8993806e0f98293dcded12516e79b27"}]},"branch":"refs/heads/master"},"3a06ce4b0a80d540912f4b74528175c707e8f1ea":{"kind":"REWORK","_number":4,"created":"2019-05-14 07:34:10.000000000","uploader":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"ref":"refs/changes/29/658329/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/29/658329/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/29/658329/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/29/658329/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/29/658329/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/29/658329/4"}}},"commit":{"parents":[{"commit":"5460291800995ffc4b0439a879a7dd41a9da1374","subject":"Update version in metadata.json","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/5460291800995ffc4b0439a879a7dd41a9da1374"}]}],"author":{"name":"Martin Magr","email":"mmagr@redhat.com","date":"2019-05-10 10:47:28.000000000","tz":120},"committer":{"name":"Martin Magr","email":"mmagr@redhat.com","date":"2019-05-14 07:33:53.000000000","tz":120},"subject":"Base profile for rsyslogd","message":"Base profile for rsyslogd\n\nAdds profile for rsyslogd composable service which aims to replace fluentd\nwith the same behaviour. This means that rsyslog will be tailing OpenStack\nlog files and forwarding it to central log collector (ELK)\n\nChange-Id: I5e5aae692482ee1de6de298eb18c49563a2d7d29\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/3a06ce4b0a80d540912f4b74528175c707e8f1ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/3a06ce4b0a80d540912f4b74528175c707e8f1ea"}]},"branch":"refs/heads/master"},"5befcb1379c9d1e68567b90689ac2cd7a6bec46e":{"kind":"REWORK","_number":5,"created":"2019-06-06 12:48:56.000000000","uploader":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"ref":"refs/changes/29/658329/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/29/658329/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/29/658329/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/29/658329/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/29/658329/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/29/658329/5"}}},"commit":{"parents":[{"commit":"5460291800995ffc4b0439a879a7dd41a9da1374","subject":"Update version in metadata.json","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/5460291800995ffc4b0439a879a7dd41a9da1374"}]}],"author":{"name":"Martin Magr","email":"mmagr@redhat.com","date":"2019-05-10 10:47:28.000000000","tz":120},"committer":{"name":"Martin Magr","email":"mmagr@redhat.com","date":"2019-06-06 12:48:41.000000000","tz":120},"subject":"Base profile for rsyslogd","message":"Base profile for rsyslogd\n\nAdds profile for rsyslogd composable service which aims to replace fluentd\nwith the same behaviour. This means that rsyslog will be tailing OpenStack\nlog files and forwarding it to central log collector (ELK)\n\nChange-Id: I5e5aae692482ee1de6de298eb18c49563a2d7d29\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/5befcb1379c9d1e68567b90689ac2cd7a6bec46e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/5befcb1379c9d1e68567b90689ac2cd7a6bec46e"}]},"branch":"refs/heads/master"},"1eaedfba4428fa23ed5590bd00fcfbed6a0ef4d2":{"kind":"REWORK","_number":6,"created":"2019-06-06 12:50:16.000000000","uploader":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"ref":"refs/changes/29/658329/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/29/658329/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/29/658329/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/29/658329/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/29/658329/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/29/658329/6"}}},"commit":{"parents":[{"commit":"5460291800995ffc4b0439a879a7dd41a9da1374","subject":"Update version in metadata.json","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/5460291800995ffc4b0439a879a7dd41a9da1374"}]}],"author":{"name":"Martin Magr","email":"mmagr@redhat.com","date":"2019-05-10 10:47:28.000000000","tz":120},"committer":{"name":"Martin Magr","email":"mmagr@redhat.com","date":"2019-06-06 12:50:05.000000000","tz":120},"subject":"Base profile for rsyslogd","message":"Base profile for rsyslogd\n\nAdds profile for rsyslogd composable service which aims to replace fluentd\nwith the same behaviour. This means that rsyslog will be tailing OpenStack\nlog files and forwarding it to central log collector (ELK)\n\nChange-Id: I5e5aae692482ee1de6de298eb18c49563a2d7d29\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/1eaedfba4428fa23ed5590bd00fcfbed6a0ef4d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/1eaedfba4428fa23ed5590bd00fcfbed6a0ef4d2"}]},"branch":"refs/heads/master"},"6b28403d6c3746611c755256dcf3446986d83438":{"kind":"REWORK","_number":7,"created":"2019-06-06 21:13:01.000000000","uploader":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"ref":"refs/changes/29/658329/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/29/658329/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/29/658329/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/29/658329/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/29/658329/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/29/658329/7"}}},"commit":{"parents":[{"commit":"5460291800995ffc4b0439a879a7dd41a9da1374","subject":"Update version in metadata.json","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/5460291800995ffc4b0439a879a7dd41a9da1374"}]}],"author":{"name":"Martin Magr","email":"mmagr@redhat.com","date":"2019-05-10 10:47:28.000000000","tz":120},"committer":{"name":"Martin Magr","email":"mmagr@redhat.com","date":"2019-06-06 21:12:49.000000000","tz":120},"subject":"Base profile for rsyslogd","message":"Base profile for rsyslogd\n\nAdds profile for rsyslogd composable service which aims to replace fluentd\nwith the same behaviour. This means that rsyslog will be tailing OpenStack\nlog files and forwarding it to central log collector (ELK)\n\nChange-Id: I5e5aae692482ee1de6de298eb18c49563a2d7d29\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/6b28403d6c3746611c755256dcf3446986d83438"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/6b28403d6c3746611c755256dcf3446986d83438"}]},"branch":"refs/heads/master"},"28e1896d643ef924db34c8f4230aeba961fb8aba":{"kind":"REWORK","_number":8,"created":"2019-06-07 08:13:52.000000000","uploader":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"ref":"refs/changes/29/658329/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/29/658329/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/29/658329/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/29/658329/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/29/658329/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/29/658329/8"}}},"commit":{"parents":[{"commit":"5460291800995ffc4b0439a879a7dd41a9da1374","subject":"Update version in metadata.json","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/5460291800995ffc4b0439a879a7dd41a9da1374"}]}],"author":{"name":"Martin Magr","email":"mmagr@redhat.com","date":"2019-05-10 10:47:28.000000000","tz":120},"committer":{"name":"Martin Magr","email":"mmagr@redhat.com","date":"2019-06-07 08:13:31.000000000","tz":120},"subject":"Base profile for rsyslogd","message":"Base profile for rsyslogd\n\nAdds profile for rsyslogd composable service which aims to replace fluentd\nwith the same behaviour. This means that rsyslog will be tailing OpenStack\nlog files and forwarding it to central log collector (ELK)\n\nChange-Id: I5e5aae692482ee1de6de298eb18c49563a2d7d29\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/28e1896d643ef924db34c8f4230aeba961fb8aba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/28e1896d643ef924db34c8f4230aeba961fb8aba"}]},"branch":"refs/heads/master"},"832b51a8e2bd4686d015e2b01ea86c5363f57e61":{"kind":"REWORK","_number":9,"created":"2019-06-07 09:15:55.000000000","uploader":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"ref":"refs/changes/29/658329/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/29/658329/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/29/658329/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/29/658329/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/29/658329/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/29/658329/9"}}},"commit":{"parents":[{"commit":"5460291800995ffc4b0439a879a7dd41a9da1374","subject":"Update version in metadata.json","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/5460291800995ffc4b0439a879a7dd41a9da1374"}]}],"author":{"name":"Martin Magr","email":"mmagr@redhat.com","date":"2019-05-10 10:47:28.000000000","tz":120},"committer":{"name":"Martin Magr","email":"mmagr@redhat.com","date":"2019-06-07 09:15:42.000000000","tz":120},"subject":"Base profile for rsyslogd","message":"Base profile for rsyslogd\n\nAdds profile for rsyslogd composable service which aims to replace fluentd\nwith the same behaviour. This means that rsyslog will be tailing OpenStack\nlog files and forwarding it to central log collector (ELK)\n\nChange-Id: I5e5aae692482ee1de6de298eb18c49563a2d7d29\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/832b51a8e2bd4686d015e2b01ea86c5363f57e61"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/832b51a8e2bd4686d015e2b01ea86c5363f57e61"}]},"branch":"refs/heads/master"},"0e5967c78bab91d2344d37abc2f5a47f158850b2":{"kind":"REWORK","_number":10,"created":"2019-06-07 10:07:39.000000000","uploader":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"ref":"refs/changes/29/658329/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/29/658329/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/29/658329/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/29/658329/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/29/658329/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/29/658329/10"}}},"commit":{"parents":[{"commit":"5460291800995ffc4b0439a879a7dd41a9da1374","subject":"Update version in metadata.json","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/5460291800995ffc4b0439a879a7dd41a9da1374"}]}],"author":{"name":"Martin Magr","email":"mmagr@redhat.com","date":"2019-05-10 10:47:28.000000000","tz":120},"committer":{"name":"Martin Magr","email":"mmagr@redhat.com","date":"2019-06-07 10:07:23.000000000","tz":120},"subject":"Base profile for rsyslogd","message":"Base profile for rsyslogd\n\nAdds profile for rsyslogd composable service which aims to replace fluentd\nwith the same behaviour. This means that rsyslog will be tailing OpenStack\nlog files and forwarding it to central log collector (ELK)\n\nChange-Id: I5e5aae692482ee1de6de298eb18c49563a2d7d29\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/0e5967c78bab91d2344d37abc2f5a47f158850b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/0e5967c78bab91d2344d37abc2f5a47f158850b2"}]},"branch":"refs/heads/master"},"333413319668beef173242e0f695739af4b4dc6e":{"kind":"REWORK","_number":11,"created":"2019-06-11 21:20:50.000000000","uploader":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"ref":"refs/changes/29/658329/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/29/658329/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/29/658329/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/29/658329/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/29/658329/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/29/658329/11"}}},"commit":{"parents":[{"commit":"5460291800995ffc4b0439a879a7dd41a9da1374","subject":"Update version in metadata.json","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/5460291800995ffc4b0439a879a7dd41a9da1374"}]}],"author":{"name":"Martin Magr","email":"mmagr@redhat.com","date":"2019-05-10 10:47:28.000000000","tz":120},"committer":{"name":"Martin Magr","email":"mmagr@redhat.com","date":"2019-06-11 21:20:35.000000000","tz":120},"subject":"Base profile for rsyslogd","message":"Base profile for rsyslogd\n\nAdds profile for rsyslogd composable service which aims to replace fluentd\nwith the same behaviour. This means that rsyslog will be tailing OpenStack\nlog files and forwarding it to central log collector (ELK)\n\nChange-Id: I5e5aae692482ee1de6de298eb18c49563a2d7d29\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/333413319668beef173242e0f695739af4b4dc6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/333413319668beef173242e0f695739af4b4dc6e"}]},"branch":"refs/heads/master"},"ff13fa1304949292985a6719b0fe16bf1c701929":{"kind":"REWORK","_number":12,"created":"2019-06-12 06:13:08.000000000","uploader":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"ref":"refs/changes/29/658329/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/29/658329/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/29/658329/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/29/658329/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/29/658329/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/29/658329/12"}}},"commit":{"parents":[{"commit":"5460291800995ffc4b0439a879a7dd41a9da1374","subject":"Update version in metadata.json","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/5460291800995ffc4b0439a879a7dd41a9da1374"}]}],"author":{"name":"Martin Magr","email":"mmagr@redhat.com","date":"2019-05-10 10:47:28.000000000","tz":120},"committer":{"name":"Martin Magr","email":"mmagr@redhat.com","date":"2019-06-12 06:12:52.000000000","tz":120},"subject":"Base profile for rsyslogd","message":"Base profile for rsyslogd\n\nAdds profile for rsyslogd composable service which aims to replace fluentd\nwith the same behaviour. This means that rsyslog will be tailing OpenStack\nlog files and forwarding it to central log collector (ELK)\n\nChange-Id: I5e5aae692482ee1de6de298eb18c49563a2d7d29\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/ff13fa1304949292985a6719b0fe16bf1c701929"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/ff13fa1304949292985a6719b0fe16bf1c701929"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
