)]}'
{"id":"openstack%2Fpuppet-tripleo~659466","triplet_id":"openstack%2Fpuppet-tripleo~master~I1325171ef60d7a7e3b57373082fcdb5487be939b","project":"openstack/puppet-tripleo","branch":"master","topic":"bug/1829338","hashtags":[],"change_id":"I1325171ef60d7a7e3b57373082fcdb5487be939b","subject":"Fix horizon firewall rules in composable roles","status":"MERGED","created":"2019-05-16 07:19:26.000000000","updated":"2019-05-24 15:41:12.000000000","submitted":"2019-05-24 15:17:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"659466-1558711026289-3d308a3e","meta_rev_id":"00c77e1d4d72e1861e38bcf23851cddd41a5b09a","_number":659466,"virtual_id_number":659466,"owner":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},{"value":0,"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-05-24 15:17:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":1,"date":"2019-05-23 07:47:58.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"recommended":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2019-05-23 05:13:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},{"value":1,"date":"2019-05-23 05:13:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"value":2,"date":"2019-05-23 15:00:37.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":2,"date":"2019-05-24 12:48:36.000000000","_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"value":1,"date":"2019-05-23 05:13:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"all":[{"value":0,"date":"2019-05-23 14:58:07.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},{"value":0,"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":1,"date":"2019-05-24 12:48:36.000000000","_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-20 18:32:18.000000000","updated_by":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"reviewer":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"state":"REVIEWER"},{"updated":"2019-05-21 06:14:10.000000000","updated_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"REVIEWER"},{"updated":"2019-05-22 09:39:17.000000000","updated_by":{"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},"reviewer":{"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},"state":"REVIEWER"},{"updated":"2019-05-23 07:47:58.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-23 14:58:07.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-05-23 15:00:37.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-05-24 12:48:36.000000000","updated_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"reviewer":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"state":"REVIEWER"},{"updated":"2019-05-24 15:17:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b5af0a273fb078fab435e87dbf7fc214dbddb4ee","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2019-05-16 07:19:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"86a2f6f2401eeb7d4c39598b3b13c0721d1de3e5","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2019-05-16 07:24:15.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"a921aebafd0660caf3391774cd28caf7d122adb1","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2019-05-16 09:04:21.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0586d5d687c68ba4dec1d2775a5e4a31e378925d","author":{"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},"date":"2019-05-16 09:43:40.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"3b34a22e7a9ebd204688d37c5d5dd63a87d8aa98","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 09:47:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/66/659466/1/check/puppet-openstack-lint-ubuntu-bionic/066eae6/ : FAILURE in 4m 04s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/66/659466/1/check/puppet-openstack-syntax-4-ubuntu-bionic/a762aef/ : SUCCESS in 4m 08s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/66/659466/1/check/puppet-openstack-syntax-5-ubuntu-bionic/9c1313a/ : SUCCESS in 3m 39s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/66/659466/1/check/puppet-openstack-syntax-6-ubuntu-bionic/4fae7a3/ : SUCCESS in 3m 36s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/66/659466/1/check/puppet-openstack-unit-5.5-centos-7/290f518/ : SUCCESS in 42m 50s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/66/659466/1/check/puppet-openstack-unit-6.0-ubuntu-bionic/0590422/ : SUCCESS in 15m 55s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/66/659466/1/check/puppet-openstack-unit-latest-ubuntu-bionic/a984c24/ : SUCCESS in 15m 58s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/66/659466/1/check/tripleo-ci-centos-7-undercloud-containers/90953ba/ : SUCCESS in 1h 27m 10s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/66/659466/1/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/c371e96/ : FAILURE in 1h 45m 03s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/66/659466/1/check/tripleo-ci-centos-7-containers-multinode/f13223b/ : FAILURE in 1h 52m 18s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/66/659466/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/6e7d46b/ : SUCCESS in 2h 02m 53s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/66/659466/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/1d138dd/ : SUCCESS in 2h 26m 29s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/66/659466/1/check/tripleo-ci-centos-7-standalone/463e528/ : SUCCESS in 1h 35m 50s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/66/659466/1/check/tripleo-ci-fedora-28-standalone/cbcfe19/ : SUCCESS in 1h 39m 00s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/66/659466/1/check/tripleo-ci-centos-7-standalone-upgrade/e1247f6/ : FAILURE in 1h 30m 59s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/66/659466/1/check/tripleo-ci-centos-7-standalone-upgrade-stein/18d5527/ : SUCCESS in 2h 02m 50s\n- tripleo-ci-centos-7-scenario001-standalone http://logs.openstack.org/66/659466/1/check/tripleo-ci-centos-7-scenario001-standalone/330106d/ : SUCCESS in 1h 37m 32s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/66/659466/1/check/tripleo-ci-centos-7-scenario003-standalone/82c8c25/ : SUCCESS in 1h 08m 42s\n- tripleo-ci-centos-7-scenario004-standalone http://logs.openstack.org/66/659466/1/check/tripleo-ci-centos-7-scenario004-standalone/8cc6fb1/ : SUCCESS in 1h 27m 08s\n- tripleo-ci-centos-7-scenario012-standalone http://logs.openstack.org/66/659466/1/check/tripleo-ci-centos-7-scenario012-standalone/bdf11da/ : FAILURE in 23m 50s (non-voting)\n- tripleo-ci-centos-7-standalone-rocky http://logs.openstack.org/66/659466/1/check/tripleo-ci-centos-7-standalone-rocky/7b14b9f/ : SUCCESS in 1h 39m 44s (non-voting)\n- tripleo-ci-centos-7-scenario008-multinode-oooq-container http://logs.openstack.org/66/659466/1/check/tripleo-ci-centos-7-scenario008-multinode-oooq-container/1e9ad1d/ : FAILURE in 1h 21m 41s (non-voting)\n- tripleo-ci-centos-7-scenario010-multinode-oooq-container http://logs.openstack.org/66/659466/1/check/tripleo-ci-centos-7-scenario010-multinode-oooq-container/ff0f9f7/ : FAILURE in 1h 42m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"254967beea4535f8cbe3aeae03ee35ef69c749f2","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-16 10:25:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/66/659466/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/d47d240/ : SUCCESS in 2h 44m 07s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/66/659466/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/b6f5383/ : SUCCESS in 3h 04m 37s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/66/659466/1/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/f4bf829/ : FAILURE in 1h 19m 49s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/66/659466/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/f88c558/ : FAILURE in 2h 14m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c988224f625dcb286fd2b9405a4819862480e53b","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2019-05-16 10:58:48.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"19f45cf5012ee154cca3727520d50485a11716c2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 13:43:46.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/66/659466/2/check/puppet-openstack-lint-ubuntu-bionic/c14a283/ : SUCCESS in 4m 21s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/66/659466/2/check/puppet-openstack-syntax-4-ubuntu-bionic/40f4db6/ : SUCCESS in 3m 34s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/66/659466/2/check/puppet-openstack-syntax-5-ubuntu-bionic/2f3ddf5/ : SUCCESS in 4m 39s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/66/659466/2/check/puppet-openstack-syntax-6-ubuntu-bionic/bdb5e99/ : SUCCESS in 3m 36s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/66/659466/2/check/puppet-openstack-unit-5.5-centos-7/21f965c/ : SUCCESS in 28m 04s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/66/659466/2/check/puppet-openstack-unit-6.0-ubuntu-bionic/f8c1603/ : SUCCESS in 15m 11s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/66/659466/2/check/puppet-openstack-unit-latest-ubuntu-bionic/04cd779/ : SUCCESS in 15m 08s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/66/659466/2/check/tripleo-ci-centos-7-undercloud-containers/0a36c7b/ : SUCCESS in 1h 20m 31s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/66/659466/2/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/52d59f3/ : FAILURE in 1h 40m 44s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/66/659466/2/check/tripleo-ci-centos-7-containers-multinode/8ff2edd/ : FAILURE in 1h 54m 50s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/66/659466/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/c32141c/ : SUCCESS in 2h 06m 11s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/66/659466/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/3e063f0/ : SUCCESS in 2h 41m 07s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/66/659466/2/check/tripleo-ci-centos-7-standalone/f5b059a/ : SUCCESS in 1h 51m 23s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/66/659466/2/check/tripleo-ci-fedora-28-standalone/8fdfc15/ : FAILURE in 1h 49m 18s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/66/659466/2/check/tripleo-ci-centos-7-standalone-upgrade/ad56394/ : FAILURE in 1h 20m 20s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/66/659466/2/check/tripleo-ci-centos-7-standalone-upgrade-stein/d3077ff/ : SUCCESS in 1h 51m 21s\n- tripleo-ci-centos-7-scenario001-standalone http://logs.openstack.org/66/659466/2/check/tripleo-ci-centos-7-scenario001-standalone/4330c3d/ : SUCCESS in 1h 37m 16s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/66/659466/2/check/tripleo-ci-centos-7-scenario003-standalone/f959e7e/ : SUCCESS in 1h 11m 04s\n- tripleo-ci-centos-7-scenario004-standalone http://logs.openstack.org/66/659466/2/check/tripleo-ci-centos-7-scenario004-standalone/7a883d4/ : SUCCESS in 1h 39m 40s\n- tripleo-ci-centos-7-scenario012-standalone http://logs.openstack.org/66/659466/2/check/tripleo-ci-centos-7-scenario012-standalone/b2d3969/ : FAILURE in 21m 52s (non-voting)\n- tripleo-ci-centos-7-standalone-rocky http://logs.openstack.org/66/659466/2/check/tripleo-ci-centos-7-standalone-rocky/e6c8333/ : SUCCESS in 1h 31m 29s (non-voting)\n- tripleo-ci-centos-7-scenario008-multinode-oooq-container http://logs.openstack.org/66/659466/2/check/tripleo-ci-centos-7-scenario008-multinode-oooq-container/0c55b17/ : FAILURE in 1h 17m 06s (non-voting)\n- tripleo-ci-centos-7-scenario010-multinode-oooq-container http://logs.openstack.org/66/659466/2/check/tripleo-ci-centos-7-scenario010-multinode-oooq-container/5ffe1f4/ : FAILURE in 2h 03m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"27073e080505efb977780a02f82606a157bc8491","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2019-05-16 19:35:48.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"5ece83c4eba92ad4c03f6da0113edd0c75af4479","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 22:09:12.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/66/659466/2/check/puppet-openstack-lint-ubuntu-bionic/1086a9e/ : SUCCESS in 4m 16s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/66/659466/2/check/puppet-openstack-syntax-4-ubuntu-bionic/33780c0/ : SUCCESS in 4m 06s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/66/659466/2/check/puppet-openstack-syntax-5-ubuntu-bionic/2aee732/ : SUCCESS in 4m 23s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/66/659466/2/check/puppet-openstack-syntax-6-ubuntu-bionic/6d844df/ : SUCCESS in 3m 23s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/66/659466/2/check/puppet-openstack-unit-5.5-centos-7/b69e448/ : SUCCESS in 30m 58s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/66/659466/2/check/puppet-openstack-unit-6.0-ubuntu-bionic/d604c3d/ : SUCCESS in 12m 27s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/66/659466/2/check/puppet-openstack-unit-latest-ubuntu-bionic/6e51da4/ : SUCCESS in 15m 51s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/66/659466/2/check/tripleo-ci-centos-7-undercloud-containers/6f16c46/ : SUCCESS in 2h 04m 16s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/66/659466/2/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/20ec3ef/ : FAILURE in 1h 51m 43s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/66/659466/2/check/tripleo-ci-centos-7-containers-multinode/15b4863/ : FAILURE in 1h 41m 45s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/66/659466/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/ec4bcd5/ : SUCCESS in 2h 09m 54s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/66/659466/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/467ddcd/ : SUCCESS in 2h 25m 59s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/66/659466/2/check/tripleo-ci-centos-7-standalone/75f2004/ : SUCCESS in 1h 50m 31s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/66/659466/2/check/tripleo-ci-fedora-28-standalone/9149f72/ : SUCCESS in 1h 44m 23s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/66/659466/2/check/tripleo-ci-centos-7-standalone-upgrade/cb7f577/ : FAILURE in 1h 36m 57s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/66/659466/2/check/tripleo-ci-centos-7-standalone-upgrade-stein/eadfa84/ : SUCCESS in 2h 00m 37s\n- tripleo-ci-centos-7-scenario001-standalone http://logs.openstack.org/66/659466/2/check/tripleo-ci-centos-7-scenario001-standalone/e31e350/ : SUCCESS in 1h 56m 50s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/66/659466/2/check/tripleo-ci-centos-7-scenario003-standalone/4182644/ : SUCCESS in 1h 09m 49s\n- tripleo-ci-centos-7-scenario004-standalone http://logs.openstack.org/66/659466/2/check/tripleo-ci-centos-7-scenario004-standalone/2601a0c/ : SUCCESS in 1h 38m 32s\n- tripleo-ci-centos-7-scenario012-standalone http://logs.openstack.org/66/659466/2/check/tripleo-ci-centos-7-scenario012-standalone/5f9f181/ : FAILURE in 23m 43s (non-voting)\n- tripleo-ci-centos-7-standalone-rocky http://logs.openstack.org/66/659466/2/check/tripleo-ci-centos-7-standalone-rocky/d57ad41/ : SUCCESS in 1h 28m 43s (non-voting)\n- tripleo-ci-centos-7-scenario008-multinode-oooq-container http://logs.openstack.org/66/659466/2/check/tripleo-ci-centos-7-scenario008-multinode-oooq-container/0902ae3/ : FAILURE in 1h 17m 00s (non-voting)\n- tripleo-ci-centos-7-scenario010-multinode-oooq-container http://logs.openstack.org/66/659466/2/check/tripleo-ci-centos-7-scenario010-multinode-oooq-container/8148c48/ : FAILURE in 1h 42m 39s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"eebe82c86646323b83e40ec995f1c3640e5ff6c1","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-16 22:32:26.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/66/659466/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/073fdaa/ : SUCCESS in 2h 42m 28s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/66/659466/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/7b68a29/ : SUCCESS in 2h 55m 31s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/66/659466/2/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/19ed279/ : FAILURE in 1h 13m 31s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/66/659466/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/5a00d29/ : FAILURE in 2h 50m 57s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bc07ecae68398f02f832ca45b9fba45d780ce37a","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2019-05-20 07:06:01.000000000","message":"Patch Set 2: Code-Review-1\n\nI need to test this on osp15 as well. for some reason it is trying to apply iptables rules in the container with this change","accounts_in_message":[],"_revision_number":2},{"id":"ed2138ad26a0319c866e1939a216e11601cfec30","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2019-05-20 14:44:37.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7b6b3c4c8ff2b1497012c94a36e812e17d1b7141","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 18:21:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/66/659466/3/check/puppet-openstack-lint-ubuntu-bionic/5034280/ : SUCCESS in 4m 41s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/66/659466/3/check/puppet-openstack-syntax-4-ubuntu-bionic/f9cef84/ : SUCCESS in 4m 22s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/66/659466/3/check/puppet-openstack-syntax-5-ubuntu-bionic/e5512c1/ : SUCCESS in 4m 56s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/66/659466/3/check/puppet-openstack-syntax-6-ubuntu-bionic/1bbb5c8/ : SUCCESS in 4m 17s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/66/659466/3/check/puppet-openstack-unit-5.5-centos-7/1cfbec6/ : SUCCESS in 31m 46s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/66/659466/3/check/puppet-openstack-unit-6.0-ubuntu-bionic/3e51d18/ : SUCCESS in 14m 36s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/66/659466/3/check/puppet-openstack-unit-latest-ubuntu-bionic/90632d1/ : SUCCESS in 13m 38s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/66/659466/3/check/tripleo-ci-centos-7-undercloud-containers/fa071d8/ : SUCCESS in 1h 52m 08s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/66/659466/3/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/08401af/ : FAILURE in 1h 11m 13s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/66/659466/3/check/tripleo-ci-centos-7-containers-multinode/0b45249/ : SUCCESS in 2h 14m 47s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/66/659466/3/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/788ecb4/ : SUCCESS in 2h 12m 59s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/66/659466/3/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/1da4fec/ : SUCCESS in 2h 25m 58s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/66/659466/3/check/tripleo-ci-centos-7-standalone/565eb5a/ : SUCCESS in 2h 08m 05s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/66/659466/3/check/tripleo-ci-fedora-28-standalone/6cdde7a/ : SUCCESS in 1h 51m 12s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/66/659466/3/check/tripleo-ci-centos-7-standalone-upgrade/c558a0e/ : FAILURE in 1h 31m 10s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/66/659466/3/check/tripleo-ci-centos-7-standalone-upgrade-stein/3188c42/ : SUCCESS in 2h 10m 52s\n- tripleo-ci-centos-7-scenario001-standalone http://logs.openstack.org/66/659466/3/check/tripleo-ci-centos-7-scenario001-standalone/f1580d0/ : SUCCESS in 1h 43m 52s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/66/659466/3/check/tripleo-ci-centos-7-scenario003-standalone/9a63780/ : SUCCESS in 1h 38m 35s\n- tripleo-ci-centos-7-scenario004-standalone http://logs.openstack.org/66/659466/3/check/tripleo-ci-centos-7-scenario004-standalone/88bfe43/ : SUCCESS in 1h 33m 52s\n- tripleo-ci-centos-7-scenario012-standalone http://logs.openstack.org/66/659466/3/check/tripleo-ci-centos-7-scenario012-standalone/2e0f9e4/ : FAILURE in 23m 22s (non-voting)\n- tripleo-ci-centos-7-scenario008-multinode-oooq-container http://logs.openstack.org/66/659466/3/check/tripleo-ci-centos-7-scenario008-multinode-oooq-container/c89408c/ : FAILURE in 1h 18m 35s (non-voting)\n- tripleo-ci-centos-7-scenario010-multinode-oooq-container http://logs.openstack.org/66/659466/3/check/tripleo-ci-centos-7-scenario010-multinode-oooq-container/feaffe8/ : FAILURE in 1h 48m 08s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b18f0791ed73dfbca266624615bab2913ff9d9ea","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-20 18:30:40.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/66/659466/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/4aee708/ : FAILURE in 2h 46m 35s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/66/659466/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/6cb25f9/ : SUCCESS in 2h 46m 22s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/66/659466/3/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/d060fc6/ : SUCCESS in 1h 20m 35s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/66/659466/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/7f67cce/ : TIMED_OUT in 3h 43m 53s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2189102d97a7db5695887b2594abf0434cfa6e10","author":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"date":"2019-05-20 18:32:18.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"666cbdd3fa10efc1ea5cfaa2ca4c07e09ab9c63b","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2019-05-21 06:14:10.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"ef5d7059c5a72c47ce7878d54a4e14f0c3498581","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2019-05-22 08:35:44.000000000","message":"Patch Set 3:\n\ncheck-rdo","accounts_in_message":[],"_revision_number":3},{"id":"11ae4575d138ef3abcb37db2fcc465828719b714","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2019-05-22 08:36:14.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"6ef7927aa39d7c54da769e5f36c17193779f22e2","author":{"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},"date":"2019-05-22 09:39:17.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"053f284842cdaba81de14a9ae2cce98863f2863d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-22 11:25:24.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/66/659466/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/4992341/ : FAILURE in 2h 36m 40s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/66/659466/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/cc32550/ : FAILURE in 2h 41m 29s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/66/659466/4/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/e3d67ff/ : SUCCESS in 1h 11m 25s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/66/659466/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/6118de5/ : FAILURE in 2h 47m 45s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"16e36f77b0198bf7f1515e571294669f7436a340","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 11:45:17.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/66/659466/4/check/puppet-openstack-lint-ubuntu-bionic/176c446/ : SUCCESS in 5m 58s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/66/659466/4/check/puppet-openstack-syntax-4-ubuntu-bionic/a6962dd/ : SUCCESS in 4m 43s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/66/659466/4/check/puppet-openstack-syntax-5-ubuntu-bionic/6bf897c/ : SUCCESS in 5m 20s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/66/659466/4/check/puppet-openstack-syntax-6-ubuntu-bionic/aea25cb/ : SUCCESS in 4m 33s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/66/659466/4/check/puppet-openstack-unit-5.5-centos-7/0fce521/ : SUCCESS in 27m 41s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/66/659466/4/check/puppet-openstack-unit-6.0-ubuntu-bionic/4b0caf9/ : SUCCESS in 15m 34s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/66/659466/4/check/puppet-openstack-unit-latest-ubuntu-bionic/c08b1b4/ : SUCCESS in 13m 55s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/66/659466/4/check/tripleo-ci-centos-7-undercloud-containers/4b0db8f/ : SUCCESS in 1h 37m 18s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/66/659466/4/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/d7752de/ : FAILURE in 1h 35m 09s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/66/659466/4/check/tripleo-ci-centos-7-containers-multinode/fade486/ : SUCCESS in 2h 12m 14s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/66/659466/4/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/2c8cf3e/ : SUCCESS in 2h 13m 10s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/66/659466/4/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/70ad37d/ : FAILURE in 2h 54m 58s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/66/659466/4/check/tripleo-ci-centos-7-standalone/9d0079c/ : FAILURE in 2h 02m 06s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/66/659466/4/check/tripleo-ci-fedora-28-standalone/4a420c2/ : SUCCESS in 1h 54m 18s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/66/659466/4/check/tripleo-ci-centos-7-standalone-upgrade/201ce0b/ : SUCCESS in 2h 04m 37s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/66/659466/4/check/tripleo-ci-centos-7-standalone-upgrade-stein/9950c05/ : SUCCESS in 1h 57m 27s\n- tripleo-ci-centos-7-scenario001-standalone http://logs.openstack.org/66/659466/4/check/tripleo-ci-centos-7-scenario001-standalone/5bd8ac1/ : SUCCESS in 1h 50m 29s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/66/659466/4/check/tripleo-ci-centos-7-scenario003-standalone/73a3fbc/ : SUCCESS in 1h 11m 06s\n- tripleo-ci-centos-7-scenario004-standalone http://logs.openstack.org/66/659466/4/check/tripleo-ci-centos-7-scenario004-standalone/e5a4277/ : SUCCESS in 1h 42m 41s\n- tripleo-ci-centos-7-scenario012-standalone http://logs.openstack.org/66/659466/4/check/tripleo-ci-centos-7-scenario012-standalone/117ba62/ : FAILURE in 25m 42s (non-voting)\n- tripleo-ci-centos-7-scenario008-multinode-oooq-container http://logs.openstack.org/66/659466/4/check/tripleo-ci-centos-7-scenario008-multinode-oooq-container/5decf6d/ : FAILURE in 1h 27m 00s (non-voting)\n- tripleo-ci-centos-7-scenario010-multinode-oooq-container http://logs.openstack.org/66/659466/4/check/tripleo-ci-centos-7-scenario010-multinode-oooq-container/fc169fe/ : FAILURE in 1h 48m 43s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"b353a8b3f74a2fc095a981b083e38e181072fae3","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2019-05-22 13:57:59.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"6b47755235777e9df846013df0492ecad99e5c84","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-22 16:52:28.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/66/659466/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/cda871f/ : FAILURE in 1h 23m 45s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/66/659466/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/ff38e4b/ : FAILURE in 2h 52m 41s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/66/659466/4/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/341876a/ : SUCCESS in 1h 16m 04s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/66/659466/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/25227f1/ : FAILURE in 1h 28m 39s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"8108ada52ba726bdc33fda9b7b1bf833c65eacc2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 19:25:47.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/66/659466/4/check/puppet-openstack-lint-ubuntu-bionic/5187637/ : SUCCESS in 4m 04s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/66/659466/4/check/puppet-openstack-syntax-4-ubuntu-bionic/20c5990/ : SUCCESS in 4m 45s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/66/659466/4/check/puppet-openstack-syntax-5-ubuntu-bionic/6a5504f/ : SUCCESS in 3m 38s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/66/659466/4/check/puppet-openstack-syntax-6-ubuntu-bionic/0a55ba0/ : SUCCESS in 3m 49s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/66/659466/4/check/puppet-openstack-unit-5.5-centos-7/310a105/ : SUCCESS in 29m 36s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/66/659466/4/check/puppet-openstack-unit-6.0-ubuntu-bionic/36af484/ : SUCCESS in 13m 42s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/66/659466/4/check/puppet-openstack-unit-latest-ubuntu-bionic/4fec94b/ : SUCCESS in 16m 09s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/66/659466/4/check/tripleo-ci-centos-7-undercloud-containers/1e0f346/ : SUCCESS in 1h 40m 46s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/66/659466/4/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/d1ba974/ : FAILURE in 1h 31m 14s (non-voting)\n- tripleo-ci-centos-7-containers-multinode finger://ze04.openstack.org/ea50ab3f17454719ae322c768460e414 : RETRY_LIMIT in 36m 17s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates finger://ze12.openstack.org/45b2b92de7f149d2baa15ff91785bc0b : RETRY_LIMIT in 37m 28s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades finger://ze08.openstack.org/ea1d3652e17b4dd8a92ec47858792fe7 : RETRY_LIMIT in 36m 09s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/66/659466/4/check/tripleo-ci-centos-7-standalone/d750f02/ : SUCCESS in 1h 48m 33s\n- tripleo-ci-fedora-28-standalone finger://ze07.openstack.org/6e50410be6fe4081b55976ceabf2d32a : RETRY_LIMIT in 34m 59s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/66/659466/4/check/tripleo-ci-centos-7-standalone-upgrade/ddc05cb/ : FAILURE in 1h 45m 07s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein finger://ze11.openstack.org/ea35bdb74af04111a9940758db41e4bb : RETRY_LIMIT in 36m 38s\n- tripleo-ci-centos-7-scenario001-standalone finger://ze08.openstack.org/e82d06fcea0a473f8ac1d34fc6e6bedd : RETRY_LIMIT in 35m 47s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/66/659466/4/check/tripleo-ci-centos-7-scenario003-standalone/5dc9da3/ : SUCCESS in 1h 18m 39s\n- tripleo-ci-centos-7-scenario004-standalone finger://ze07.openstack.org/50f2bab6f148433d9ab46af1992b33c9 : RETRY_LIMIT in 35m 43s\n- tripleo-ci-centos-7-scenario012-standalone http://logs.openstack.org/66/659466/4/check/tripleo-ci-centos-7-scenario012-standalone/73d79bd/ : FAILURE in 24m 23s (non-voting)\n- tripleo-ci-centos-7-scenario008-multinode-oooq-container finger://ze01.openstack.org/17ef86c0043d4b8d89582ee58cd61cf3 : RETRY_LIMIT in 36m 04s (non-voting)\n- tripleo-ci-centos-7-scenario010-multinode-oooq-container finger://ze04.openstack.org/1a2c71d05c7f47d58f03938f8749ad4a : RETRY_LIMIT in 36m 03s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"79ed63ddf24484b995a3f08e6e7fb3cdb3120dac","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2019-05-23 05:13:56.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"609fa366682aca9b3eea72af79411f34579ab147","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-23 07:47:58.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/66/659466/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/c53bca5/ : SUCCESS in 2h 23m 11s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/66/659466/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/9670e99/ : SUCCESS in 2h 30m 20s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/66/659466/5/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/1d94a32/ : SUCCESS in 1h 15m 12s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/66/659466/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/1d327ea/ : FAILURE in 2h 32m 22s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"2ea1bb195dfd27da14290bbb02d05df51640cc80","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 07:57:18.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/66/659466/5/check/puppet-openstack-lint-ubuntu-bionic/ac068f5/ : SUCCESS in 4m 14s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/66/659466/5/check/puppet-openstack-syntax-4-ubuntu-bionic/5bd7e7f/ : SUCCESS in 4m 58s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/66/659466/5/check/puppet-openstack-syntax-5-ubuntu-bionic/028185b/ : SUCCESS in 4m 48s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/66/659466/5/check/puppet-openstack-syntax-6-ubuntu-bionic/f22650c/ : SUCCESS in 5m 28s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/66/659466/5/check/puppet-openstack-unit-5.5-centos-7/6ad6e03/ : SUCCESS in 26m 53s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/66/659466/5/check/puppet-openstack-unit-6.0-ubuntu-bionic/ad45214/ : SUCCESS in 12m 01s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/66/659466/5/check/puppet-openstack-unit-latest-ubuntu-bionic/bb401dd/ : SUCCESS in 15m 58s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/66/659466/5/check/tripleo-ci-centos-7-undercloud-containers/41d96d7/ : SUCCESS in 1h 40m 00s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/66/659466/5/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/5df7cd6/ : FAILURE in 1h 44m 32s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/66/659466/5/check/tripleo-ci-centos-7-containers-multinode/07080b2/ : SUCCESS in 1h 58m 27s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/66/659466/5/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/495e519/ : SUCCESS in 2h 08m 29s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/66/659466/5/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/e9cfc97/ : SUCCESS in 2h 26m 51s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/66/659466/5/check/tripleo-ci-centos-7-standalone/88461c2/ : SUCCESS in 1h 44m 22s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/66/659466/5/check/tripleo-ci-fedora-28-standalone/17a610f/ : SUCCESS in 1h 40m 25s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/66/659466/5/check/tripleo-ci-centos-7-standalone-upgrade/ccabad9/ : SUCCESS in 1h 59m 26s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/66/659466/5/check/tripleo-ci-centos-7-standalone-upgrade-stein/437a834/ : SUCCESS in 2h 02m 01s\n- tripleo-ci-centos-7-scenario001-standalone http://logs.openstack.org/66/659466/5/check/tripleo-ci-centos-7-scenario001-standalone/e7c08dd/ : SUCCESS in 1h 29m 05s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/66/659466/5/check/tripleo-ci-centos-7-scenario003-standalone/70af206/ : SUCCESS in 1h 13m 56s\n- tripleo-ci-centos-7-scenario004-standalone http://logs.openstack.org/66/659466/5/check/tripleo-ci-centos-7-scenario004-standalone/7ecb571/ : SUCCESS in 1h 46m 56s\n- tripleo-ci-centos-7-scenario012-standalone http://logs.openstack.org/66/659466/5/check/tripleo-ci-centos-7-scenario012-standalone/29c1f64/ : FAILURE in 28m 30s (non-voting)\n- tripleo-ci-centos-7-scenario008-multinode-oooq-container http://logs.openstack.org/66/659466/5/check/tripleo-ci-centos-7-scenario008-multinode-oooq-container/624dc8f/ : FAILURE in 1h 21m 26s (non-voting)\n- tripleo-ci-centos-7-scenario010-multinode-oooq-container http://logs.openstack.org/66/659466/5/check/tripleo-ci-centos-7-scenario010-multinode-oooq-container/5ae1038/ : FAILURE in 1h 59m 48s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"4605e3d5ba070018306613e313fada535cb79268","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-05-23 15:00:37.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"b4c4d5c5932449182e91e6b10b04df579d6273ef","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2019-05-24 12:48:36.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"f046036d08ab260c3ef8562ec26f04f66acf2f6e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 12:48:49.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"83adc85b8a6b394d1cb244880841a24f939bfdf7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 15:17:05.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/66/659466/5/gate/puppet-openstack-lint-ubuntu-bionic/76250cd/ : SUCCESS in 4m 08s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/66/659466/5/gate/puppet-openstack-syntax-4-ubuntu-bionic/d728874/ : SUCCESS in 3m 58s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/66/659466/5/gate/puppet-openstack-syntax-5-ubuntu-bionic/691a4c0/ : SUCCESS in 3m 50s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/66/659466/5/gate/puppet-openstack-syntax-6-ubuntu-bionic/3117f23/ : SUCCESS in 3m 57s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/66/659466/5/gate/puppet-openstack-unit-5.5-centos-7/6b6255f/ : SUCCESS in 32m 50s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/66/659466/5/gate/puppet-openstack-unit-6.0-ubuntu-bionic/61b31f0/ : SUCCESS in 18m 39s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/66/659466/5/gate/tripleo-ci-centos-7-undercloud-containers/b52690b/ : SUCCESS in 1h 26m 15s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/66/659466/5/gate/tripleo-ci-centos-7-containers-multinode/2e714a4/ : SUCCESS in 1h 57m 47s\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/66/659466/5/gate/tripleo-ci-centos-7-standalone/bb1a5cc/ : SUCCESS in 1h 52m 48s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/66/659466/5/gate/tripleo-ci-fedora-28-standalone/22f8824/ : SUCCESS in 1h 39m 00s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/66/659466/5/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/d26a815/ : SUCCESS in 2h 21m 26s\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/66/659466/5/gate/tripleo-ci-centos-7-standalone-upgrade-stein/be4f1e2/ : SUCCESS in 1h 57m 30s\n- tripleo-ci-centos-7-scenario001-standalone http://logs.openstack.org/66/659466/5/gate/tripleo-ci-centos-7-scenario001-standalone/6d17ace/ : SUCCESS in 1h 32m 24s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/66/659466/5/gate/tripleo-ci-centos-7-scenario003-standalone/9ecfc1b/ : SUCCESS in 1h 08m 55s\n- tripleo-ci-centos-7-scenario004-standalone http://logs.openstack.org/66/659466/5/gate/tripleo-ci-centos-7-scenario004-standalone/b000a3c/ : SUCCESS in 1h 37m 40s","accounts_in_message":[],"_revision_number":5},{"id":"fa76ffcfbbebebacc188fc5ddb8e2112e812591c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 15:17:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"2b8972b1dfa468f037dd9a90ee13fcc032786b82","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2019-05-24 15:41:12.000000000","message":"Patch Set 5: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit f58d8af343c24a709fdfb735534e2dc6b4209338","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"6c2e164adaa0bf0bbc38a3d69e16db105a5519aa","revisions":{"d0258ea36cb5e96b8a72c04029b5fb1c98aed0eb":{"kind":"REWORK","_number":1,"created":"2019-05-16 07:19:26.000000000","uploader":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"ref":"refs/changes/66/659466/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/66/659466/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/66/659466/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/66/659466/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/66/659466/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/66/659466/1"}}},"commit":{"parents":[{"commit":"e32f2ef8aad090590f0ee580fcb075af2248351e","subject":"Merge \"Let fluentd not create /etc/rsyslog.d/fluentd.conf file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/e32f2ef8aad090590f0ee580fcb075af2248351e"}]}],"author":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2019-05-16 07:12:50.000000000","tz":120},"committer":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2019-05-16 07:18:22.000000000","tz":120},"subject":"Fix horizon firewall rules in composable roles","message":"Fix horizon firewall rules in composable roles\n\nAtm horizon haproxy firewall rules obfuscate any other rule defined via\nthe tripleo.haproxy.firewall_rules key.\n\nThings broke with https://review.opendev.org/#/c/625600/.  The reason\nthat was pushed is that in composable roles, when splitting off horizon\naway from where haproxy runs, we would not have the proper iptables rules\non the haproxy role. This was due to the fact that we had\nthe following code:\n      service_config_settings:\n        haproxy:\n          tripleo.horizon.firewall_rules:\n            \u0027127 horizon\u0027:\n              dport:\n                - 80\n                - 443\n\nThe above code never worked as explained in\n3f8ce6fd96bc4f28a052b4c87a19b4b152734091 and so we fixed it by setting\nthe proper tripleo.haproxy.firewall_rules key. The issue is that rules\nfor haproxy should just never have been set at all via\nservice_config_settings keys in the first place. As demonstrated with\nthis bug, the merging of hiera dictionaries will mess us up and we\u0027ll\nend up overwriting other keys. Haproxy stats access has this:\noutputs:\n  role_data:\n    description: Role data for the HAproxy role.\n    value:\n      service_name: haproxy\n      monitoring_subscription: {get_param: MonitoringSubscriptionHaproxy}\n      config_settings:\n        map_merge:\n          - tripleo.haproxy.firewall_rules:\n              \u0027107 haproxy stats\u0027:\n                dport: 1993\n\nAnd since hiera will return the horizon settings for\ntripleo.haproxy.firewall_rules which won\u0027t be deep merged with the\nfirewall rules from haproxy stats and so rule \u0027107 haproxy stats\u0027 will\nnever be present.\n\nRules for haproxy need to happen in puppet-tripleo/manifests/haproxy*.\nNormally they do, the exception is horizon which uses a specialized\nhorizon_endpoint.pp manifest which does not trigger these rules.\n\nLet\u0027s create the firewall rules in haproxy/horizon_endpoint.pp like we\ndo for all other endpoints.\n\nTested and correctly got:\n[root@controller-0 ~]# iptables -nvL |grep hor\n    0     0 ACCEPT     tcp  --  *      *       0.0.0.0/0            0.0.0.0/0            multiport dports 80 state NEW /* 100 horizon_haproxy ipv4 */\n    0     0 ACCEPT     tcp  --  *      *       0.0.0.0/0            0.0.0.0/0            multiport dports 443 state NEW /* 100 horizon_haproxy_ssl ipv4 */\n    0     0 ACCEPT     tcp  --  *      *       0.0.0.0/0            0.0.0.0/0            multiport dports 80,443 state NEW /* 126 horizon ipv4 */\n\nChange-Id: I1325171ef60d7a7e3b57373082fcdb5487be939b\nRelated-Bug: #1829338\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/d0258ea36cb5e96b8a72c04029b5fb1c98aed0eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/d0258ea36cb5e96b8a72c04029b5fb1c98aed0eb"}]},"branch":"refs/heads/master"},"fb54910263bcc6cd085b6da6b7eedec79d6f273e":{"kind":"REWORK","_number":2,"created":"2019-05-16 10:58:48.000000000","uploader":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"ref":"refs/changes/66/659466/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/66/659466/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/66/659466/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/66/659466/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/66/659466/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/66/659466/2"}}},"commit":{"parents":[{"commit":"e32f2ef8aad090590f0ee580fcb075af2248351e","subject":"Merge \"Let fluentd not create /etc/rsyslog.d/fluentd.conf file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/e32f2ef8aad090590f0ee580fcb075af2248351e"}]}],"author":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2019-05-16 07:12:50.000000000","tz":120},"committer":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2019-05-16 10:58:32.000000000","tz":120},"subject":"Fix horizon firewall rules in composable roles","message":"Fix horizon firewall rules in composable roles\n\nAtm horizon haproxy firewall rules obfuscate any other rule defined via\nthe tripleo.haproxy.firewall_rules key.\n\nThings broke with https://review.opendev.org/#/c/625600/.  The reason\nthat was pushed is that in composable roles, when splitting off horizon\naway from where haproxy runs, we would not have the proper iptables rules\non the haproxy role. This was due to the fact that we had\nthe following code:\n      service_config_settings:\n        haproxy:\n          tripleo.horizon.firewall_rules:\n            \u0027127 horizon\u0027:\n              dport:\n                - 80\n                - 443\n\nThe above code never worked as explained in\n3f8ce6fd96bc4f28a052b4c87a19b4b152734091 and so we fixed it by setting\nthe proper tripleo.haproxy.firewall_rules key. The issue is that rules\nfor haproxy should just never have been set at all via\nservice_config_settings keys in the first place. As demonstrated with\nthis bug, the merging of hiera dictionaries will mess us up and we\u0027ll\nend up overwriting other keys. Haproxy stats access has this:\noutputs:\n  role_data:\n    description: Role data for the HAproxy role.\n    value:\n      service_name: haproxy\n      monitoring_subscription: {get_param: MonitoringSubscriptionHaproxy}\n      config_settings:\n        map_merge:\n          - tripleo.haproxy.firewall_rules:\n              \u0027107 haproxy stats\u0027:\n                dport: 1993\n\nAnd since hiera will return the horizon settings for\ntripleo.haproxy.firewall_rules which won\u0027t be deep merged with the\nfirewall rules from haproxy stats and so rule \u0027107 haproxy stats\u0027 will\nnever be present.\n\nRules for haproxy need to happen in puppet-tripleo/manifests/haproxy*.\nNormally they do, the exception is horizon which uses a specialized\nhorizon_endpoint.pp manifest which does not trigger these rules.\n\nLet\u0027s create the firewall rules in haproxy/horizon_endpoint.pp like we\ndo for all other endpoints.\n\nTested and correctly got:\n[root@controller-0 ~]# iptables -nvL |grep hor\n    0     0 ACCEPT     tcp  --  *      *       0.0.0.0/0            0.0.0.0/0            multiport dports 80 state NEW /* 100 horizon_haproxy ipv4 */\n    0     0 ACCEPT     tcp  --  *      *       0.0.0.0/0            0.0.0.0/0            multiport dports 443 state NEW /* 100 horizon_haproxy_ssl ipv4 */\n    0     0 ACCEPT     tcp  --  *      *       0.0.0.0/0            0.0.0.0/0            multiport dports 80,443 state NEW /* 126 horizon ipv4 */\n\nChange-Id: I1325171ef60d7a7e3b57373082fcdb5487be939b\nRelated-Bug: #1829338\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/fb54910263bcc6cd085b6da6b7eedec79d6f273e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/fb54910263bcc6cd085b6da6b7eedec79d6f273e"}]},"branch":"refs/heads/master"},"25d7fbbc4d39d0b687fbe6f51fa80adb1a1a0190":{"kind":"REWORK","_number":3,"created":"2019-05-20 14:44:37.000000000","uploader":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"ref":"refs/changes/66/659466/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/66/659466/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/66/659466/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/66/659466/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/66/659466/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/66/659466/3"}}},"commit":{"parents":[{"commit":"e32f2ef8aad090590f0ee580fcb075af2248351e","subject":"Merge \"Let fluentd not create /etc/rsyslog.d/fluentd.conf file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/e32f2ef8aad090590f0ee580fcb075af2248351e"}]}],"author":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2019-05-16 07:12:50.000000000","tz":120},"committer":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2019-05-20 14:44:21.000000000","tz":120},"subject":"Fix horizon firewall rules in composable roles","message":"Fix horizon firewall rules in composable roles\n\nAtm horizon haproxy firewall rules obfuscate any other rule defined via\nthe tripleo.haproxy.firewall_rules key.\n\nThings broke with https://review.opendev.org/#/c/625600/.  The reason\nthat was pushed is that in composable roles, when splitting off horizon\naway from where haproxy runs, we would not have the proper iptables rules\non the haproxy role. This was due to the fact that we had\nthe following code:\n      service_config_settings:\n        haproxy:\n          tripleo.horizon.firewall_rules:\n            \u0027127 horizon\u0027:\n              dport:\n                - 80\n                - 443\n\nThe above code never worked as explained in\n3f8ce6fd96bc4f28a052b4c87a19b4b152734091 and so we fixed it by setting\nthe proper tripleo.haproxy.firewall_rules key. The issue is that rules\nfor haproxy should just never have been set at all via\nservice_config_settings keys in the first place. As demonstrated with\nthis bug, the merging of hiera dictionaries will mess us up and we\u0027ll\nend up overwriting other keys. Haproxy stats access has this:\noutputs:\n  role_data:\n    description: Role data for the HAproxy role.\n    value:\n      service_name: haproxy\n      monitoring_subscription: {get_param: MonitoringSubscriptionHaproxy}\n      config_settings:\n        map_merge:\n          - tripleo.haproxy.firewall_rules:\n              \u0027107 haproxy stats\u0027:\n                dport: 1993\n\nAnd since hiera will return the horizon settings for\ntripleo.haproxy.firewall_rules which won\u0027t be deep merged with the\nfirewall rules from haproxy stats and so rule \u0027107 haproxy stats\u0027 will\nnever be present.\n\nRules for haproxy need to happen in puppet-tripleo/manifests/haproxy*.\nNormally they do, the exception is horizon which uses a specialized\nhorizon_endpoint.pp manifest which does not trigger these rules.\n\nLet\u0027s create the firewall rules in haproxy/horizon_endpoint.pp like we\ndo for all other endpoints.\n\nTested and correctly got:\n[root@controller-0 ~]# iptables -nvL |grep hor\n    0     0 ACCEPT     tcp  --  *      *       0.0.0.0/0            0.0.0.0/0            multiport dports 80 state NEW /* 100 horizon_haproxy ipv4 */\n    0     0 ACCEPT     tcp  --  *      *       0.0.0.0/0            0.0.0.0/0            multiport dports 443 state NEW /* 100 horizon_haproxy_ssl ipv4 */\n    0     0 ACCEPT     tcp  --  *      *       0.0.0.0/0            0.0.0.0/0            multiport dports 80,443 state NEW /* 126 horizon ipv4 */\n\nChange-Id: I1325171ef60d7a7e3b57373082fcdb5487be939b\nRelated-Bug: #1829338\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/25d7fbbc4d39d0b687fbe6f51fa80adb1a1a0190"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/25d7fbbc4d39d0b687fbe6f51fa80adb1a1a0190"}]},"branch":"refs/heads/master"},"3bf16934d8f9022971a7998f107303d00e5ccf0a":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-05-22 08:36:14.000000000","uploader":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"ref":"refs/changes/66/659466/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/66/659466/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/66/659466/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/66/659466/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/66/659466/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/66/659466/4"}}},"commit":{"parents":[{"commit":"68c445e7eac58aa6e69660d678dcdc2a4c091086","subject":"Merge \"Include \u0027octavia\u0027 driver on ML2/OVN deployments\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/68c445e7eac58aa6e69660d678dcdc2a4c091086"}]}],"author":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2019-05-16 07:12:50.000000000","tz":120},"committer":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2019-05-22 08:36:14.000000000","tz":0},"subject":"Fix horizon firewall rules in composable roles","message":"Fix horizon firewall rules in composable roles\n\nAtm horizon haproxy firewall rules obfuscate any other rule defined via\nthe tripleo.haproxy.firewall_rules key.\n\nThings broke with https://review.opendev.org/#/c/625600/.  The reason\nthat was pushed is that in composable roles, when splitting off horizon\naway from where haproxy runs, we would not have the proper iptables rules\non the haproxy role. This was due to the fact that we had\nthe following code:\n      service_config_settings:\n        haproxy:\n          tripleo.horizon.firewall_rules:\n            \u0027127 horizon\u0027:\n              dport:\n                - 80\n                - 443\n\nThe above code never worked as explained in\n3f8ce6fd96bc4f28a052b4c87a19b4b152734091 and so we fixed it by setting\nthe proper tripleo.haproxy.firewall_rules key. The issue is that rules\nfor haproxy should just never have been set at all via\nservice_config_settings keys in the first place. As demonstrated with\nthis bug, the merging of hiera dictionaries will mess us up and we\u0027ll\nend up overwriting other keys. Haproxy stats access has this:\noutputs:\n  role_data:\n    description: Role data for the HAproxy role.\n    value:\n      service_name: haproxy\n      monitoring_subscription: {get_param: MonitoringSubscriptionHaproxy}\n      config_settings:\n        map_merge:\n          - tripleo.haproxy.firewall_rules:\n              \u0027107 haproxy stats\u0027:\n                dport: 1993\n\nAnd since hiera will return the horizon settings for\ntripleo.haproxy.firewall_rules which won\u0027t be deep merged with the\nfirewall rules from haproxy stats and so rule \u0027107 haproxy stats\u0027 will\nnever be present.\n\nRules for haproxy need to happen in puppet-tripleo/manifests/haproxy*.\nNormally they do, the exception is horizon which uses a specialized\nhorizon_endpoint.pp manifest which does not trigger these rules.\n\nLet\u0027s create the firewall rules in haproxy/horizon_endpoint.pp like we\ndo for all other endpoints.\n\nTested and correctly got:\n[root@controller-0 ~]# iptables -nvL |grep hor\n    0     0 ACCEPT     tcp  --  *      *       0.0.0.0/0            0.0.0.0/0            multiport dports 80 state NEW /* 100 horizon_haproxy ipv4 */\n    0     0 ACCEPT     tcp  --  *      *       0.0.0.0/0            0.0.0.0/0            multiport dports 443 state NEW /* 100 horizon_haproxy_ssl ipv4 */\n    0     0 ACCEPT     tcp  --  *      *       0.0.0.0/0            0.0.0.0/0            multiport dports 80,443 state NEW /* 126 horizon ipv4 */\n\nChange-Id: I1325171ef60d7a7e3b57373082fcdb5487be939b\nRelated-Bug: #1829338\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/3bf16934d8f9022971a7998f107303d00e5ccf0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/3bf16934d8f9022971a7998f107303d00e5ccf0a"}]},"branch":"refs/heads/master"},"6c2e164adaa0bf0bbc38a3d69e16db105a5519aa":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2019-05-23 05:13:56.000000000","uploader":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"ref":"refs/changes/66/659466/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/66/659466/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/66/659466/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/66/659466/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/66/659466/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/66/659466/5"}}},"commit":{"parents":[{"commit":"8ff689bc835836799fe4b51522aa2beb9d125cad","subject":"Make sure to set umask 022 for nova_migration_target logins","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/8ff689bc835836799fe4b51522aa2beb9d125cad"}]}],"author":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2019-05-16 07:12:50.000000000","tz":120},"committer":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2019-05-23 05:13:56.000000000","tz":0},"subject":"Fix horizon firewall rules in composable roles","message":"Fix horizon firewall rules in composable roles\n\nAtm horizon haproxy firewall rules obfuscate any other rule defined via\nthe tripleo.haproxy.firewall_rules key.\n\nThings broke with https://review.opendev.org/#/c/625600/.  The reason\nthat was pushed is that in composable roles, when splitting off horizon\naway from where haproxy runs, we would not have the proper iptables rules\non the haproxy role. This was due to the fact that we had\nthe following code:\n      service_config_settings:\n        haproxy:\n          tripleo.horizon.firewall_rules:\n            \u0027127 horizon\u0027:\n              dport:\n                - 80\n                - 443\n\nThe above code never worked as explained in\n3f8ce6fd96bc4f28a052b4c87a19b4b152734091 and so we fixed it by setting\nthe proper tripleo.haproxy.firewall_rules key. The issue is that rules\nfor haproxy should just never have been set at all via\nservice_config_settings keys in the first place. As demonstrated with\nthis bug, the merging of hiera dictionaries will mess us up and we\u0027ll\nend up overwriting other keys. Haproxy stats access has this:\noutputs:\n  role_data:\n    description: Role data for the HAproxy role.\n    value:\n      service_name: haproxy\n      monitoring_subscription: {get_param: MonitoringSubscriptionHaproxy}\n      config_settings:\n        map_merge:\n          - tripleo.haproxy.firewall_rules:\n              \u0027107 haproxy stats\u0027:\n                dport: 1993\n\nAnd since hiera will return the horizon settings for\ntripleo.haproxy.firewall_rules which won\u0027t be deep merged with the\nfirewall rules from haproxy stats and so rule \u0027107 haproxy stats\u0027 will\nnever be present.\n\nRules for haproxy need to happen in puppet-tripleo/manifests/haproxy*.\nNormally they do, the exception is horizon which uses a specialized\nhorizon_endpoint.pp manifest which does not trigger these rules.\n\nLet\u0027s create the firewall rules in haproxy/horizon_endpoint.pp like we\ndo for all other endpoints.\n\nTested and correctly got:\n[root@controller-0 ~]# iptables -nvL |grep hor\n    0     0 ACCEPT     tcp  --  *      *       0.0.0.0/0            0.0.0.0/0            multiport dports 80 state NEW /* 100 horizon_haproxy ipv4 */\n    0     0 ACCEPT     tcp  --  *      *       0.0.0.0/0            0.0.0.0/0            multiport dports 443 state NEW /* 100 horizon_haproxy_ssl ipv4 */\n    0     0 ACCEPT     tcp  --  *      *       0.0.0.0/0            0.0.0.0/0            multiport dports 80,443 state NEW /* 126 horizon ipv4 */\n\nChange-Id: I1325171ef60d7a7e3b57373082fcdb5487be939b\nRelated-Bug: #1829338\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/6c2e164adaa0bf0bbc38a3d69e16db105a5519aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/6c2e164adaa0bf0bbc38a3d69e16db105a5519aa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
