)]}'
{"id":"openstack%2Fpuppet-tripleo~659532","triplet_id":"openstack%2Fpuppet-tripleo~master~I5b29a51d7edace30e6357a2af95d1af4596ddfd3","project":"openstack/puppet-tripleo","branch":"master","hashtags":[],"change_id":"I5b29a51d7edace30e6357a2af95d1af4596ddfd3","subject":"Include \u0027octavia\u0027 driver on ML2/OVN deployments","status":"MERGED","created":"2019-05-16 11:27:31.000000000","updated":"2019-05-21 20:20:35.000000000","submitted":"2019-05-21 20:03:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"659532-1558468994257-db3b04f4","meta_rev_id":"bbfce5d90866e6463e1d9583cf21ebf627346eba","_number":659532,"virtual_id_number":659532,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"value":2,"date":"2019-05-21 20:03:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2019-05-20 12:50:41.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2019-05-21 15:48:53.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2019-05-20 13:46:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2019-05-21 17:21:23.000000000","_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-05-20 13:24:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"value":0,"date":"2019-05-20 13:15:49.000000000","_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2019-05-21 17:21:26.000000000","_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"date":"2019-05-20 12:54:32.000000000","_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-20 12:50:41.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-20 12:54:32.000000000","updated_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"reviewer":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"state":"REVIEWER"},{"updated":"2019-05-20 13:15:49.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2019-05-20 13:24:20.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-05-20 13:46:48.000000000","updated_by":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"reviewer":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"state":"REVIEWER"},{"updated":"2019-05-21 15:48:53.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-05-21 17:21:26.000000000","updated_by":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"reviewer":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"state":"REVIEWER"},{"updated":"2019-05-21 20:03:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c6d900785f8ea6edf270878519f725b639ac986b","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-05-16 11:27:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"54b3b0e8b83e59928350df29d5596287c6afc4c2","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-05-16 13:44:36.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"583e4840d5b992ffb8a25f41fef1daacc6d549fc","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-05-16 13:48:26.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b0794f7391124718b95a3afd409a0c29eca3bf34","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-05-16 16:52:19.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"93104f823a5008910db074f8c9918d3d07800175","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-16 19:45:21.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/32/659532/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/2a0eea9/ : SUCCESS in 2h 47m 33s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/32/659532/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/560ad6a/ : SUCCESS in 2h 50m 59s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/32/659532/4/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/a5c1829/ : SUCCESS in 1h 19m 04s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/32/659532/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/06bcc68/ : FAILURE in 2h 51m 03s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"44d6233f46cb7069e9ddffe9e58a326ec3767dcc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 20:19:18.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/32/659532/4/check/puppet-openstack-lint-ubuntu-bionic/f1d207e/ : SUCCESS in 6m 08s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/32/659532/4/check/puppet-openstack-syntax-4-ubuntu-bionic/43ea872/ : FAILURE in 3m 33s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/32/659532/4/check/puppet-openstack-syntax-5-ubuntu-bionic/5fcc233/ : FAILURE in 4m 08s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/32/659532/4/check/puppet-openstack-syntax-6-ubuntu-bionic/de1c99e/ : FAILURE in 3m 58s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/32/659532/4/check/puppet-openstack-unit-5.5-centos-7/e4b4105/ : FAILURE in 52m 28s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/32/659532/4/check/puppet-openstack-unit-6.0-ubuntu-bionic/25090c3/ : FAILURE in 15m 16s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/32/659532/4/check/puppet-openstack-unit-latest-ubuntu-bionic/39024dd/ : FAILURE in 15m 52s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/32/659532/4/check/tripleo-ci-centos-7-undercloud-containers/4a8b215/ : SUCCESS in 1h 27m 43s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/32/659532/4/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/1d97d07/ : FAILURE in 1h 51m 57s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/32/659532/4/check/tripleo-ci-centos-7-containers-multinode/21cbe41/ : SUCCESS in 2h 09m 23s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/32/659532/4/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/9b00f74/ : SUCCESS in 2h 07m 01s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/32/659532/4/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/4bf5e41/ : SUCCESS in 2h 23m 26s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/32/659532/4/check/tripleo-ci-centos-7-standalone/346545e/ : SUCCESS in 1h 46m 40s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/32/659532/4/check/tripleo-ci-fedora-28-standalone/fee558b/ : FAILURE in 1h 42m 29s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/32/659532/4/check/tripleo-ci-centos-7-standalone-upgrade/ac271aa/ : FAILURE in 1h 32m 59s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/32/659532/4/check/tripleo-ci-centos-7-standalone-upgrade-stein/6597133/ : SUCCESS in 2h 15m 08s\n- tripleo-ci-centos-7-scenario010-standalone http://logs.openstack.org/32/659532/4/check/tripleo-ci-centos-7-scenario010-standalone/1e1b37e/ : FAILURE in 42m 28s (non-voting)\n- tripleo-ci-centos-7-standalone-rocky http://logs.openstack.org/32/659532/4/check/tripleo-ci-centos-7-standalone-rocky/9d561e3/ : SUCCESS in 1h 39m 37s (non-voting)\n- tripleo-ci-centos-7-scenario010-multinode-oooq-container http://logs.openstack.org/32/659532/4/check/tripleo-ci-centos-7-scenario010-multinode-oooq-container/f7cc8ee/ : FAILURE in 1h 38m 01s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"cc248168e28d1f97daf9c5ce8a3582a0d67f048f","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-05-20 09:16:06.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"7528036068c8f095839fb33e0c0fb3b1fcfc36b9","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-20 09:36:08.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/32/659532/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/1ffd097/ : RETRY_LIMIT in 4m 04s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/32/659532/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/b68618c/ : RETRY_LIMIT in 5m 19s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/32/659532/5/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/ed43b91/ : FAILURE in 3m 52s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/32/659532/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/b403263/ : RETRY_LIMIT in 5m 34s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"674b236ea1060f1e6c27e25bc221a2340367110b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 09:49:11.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/32/659532/5/check/puppet-openstack-lint-ubuntu-bionic/3436b9d/ : SUCCESS in 5m 34s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/32/659532/5/check/puppet-openstack-syntax-4-ubuntu-bionic/eb2c075/ : SUCCESS in 6m 25s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/32/659532/5/check/puppet-openstack-syntax-5-ubuntu-bionic/864af5d/ : SUCCESS in 5m 44s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/32/659532/5/check/puppet-openstack-syntax-6-ubuntu-bionic/511a471/ : SUCCESS in 4m 05s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/32/659532/5/check/puppet-openstack-unit-5.5-centos-7/2feb11f/ : FAILURE in 28m 23s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/32/659532/5/check/puppet-openstack-unit-6.0-ubuntu-bionic/594a131/ : FAILURE in 19m 54s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/32/659532/5/check/puppet-openstack-unit-latest-ubuntu-bionic/2807671/ : FAILURE in 20m 36s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/32/659532/5/check/tripleo-ci-centos-7-undercloud-containers/94bb9a6/ : FAILURE in 8m 56s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/32/659532/5/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/8cfa20f/ : FAILURE in 9m 12s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/32/659532/5/check/tripleo-ci-centos-7-containers-multinode/701dd61/ : FAILURE in 14m 07s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/32/659532/5/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/2676d12/ : FAILURE in 10m 05s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/32/659532/5/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/6199b68/ : FAILURE in 14m 03s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/32/659532/5/check/tripleo-ci-centos-7-standalone/5cbedb3/ : FAILURE in 10m 45s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/32/659532/5/check/tripleo-ci-fedora-28-standalone/dbbd343/ : FAILURE in 12m 53s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/32/659532/5/check/tripleo-ci-centos-7-standalone-upgrade/f4e262d/ : FAILURE in 10m 49s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/32/659532/5/check/tripleo-ci-centos-7-standalone-upgrade-stein/53cea2e/ : FAILURE in 11m 48s\n- tripleo-ci-centos-7-scenario010-standalone http://logs.openstack.org/32/659532/5/check/tripleo-ci-centos-7-scenario010-standalone/39481c8/ : FAILURE in 11m 59s (non-voting)\n- tripleo-ci-centos-7-standalone-rocky http://logs.openstack.org/32/659532/5/check/tripleo-ci-centos-7-standalone-rocky/d59f24b/ : FAILURE in 13m 18s (non-voting)\n- tripleo-ci-centos-7-scenario010-multinode-oooq-container http://logs.openstack.org/32/659532/5/check/tripleo-ci-centos-7-scenario010-multinode-oooq-container/4d5bd3d/ : FAILURE in 11m 52s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"92e295c3a7c673350b7977b1ba35f58bc9a86f0d","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-05-20 10:01:02.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"7138a08987d53574cec4f611853ec12b5acbfa58","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-05-20 10:16:46.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"1545bb586167e85abedfcacab83050d782f4b8da","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-20 12:50:41.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/32/659532/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/db99adc/ : SUCCESS in 2h 21m 41s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/32/659532/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/293b785/ : SUCCESS in 2h 32m 19s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/32/659532/7/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/6c5b34b/ : SUCCESS in 1h 12m 19s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/32/659532/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/fcb9031/ : FAILURE in 2h 29m 38s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"03acac1ef21a113fd450051afc2bef8bb03a8683","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 12:51:05.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/32/659532/7/check/puppet-openstack-lint-ubuntu-bionic/bf3c75c/ : SUCCESS in 4m 48s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/32/659532/7/check/puppet-openstack-syntax-4-ubuntu-bionic/d188fd6/ : SUCCESS in 4m 06s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/32/659532/7/check/puppet-openstack-syntax-5-ubuntu-bionic/5fe1af4/ : SUCCESS in 4m 06s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/32/659532/7/check/puppet-openstack-syntax-6-ubuntu-bionic/877588c/ : SUCCESS in 3m 56s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/32/659532/7/check/puppet-openstack-unit-5.5-centos-7/1a0e2db/ : SUCCESS in 27m 46s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/32/659532/7/check/puppet-openstack-unit-6.0-ubuntu-bionic/2b08b02/ : SUCCESS in 15m 03s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/32/659532/7/check/puppet-openstack-unit-latest-ubuntu-bionic/efc9620/ : SUCCESS in 14m 01s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/32/659532/7/check/tripleo-ci-centos-7-undercloud-containers/f8b48d9/ : SUCCESS in 1h 26m 53s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/32/659532/7/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/3a759ba/ : FAILURE in 1h 16m 34s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/32/659532/7/check/tripleo-ci-centos-7-containers-multinode/0a3e60c/ : SUCCESS in 2h 10m 14s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/32/659532/7/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/2f99a0d/ : SUCCESS in 2h 14m 22s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/32/659532/7/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/2fcdc42/ : SUCCESS in 2h 11m 27s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/32/659532/7/check/tripleo-ci-centos-7-standalone/7903aeb/ : SUCCESS in 1h 49m 18s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/32/659532/7/check/tripleo-ci-fedora-28-standalone/6d5fd93/ : SUCCESS in 1h 42m 27s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/32/659532/7/check/tripleo-ci-centos-7-standalone-upgrade/3d71468/ : FAILURE in 1h 31m 41s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/32/659532/7/check/tripleo-ci-centos-7-standalone-upgrade-stein/6b22d5c/ : SUCCESS in 2h 03m 26s\n- tripleo-ci-centos-7-scenario010-standalone http://logs.openstack.org/32/659532/7/check/tripleo-ci-centos-7-scenario010-standalone/c109c5a/ : SUCCESS in 1h 11m 30s (non-voting)\n- tripleo-ci-centos-7-standalone-rocky http://logs.openstack.org/32/659532/7/check/tripleo-ci-centos-7-standalone-rocky/b6c3267/ : SUCCESS in 1h 26m 10s (non-voting)\n- tripleo-ci-centos-7-scenario010-multinode-oooq-container http://logs.openstack.org/32/659532/7/check/tripleo-ci-centos-7-scenario010-multinode-oooq-container/8f6b592/ : FAILURE in 1h 53m 34s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"75fc00863a5591622d61cb94995c07021bc0c235","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-05-20 13:24:20.000000000","message":"Patch Set 7: Code-Review+1\n\nLGTM","accounts_in_message":[],"_revision_number":7},{"id":"54fde59d4b6946b9623d00f4c783fdbe35f70dce","author":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"date":"2019-05-20 13:46:48.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"5641acd5591906b1445fb57a89e0ca454cc839d6","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-05-21 15:48:53.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"ff3d63927a8ecc5dcdee33b18ff96ce69e10a4c6","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2019-05-21 17:21:23.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"add797767465026c17e6a1feed290ffcec26ceac","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2019-05-21 17:21:26.000000000","message":"Patch Set 7: Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"03ebb11626f628f209dbe779a5f5c88979f74070","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 17:21:41.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"55e9bea67301e2fc765525b7b8439db7fc8974a3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 20:03:13.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/32/659532/7/gate/puppet-openstack-lint-ubuntu-bionic/84af99b/ : SUCCESS in 6m 57s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/32/659532/7/gate/puppet-openstack-syntax-4-ubuntu-bionic/9fafc11/ : SUCCESS in 6m 37s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/32/659532/7/gate/puppet-openstack-syntax-5-ubuntu-bionic/8479efd/ : SUCCESS in 4m 40s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/32/659532/7/gate/puppet-openstack-syntax-6-ubuntu-bionic/94e8641/ : SUCCESS in 6m 30s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/32/659532/7/gate/puppet-openstack-unit-5.5-centos-7/5b1854c/ : SUCCESS in 28m 27s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/32/659532/7/gate/puppet-openstack-unit-6.0-ubuntu-bionic/93436c6/ : SUCCESS in 15m 14s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/32/659532/7/gate/tripleo-ci-centos-7-undercloud-containers/4614dd1/ : SUCCESS in 1h 42m 00s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/32/659532/7/gate/tripleo-ci-centos-7-containers-multinode/b949ba4/ : SUCCESS in 2h 28m 50s\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/32/659532/7/gate/tripleo-ci-centos-7-standalone/d6a76ce/ : SUCCESS in 1h 46m 43s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/32/659532/7/gate/tripleo-ci-fedora-28-standalone/4199d70/ : SUCCESS in 2h 02m 49s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/32/659532/7/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/a7ed1ae/ : SUCCESS in 2h 38m 51s\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/32/659532/7/gate/tripleo-ci-centos-7-standalone-upgrade-stein/cc923d3/ : SUCCESS in 1h 57m 04s","accounts_in_message":[],"_revision_number":7},{"id":"73a4e6cab33ab90ede275df3704210017ff26150","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 20:03:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"bbfce5d90866e6463e1d9583cf21ebf627346eba","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-05-21 20:20:35.000000000","message":"Patch Set 7: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 5a48995c19cea4b11d813bd9aeb54ff88276c8a3","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"1ae89eb002a0e0f59a442f3ac78db2b3a5b57635","revisions":{"f32fa545104202945398ceb0bc1571f8cd2ef784":{"kind":"REWORK","_number":1,"created":"2019-05-16 11:27:31.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/32/659532/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/32/659532/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/32/659532/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/32/659532/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/32/659532/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/32/659532/1"}}},"commit":{"parents":[{"commit":"e32f2ef8aad090590f0ee580fcb075af2248351e","subject":"Merge \"Let fluentd not create /etc/rsyslog.d/fluentd.conf file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/e32f2ef8aad090590f0ee580fcb075af2248351e"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-05-16 11:22:30.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-05-16 11:22:30.000000000","tz":120},"subject":"Add the \u0027octavia\u0027 driver on ML2/OVN deployments","message":"Add the \u0027octavia\u0027 driver on ML2/OVN deployments\n\nBy default, both the \u0027amphora\u0027 and \u0027octavia\u0027 provider drivers are\nenabled in Octavia API. A previous change in puppet-tripleo included the\n\u0027ovn\u0027 driver when on Neutron ML2/OVN deployments but in doing so omitted\nthe \u0027octavia\u0027 driver. This patch ensures the \u0027octavia\u0027 driver remains\nenabled as is needed for Octavia Tempest tests.\n\nChange-Id: I5b29a51d7edace30e6357a2af95d1af4596ddfd3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/f32fa545104202945398ceb0bc1571f8cd2ef784"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/f32fa545104202945398ceb0bc1571f8cd2ef784"}]},"branch":"refs/heads/master"},"531f51df69069b7fcd2992a78f3fed93a7b7f39c":{"kind":"REWORK","_number":2,"created":"2019-05-16 13:44:36.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/32/659532/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/32/659532/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/32/659532/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/32/659532/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/32/659532/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/32/659532/2"}}},"commit":{"parents":[{"commit":"e32f2ef8aad090590f0ee580fcb075af2248351e","subject":"Merge \"Let fluentd not create /etc/rsyslog.d/fluentd.conf file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/e32f2ef8aad090590f0ee580fcb075af2248351e"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-05-16 11:22:30.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-05-16 13:43:20.000000000","tz":120},"subject":"Add the \u0027octavia\u0027 driver on ML2/OVN deployments","message":"Add the \u0027octavia\u0027 driver on ML2/OVN deployments\n\nBy default, both the \u0027amphora\u0027 and \u0027octavia\u0027 provider drivers are\nenabled in Octavia API. A previous change in puppet-tripleo included the\n\u0027ovn\u0027 driver when on Neutron ML2/OVN deployments but in doing so omitted\nthe \u0027octavia\u0027 driver. This patch ensures the \u0027octavia\u0027 driver remains\nenabled as is needed for Octavia Tempest tests.\n\nChange-Id: I5b29a51d7edace30e6357a2af95d1af4596ddfd3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/531f51df69069b7fcd2992a78f3fed93a7b7f39c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/531f51df69069b7fcd2992a78f3fed93a7b7f39c"}]},"branch":"refs/heads/master"},"72ceeb76da9ae911f16cce72d95be1a538700eb6":{"kind":"REWORK","_number":3,"created":"2019-05-16 13:48:26.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/32/659532/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/32/659532/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/32/659532/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/32/659532/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/32/659532/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/32/659532/3"}}},"commit":{"parents":[{"commit":"e32f2ef8aad090590f0ee580fcb075af2248351e","subject":"Merge \"Let fluentd not create /etc/rsyslog.d/fluentd.conf file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/e32f2ef8aad090590f0ee580fcb075af2248351e"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-05-16 11:22:30.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-05-16 13:48:13.000000000","tz":120},"subject":"Add the \u0027octavia\u0027 driver on ML2/OVN deployments","message":"Add the \u0027octavia\u0027 driver on ML2/OVN deployments\n\nBy default, both the \u0027amphora\u0027 and \u0027octavia\u0027 provider drivers are\nenabled in Octavia API. A previous change in puppet-tripleo included the\n\u0027ovn\u0027 driver when on Neutron ML2/OVN deployments but in doing so omitted\nthe \u0027octavia\u0027 driver. This patch ensures the \u0027octavia\u0027 driver remains\nenabled as is needed for Octavia Tempest tests.\n\nChange-Id: I5b29a51d7edace30e6357a2af95d1af4596ddfd3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/72ceeb76da9ae911f16cce72d95be1a538700eb6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/72ceeb76da9ae911f16cce72d95be1a538700eb6"}]},"branch":"refs/heads/master"},"d9cc88b341e258b4582916ecaaa5c8a88ad49754":{"kind":"REWORK","_number":4,"created":"2019-05-16 16:52:19.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/32/659532/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/32/659532/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/32/659532/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/32/659532/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/32/659532/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/32/659532/4"}}},"commit":{"parents":[{"commit":"e32f2ef8aad090590f0ee580fcb075af2248351e","subject":"Merge \"Let fluentd not create /etc/rsyslog.d/fluentd.conf file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/e32f2ef8aad090590f0ee580fcb075af2248351e"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-05-16 11:22:30.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-05-16 16:52:09.000000000","tz":120},"subject":"Add the \u0027octavia\u0027 driver on ML2/OVN deployments","message":"Add the \u0027octavia\u0027 driver on ML2/OVN deployments\n\nBy default, both the \u0027amphora\u0027 and \u0027octavia\u0027 provider drivers are\nenabled in Octavia API. A previous change in puppet-tripleo included the\n\u0027ovn\u0027 driver when on Neutron ML2/OVN deployments but in doing so omitted\nthe \u0027octavia\u0027 driver. This patch ensures the \u0027octavia\u0027 driver remains\nenabled as is needed for Octavia Tempest tests.\n\nChange-Id: I5b29a51d7edace30e6357a2af95d1af4596ddfd3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/d9cc88b341e258b4582916ecaaa5c8a88ad49754"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/d9cc88b341e258b4582916ecaaa5c8a88ad49754"}]},"branch":"refs/heads/master"},"647667f4f30fa21204c8090ff47ecaf61f6c3d22":{"kind":"REWORK","_number":5,"created":"2019-05-20 09:16:06.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/32/659532/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/32/659532/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/32/659532/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/32/659532/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/32/659532/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/32/659532/5"}}},"commit":{"parents":[{"commit":"e32f2ef8aad090590f0ee580fcb075af2248351e","subject":"Merge \"Let fluentd not create /etc/rsyslog.d/fluentd.conf file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/e32f2ef8aad090590f0ee580fcb075af2248351e"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-05-16 11:22:30.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-05-20 09:15:54.000000000","tz":120},"subject":"Add the \u0027octavia\u0027 driver on ML2/OVN deployments","message":"Add the \u0027octavia\u0027 driver on ML2/OVN deployments\n\nBy default, both the \u0027amphora\u0027 and \u0027octavia\u0027 provider drivers are\nenabled in Octavia API. A previous change in puppet-tripleo included the\n\u0027ovn\u0027 driver when on Neutron ML2/OVN deployments but in doing so omitted\nthe \u0027octavia\u0027 driver. This patch ensures the \u0027octavia\u0027 driver remains\nenabled as is needed for Octavia Tempest tests.\n\nChange-Id: I5b29a51d7edace30e6357a2af95d1af4596ddfd3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/647667f4f30fa21204c8090ff47ecaf61f6c3d22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/647667f4f30fa21204c8090ff47ecaf61f6c3d22"}]},"branch":"refs/heads/master"},"30721fabb1f91697ace2806a8e986f813d783c86":{"kind":"REWORK","_number":6,"created":"2019-05-20 10:01:02.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/32/659532/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/32/659532/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/32/659532/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/32/659532/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/32/659532/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/32/659532/6"}}},"commit":{"parents":[{"commit":"e32f2ef8aad090590f0ee580fcb075af2248351e","subject":"Merge \"Let fluentd not create /etc/rsyslog.d/fluentd.conf file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/e32f2ef8aad090590f0ee580fcb075af2248351e"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-05-16 11:22:30.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-05-20 09:59:56.000000000","tz":120},"subject":"Add the \u0027octavia\u0027 driver on ML2/OVN deployments","message":"Add the \u0027octavia\u0027 driver on ML2/OVN deployments\n\nBy default, both the \u0027amphora\u0027 and \u0027octavia\u0027 provider drivers are\nenabled in Octavia API. A previous change in puppet-tripleo included the\n\u0027ovn\u0027 driver when on Neutron ML2/OVN deployments but in doing so omitted\nthe \u0027octavia\u0027 driver. This patch ensures the \u0027octavia\u0027 driver remains\nenabled as is needed for Octavia Tempest tests.\n\nChange-Id: I5b29a51d7edace30e6357a2af95d1af4596ddfd3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/30721fabb1f91697ace2806a8e986f813d783c86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/30721fabb1f91697ace2806a8e986f813d783c86"}]},"branch":"refs/heads/master"},"1ae89eb002a0e0f59a442f3ac78db2b3a5b57635":{"kind":"REWORK","_number":7,"created":"2019-05-20 10:16:46.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/32/659532/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/32/659532/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/32/659532/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/32/659532/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/32/659532/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/32/659532/7"}}},"commit":{"parents":[{"commit":"e32f2ef8aad090590f0ee580fcb075af2248351e","subject":"Merge \"Let fluentd not create /etc/rsyslog.d/fluentd.conf file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/e32f2ef8aad090590f0ee580fcb075af2248351e"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-05-16 11:22:30.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2019-05-20 10:16:26.000000000","tz":120},"subject":"Include \u0027octavia\u0027 driver on ML2/OVN deployments","message":"Include \u0027octavia\u0027 driver on ML2/OVN deployments\n\nBy default, both the \u0027amphora\u0027 and \u0027octavia\u0027 provider drivers are\nenabled in Octavia API. A previous change in puppet-tripleo included the\n\u0027ovn\u0027 driver when on Neutron ML2/OVN deployments but in doing so omitted\nthe \u0027octavia\u0027 driver. This patch ensures the \u0027octavia\u0027 driver remains\nenabled as is needed for Octavia Tempest tests.\n\nChange-Id: I5b29a51d7edace30e6357a2af95d1af4596ddfd3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/1ae89eb002a0e0f59a442f3ac78db2b3a5b57635"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/1ae89eb002a0e0f59a442f3ac78db2b3a5b57635"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
