)]}'
{"id":"openstack%2Fpuppet-tripleo~659850","triplet_id":"openstack%2Fpuppet-tripleo~stable%2Fqueens~I248859bf0e4b70e3a057e96b5fb74be64f4008ed","project":"openstack/puppet-tripleo","branch":"stable/queens","hashtags":[],"change_id":"I248859bf0e4b70e3a057e96b5fb74be64f4008ed","subject":"Refactor collectd/gnocchi integration","status":"MERGED","created":"2019-05-17 17:30:16.000000000","updated":"2019-06-06 20:38:04.000000000","submitted":"2019-06-01 16:46:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"659850-1559407599494-946adf98","meta_rev_id":"157569aad5e0aa2176bcfb61f5c152ad4062cc6a","_number":659850,"virtual_id_number":659850,"owner":{"_account_id":28504,"name":"Vinay Kapalavai","email":"vkapalav@redhat.com","username":"vkapalav"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-06-01 16:46:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-05-31 22:05:00.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},{"value":2,"date":"2019-05-31 19:59:35.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},{"value":1,"date":"2019-06-01 14:34:30.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-17 17:30:16.000000000","updated_by":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"reviewer":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"state":"REVIEWER"},{"updated":"2019-05-31 22:05:00.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-01 14:34:30.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-06-01 16:46:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d512cd94882c493e0a13ced4911d61325b89c702","author":{"_account_id":28504,"name":"Vinay Kapalavai","email":"vkapalav@redhat.com","username":"vkapalav"},"date":"2019-05-17 17:30:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"314df33add3728c7eddccc84399da600bb65b254","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-17 19:38:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/50/659850/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/48f5ab9/ : SUCCESS in 2h 07m 34s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/50/659850/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/736296a/ : SUCCESS in 2h 00m 20s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/50/659850/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/38fe83c/ : FAILURE in 1h 57m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3ec4070db9f4d828e4b6024bf5f167a1de444fa9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 19:44:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-oooq http://logs.openstack.org/50/659850/1/check/tripleo-ci-centos-7-undercloud-oooq/6899a4d/ : SUCCESS in 1h 12m 12s\n- tripleo-ci-centos-7-undercloud-upgrades http://logs.openstack.org/50/659850/1/check/tripleo-ci-centos-7-undercloud-upgrades/6747f4c/ : SUCCESS in 1h 22m 13s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/50/659850/1/check/tripleo-ci-centos-7-containers-multinode/adc25ab/ : SUCCESS in 2h 07m 40s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/50/659850/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/0f4d398/ : FAILURE in 1h 51m 02s (non-voting)\n- puppet-openstack-lint http://logs.openstack.org/50/659850/1/check/puppet-openstack-lint/ece44b7/ : SUCCESS in 4m 02s\n- puppet-openstack-syntax-4 http://logs.openstack.org/50/659850/1/check/puppet-openstack-syntax-4/1958ecc/ : SUCCESS in 4m 37s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/50/659850/1/check/puppet-openstack-unit-4.8-centos-7/426fefd/ : SUCCESS in 29m 02s\n- tripleo-ci-centos-7-scenario001-multinode-oooq-container http://logs.openstack.org/50/659850/1/check/tripleo-ci-centos-7-scenario001-multinode-oooq-container/393e202/ : FAILURE in 1h 47m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bf53985371d613bd0c1902a68c75a7c53661a7d4","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-05-22 19:53:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fefd83c4f2ad83b190b6502f49620d86ec07ae83","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 19:54:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1b86fc532999c1c6968de30c40e8eeaa332e288c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 21:19:35.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"358e1bf3bbd731e9ebed69fdd9c51a4be9c415e3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 01:39:30.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-oooq http://logs.openstack.org/50/659850/1/gate/tripleo-ci-centos-7-undercloud-oooq/adfa212/ : FAILURE in 1h 08m 24s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/50/659850/1/gate/tripleo-ci-centos-7-containers-multinode/94443b9/ : SUCCESS in 1h 56m 10s\n- puppet-openstack-lint http://logs.openstack.org/50/659850/1/gate/puppet-openstack-lint/0527760/ : SUCCESS in 4m 52s\n- puppet-openstack-syntax-4 http://logs.openstack.org/50/659850/1/gate/puppet-openstack-syntax-4/665cbb0/ : SUCCESS in 4m 28s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/50/659850/1/gate/puppet-openstack-unit-4.8-centos-7/d923dd0/ : SUCCESS in 31m 39s","accounts_in_message":[],"_revision_number":1},{"id":"f291ffcbadcdce4d172079ce6d1f2d8b22e1fd56","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-05-23 14:14:57.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"cbf89b7ed6abc83cbee6078777023a9ce203a270","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-23 16:57:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/50/659850/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/a71f7d3/ : SUCCESS in 2h 07m 18s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/50/659850/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/7f50037/ : SUCCESS in 2h 05m 25s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/50/659850/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/1d35678/ : FAILURE in 2h 02m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"289a756f1894b2361a52e43deb0413398c37088e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 17:27:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-oooq http://logs.openstack.org/50/659850/1/check/tripleo-ci-centos-7-undercloud-oooq/49513c9/ : FAILURE in 58m 34s\n- tripleo-ci-centos-7-undercloud-upgrades http://logs.openstack.org/50/659850/1/check/tripleo-ci-centos-7-undercloud-upgrades/90a43dc/ : SUCCESS in 1h 17m 43s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/50/659850/1/check/tripleo-ci-centos-7-containers-multinode/365811e/ : SUCCESS in 2h 16m 49s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/50/659850/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/4aa1a7a/ : FAILURE in 1h 46m 59s (non-voting)\n- puppet-openstack-lint http://logs.openstack.org/50/659850/1/check/puppet-openstack-lint/a05c0e5/ : SUCCESS in 4m 13s\n- puppet-openstack-syntax-4 http://logs.openstack.org/50/659850/1/check/puppet-openstack-syntax-4/6130c17/ : SUCCESS in 3m 42s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/50/659850/1/check/puppet-openstack-unit-4.8-centos-7/86ebc4d/ : SUCCESS in 31m 14s\n- tripleo-ci-centos-7-scenario001-multinode-oooq-container http://logs.openstack.org/50/659850/1/check/tripleo-ci-centos-7-scenario001-multinode-oooq-container/7666a5c/ : FAILURE in 1h 42m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f0e596b520b108f47b33119bf28d723777a45613","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-05-30 23:47:28.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8ed13f0eeddd7be730ebf47740421d96e5b14eaa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-31 01:59:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-oooq http://logs.openstack.org/50/659850/1/check/tripleo-ci-centos-7-undercloud-oooq/6ba960a/ : SUCCESS in 1h 08m 12s\n- tripleo-ci-centos-7-undercloud-upgrades http://logs.openstack.org/50/659850/1/check/tripleo-ci-centos-7-undercloud-upgrades/819840f/ : SUCCESS in 1h 18m 51s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/50/659850/1/check/tripleo-ci-centos-7-containers-multinode/5530762/ : SUCCESS in 2h 02m 39s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/50/659850/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/25d20fe/ : FAILURE in 1h 51m 47s (non-voting)\n- puppet-openstack-lint http://logs.openstack.org/50/659850/1/check/puppet-openstack-lint/76441c4/ : SUCCESS in 4m 26s\n- puppet-openstack-syntax-4 http://logs.openstack.org/50/659850/1/check/puppet-openstack-syntax-4/a90d5be/ : SUCCESS in 6m 23s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/50/659850/1/check/puppet-openstack-unit-4.8-centos-7/7539185/ : SUCCESS in 32m 37s\n- tripleo-ci-centos-7-scenario001-multinode-oooq-container http://logs.openstack.org/50/659850/1/check/tripleo-ci-centos-7-scenario001-multinode-oooq-container/9f3e6e2/ : FAILURE in 1h 53m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"351863db096fcbb921ad27b2d848a4f3a9ec38cf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-31 02:00:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d8680d46d27a448980a5ccf16b705b5c6ea615ae","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-31 02:20:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 finger://ze05.softwarefactory-project.io:7979/4338d7d3e13e421cae56ed2421437884 : RETRY_LIMIT in 6s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/50/659850/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/09369e0/ : SUCCESS in 2h 05m 12s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/50/659850/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/44724e0/ : FAILURE in 2h 12m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5a83472134183c027aa65f53f858fe33e8868d32","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-31 08:43:11.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-oooq http://logs.openstack.org/50/659850/1/gate/tripleo-ci-centos-7-undercloud-oooq/6727b45/ : FAILURE in 1h 14m 27s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/50/659850/1/gate/tripleo-ci-centos-7-containers-multinode/4f53947/ : SUCCESS in 2h 01m 41s\n- puppet-openstack-lint http://logs.openstack.org/50/659850/1/gate/puppet-openstack-lint/828f4bd/ : SUCCESS in 6m 47s\n- puppet-openstack-syntax-4 http://logs.openstack.org/50/659850/1/gate/puppet-openstack-syntax-4/b44d3c4/ : SUCCESS in 4m 01s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/50/659850/1/gate/puppet-openstack-unit-4.8-centos-7/3f250b4/ : SUCCESS in 57m 54s","accounts_in_message":[],"_revision_number":1},{"id":"eae9edc2972ff705b5d6310c977473948b710965","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-05-31 13:46:32.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"de6422c9483b6593723c98996031bd85f88925ff","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-31 16:12:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/50/659850/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/4a0a7df/ : SUCCESS in 2h 07m 02s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/50/659850/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/a87b9c5/ : SUCCESS in 2h 22m 52s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/50/659850/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/5984620/ : FAILURE in 1h 55m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d7640c984d5d3cbfa7d3bc8de6b3528febd44ac2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-31 17:40:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-oooq http://logs.openstack.org/50/659850/1/check/tripleo-ci-centos-7-undercloud-oooq/b80344d/ : SUCCESS in 1h 12m 38s\n- tripleo-ci-centos-7-undercloud-upgrades http://logs.openstack.org/50/659850/1/check/tripleo-ci-centos-7-undercloud-upgrades/3807277/ : SUCCESS in 1h 23m 06s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/50/659850/1/check/tripleo-ci-centos-7-containers-multinode/8d464d0/ : SUCCESS in 1h 59m 57s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/50/659850/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/12d9da5/ : FAILURE in 1h 51m 20s (non-voting)\n- puppet-openstack-lint http://logs.openstack.org/50/659850/1/check/puppet-openstack-lint/a5f6311/ : SUCCESS in 5m 45s\n- puppet-openstack-syntax-4 http://logs.openstack.org/50/659850/1/check/puppet-openstack-syntax-4/b88fc23/ : SUCCESS in 4m 01s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/50/659850/1/check/puppet-openstack-unit-4.8-centos-7/9f14e4f/ : SUCCESS in 31m 36s\n- tripleo-ci-centos-7-scenario001-multinode-oooq-container http://logs.openstack.org/50/659850/1/check/tripleo-ci-centos-7-scenario001-multinode-oooq-container/615801c/ : FAILURE in 1h 44m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4677774e6855218339bfac512bd3813282305af2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-31 17:40:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b087e556586bee99154662ae55a50c2f12b90fa1","author":{"_account_id":28504,"name":"Vinay Kapalavai","email":"vkapalav@redhat.com","username":"vkapalav"},"date":"2019-05-31 19:59:35.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"ee6c5e254362d81608cfbecd01850a16f6525bcd","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-31 22:05:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/50/659850/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/1941697/ : SUCCESS in 2h 03m 35s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/50/659850/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/701fb1c/ : SUCCESS in 2h 03m 42s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/50/659850/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/620e787/ : FAILURE in 1h 54m 35s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a4bbdb5ed408cfab6c275737afff704d7337a25d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-01 01:10:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-oooq http://logs.openstack.org/50/659850/2/check/tripleo-ci-centos-7-undercloud-oooq/d3e39bc/ : SUCCESS in 1h 09m 20s\n- tripleo-ci-centos-7-undercloud-upgrades http://logs.openstack.org/50/659850/2/check/tripleo-ci-centos-7-undercloud-upgrades/2fedf67/ : SUCCESS in 1h 26m 37s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/50/659850/2/check/tripleo-ci-centos-7-containers-multinode/1f5d65a/ : SUCCESS in 2h 00m 49s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/50/659850/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/46fdc9e/ : FAILURE in 1h 48m 44s (non-voting)\n- puppet-openstack-lint http://logs.openstack.org/50/659850/2/check/puppet-openstack-lint/b5bf30e/ : SUCCESS in 4m 08s\n- puppet-openstack-syntax-4 http://logs.openstack.org/50/659850/2/check/puppet-openstack-syntax-4/9f3d68b/ : SUCCESS in 3m 55s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/50/659850/2/check/puppet-openstack-unit-4.8-centos-7/1a27e9a/ : SUCCESS in 31m 45s\n- tripleo-ci-centos-7-scenario001-multinode-oooq-container http://logs.openstack.org/50/659850/2/check/tripleo-ci-centos-7-scenario001-multinode-oooq-container/5b9c82c/ : FAILURE in 1h 50m 07s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7f01c9a445bc4d70e7cd036775d5b5b3bce1bfd2","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-01 14:34:30.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"36834cc818eeca5b055a0c989e7f8d2820d449d7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-01 14:34:43.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ab37e5e8dc46aa57d655e1bb4f1c35f35d744310","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-01 16:46:39.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-undercloud-oooq http://logs.openstack.org/50/659850/2/gate/tripleo-ci-centos-7-undercloud-oooq/9ea0d01/ : SUCCESS in 1h 07m 16s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/50/659850/2/gate/tripleo-ci-centos-7-containers-multinode/c7ecdf4/ : SUCCESS in 2h 06m 39s\n- puppet-openstack-lint http://logs.openstack.org/50/659850/2/gate/puppet-openstack-lint/a9c4b55/ : SUCCESS in 3m 52s\n- puppet-openstack-syntax-4 http://logs.openstack.org/50/659850/2/gate/puppet-openstack-syntax-4/a902c24/ : SUCCESS in 3m 32s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/50/659850/2/gate/puppet-openstack-unit-4.8-centos-7/8123f91/ : SUCCESS in 29m 36s","accounts_in_message":[],"_revision_number":2},{"id":"8e0ef033788c205cb9fb9bf7610a825314487202","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-01 16:46:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"157569aad5e0aa2176bcfb61f5c152ad4062cc6a","author":{"_account_id":28504,"name":"Vinay Kapalavai","email":"vkapalav@redhat.com","username":"vkapalav"},"date":"2019-06-06 20:38:04.000000000","message":"Created a revert of this change as I8a06c25d9ef73d72c5ea30086ffa340d82ab1dfe","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2a9f0d89f10216f5d3915555f1b57faa9cddf8bc","revisions":{"5e5d755a88da72f33bd2be5cd58a4f4dca90b0ad":{"kind":"REWORK","_number":1,"created":"2019-05-17 17:30:16.000000000","uploader":{"_account_id":28504,"name":"Vinay Kapalavai","email":"vkapalav@redhat.com","username":"vkapalav"},"ref":"refs/changes/50/659850/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/50/659850/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/50/659850/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/50/659850/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/50/659850/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/50/659850/1"}}},"commit":{"parents":[{"commit":"6b3573b00fe5dc9b98e1c6cf3f8a0188715430bd","subject":"Add an Endpoint in ceilometer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/6b3573b00fe5dc9b98e1c6cf3f8a0188715430bd"}]}],"author":{"name":"Martin Magr","email":"mmagr@redhat.com","date":"2019-03-20 11:53:53.000000000","tz":60},"committer":{"name":"Nagasai Vinaykumar Kapalavai","email":"vkapalav@redhat.com","date":"2019-05-17 17:30:09.000000000","tz":-240},"subject":"Refactor collectd/gnocchi integration","message":"Refactor collectd/gnocchi integration\n\nCurrent implementation of collect_gnocchi plugin configuration blocks\nusage of other python-based plugins. This patch leverages puppet-collectd\nclasses properly to enable configuration of more than one python-based\nplugin together.\n\nChange-Id: I248859bf0e4b70e3a057e96b5fb74be64f4008ed\n(cherry picked from commit cfa969dac63ab1aafd7343091302c089c5cd0873)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/5e5d755a88da72f33bd2be5cd58a4f4dca90b0ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/5e5d755a88da72f33bd2be5cd58a4f4dca90b0ad"}]},"branch":"refs/heads/stable/queens"},"2a9f0d89f10216f5d3915555f1b57faa9cddf8bc":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-05-31 19:59:35.000000000","uploader":{"_account_id":28504,"name":"Vinay Kapalavai","email":"vkapalav@redhat.com","username":"vkapalav"},"ref":"refs/changes/50/659850/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/50/659850/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/50/659850/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/50/659850/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/50/659850/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/50/659850/2"}}},"commit":{"parents":[{"commit":"cdc7c11e936fed893b8f847beb9287a564186508","subject":"Merge \"Add nova::cinder to nova compute\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/cdc7c11e936fed893b8f847beb9287a564186508"}]}],"author":{"name":"Martin Magr","email":"mmagr@redhat.com","date":"2019-03-20 11:53:53.000000000","tz":60},"committer":{"name":"Nagasai Vinaykumar Kapalavai","email":"vkapalav@redhat.com","date":"2019-05-31 19:58:17.000000000","tz":-240},"subject":"Refactor collectd/gnocchi integration","message":"Refactor collectd/gnocchi integration\n\nCurrent implementation of collect_gnocchi plugin configuration blocks\nusage of other python-based plugins. This patch leverages puppet-collectd\nclasses properly to enable configuration of more than one python-based\nplugin together.\n\nChange-Id: I248859bf0e4b70e3a057e96b5fb74be64f4008ed\n(cherry picked from commit cfa969dac63ab1aafd7343091302c089c5cd0873)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/2a9f0d89f10216f5d3915555f1b57faa9cddf8bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/2a9f0d89f10216f5d3915555f1b57faa9cddf8bc"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
