)]}'
{"id":"openstack%2Fpuppet-tripleo~662983","triplet_id":"openstack%2Fpuppet-tripleo~stable%2Fqueens~Ib8e7ef5b58cc0a6d125502081c1459ccf6648ee2","project":"openstack/puppet-tripleo","branch":"stable/queens","topic":"bp/manila-external-ganesha","hashtags":[],"change_id":"Ib8e7ef5b58cc0a6d125502081c1459ccf6648ee2","subject":"Allow external Ganesha for the cephfs manila backend","status":"MERGED","created":"2019-06-04 09:21:21.000000000","updated":"2019-06-06 02:10:00.000000000","submitted":"2019-06-06 02:10:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"662983-1559787000477-5574ccc7","meta_rev_id":"d7afd3ad6a356338b5fce5a30254dfd6ab26b971","_number":662983,"virtual_id_number":662983,"owner":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":2,"date":"2019-06-06 02:10:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-04 11:30:05.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"all":[{"value":2,"date":"2019-06-05 20:42:01.000000000","_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":1,"date":"2019-06-05 07:30:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"all":[{"value":1,"date":"2019-06-05 20:42:01.000000000","_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},{"value":0,"date":"2019-06-04 09:21:21.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-04 09:21:21.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2019-06-04 11:30:05.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-05 07:30:47.000000000","updated_by":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"reviewer":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"state":"REVIEWER"},{"updated":"2019-06-05 20:42:01.000000000","updated_by":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"reviewer":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"state":"REVIEWER"},{"updated":"2019-06-06 02:10:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7609cdc020c82c01dae18de174d8db9657f381ff","author":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"date":"2019-06-04 09:21:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"30eee1ed2e4668f4b7df81d7bb76d713415922b3","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-04 11:30:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/83/662983/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/13189a6/ : FAILURE in 2h 07m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"efaeecc58b3b128aaaccff9d40a28b9aa100628c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-04 12:12:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-oooq http://logs.openstack.org/83/662983/1/check/tripleo-ci-centos-7-undercloud-oooq/25073c5/ : SUCCESS in 1h 14m 20s\n- tripleo-ci-centos-7-undercloud-upgrades http://logs.openstack.org/83/662983/1/check/tripleo-ci-centos-7-undercloud-upgrades/e468230/ : SUCCESS in 1h 21m 02s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/83/662983/1/check/tripleo-ci-centos-7-containers-multinode/2d5fd2f/ : SUCCESS in 2h 00m 21s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/83/662983/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/c0c900c/ : FAILURE in 1h 55m 44s (non-voting)\n- puppet-openstack-lint http://logs.openstack.org/83/662983/1/check/puppet-openstack-lint/c9ec810/ : SUCCESS in 5m 29s\n- puppet-openstack-syntax-4 http://logs.openstack.org/83/662983/1/check/puppet-openstack-syntax-4/43b1a90/ : SUCCESS in 4m 19s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/83/662983/1/check/puppet-openstack-unit-4.8-centos-7/19ab041/ : SUCCESS in 33m 24s\n- tripleo-ci-centos-7-scenario004-multinode-oooq-container http://logs.openstack.org/83/662983/1/check/tripleo-ci-centos-7-scenario004-multinode-oooq-container/c656ff7/ : SUCCESS in 2h 37m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e9db42d9bceec74ab406fc8c1f8e60f9c66719e8","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2019-06-05 07:30:47.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"0cb1121494637e50ab5cc13d1ddfbff224bc8a3c","author":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"date":"2019-06-05 20:42:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1b659522fd88ab2681ab873140f1a95b1861dc18","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 20:42:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"da6b634fc994e34b66dcba36c8f670bc902ef115","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 02:10:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"abbe80d346f80e102e61ec62f90f9be6678c169e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 02:10:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-undercloud-oooq http://logs.openstack.org/83/662983/1/gate/tripleo-ci-centos-7-undercloud-oooq/8f57479/ : SUCCESS in 1h 17m 29s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/83/662983/1/gate/tripleo-ci-centos-7-containers-multinode/23696d4/ : SUCCESS in 2h 09m 05s\n- puppet-openstack-lint http://logs.openstack.org/83/662983/1/gate/puppet-openstack-lint/afbd41b/ : SUCCESS in 5m 16s\n- puppet-openstack-syntax-4 http://logs.openstack.org/83/662983/1/gate/puppet-openstack-syntax-4/9327672/ : SUCCESS in 4m 27s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/83/662983/1/gate/puppet-openstack-unit-4.8-centos-7/dd0213a/ : SUCCESS in 31m 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"06be0e18123b0c242fcc8464241246fd9b9c7c1c","revisions":{"06be0e18123b0c242fcc8464241246fd9b9c7c1c":{"kind":"REWORK","_number":1,"created":"2019-06-04 09:21:21.000000000","uploader":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"ref":"refs/changes/83/662983/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/83/662983/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/83/662983/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/83/662983/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/83/662983/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/83/662983/1"}}},"commit":{"parents":[{"commit":"51ddc3a4c322312bbe26dda69b50cd7769e4ea16","subject":"Switch metrics qdrs to edge mode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/51ddc3a4c322312bbe26dda69b50cd7769e4ea16"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2019-05-29 19:40:19.000000000","tz":-420},"committer":{"name":"Giulio Fidente","email":"gfidente@redhat.com","date":"2019-06-04 09:21:03.000000000","tz":120},"subject":"Allow external Ganesha for the cephfs manila backend","message":"Allow external Ganesha for the cephfs manila backend\n\nParameters are added here [1]; however one can use\nControllerExtraConfig to configure the missing params.\n\n[1] https://review.opendev.org/#/c/661287/\n\nSee: blueprint manila-external-ganesha\nCo-Authored-By: Giulio Fidente \u003cgfidente@redhat.com\u003e\nChange-Id: Ib8e7ef5b58cc0a6d125502081c1459ccf6648ee2\n(cherry picked from commit da474f17fb708c64d634aef7f8649972c6206f5a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/06be0e18123b0c242fcc8464241246fd9b9c7c1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/06be0e18123b0c242fcc8464241246fd9b9c7c1c"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
