)]}'
{"id":"openstack%2Fpuppet-tripleo~662985","triplet_id":"openstack%2Fpuppet-tripleo~stable%2Frocky~I1741721ede9f9f469e2554b409ec6e0e6c7c22a2","project":"openstack/puppet-tripleo","branch":"stable/rocky","topic":"bp/manila-external-ganesha","hashtags":[],"change_id":"I1741721ede9f9f469e2554b409ec6e0e6c7c22a2","subject":"Add support for external Ganesha service in Manila Share profile","status":"MERGED","created":"2019-06-04 09:25:22.000000000","updated":"2019-06-06 00:35:02.000000000","submitted":"2019-06-06 00:35:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"662985-1559781302051-2f66b382","meta_rev_id":"46e9fa861899901465bf51b34d9a004db6e78c15","_number":662985,"virtual_id_number":662985,"owner":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},{"value":0,"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":2,"date":"2019-06-06 00:35:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-04 11:59:11.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"all":[{"value":2,"date":"2019-06-05 20:24:30.000000000","_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},{"value":1,"date":"2019-06-05 07:29:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"all":[{"value":1,"date":"2019-06-05 20:24:30.000000000","_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},{"value":0,"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-04 11:59:11.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-05 07:29:57.000000000","updated_by":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"reviewer":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"state":"REVIEWER"},{"updated":"2019-06-05 20:24:30.000000000","updated_by":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"reviewer":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"state":"REVIEWER"},{"updated":"2019-06-06 00:35:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6353ab54b4700b746fbc1ecb8626daf58714ade5","author":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"date":"2019-06-04 09:25:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7c7f85d85a97d36f5a79a45f618902bceb3342bd","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-04 11:59:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/85/662985/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/68510a3/ : SUCCESS in 2h 32m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"70e3a98790d0cfe669adb4aeb087d67837633ac1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-04 13:07:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-oooq http://logs.openstack.org/85/662985/1/check/tripleo-ci-centos-7-undercloud-oooq/5526cff/ : SUCCESS in 1h 36m 43s\n- tripleo-ci-centos-7-undercloud-upgrades http://logs.openstack.org/85/662985/1/check/tripleo-ci-centos-7-undercloud-upgrades/df43385/ : SUCCESS in 2h 22m 58s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/85/662985/1/check/tripleo-ci-centos-7-undercloud-containers/1563c9f/ : SUCCESS in 1h 31m 05s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/85/662985/1/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/6672606/ : SUCCESS in 2h 09m 05s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/85/662985/1/check/tripleo-ci-centos-7-containers-multinode/1133424/ : SUCCESS in 2h 05m 58s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/85/662985/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/ec0e735/ : SUCCESS in 2h 09m 12s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/85/662985/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/68a6866/ : SUCCESS in 2h 44m 28s (non-voting)\n- tripleo-ci-centos-7-standalone-rocky http://logs.openstack.org/85/662985/1/check/tripleo-ci-centos-7-standalone-rocky/a064f8a/ : SUCCESS in 1h 44m 35s\n- puppet-openstack-lint http://logs.openstack.org/85/662985/1/check/puppet-openstack-lint/fae1bb7/ : SUCCESS in 5m 25s\n- puppet-openstack-syntax-4 http://logs.openstack.org/85/662985/1/check/puppet-openstack-syntax-4/13ea5ac/ : SUCCESS in 5m 34s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/85/662985/1/check/puppet-openstack-unit-4.8-centos-7/f4d2d75/ : SUCCESS in 32m 54s\n- tripleo-ci-centos-7-scenario004-multinode-oooq-container http://logs.openstack.org/85/662985/1/check/tripleo-ci-centos-7-scenario004-multinode-oooq-container/fad8951/ : SUCCESS in 2h 35m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"52b23293d149126d9ad321bd83ce92a413242937","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2019-06-05 07:29:57.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b52f770a63e7218c60733a3e1cd5f07919ccbcd7","author":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"date":"2019-06-05 20:24:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nclean cherry pick","accounts_in_message":[],"_revision_number":1},{"id":"0ae96f4509e88d17ec13c655eb417b837481c83d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 20:24:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e4d0aebfc090fe3ef962a7c7e1d88408d5b17a23","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 00:35:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-undercloud-oooq http://logs.openstack.org/85/662985/1/gate/tripleo-ci-centos-7-undercloud-oooq/3b0dd28/ : SUCCESS in 1h 15m 22s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/85/662985/1/gate/tripleo-ci-centos-7-undercloud-containers/09916dc/ : SUCCESS in 1h 25m 01s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/85/662985/1/gate/tripleo-ci-centos-7-containers-multinode/d535b10/ : SUCCESS in 2h 10m 53s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/85/662985/1/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/4281640/ : SUCCESS in 2h 27m 06s\n- puppet-openstack-lint http://logs.openstack.org/85/662985/1/gate/puppet-openstack-lint/3fb580f/ : SUCCESS in 4m 28s\n- puppet-openstack-syntax-4 http://logs.openstack.org/85/662985/1/gate/puppet-openstack-syntax-4/b678497/ : SUCCESS in 4m 10s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/85/662985/1/gate/puppet-openstack-unit-4.8-centos-7/a6fb6a4/ : SUCCESS in 30m 34s","accounts_in_message":[],"_revision_number":1},{"id":"5ddf2e81b4909c81046283a85b8e2d15eca79867","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 00:35:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7ae85edf954e1222ceadfade90a04343cbf920d7","revisions":{"7ae85edf954e1222ceadfade90a04343cbf920d7":{"kind":"REWORK","_number":1,"created":"2019-06-04 09:25:22.000000000","uploader":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"ref":"refs/changes/85/662985/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/85/662985/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/85/662985/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/85/662985/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/85/662985/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/85/662985/1"}}},"commit":{"parents":[{"commit":"d7fe21baf5c8b09b0fc9f9a4338997415ec82514","subject":"Merge \"Fix tripleo::haproxy::stats to be more correct and flexible\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/d7fe21baf5c8b09b0fc9f9a4338997415ec82514"}]}],"author":{"name":"Giulio Fidente","email":"gfidente@redhat.com","date":"2019-05-24 13:13:47.000000000","tz":120},"committer":{"name":"Giulio Fidente","email":"gfidente@redhat.com","date":"2019-06-04 09:23:10.000000000","tz":120},"subject":"Add support for external Ganesha service in Manila Share profile","message":"Add support for external Ganesha service in Manila Share profile\n\nWe might want to connect Manila Share to an external Ganesha service\nand in that case need to provide additional configs so that we can\nconnect to the remote node via SSH.\n\nChange-Id: I1741721ede9f9f469e2554b409ec6e0e6c7c22a2\nImplements: blueprint manila-external-ganesha\nDepends-On: I31fff6a1b6fede860b8aaee0079183924e8b31d6\n(cherry picked from commit dabfc9e1e3199d59d79b912a1d0c240470ef6edb)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/7ae85edf954e1222ceadfade90a04343cbf920d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/7ae85edf954e1222ceadfade90a04343cbf920d7"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
