)]}'
{"id":"openstack%2Fpuppet-tripleo~663288","triplet_id":"openstack%2Fpuppet-tripleo~stable%2Frocky~I572d53bb6977e16d1deffe0127e17e48a8cda29c","project":"openstack/puppet-tripleo","branch":"stable/rocky","hashtags":[],"change_id":"I572d53bb6977e16d1deffe0127e17e48a8cda29c","subject":"Only request certs in step 1","status":"MERGED","created":"2019-06-05 09:35:35.000000000","updated":"2019-06-08 01:45:54.000000000","submitted":"2019-06-08 01:45:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"663288-1559958354685-d9d96933","meta_rev_id":"4e54181b4b674d150c5610fb912dedba8a887511","_number":663288,"virtual_id_number":663288,"owner":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":0,"_account_id":11589,"name":"Harry Rybacki","email":"hrybacki@redhat.com","username":"hrybacki"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":9098,"name":"Nathan Kinder","email":"nkinder@redhat.com","username":"nkinder"},{"value":2,"date":"2019-06-08 01:45:54.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2019-06-06 17:24:51.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":0,"_account_id":11589,"name":"Harry Rybacki","email":"hrybacki@redhat.com","username":"hrybacki"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2019-06-06 21:55:47.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":1,"date":"2019-06-07 11:12:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9098,"name":"Nathan Kinder","email":"nkinder@redhat.com","username":"nkinder"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":0,"_account_id":11589,"name":"Harry Rybacki","email":"hrybacki@redhat.com","username":"hrybacki"},{"value":0,"date":"2019-06-07 10:53:39.000000000","_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2019-06-07 14:02:15.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":9098,"name":"Nathan Kinder","email":"nkinder@redhat.com","username":"nkinder"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"_account_id":9098,"name":"Nathan Kinder","email":"nkinder@redhat.com","username":"nkinder"},{"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},{"_account_id":11589,"name":"Harry Rybacki","email":"hrybacki@redhat.com","username":"hrybacki"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-05 09:35:35.000000000","updated_by":{"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},"reviewer":{"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},"state":"REVIEWER"},{"updated":"2019-06-05 10:39:41.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-06-05 10:39:58.000000000","updated_by":{"_account_id":11589,"name":"Harry Rybacki","email":"hrybacki@redhat.com","username":"hrybacki"},"reviewer":{"_account_id":11589,"name":"Harry Rybacki","email":"hrybacki@redhat.com","username":"hrybacki"},"state":"REVIEWER"},{"updated":"2019-06-05 10:40:02.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"},{"updated":"2019-06-05 10:42:29.000000000","updated_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"reviewer":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"state":"REVIEWER"},{"updated":"2019-06-06 17:24:51.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-07 10:53:39.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2019-06-07 11:12:15.000000000","updated_by":{"_account_id":9098,"name":"Nathan Kinder","email":"nkinder@redhat.com","username":"nkinder"},"reviewer":{"_account_id":9098,"name":"Nathan Kinder","email":"nkinder@redhat.com","username":"nkinder"},"state":"REVIEWER"},{"updated":"2019-06-07 14:02:15.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-06-08 01:45:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1c39e9c9d916b5ad3587e1a69edec5a7d6752bfe","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2019-06-05 09:35:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f48b5915422a96c63f8ceea4372a51723842f800","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-05 12:02:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/88/663288/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/f3fb8af/ : SUCCESS in 2h 14m 57s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/88/663288/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/5aa1665/ : SUCCESS in 2h 13m 17s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/88/663288/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/edf5a62/ : SUCCESS in 2h 25m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3395779e84dc782f995f2ee327bb912e768ed8ec","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 12:51:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-oooq http://logs.openstack.org/88/663288/1/check/tripleo-ci-centos-7-undercloud-oooq/37d98d6/ : FAILURE in 21m 03s\n- tripleo-ci-centos-7-undercloud-upgrades http://logs.openstack.org/88/663288/1/check/tripleo-ci-centos-7-undercloud-upgrades/390b0a1/ : SUCCESS in 1h 59m 10s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/88/663288/1/check/tripleo-ci-centos-7-undercloud-containers/152b02e/ : SUCCESS in 1h 34m 02s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/88/663288/1/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/524865c/ : SUCCESS in 2h 05m 46s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/88/663288/1/check/tripleo-ci-centos-7-containers-multinode/ff3501e/ : SUCCESS in 2h 05m 58s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/88/663288/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/d60d184/ : SUCCESS in 2h 20m 16s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/88/663288/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/b7c9822/ : SUCCESS in 2h 44m 48s (non-voting)\n- tripleo-ci-centos-7-standalone-rocky http://logs.openstack.org/88/663288/1/check/tripleo-ci-centos-7-standalone-rocky/8aa2a79/ : SUCCESS in 1h 37m 34s\n- puppet-openstack-lint http://logs.openstack.org/88/663288/1/check/puppet-openstack-lint/180d2f4/ : SUCCESS in 6m 36s\n- puppet-openstack-syntax-4 http://logs.openstack.org/88/663288/1/check/puppet-openstack-syntax-4/1cdf8fa/ : SUCCESS in 6m 42s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/88/663288/1/check/puppet-openstack-unit-4.8-centos-7/1a32bf8/ : SUCCESS in 32m 20s","accounts_in_message":[],"_revision_number":1},{"id":"46684517cee363daeb07cd39b55cd33c27432f0d","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2019-06-05 13:18:18.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2ff53c4d65dbf39d19206709e755020503f7a2c1","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-05 13:41:09.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d20ee7c239fee756ee83ec3c1fd75e75d44d3e89","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-05 15:51:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/88/663288/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/e7b619c/ : SUCCESS in 2h 09m 31s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/88/663288/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/3181076/ : SUCCESS in 2h 11m 24s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/88/663288/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/d6183f8/ : SUCCESS in 2h 31m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"662574c91af086a03e163db0d4de2cf0e38a65c3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 20:05:39.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-oooq http://logs.openstack.org/88/663288/1/check/tripleo-ci-centos-7-undercloud-oooq/d7276c2/ : FAILURE in 23m 40s\n- tripleo-ci-centos-7-undercloud-upgrades http://logs.openstack.org/88/663288/1/check/tripleo-ci-centos-7-undercloud-upgrades/fd9cc94/ : SUCCESS in 1h 59m 56s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/88/663288/1/check/tripleo-ci-centos-7-undercloud-containers/cc7ab06/ : SUCCESS in 1h 36m 52s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/88/663288/1/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/dd01080/ : SUCCESS in 1h 49m 11s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/88/663288/1/check/tripleo-ci-centos-7-containers-multinode/0977ab8/ : SUCCESS in 2h 09m 12s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/88/663288/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/7aac81c/ : SUCCESS in 2h 37m 56s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/88/663288/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/a5a5440/ : SUCCESS in 2h 44m 21s (non-voting)\n- tripleo-ci-centos-7-standalone-rocky http://logs.openstack.org/88/663288/1/check/tripleo-ci-centos-7-standalone-rocky/0bffa66/ : SUCCESS in 1h 43m 06s\n- puppet-openstack-lint http://logs.openstack.org/88/663288/1/check/puppet-openstack-lint/14bf358/ : SUCCESS in 4m 47s\n- puppet-openstack-syntax-4 http://logs.openstack.org/88/663288/1/check/puppet-openstack-syntax-4/5dd7450/ : SUCCESS in 4m 17s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/88/663288/1/check/puppet-openstack-unit-4.8-centos-7/d20c020/ : SUCCESS in 32m 12s","accounts_in_message":[],"_revision_number":1},{"id":"2103d38bbbcb0366192543840dc53f9647a4c447","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-05 20:15:21.000000000","message":"Patch Set 1: Code-Review-1\n\nOh this breaks instack-undercloud because we have no concept of step","accounts_in_message":[],"_revision_number":1},{"id":"658f9cc8eeb8a078ec2e5b5d5afcc1a27064cddf","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2019-06-06 11:57:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b040905b4312271cd3733fd57bf09c080f167d7a","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2019-06-06 11:59:29.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"26f21552da06ac9856725586b13dd22424378176","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2019-06-06 12:00:17.000000000","message":"Patch Set 3:\n\nAdded default value to cope with instack-undercloud.","accounts_in_message":[],"_revision_number":3},{"id":"d0c07d2ad394ab16f1e9ea25b7e1ab6d32de635a","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-06 13:28:25.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f734420414704270c8999e82f4ac26293aaab7a5","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2019-06-06 13:44:20.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"df57a21c9101537508eabb1aff38ef314923c682","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-06 17:24:51.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/88/663288/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/133e553/ : TIMED_OUT in 3h 39m 45s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/88/663288/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/6b3d2ea/ : SUCCESS in 2h 13m 30s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/88/663288/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/90b77e2/ : SUCCESS in 2h 50m 10s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"a0e37c798ce971955120c9b863b8be31f8874148","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 21:39:27.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-oooq http://logs.openstack.org/88/663288/4/check/tripleo-ci-centos-7-undercloud-oooq/e1d6961/ : SUCCESS in 1h 03m 16s\n- tripleo-ci-centos-7-undercloud-upgrades http://logs.openstack.org/88/663288/4/check/tripleo-ci-centos-7-undercloud-upgrades/e374530/ : SUCCESS in 1h 58m 08s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/88/663288/4/check/tripleo-ci-centos-7-undercloud-containers/2e2bd2f/ : SUCCESS in 1h 40m 59s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/88/663288/4/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/a07e696/ : SUCCESS in 1h 51m 33s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/88/663288/4/check/tripleo-ci-centos-7-containers-multinode/e2c5c9a/ : SUCCESS in 2h 15m 38s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/88/663288/4/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/87336ac/ : SUCCESS in 2h 19m 33s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/88/663288/4/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/d62cb8e/ : SUCCESS in 2h 41m 09s (non-voting)\n- tripleo-ci-centos-7-standalone-rocky http://logs.openstack.org/88/663288/4/check/tripleo-ci-centos-7-standalone-rocky/5f3837f/ : SUCCESS in 1h 39m 04s\n- puppet-openstack-lint http://logs.openstack.org/88/663288/4/check/puppet-openstack-lint/bbfb9b9/ : SUCCESS in 4m 31s\n- puppet-openstack-syntax-4 http://logs.openstack.org/88/663288/4/check/puppet-openstack-syntax-4/524b667/ : SUCCESS in 5m 06s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/88/663288/4/check/puppet-openstack-unit-4.8-centos-7/8f48739/ : SUCCESS in 35m 00s","accounts_in_message":[],"_revision_number":4},{"id":"fa8132bced70e9496ed4685bb0cea30ccccc18f7","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-06 21:55:47.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"dac34ae8f84c6e6263e45c5d2a3a49c4a5bb5484","author":{"_account_id":9098,"name":"Nathan Kinder","email":"nkinder@redhat.com","username":"nkinder"},"date":"2019-06-07 11:12:15.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"b41940a6eaea3ba4824d41e91a589abe930cd043","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-07 14:02:15.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"6a310d427cafd6878b75edfa185352d429d56ec9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 14:02:32.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"de034a5b8feaec0dcbdf51f8f443a679a8ea067b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 21:06:31.000000000","message":"Patch Set 4:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"6a05fd5d3f01983e1da78e598a6cb288ef995249","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-08 01:45:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"4e54181b4b674d150c5610fb912dedba8a887511","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-08 01:45:54.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-undercloud-oooq http://logs.openstack.org/88/663288/4/gate/tripleo-ci-centos-7-undercloud-oooq/66601cb/ : SUCCESS in 1h 07m 26s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/88/663288/4/gate/tripleo-ci-centos-7-undercloud-containers/32eca75/ : SUCCESS in 1h 34m 40s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/88/663288/4/gate/tripleo-ci-centos-7-containers-multinode/dbd3cd0/ : SUCCESS in 2h 04m 44s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/88/663288/4/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/f649b17/ : SUCCESS in 2h 25m 05s\n- puppet-openstack-lint http://logs.openstack.org/88/663288/4/gate/puppet-openstack-lint/b830d24/ : SUCCESS in 4m 43s\n- puppet-openstack-syntax-4 http://logs.openstack.org/88/663288/4/gate/puppet-openstack-syntax-4/93ffd05/ : SUCCESS in 4m 13s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/88/663288/4/gate/puppet-openstack-unit-4.8-centos-7/2b39c05/ : SUCCESS in 35m 41s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"efc1f0245bdc3afcc5a214060f84441f5496e592","revisions":{"b3ffee291be6d5c7364d1b562043cb5587212aa0":{"kind":"REWORK","_number":1,"created":"2019-06-05 09:35:35.000000000","uploader":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"ref":"refs/changes/88/663288/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/88/663288/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/88/663288/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/88/663288/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/88/663288/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/88/663288/1"}}},"commit":{"parents":[{"commit":"d7fe21baf5c8b09b0fc9f9a4338997415ec82514","subject":"Merge \"Fix tripleo::haproxy::stats to be more correct and flexible\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/d7fe21baf5c8b09b0fc9f9a4338997415ec82514"}]}],"author":{"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","date":"2019-01-30 13:39:23.000000000","tz":120},"committer":{"name":"Moisés Guimarães de Medeiros","email":"moguimar@redhat.com","date":"2019-06-05 09:35:24.000000000","tz":120},"subject":"Only request certs in step 1","message":"Only request certs in step 1\n\nThe certmonger-user service used to run in every step. This limits it to\nonly step 1.\n\nChange-Id: I572d53bb6977e16d1deffe0127e17e48a8cda29c\n(cherry picked from commit 98e9879fd55194ed86c294158cabbf21ec1c903f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/b3ffee291be6d5c7364d1b562043cb5587212aa0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/b3ffee291be6d5c7364d1b562043cb5587212aa0"}]},"branch":"refs/heads/stable/rocky"},"9c6ff0df5a9c2d7f30b28a8db3ae5ed2a279f8ca":{"kind":"REWORK","_number":2,"created":"2019-06-06 11:57:58.000000000","uploader":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"ref":"refs/changes/88/663288/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/88/663288/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/88/663288/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/88/663288/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/88/663288/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/88/663288/2"}}},"commit":{"parents":[{"commit":"4c390cf8950e2e783f277f0e93977bd065ba1b20","subject":"Merge \"Allow external Ganesha for the cephfs manila backend\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/4c390cf8950e2e783f277f0e93977bd065ba1b20"}]}],"author":{"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","date":"2019-01-30 13:39:23.000000000","tz":120},"committer":{"name":"Moisés Guimarães de Medeiros","email":"moguimar@redhat.com","date":"2019-06-06 11:57:49.000000000","tz":120},"subject":"Only request certs in step 1","message":"Only request certs in step 1\n\nThe certmonger-user service used to run in every step. This limits it to\nonly step 1.\n\nChange-Id: I572d53bb6977e16d1deffe0127e17e48a8cda29c\n(cherry picked from commit 98e9879fd55194ed86c294158cabbf21ec1c903f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/9c6ff0df5a9c2d7f30b28a8db3ae5ed2a279f8ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/9c6ff0df5a9c2d7f30b28a8db3ae5ed2a279f8ca"}]},"branch":"refs/heads/stable/rocky"},"27bcfb74c0c917ba0cb3ab4908ade8f052525179":{"kind":"REWORK","_number":3,"created":"2019-06-06 11:59:29.000000000","uploader":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"ref":"refs/changes/88/663288/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/88/663288/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/88/663288/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/88/663288/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/88/663288/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/88/663288/3"}}},"commit":{"parents":[{"commit":"4c390cf8950e2e783f277f0e93977bd065ba1b20","subject":"Merge \"Allow external Ganesha for the cephfs manila backend\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/4c390cf8950e2e783f277f0e93977bd065ba1b20"}]}],"author":{"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","date":"2019-01-30 13:39:23.000000000","tz":120},"committer":{"name":"Moisés Guimarães de Medeiros","email":"moguimar@redhat.com","date":"2019-06-06 11:59:21.000000000","tz":120},"subject":"Only request certs in step 1","message":"Only request certs in step 1\n\nThe certmonger-user service used to run in every step. This limits it to\nonly step 1.\n\nChange-Id: I572d53bb6977e16d1deffe0127e17e48a8cda29c\n(cherry picked from commit 98e9879fd55194ed86c294158cabbf21ec1c903f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/27bcfb74c0c917ba0cb3ab4908ade8f052525179"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/27bcfb74c0c917ba0cb3ab4908ade8f052525179"}]},"branch":"refs/heads/stable/rocky"},"efc1f0245bdc3afcc5a214060f84441f5496e592":{"kind":"REWORK","_number":4,"created":"2019-06-06 13:44:20.000000000","uploader":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"ref":"refs/changes/88/663288/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/88/663288/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/88/663288/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/88/663288/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/88/663288/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/88/663288/4"}}},"commit":{"parents":[{"commit":"4c390cf8950e2e783f277f0e93977bd065ba1b20","subject":"Merge \"Allow external Ganesha for the cephfs manila backend\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/4c390cf8950e2e783f277f0e93977bd065ba1b20"}]}],"author":{"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","date":"2019-01-30 13:39:23.000000000","tz":120},"committer":{"name":"Moisés Guimarães de Medeiros","email":"moguimar@redhat.com","date":"2019-06-06 13:44:06.000000000","tz":120},"subject":"Only request certs in step 1","message":"Only request certs in step 1\n\nThe certmonger-user service used to run in every step. This limits it to\nonly step 1.\n\nChange-Id: I572d53bb6977e16d1deffe0127e17e48a8cda29c\n(cherry picked from commit 98e9879fd55194ed86c294158cabbf21ec1c903f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/efc1f0245bdc3afcc5a214060f84441f5496e592"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/efc1f0245bdc3afcc5a214060f84441f5496e592"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
