)]}'
{"id":"openstack%2Fpuppet-tripleo~663874","triplet_id":"openstack%2Fpuppet-tripleo~stable%2Fqueens~I119fa1da61f2385c672391beb4071f0ac2731f21","project":"openstack/puppet-tripleo","branch":"stable/queens","hashtags":[],"change_id":"I119fa1da61f2385c672391beb4071f0ac2731f21","subject":"Only request certs in step 1","status":"MERGED","created":"2019-06-07 09:51:37.000000000","updated":"2019-06-11 03:24:14.000000000","submitted":"2019-06-11 03:24:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"663874-1560223454712-8840775f","meta_rev_id":"599aaac62de1373e75c7cb5c9463c879a8c911cc","_number":663874,"virtual_id_number":663874,"owner":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-06-11 03:24:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-07 11:59:02.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"recommended":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2019-06-10 06:22:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},{"value":2,"date":"2019-06-07 14:02:29.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"date":"2019-06-07 10:54:04.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"date":"2019-06-07 09:51:37.000000000","_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},{"value":1,"date":"2019-06-10 13:27:00.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"date":"2019-06-07 10:53:51.000000000","_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-07 09:51:37.000000000","updated_by":{"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},"reviewer":{"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},"state":"REVIEWER"},{"updated":"2019-06-07 10:53:51.000000000","updated_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"reviewer":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"state":"REVIEWER"},{"updated":"2019-06-07 10:54:04.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-06-07 11:59:02.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-10 06:22:43.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2019-06-10 13:27:00.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-06-11 03:24:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b84fc54d301e788ebb627063cf0bef418b33ccdb","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2019-06-07 09:51:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"703b6ad66f12a8e4a945a16b9611c391f0b6fc31","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-07 11:59:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/74/663874/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/e16d4a4/ : SUCCESS in 2h 06m 34s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/74/663874/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/3764cc4/ : SUCCESS in 2h 00m 10s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/74/663874/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/9dd19a1/ : FAILURE in 1h 58m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e99e137ea8b0b59d5ccda280e7ec85617404c127","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 13:06:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-oooq http://logs.openstack.org/74/663874/1/check/tripleo-ci-centos-7-undercloud-oooq/ceea8e8/ : SUCCESS in 1h 12m 55s\n- tripleo-ci-centos-7-undercloud-upgrades http://logs.openstack.org/74/663874/1/check/tripleo-ci-centos-7-undercloud-upgrades/ade89e4/ : SUCCESS in 1h 16m 58s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/74/663874/1/check/tripleo-ci-centos-7-containers-multinode/28e2266/ : SUCCESS in 2h 00m 21s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/74/663874/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/759525c/ : FAILURE in 1h 49m 39s (non-voting)\n- puppet-openstack-lint http://logs.openstack.org/74/663874/1/check/puppet-openstack-lint/16d3405/ : SUCCESS in 4m 00s\n- puppet-openstack-syntax-4 http://logs.openstack.org/74/663874/1/check/puppet-openstack-syntax-4/be15ae9/ : SUCCESS in 3m 47s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/74/663874/1/check/puppet-openstack-unit-4.8-centos-7/2796b60/ : SUCCESS in 28m 10s","accounts_in_message":[],"_revision_number":1},{"id":"cc4e83af23a556ad49b5607db838ed31f34fed1b","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-07 14:02:29.000000000","message":"Patch Set 1: Code-Review+2\n\nwaiting for https://review.opendev.org/#/c/663288/ to land","accounts_in_message":[],"_revision_number":1},{"id":"df57475d57158478a26d6829154aad1ee2950ba7","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2019-06-10 06:22:43.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f9933ed02e7b0ca69ddf71b78e7766baeefcf9ac","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2019-06-10 06:22:55.000000000","message":"Patch Set 1:\n\nJobs passing this time","accounts_in_message":[],"_revision_number":1},{"id":"b6f6dda9c33249157c8800425c3430de0fbde8bc","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-10 13:27:00.000000000","message":"Patch Set 1: Workflow+1\n\nrocky version has landed","accounts_in_message":[],"_revision_number":1},{"id":"81ca182fa4eae5a394c7b8ae15019729cc22e482","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 13:27:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"87ddd9b2eb05c36966044a8bb9b3e9d0ebf1d364","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 03:24:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-undercloud-oooq http://logs.openstack.org/74/663874/1/gate/tripleo-ci-centos-7-undercloud-oooq/7c1349f/ : SUCCESS in 1h 22m 05s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/74/663874/1/gate/tripleo-ci-centos-7-containers-multinode/6540027/ : SUCCESS in 2h 10m 56s\n- puppet-openstack-lint http://logs.openstack.org/74/663874/1/gate/puppet-openstack-lint/76f1f62/ : SUCCESS in 4m 33s\n- puppet-openstack-syntax-4 http://logs.openstack.org/74/663874/1/gate/puppet-openstack-syntax-4/0351812/ : SUCCESS in 4m 13s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/74/663874/1/gate/puppet-openstack-unit-4.8-centos-7/4389c17/ : SUCCESS in 31m 06s","accounts_in_message":[],"_revision_number":1},{"id":"599aaac62de1373e75c7cb5c9463c879a8c911cc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 03:24:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"78f21fcf3ba3c5c2586d68c521440022dcd57305","revisions":{"78f21fcf3ba3c5c2586d68c521440022dcd57305":{"kind":"REWORK","_number":1,"created":"2019-06-07 09:51:37.000000000","uploader":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"ref":"refs/changes/74/663874/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/74/663874/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/74/663874/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/74/663874/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/74/663874/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/74/663874/1"}}},"commit":{"parents":[{"commit":"3e3af60814c2f3a83f9d0828e9be11e58feafc89","subject":"Merge \"Revert \"Refactor collectd/gnocchi integration\"\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/3e3af60814c2f3a83f9d0828e9be11e58feafc89"}]}],"author":{"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","date":"2019-01-30 13:39:23.000000000","tz":120},"committer":{"name":"Moisés Guimarães de Medeiros","email":"moguimar@redhat.com","date":"2019-06-07 09:51:25.000000000","tz":120},"subject":"Only request certs in step 1","message":"Only request certs in step 1\n\nThe certmonger-user service used to run in every step. This limits it to\nonly step 1.\n\n(cherry picked from commit 98e9879fd55194ed86c294158cabbf21ec1c903f)\nSigned-off-by: Moisés Guimarães de Medeiros \u003cmoguimar@redhat.com\u003e\n\nChange-Id: I119fa1da61f2385c672391beb4071f0ac2731f21\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/78f21fcf3ba3c5c2586d68c521440022dcd57305"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/78f21fcf3ba3c5c2586d68c521440022dcd57305"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
