)]}'
{"id":"openstack%2Fpuppet-tripleo~668544","triplet_id":"openstack%2Fpuppet-tripleo~master~If1be877b265f36078a5e66a13b7ab55db699af0f","project":"openstack/puppet-tripleo","branch":"master","topic":"format-markdown","hashtags":[],"change_id":"If1be877b265f36078a5e66a13b7ab55db699af0f","subject":"Force to use markdown to prevent pypi issue","status":"MERGED","created":"2019-07-01 18:45:03.000000000","updated":"2019-07-15 19:47:03.000000000","submitted":"2019-07-15 19:47:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"668544-1563220023353-9977374c","meta_rev_id":"9d74ac8dd5515b31f8ab5750a877ccd6c116a3e9","_number":668544,"virtual_id_number":668544,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-07-15 19:47:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2019-07-15 13:41:46.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2019-07-15 16:39:41.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2019-07-15 16:39:41.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-15 13:41:46.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-07-15 16:39:41.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-07-15 19:47:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e09da469d73a4597af9426e9c498a912cf72d456","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-07-01 18:45:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3d403d9f6f384fb3b9e272a12563e2c37ed00f00","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-02 02:04:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/44/668544/1/check/puppet-openstack-lint-ubuntu-bionic/20867ce/ : SUCCESS in 6m 03s\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/44/668544/1/check/tripleo-ci-centos-7-standalone/500954d/ : SUCCESS in 1h 51m 17s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/44/668544/1/check/tripleo-ci-fedora-28-standalone/faa6d06/ : SUCCESS in 1h 53m 19s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/44/668544/1/check/tripleo-ci-centos-7-standalone-upgrade/8a66efa/ : SUCCESS in 2h 29m 45s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/44/668544/1/check/tripleo-ci-centos-7-standalone-upgrade-stein/fde8988/ : SUCCESS in 2h 05m 51s","accounts_in_message":[],"_revision_number":1},{"id":"9d1f060320978828db16ddc76e2397f7cd458731","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-07-08 19:30:01.000000000","message":"Patch Set 1:\n\nWe don\u0027t publish puppet-* to pypi so not sure this is necessary","accounts_in_message":[],"_revision_number":1},{"id":"7a8181ea03d750bd5354eb8949f77bf78515f901","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-07-15 13:41:46.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2cdd3109f37dfe6ac78bbd800e0813bbfc63579c","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-07-15 16:39:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3820c2bef964ceb18a8769cfe2abff100b0bc9e6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-15 17:17:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"69a804a4439e869b9449827dad96f34156f287a3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-15 19:47:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"86069874b05bb831ee0c4608188a2862f502d8b0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-15 19:47:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/44/668544/1/gate/puppet-openstack-lint-ubuntu-bionic/94f0c73/ : SUCCESS in 5m 31s\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/44/668544/1/gate/tripleo-ci-centos-7-standalone/9172b46/ : SUCCESS in 1h 52m 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"03dd75eb17ae04044476c2f2e3180cb7d1b2d40a","revisions":{"03dd75eb17ae04044476c2f2e3180cb7d1b2d40a":{"kind":"REWORK","_number":1,"created":"2019-07-01 18:45:03.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/44/668544/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/44/668544/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/44/668544/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/44/668544/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/44/668544/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/44/668544/1"}}},"commit":{"parents":[{"commit":"f30590e8d42789e0ffe510b4f565a1914b3a5ba1","subject":"Merge \"Allow enabling custom backends with manila\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/f30590e8d42789e0ffe510b4f565a1914b3a5ba1"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2019-07-01 18:44:46.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2019-07-01 18:45:00.000000000","tz":120},"subject":"Force to use markdown to prevent pypi issue","message":"Force to use markdown to prevent pypi issue\n\nRecently the pypi team have reintroduce some deactivated tests\non package deployment and especially tests concerning long_description\nsyntaxe.\n\nIf projects don\u0027t specify that they use the markdown format the\npackage deployment will fail if something went wrong in the markdown format.\n\nBy example if a project use inline literal this can be an issue due to\nthe fact that if `long_description_content_type` is missing [1]. The\ndefault format will be `text/x-rst` where code block is different from\nmarkdown.\n\nThese changes force to use the markdown format to avoid this kind of issue\non project who use `README` at markdown format.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-July/007459.html\n\nFor more details:\nhttps://github.com/pypa/warehouse/issues/5890#issuecomment-494868157\n\nChange-Id: If1be877b265f36078a5e66a13b7ab55db699af0f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/03dd75eb17ae04044476c2f2e3180cb7d1b2d40a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/03dd75eb17ae04044476c2f2e3180cb7d1b2d40a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
