)]}'
{"id":"openstack%2Fpuppet-tripleo~669799","triplet_id":"openstack%2Fpuppet-tripleo~stable%2Fstein~I798304a472df41b86a88611c97c2c99131faa0ad","project":"openstack/puppet-tripleo","branch":"stable/stein","topic":"add_dateext-stable/stein","hashtags":[],"change_id":"I798304a472df41b86a88611c97c2c99131faa0ad","subject":"Add dateext and related parameters for logrotate","status":"MERGED","created":"2019-07-09 06:44:07.000000000","updated":"2019-07-10 23:50:17.000000000","submitted":"2019-07-10 23:47:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"669799-1562802439870-6e4dd5bd","meta_rev_id":"41f9b25aaf968309a6f72d392d3b38b10ffc4cbd","_number":669799,"virtual_id_number":669799,"owner":{"_account_id":25613,"name":"Keigo Noha","email":"knoha@redhat.com","username":"knoha"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-07-10 23:47:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2019-07-09 10:35:31.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"recommended":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"all":[{"value":1,"date":"2019-07-09 14:04:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"value":2,"date":"2019-07-10 20:51:03.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"value":1,"date":"2019-07-10 20:51:03.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-09 10:35:31.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-09 14:04:59.000000000","updated_by":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"reviewer":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"state":"REVIEWER"},{"updated":"2019-07-10 20:51:03.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-07-10 23:47:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"efb8c39081560db642514b1230e6512335dd43d4","author":{"_account_id":25613,"name":"Keigo Noha","email":"knoha@redhat.com","username":"knoha"},"date":"2019-07-09 06:44:07.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"5416b2641d7cb5f2a7a5608e570e02540a8ed578","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-09 10:05:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/99/669799/1/check/puppet-openstack-lint-ubuntu-bionic/a794c0e/ : SUCCESS in 5m 30s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/99/669799/1/check/puppet-openstack-syntax-4-ubuntu-bionic/5ad2dfe/ : SUCCESS in 6m 06s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/99/669799/1/check/puppet-openstack-syntax-5-ubuntu-bionic/df7de0b/ : SUCCESS in 5m 05s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/99/669799/1/check/puppet-openstack-syntax-6-ubuntu-bionic/afc4bd7/ : SUCCESS in 4m 16s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/99/669799/1/check/puppet-openstack-unit-4.8-centos-7/1ff510b/ : SUCCESS in 32m 34s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/99/669799/1/check/puppet-openstack-unit-5.5-centos-7/4313b17/ : SUCCESS in 29m 41s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/99/669799/1/check/tripleo-ci-centos-7-undercloud-containers/f32e288/ : SUCCESS in 1h 30m 32s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/99/669799/1/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/8f70ffb/ : FAILURE in 1h 36m 10s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/99/669799/1/check/tripleo-ci-centos-7-containers-multinode/3984e3c/ : SUCCESS in 2h 21m 31s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/99/669799/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/5699646/ : SUCCESS in 2h 28m 33s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/99/669799/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/78cf2dd/ : SUCCESS in 2h 24m 08s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/99/669799/1/check/tripleo-ci-centos-7-standalone/03903aa/ : SUCCESS in 1h 49m 49s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/99/669799/1/check/tripleo-ci-fedora-28-standalone/4068b00/ : SUCCESS in 2h 07m 48s (non-voting)\n- build-openstack-releasenotes http://logs.openstack.org/99/669799/1/check/build-openstack-releasenotes/66b2268/html/ : SUCCESS in 3m 35s\n- tripleo-ci-centos-7-scenario001-standalone http://logs.openstack.org/99/669799/1/check/tripleo-ci-centos-7-scenario001-standalone/ae21537/ : SUCCESS in 1h 54m 55s","accounts_in_message":[],"_revision_number":1},{"id":"3451a3e0f2d10b4b5582beb0fd9e2e1d83b783ec","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-09 10:35:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/99/669799/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/7aa03f1/ : FAILURE in 1h 39m 02s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 finger://ze02.softwarefactory-project.io:7979/a12bb884eba346e1af7c21e052f6c786 : POST_FAILURE in 3h 30m 29s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 finger://ze01.softwarefactory-project.io:7979/b80b7a69f3034b2dbfa354c0b5c7c7f2 : POST_FAILURE in 3h 49m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7b4905464351f69cbb163745b4178c891c0a04b9","author":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"date":"2019-07-09 14:04:59.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"8b0e61bec815cbcf7d42ab6ccce18d8ab7817aed","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-07-10 15:02:12.000000000","message":"Patch Set 1:\n\nrecheck (checking something)","accounts_in_message":[],"_revision_number":1},{"id":"c1fd246e3a537fa486271171ca507b5d4e4949ad","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-10 16:18:05.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/99/669799/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/7b6de5f/ : FAILURE in 1h 12m 54s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 : NODE_FAILURE in 0s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8cbfdbb1da70074e418dd32a522eb6c38698f170","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 20:32:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/99/669799/1/check/puppet-openstack-lint-ubuntu-bionic/9457cf7/ : SUCCESS in 5m 10s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/99/669799/1/check/puppet-openstack-syntax-4-ubuntu-bionic/7d9e052/ : SUCCESS in 4m 40s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/99/669799/1/check/puppet-openstack-syntax-5-ubuntu-bionic/1cee807/ : SUCCESS in 4m 39s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/99/669799/1/check/puppet-openstack-syntax-6-ubuntu-bionic/079bb8a/ : SUCCESS in 5m 12s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/99/669799/1/check/puppet-openstack-unit-4.8-centos-7/a8ffa08/ : SUCCESS in 39m 00s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/99/669799/1/check/puppet-openstack-unit-5.5-centos-7/0ae960c/ : SUCCESS in 29m 05s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/99/669799/1/check/tripleo-ci-centos-7-undercloud-containers/4aabe67/ : SUCCESS in 1h 50m 20s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/99/669799/1/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/911f304/ : FAILURE in 1h 17m 38s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/99/669799/1/check/tripleo-ci-centos-7-containers-multinode/05ec3de/ : SUCCESS in 2h 20m 42s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/99/669799/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/eba4c0b/ : SUCCESS in 2h 32m 48s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/99/669799/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/45c01d3/ : SUCCESS in 2h 53m 41s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/99/669799/1/check/tripleo-ci-centos-7-standalone/432850d/ : SUCCESS in 1h 52m 39s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/99/669799/1/check/tripleo-ci-fedora-28-standalone/23fc287/ : SUCCESS in 1h 58m 39s (non-voting)\n- build-openstack-releasenotes http://logs.openstack.org/99/669799/1/check/build-openstack-releasenotes/2596d9f/html/ : SUCCESS in 3m 36s\n- tripleo-ci-centos-7-scenario001-standalone http://logs.openstack.org/99/669799/1/check/tripleo-ci-centos-7-scenario001-standalone/57cca56/ : SUCCESS in 1h 57m 43s","accounts_in_message":[],"_revision_number":1},{"id":"e6e488e8f7d85cf430065e732de4eed90114d075","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-07-10 20:51:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7f53d85cfaf58bc03a5d71a4f36bd8d45f1f4054","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 20:53:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1daa13509d6b70b57527badf069a6d38eaa53648","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 23:47:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c493c133967f1f7423bac357805c1d6abe9e8026","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 23:47:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/99/669799/1/gate/puppet-openstack-lint-ubuntu-bionic/6e86cd6/ : SUCCESS in 5m 44s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/99/669799/1/gate/puppet-openstack-syntax-4-ubuntu-bionic/ae192d8/ : SUCCESS in 4m 48s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/99/669799/1/gate/puppet-openstack-syntax-5-ubuntu-bionic/b77ac90/ : SUCCESS in 5m 08s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/99/669799/1/gate/puppet-openstack-syntax-6-ubuntu-bionic/6312ee2/ : SUCCESS in 4m 37s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/99/669799/1/gate/puppet-openstack-unit-4.8-centos-7/0e2ae1f/ : SUCCESS in 36m 59s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/99/669799/1/gate/puppet-openstack-unit-5.5-centos-7/51bf4bb/ : SUCCESS in 30m 58s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/99/669799/1/gate/tripleo-ci-centos-7-undercloud-containers/4fa37bc/ : SUCCESS in 1h 47m 41s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/99/669799/1/gate/tripleo-ci-centos-7-containers-multinode/ed26e82/ : SUCCESS in 1h 59m 45s\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/99/669799/1/gate/tripleo-ci-centos-7-standalone/6e10bf6/ : SUCCESS in 1h 36m 11s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/99/669799/1/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/acbd0b5/ : SUCCESS in 2h 35m 52s\n- build-openstack-releasenotes http://logs.openstack.org/99/669799/1/gate/build-openstack-releasenotes/7ec3ffc/html/ : SUCCESS in 4m 35s\n- tripleo-ci-centos-7-scenario001-standalone http://logs.openstack.org/99/669799/1/gate/tripleo-ci-centos-7-scenario001-standalone/3867eab/ : SUCCESS in 1h 52m 08s","accounts_in_message":[],"_revision_number":1},{"id":"2b3fb7054810ef98e42f863bd64b738e55308517","author":{"_account_id":25613,"name":"Keigo Noha","email":"knoha@redhat.com","username":"knoha"},"date":"2019-07-10 23:50:17.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/rocky as commit 9d62bf52a4e3b3d095a6a8231cd0508d80a301ad","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"964fe8cc62192e180a93f215e8bf038cf7ba0ad8","revisions":{"964fe8cc62192e180a93f215e8bf038cf7ba0ad8":{"kind":"REWORK","_number":1,"created":"2019-07-09 06:44:07.000000000","uploader":{"_account_id":25613,"name":"Keigo Noha","email":"knoha@redhat.com","username":"knoha"},"ref":"refs/changes/99/669799/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/99/669799/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/99/669799/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/99/669799/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/99/669799/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/99/669799/1"}}},"commit":{"parents":[{"commit":"1c96d897663c5160754b8e56407cc1d8b10ad97b","subject":"Fix amphora provider driver description","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/1c96d897663c5160754b8e56407cc1d8b10ad97b"}]}],"author":{"name":"Keigo Noha","email":"knoha@redhat.com","date":"2019-07-08 05:34:47.000000000","tz":540},"committer":{"name":"Keigo Noha","email":"knoha@redhat.com","date":"2019-07-09 06:44:07.000000000","tz":0},"subject":"Add dateext and related parameters for logrotate","message":"Add dateext and related parameters for logrotate\n\nlogrotate.pp should support dateext and related parameters.\nBy this change, a filename of a rotated file can be easily distinguished\nby rotated date.\n\nChange-Id: I798304a472df41b86a88611c97c2c99131faa0ad\n(cherry picked from commit 1eafeb67bf6d613d891fc26ee4452cded8115741)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/964fe8cc62192e180a93f215e8bf038cf7ba0ad8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/964fe8cc62192e180a93f215e8bf038cf7ba0ad8"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
