)]}'
{"id":"openstack%2Fpuppet-tripleo~669902","triplet_id":"openstack%2Fpuppet-tripleo~stable%2Fqueens~Ic77281cc69230b77224421e3d79d93803ea18bad","project":"openstack/puppet-tripleo","branch":"stable/queens","topic":"kernel_disable_ipv6-stable/queens","hashtags":[],"change_id":"Ic77281cc69230b77224421e3d79d93803ea18bad","subject":"Support TLS deployments with KernelDisableIPv6 enabled","status":"MERGED","created":"2019-07-09 16:06:21.000000000","updated":"2019-07-10 11:26:50.000000000","submitted":"2019-07-10 03:16:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"669902-1562728598925-83a8d452","meta_rev_id":"0c101f4f85b5d2d9a7ac2839c7b7f71491d577fd","_number":669902,"virtual_id_number":669902,"owner":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":2,"date":"2019-07-10 03:16:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2019-07-09 16:11:09.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2019-07-09 16:16:12.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":1,"date":"2019-07-09 16:16:12.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"date":"2019-07-09 16:06:21.000000000","_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-09 16:06:21.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"},{"updated":"2019-07-09 16:11:09.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-09 16:16:12.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-07-10 03:16:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5ce1239114943aea4405f3b62940256310319c4a","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-07-09 16:06:21.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"9c35c34e0bfffa8478ddd38413c405b0bb219a51","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-09 16:11:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 : NODE_FAILURE in 0s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 : NODE_FAILURE in 0s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c7ef6eafadee55c83bd818ad50de8c9e4e155254","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-07-09 16:16:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"411c3c76c057212e27665b7b183aab939c0d764b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-09 21:27:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-oooq http://logs.openstack.org/02/669902/1/check/tripleo-ci-centos-7-undercloud-oooq/6b236ed/ : SUCCESS in 1h 22m 31s\n- tripleo-ci-centos-7-undercloud-upgrades http://logs.openstack.org/02/669902/1/check/tripleo-ci-centos-7-undercloud-upgrades/694bb7b/ : SUCCESS in 1h 28m 20s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/02/669902/1/check/tripleo-ci-centos-7-containers-multinode/83336d9/ : SUCCESS in 2h 10m 59s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/02/669902/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/6e1f015/ : FAILURE in 2h 13m 20s (non-voting)\n- puppet-openstack-lint http://logs.openstack.org/02/669902/1/check/puppet-openstack-lint/8b449af/ : SUCCESS in 6m 31s\n- puppet-openstack-syntax-4 http://logs.openstack.org/02/669902/1/check/puppet-openstack-syntax-4/1686f9a/ : SUCCESS in 6m 02s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/02/669902/1/check/puppet-openstack-unit-4.8-centos-7/295ad29/ : SUCCESS in 31m 40s","accounts_in_message":[],"_revision_number":1},{"id":"82f190da1fdf4bf822b7629cf4d3964c5afce891","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-09 21:27:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c9e503edbbc2ba67f6e2afd8d30d7993e7fd5045","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-09 23:12:08.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bff7e575a0ce9389c529719b5a89764aaf86b4d8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 03:16:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-undercloud-oooq http://logs.openstack.org/02/669902/1/gate/tripleo-ci-centos-7-undercloud-oooq/d0a2147/ : SUCCESS in 1h 13m 15s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/02/669902/1/gate/tripleo-ci-centos-7-containers-multinode/f01432f/ : SUCCESS in 2h 27m 45s\n- puppet-openstack-lint http://logs.openstack.org/02/669902/1/gate/puppet-openstack-lint/b3a23c2/ : SUCCESS in 4m 19s\n- puppet-openstack-syntax-4 http://logs.openstack.org/02/669902/1/gate/puppet-openstack-syntax-4/6686849/ : SUCCESS in 5m 08s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/02/669902/1/gate/puppet-openstack-unit-4.8-centos-7/a457464/ : SUCCESS in 30m 34s","accounts_in_message":[],"_revision_number":1},{"id":"79a06cf5d8cb109767a3fce40a78e4be37b31941","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 03:16:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"84651bf7e16a97058aafc4e9063fb00dedff5bce","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-07-10 11:26:41.000000000","message":"Patch Set 1:\n\npossible it broke featureset039 all TLS job for queens","accounts_in_message":[],"_revision_number":1},{"id":"37cb2512583d2e748a212f20d9f531a66a859362","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-07-10 11:26:50.000000000","message":"Patch Set 1:\n\ncheck-rdo","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9f6cbedae62752bcd330919c39900483fb375e1f","revisions":{"9f6cbedae62752bcd330919c39900483fb375e1f":{"kind":"REWORK","_number":1,"created":"2019-07-09 16:06:21.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/02/669902/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/02/669902/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/02/669902/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/02/669902/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/02/669902/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/02/669902/1"}}},"commit":{"parents":[{"commit":"0aaa58e0949a30eb5754366c5cf2c5b4f11cc088","subject":"Merge \"Fix interface_for_ip for newer facter versions\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/0aaa58e0949a30eb5754366c5cf2c5b4f11cc088"}]}],"author":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2019-07-03 16:47:07.000000000","tz":120},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-07-09 16:06:21.000000000","tz":0},"subject":"Support TLS deployments with KernelDisableIPv6 enabled","message":"Support TLS deployments with KernelDisableIPv6 enabled\n\nThis will listen on 127.0.0.1 in case ipv6 is disabled.\nThe localhost_address is set in t-h-t kernel-baremetal-puppet.yaml\nin a related patch.\n\nChange-Id: Ic77281cc69230b77224421e3d79d93803ea18bad\nNeeded-By: Ide761c21dc87dadc722e27c9b8a7b68194164cb2\nRelated: rhbz#1703460\n(cherry picked from commit 5061ed8b7b8419f1302d425e05b7f80f4d6d051b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/9f6cbedae62752bcd330919c39900483fb375e1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/9f6cbedae62752bcd330919c39900483fb375e1f"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
