)]}'
{"id":"openstack%2Fpuppet-tripleo~670184","triplet_id":"openstack%2Fpuppet-tripleo~master~I703b5a429f48063474aace85bc45d948f5c91435","project":"openstack/puppet-tripleo","branch":"master","topic":"pcmk_tls","hashtags":[],"change_id":"I703b5a429f48063474aace85bc45d948f5c91435","subject":"Initial support for tls_priorities","status":"MERGED","created":"2019-07-10 20:26:06.000000000","updated":"2019-08-15 06:02:41.000000000","submitted":"2019-07-29 08:39:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"670184-1564389580912-652102b1","meta_rev_id":"dbbc41c7911a60417a2c899647d30e4a3faee1a7","_number":670184,"virtual_id_number":670184,"owner":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-07-29 08:39:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-07-28 09:49:20.000000000","_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":-1,"date":"2019-07-27 11:39:22.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2019-07-28 21:47:28.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2019-07-29 05:59:31.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-27 07:59:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2019-07-29 05:59:31.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-24 05:34:45.000000000","updated_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"REVIEWER"},{"updated":"2019-07-27 11:39:22.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-28 21:47:28.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-07-29 05:59:31.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-07-29 08:39:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a94142c4dd714bdd465b4ab7ee08b8dd16b83120","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2019-07-10 20:26:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4d4aec8304050a201b1c0d45e6c8dab3c5b7cf79","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-10 23:10:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/84/670184/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/3a2ee00/ : SUCCESS in 2h 40m 47s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/84/670184/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/4319dcf/ : FAILURE in 2h 20m 53s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/84/670184/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/63866f7/ : FAILURE in 2h 10m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d2efa4e812f197be9da63aa4b65ed1d93a140414","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 01:55:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/84/670184/1/check/puppet-openstack-lint-ubuntu-bionic/93d39a7/ : SUCCESS in 4m 49s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/84/670184/1/check/puppet-openstack-syntax-5-ubuntu-bionic/25150c1/ : SUCCESS in 5m 25s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/84/670184/1/check/puppet-openstack-syntax-6-ubuntu-bionic/211fa27/ : SUCCESS in 5m 03s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/84/670184/1/check/puppet-openstack-unit-5.5-centos-7/4b1219d/ : FAILURE in 27m 42s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/84/670184/1/check/puppet-openstack-unit-6.0-ubuntu-bionic/96f9181/ : FAILURE in 16m 31s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/84/670184/1/check/puppet-openstack-unit-latest-ubuntu-bionic/37ed5cf/ : FAILURE in 16m 26s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/84/670184/1/check/tripleo-ci-centos-7-undercloud-containers/8bfb4be/ : SUCCESS in 1h 40m 42s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/84/670184/1/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/985571f/ : FAILURE in 2h 08m 30s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/84/670184/1/check/tripleo-ci-centos-7-containers-multinode/6ec8da1/ : SUCCESS in 2h 37m 02s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/84/670184/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/c48d76d/ : SUCCESS in 2h 09m 50s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/84/670184/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/48e0093/ : FAILURE in 2h 30m 03s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/84/670184/1/check/tripleo-ci-centos-7-standalone/66a1007/ : FAILURE in 56m 36s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/84/670184/1/check/tripleo-ci-fedora-28-standalone/6f469ef/ : SUCCESS in 1h 32m 10s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/84/670184/1/check/tripleo-ci-centos-7-standalone-upgrade/651be3b/ : FAILURE in 1h 53m 14s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/84/670184/1/check/tripleo-ci-centos-7-standalone-upgrade-stein/3d9708e/ : SUCCESS in 2h 27m 09s\n- tripleo-ci-centos-7-scenario001-standalone http://logs.openstack.org/84/670184/1/check/tripleo-ci-centos-7-scenario001-standalone/e34c382/ : SUCCESS in 1h 48m 28s\n- tripleo-ci-centos-7-scenario002-standalone http://logs.openstack.org/84/670184/1/check/tripleo-ci-centos-7-scenario002-standalone/24fc0c8/ : SUCCESS in 1h 38m 25s\n- tripleo-ci-centos-7-scenario004-standalone http://logs.openstack.org/84/670184/1/check/tripleo-ci-centos-7-scenario004-standalone/55fb49f/ : SUCCESS in 2h 08m 33s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/84/670184/1/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/2fb5690/ : SUCCESS in 2h 33m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2b4f95e129f35b493478a52b3f719f991fbd6c09","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2019-07-11 06:48:50.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"6f4b382c1b60d83c4254bd03090140daf7657d4b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-11 09:23:34.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/84/670184/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/9e244ff/ : FAILURE in 1h 50m 00s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/84/670184/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/5153bef/ : SUCCESS in 2h 32m 48s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/84/670184/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/e2035a4/ : FAILURE in 2h 25m 40s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"cab5c6b93c168096f806498185a50d891f31980b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 09:44:14.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/84/670184/2/check/puppet-openstack-lint-ubuntu-bionic/7bb9392/ : SUCCESS in 5m 21s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/84/670184/2/check/puppet-openstack-syntax-5-ubuntu-bionic/1c7878a/ : SUCCESS in 6m 24s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/84/670184/2/check/puppet-openstack-syntax-6-ubuntu-bionic/593bfcb/ : SUCCESS in 6m 42s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/84/670184/2/check/puppet-openstack-unit-5.5-centos-7/7ae9423/ : FAILURE in 31m 51s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/84/670184/2/check/puppet-openstack-unit-6.0-ubuntu-bionic/554a3b2/ : FAILURE in 16m 27s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/84/670184/2/check/puppet-openstack-unit-latest-ubuntu-bionic/bc49317/ : FAILURE in 14m 49s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/84/670184/2/check/tripleo-ci-centos-7-undercloud-containers/e7cf96a/ : SUCCESS in 1h 45m 46s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/84/670184/2/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/bce0f64/ : FAILURE in 2h 07m 34s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/84/670184/2/check/tripleo-ci-centos-7-containers-multinode/9b37268/ : SUCCESS in 2h 08m 19s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/84/670184/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/db9d0ec/ : SUCCESS in 2h 24m 25s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/84/670184/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/135de6a/ : FAILURE in 2h 32m 58s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/84/670184/2/check/tripleo-ci-centos-7-standalone/d2f5d33/ : SUCCESS in 1h 48m 39s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/84/670184/2/check/tripleo-ci-fedora-28-standalone/0453592/ : SUCCESS in 1h 54m 44s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/84/670184/2/check/tripleo-ci-centos-7-standalone-upgrade/ec3e6b8/ : FAILURE in 2h 06m 19s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/84/670184/2/check/tripleo-ci-centos-7-standalone-upgrade-stein/f715767/ : SUCCESS in 2h 14m 48s\n- tripleo-ci-centos-7-scenario001-standalone http://logs.openstack.org/84/670184/2/check/tripleo-ci-centos-7-scenario001-standalone/0ad7782/ : SUCCESS in 2h 10m 17s\n- tripleo-ci-centos-7-scenario002-standalone http://logs.openstack.org/84/670184/2/check/tripleo-ci-centos-7-scenario002-standalone/2cd66c6/ : SUCCESS in 1h 36m 58s\n- tripleo-ci-centos-7-scenario004-standalone http://logs.openstack.org/84/670184/2/check/tripleo-ci-centos-7-scenario004-standalone/60592bc/ : SUCCESS in 1h 47m 20s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/84/670184/2/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/9f9a5e0/ : SUCCESS in 2h 48m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f3e656f673a12413a0dd535b0f3eea8336e689eb","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2019-07-17 17:42:45.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"851766558fd7ab41282714c50894bdd39c25d975","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-17 20:31:24.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/84/670184/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/0de4219/ : SUCCESS in 2h 30m 02s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/84/670184/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/bb91b53/ : SUCCESS in 2h 41m 00s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/84/670184/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/7d17443/ : FAILURE in 2h 47m 08s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ff34a52a6425f70f4c2254713e0d600998c89692","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 20:34:46.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/84/670184/3/check/puppet-openstack-lint-ubuntu-bionic/fb511b9/ : SUCCESS in 4m 46s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/84/670184/3/check/puppet-openstack-syntax-5-ubuntu-bionic/1a996b9/ : SUCCESS in 4m 12s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/84/670184/3/check/puppet-openstack-syntax-6-ubuntu-bionic/d6ec6e1/ : SUCCESS in 4m 41s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/84/670184/3/check/puppet-openstack-unit-5.5-centos-7/db53c19/ : SUCCESS in 28m 59s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/84/670184/3/check/puppet-openstack-unit-6.0-ubuntu-bionic/f897702/ : SUCCESS in 17m 48s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/84/670184/3/check/puppet-openstack-unit-latest-ubuntu-bionic/9278eb7/ : SUCCESS in 19m 53s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/84/670184/3/check/tripleo-ci-centos-7-undercloud-containers/5ad8afa/ : SUCCESS in 1h 36m 57s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/84/670184/3/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/a08e27b/ : FAILURE in 2h 06m 24s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/84/670184/3/check/tripleo-ci-centos-7-containers-multinode/38cc34b/ : FAILURE in 2h 06m 21s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/84/670184/3/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/acaaf4f/ : SUCCESS in 2h 11m 28s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/84/670184/3/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/417ad80/ : FAILURE in 2h 40m 58s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/84/670184/3/check/tripleo-ci-centos-7-standalone/4c8ada5/ : FAILURE in 1h 06m 15s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/84/670184/3/check/tripleo-ci-fedora-28-standalone/02fa58e/ : SUCCESS in 1h 58m 22s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/84/670184/3/check/tripleo-ci-centos-7-standalone-upgrade/af7ea21/ : FAILURE in 2h 03m 46s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone http://logs.openstack.org/84/670184/3/check/tripleo-ci-centos-7-scenario001-standalone/8cf8e84/ : FAILURE in 1h 31m 05s\n- tripleo-ci-centos-7-scenario002-standalone http://logs.openstack.org/84/670184/3/check/tripleo-ci-centos-7-scenario002-standalone/a6bb784/ : FAILURE in 1h 23m 29s\n- tripleo-ci-centos-7-scenario004-standalone http://logs.openstack.org/84/670184/3/check/tripleo-ci-centos-7-scenario004-standalone/eec501f/ : FAILURE in 1h 50m 18s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/84/670184/3/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/c76a2fe/ : SUCCESS in 2h 28m 40s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"562d4e165de3436381fcfffcd526a6909afa90f8","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2019-07-18 07:23:30.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"c8b5195d08ed21355da70ae525e69dfe74e5cf97","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-18 10:38:45.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/84/670184/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/4e55b79/ : SUCCESS in 2h 37m 26s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/84/670184/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/c61a2d3/ : SUCCESS in 3h 03m 10s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/84/670184/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/ed62882/ : SUCCESS in 2h 48m 06s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2232a88caebcf7bd04c371e0b4c2024fb31f0bae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-18 11:05:54.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/84/670184/3/check/puppet-openstack-lint-ubuntu-bionic/254fa1e/ : SUCCESS in 4m 51s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/84/670184/3/check/puppet-openstack-syntax-5-ubuntu-bionic/8b27d8d/ : SUCCESS in 4m 35s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/84/670184/3/check/puppet-openstack-syntax-6-ubuntu-bionic/1c76c5a/ : SUCCESS in 4m 56s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/84/670184/3/check/puppet-openstack-unit-5.5-centos-7/79bfb23/ : SUCCESS in 26m 43s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/84/670184/3/check/puppet-openstack-unit-6.0-ubuntu-bionic/266967c/ : SUCCESS in 13m 54s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/84/670184/3/check/puppet-openstack-unit-latest-ubuntu-bionic/69dbc25/ : SUCCESS in 16m 33s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/84/670184/3/check/tripleo-ci-centos-7-undercloud-containers/e091765/ : SUCCESS in 1h 38m 48s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/84/670184/3/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/97c58e6/ : FAILURE in 2h 03m 10s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/84/670184/3/check/tripleo-ci-centos-7-containers-multinode/f2bc8b0/ : SUCCESS in 2h 42m 27s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/84/670184/3/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/ef0c025/ : SUCCESS in 2h 27m 14s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/84/670184/3/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/733ce03/ : FAILURE in 2h 57m 26s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/84/670184/3/check/tripleo-ci-centos-7-standalone/103eb6a/ : SUCCESS in 1h 59m 09s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/84/670184/3/check/tripleo-ci-fedora-28-standalone/fdef8e2/ : FAILURE in 1h 52m 22s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/84/670184/3/check/tripleo-ci-centos-7-standalone-upgrade/f76c60d/ : FAILURE in 1h 58m 21s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone http://logs.openstack.org/84/670184/3/check/tripleo-ci-centos-7-scenario001-standalone/e17aeed/ : SUCCESS in 1h 41m 05s\n- tripleo-ci-centos-7-scenario002-standalone http://logs.openstack.org/84/670184/3/check/tripleo-ci-centos-7-scenario002-standalone/b7c54c6/ : SUCCESS in 1h 36m 02s\n- tripleo-ci-centos-7-scenario004-standalone http://logs.openstack.org/84/670184/3/check/tripleo-ci-centos-7-scenario004-standalone/87f521e/ : SUCCESS in 1h 48m 36s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/84/670184/3/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/b194e1d/ : SUCCESS in 2h 46m 27s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b0ac09436de9cb567a53f95039c2048b97c06aa1","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2019-07-22 15:46:52.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"f79044276c7962cad0fcbdf144bb19d1db373e9e","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2019-07-22 15:48:47.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"e8e07661c99889128eab9844958b751c2f935221","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-22 18:34:35.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/84/670184/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/188a5ba/ : SUCCESS in 2h 31m 38s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/84/670184/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/5ea8f29/ : SUCCESS in 2h 34m 07s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/84/670184/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/e7d1d5e/ : SUCCESS in 2h 44m 16s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"e2c8d8c6e743d3531988becb9f285f8c81332e42","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 19:06:52.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/84/670184/5/check/puppet-openstack-lint-ubuntu-bionic/30536be/ : SUCCESS in 6m 10s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/84/670184/5/check/puppet-openstack-syntax-5-ubuntu-bionic/890e388/ : SUCCESS in 5m 40s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/84/670184/5/check/puppet-openstack-syntax-6-ubuntu-bionic/f793f01/ : SUCCESS in 6m 21s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/84/670184/5/check/puppet-openstack-unit-5.5-centos-7/c651597/ : SUCCESS in 31m 18s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/84/670184/5/check/puppet-openstack-unit-6.0-ubuntu-bionic/af2b12b/ : SUCCESS in 18m 37s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/84/670184/5/check/puppet-openstack-unit-latest-ubuntu-bionic/e1c70cf/ : SUCCESS in 20m 50s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/84/670184/5/check/tripleo-ci-centos-7-undercloud-containers/929aff0/ : SUCCESS in 1h 38m 33s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/84/670184/5/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/ebed9d0/ : FAILURE in 2h 01m 16s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/84/670184/5/check/tripleo-ci-centos-7-containers-multinode/0ce1da4/ : FAILURE in 10m 27s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/84/670184/5/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/afeee7a/ : SUCCESS in 2h 24m 07s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/84/670184/5/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/516353f/ : SUCCESS in 2h 25m 49s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/84/670184/5/check/tripleo-ci-centos-7-standalone/a3908e8/ : SUCCESS in 1h 49m 34s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/84/670184/5/check/tripleo-ci-fedora-28-standalone/4d1f0db/ : SUCCESS in 1h 31m 11s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/84/670184/5/check/tripleo-ci-centos-7-standalone-upgrade/ff69268/ : FAILURE in 2h 00m 45s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone http://logs.openstack.org/84/670184/5/check/tripleo-ci-centos-7-scenario001-standalone/4947bab/ : FAILURE in 1h 15m 29s\n- tripleo-ci-centos-7-scenario002-standalone http://logs.openstack.org/84/670184/5/check/tripleo-ci-centos-7-scenario002-standalone/ff16cc0/ : SUCCESS in 1h 31m 38s\n- tripleo-ci-centos-7-scenario004-standalone http://logs.openstack.org/84/670184/5/check/tripleo-ci-centos-7-scenario004-standalone/e1dc867/ : SUCCESS in 1h 41m 25s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/84/670184/5/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/4a68ffe/ : SUCCESS in 2h 14m 55s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"6cda6e2073b448dc2516ef61fe3b2d35b3142128","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2019-07-22 21:05:17.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"5a6b8130cc10aa4084ee1ac057b5ee06aec6bb18","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 23:49:44.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/84/670184/5/check/puppet-openstack-lint-ubuntu-bionic/1dd4a76/ : SUCCESS in 4m 32s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/84/670184/5/check/puppet-openstack-syntax-5-ubuntu-bionic/27171e4/ : SUCCESS in 4m 50s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/84/670184/5/check/puppet-openstack-syntax-6-ubuntu-bionic/b75c794/ : SUCCESS in 4m 43s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/84/670184/5/check/puppet-openstack-unit-5.5-centos-7/4a74592/ : SUCCESS in 30m 59s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/84/670184/5/check/puppet-openstack-unit-6.0-ubuntu-bionic/166d3ab/ : SUCCESS in 17m 21s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/84/670184/5/check/puppet-openstack-unit-latest-ubuntu-bionic/d0de6f8/ : SUCCESS in 16m 50s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/84/670184/5/check/tripleo-ci-centos-7-undercloud-containers/a2d9511/ : SUCCESS in 1h 19m 23s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/84/670184/5/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/d25ce2f/ : FAILURE in 2h 20m 12s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/84/670184/5/check/tripleo-ci-centos-7-containers-multinode/b5df207/ : SUCCESS in 2h 05m 57s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/84/670184/5/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/2da4e76/ : SUCCESS in 2h 08m 20s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/84/670184/5/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/c815016/ : SUCCESS in 2h 37m 46s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/84/670184/5/check/tripleo-ci-centos-7-standalone/c530cf0/ : SUCCESS in 1h 43m 59s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/84/670184/5/check/tripleo-ci-fedora-28-standalone/345dcab/ : SUCCESS in 1h 53m 28s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/84/670184/5/check/tripleo-ci-centos-7-standalone-upgrade/5680af9/ : FAILURE in 2h 05m 38s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone http://logs.openstack.org/84/670184/5/check/tripleo-ci-centos-7-scenario001-standalone/b15dbac/ : FAILURE in 1h 02m 30s\n- tripleo-ci-centos-7-scenario002-standalone http://logs.openstack.org/84/670184/5/check/tripleo-ci-centos-7-scenario002-standalone/751e8f7/ : SUCCESS in 1h 32m 16s\n- tripleo-ci-centos-7-scenario004-standalone http://logs.openstack.org/84/670184/5/check/tripleo-ci-centos-7-scenario004-standalone/dcdaab4/ : SUCCESS in 1h 43m 23s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/84/670184/5/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/bbb7768/ : SUCCESS in 2h 21m 11s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"31747ffd4ce25ff2a7132f7cfe2cb8501a252318","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-23 00:17:46.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/84/670184/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/0276375/ : SUCCESS in 2h 27m 46s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/84/670184/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/09f6e7d/ : SUCCESS in 2h 39m 09s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/84/670184/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/06ff505/ : SUCCESS in 3h 11m 03s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"3d7e251a9f4de0561b99628525b03006b5311f9a","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2019-07-24 05:29:19.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"8cd1c42c2a7b1ed906329429f303383c5b16e0d4","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2019-07-24 05:34:45.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"7e80e2e97972661063b4bf20bfb453ed75a23a75","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 08:06:56.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/84/670184/6/check/puppet-openstack-lint-ubuntu-bionic/be1bdf3/ : SUCCESS in 4m 44s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/84/670184/6/check/puppet-openstack-syntax-5-ubuntu-bionic/8247b00/ : SUCCESS in 4m 54s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/84/670184/6/check/puppet-openstack-syntax-6-ubuntu-bionic/c5b89b1/ : SUCCESS in 4m 15s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/84/670184/6/check/puppet-openstack-unit-5.5-centos-7/8a3f4ce/ : SUCCESS in 25m 47s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/84/670184/6/check/puppet-openstack-unit-6.0-ubuntu-bionic/bc25e87/ : SUCCESS in 15m 39s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/84/670184/6/check/puppet-openstack-unit-latest-ubuntu-bionic/3fa195f/ : SUCCESS in 15m 28s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/84/670184/6/check/tripleo-ci-centos-7-undercloud-containers/2565bac/ : SUCCESS in 1h 32m 26s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/84/670184/6/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/1ceda63/ : FAILURE in 2h 05m 23s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/84/670184/6/check/tripleo-ci-centos-7-containers-multinode/3494d1b/ : FAILURE in 2h 01m 14s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/84/670184/6/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/21dcf04/ : SUCCESS in 2h 14m 40s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/84/670184/6/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/de91d46/ : SUCCESS in 2h 32m 40s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/84/670184/6/check/tripleo-ci-centos-7-standalone/b37b442/ : FAILURE in 1h 09m 23s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/84/670184/6/check/tripleo-ci-fedora-28-standalone/287f230/ : SUCCESS in 1h 49m 10s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/84/670184/6/check/tripleo-ci-centos-7-standalone-upgrade/183ba0e/ : FAILURE in 2h 06m 57s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone http://logs.openstack.org/84/670184/6/check/tripleo-ci-centos-7-scenario001-standalone/dab6f5b/ : FAILURE in 1h 37m 09s\n- tripleo-ci-centos-7-scenario002-standalone http://logs.openstack.org/84/670184/6/check/tripleo-ci-centos-7-scenario002-standalone/a6e8fbe/ : FAILURE in 1h 33m 42s\n- tripleo-ci-centos-7-scenario004-standalone http://logs.openstack.org/84/670184/6/check/tripleo-ci-centos-7-scenario004-standalone/626d05c/ : FAILURE in 1h 50m 54s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/84/670184/6/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/f59de24/ : SUCCESS in 2h 26m 18s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"a0493c96944b2c773842964766701612b70833b5","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-24 09:05:44.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/84/670184/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/a242d20/ : FAILURE in 3h 08m 57s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/84/670184/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/7658cec/ : SUCCESS in 2h 28m 53s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/84/670184/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/ec24cd1/ : FAILURE in 3h 34m 54s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"ab9540287cad1384d6e41d6b2952f371c5260f07","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2019-07-24 13:06:06.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"942d3a096b32688d98c3f0692be342b301fe5328","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 15:59:18.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/84/670184/6/check/puppet-openstack-lint-ubuntu-bionic/f82c933/ : SUCCESS in 7m 44s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/84/670184/6/check/puppet-openstack-syntax-5-ubuntu-bionic/812afa2/ : SUCCESS in 5m 06s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/84/670184/6/check/puppet-openstack-syntax-6-ubuntu-bionic/8c8ff72/ : SUCCESS in 5m 57s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/84/670184/6/check/puppet-openstack-unit-5.5-centos-7/6350131/ : SUCCESS in 38m 24s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/84/670184/6/check/puppet-openstack-unit-6.0-ubuntu-bionic/22e081f/ : SUCCESS in 15m 49s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/84/670184/6/check/puppet-openstack-unit-latest-ubuntu-bionic/5a336e0/ : SUCCESS in 18m 40s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/84/670184/6/check/tripleo-ci-centos-7-undercloud-containers/3e5caa3/ : SUCCESS in 1h 38m 45s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/84/670184/6/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/daf2eff/ : FAILURE in 2h 04m 34s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/84/670184/6/check/tripleo-ci-centos-7-containers-multinode/045a098/ : SUCCESS in 2h 36m 24s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/84/670184/6/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/1372dcc/ : SUCCESS in 2h 25m 55s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/84/670184/6/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/a2c882e/ : SUCCESS in 2h 46m 38s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/84/670184/6/check/tripleo-ci-centos-7-standalone/6f08ce3/ : SUCCESS in 2h 01m 36s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/84/670184/6/check/tripleo-ci-fedora-28-standalone/d249c9c/ : SUCCESS in 1h 48m 00s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/84/670184/6/check/tripleo-ci-centos-7-standalone-upgrade/80e21b4/ : FAILURE in 1h 57m 10s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone http://logs.openstack.org/84/670184/6/check/tripleo-ci-centos-7-scenario001-standalone/d4678cf/ : SUCCESS in 2h 06m 11s\n- tripleo-ci-centos-7-scenario002-standalone http://logs.openstack.org/84/670184/6/check/tripleo-ci-centos-7-scenario002-standalone/061feb6/ : SUCCESS in 1h 41m 54s\n- tripleo-ci-centos-7-scenario004-standalone http://logs.openstack.org/84/670184/6/check/tripleo-ci-centos-7-scenario004-standalone/cdbc138/ : FAILURE in 1h 14m 30s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/84/670184/6/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/f2916db/ : SUCCESS in 2h 20m 56s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"0e64264f0abfd673f953910b600191278658695c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-24 17:05:47.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/84/670184/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/b2cdee9/ : SUCCESS in 3h 23m 16s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/84/670184/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/b4d7188/ : SUCCESS in 2h 39m 23s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 finger://ze02.softwarefactory-project.io:7979/567a621016bc4b00a1c33b049608d53f : POST_FAILURE in 3h 58m 04s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"88a2f4e9cfa7a8501f3ee551981e59d076f24ea7","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2019-07-25 04:33:13.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"98ff13c1f7e189725161de3bad56e589166b0cfb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-25 07:23:05.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/84/670184/6/check/puppet-openstack-lint-ubuntu-bionic/99d3962/ : SUCCESS in 4m 50s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/84/670184/6/check/puppet-openstack-syntax-5-ubuntu-bionic/cc018a1/ : SUCCESS in 4m 46s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/84/670184/6/check/puppet-openstack-syntax-6-ubuntu-bionic/3c620f9/ : SUCCESS in 6m 01s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/84/670184/6/check/puppet-openstack-unit-5.5-centos-7/e21cdc5/ : SUCCESS in 33m 41s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/84/670184/6/check/puppet-openstack-unit-6.0-ubuntu-bionic/1897141/ : SUCCESS in 18m 16s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/84/670184/6/check/puppet-openstack-unit-latest-ubuntu-bionic/de7f496/ : SUCCESS in 14m 26s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/84/670184/6/check/tripleo-ci-centos-7-undercloud-containers/36ab2c1/ : SUCCESS in 1h 40m 12s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/84/670184/6/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/b2368af/ : FAILURE in 2h 08m 31s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/84/670184/6/check/tripleo-ci-centos-7-containers-multinode/8943de1/ : FAILURE in 2h 27m 42s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/84/670184/6/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/f98069d/ : SUCCESS in 2h 23m 02s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/84/670184/6/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/ec553d3/ : SUCCESS in 2h 44m 35s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/84/670184/6/check/tripleo-ci-centos-7-standalone/baa6f6a/ : SUCCESS in 1h 58m 06s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/84/670184/6/check/tripleo-ci-fedora-28-standalone/db028ba/ : SUCCESS in 1h 43m 25s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/84/670184/6/check/tripleo-ci-centos-7-standalone-upgrade/2229945/ : FAILURE in 2h 10m 50s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone http://logs.openstack.org/84/670184/6/check/tripleo-ci-centos-7-scenario001-standalone/5d91c1f/ : SUCCESS in 1h 55m 57s\n- tripleo-ci-centos-7-scenario002-standalone http://logs.openstack.org/84/670184/6/check/tripleo-ci-centos-7-scenario002-standalone/30600f6/ : SUCCESS in 1h 31m 40s\n- tripleo-ci-centos-7-scenario004-standalone http://logs.openstack.org/84/670184/6/check/tripleo-ci-centos-7-scenario004-standalone/fab62c9/ : SUCCESS in 1h 45m 48s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/84/670184/6/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/3d8a963/ : SUCCESS in 2h 31m 36s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"27a2463f062d795f55a1813d11ecd637e8145eec","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-25 07:25:00.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/84/670184/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/7e45341/ : SUCCESS in 2h 39m 43s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/84/670184/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/778ce90/ : SUCCESS in 2h 44m 20s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/84/670184/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/3d09d32/ : SUCCESS in 2h 50m 34s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"f0b3bb4275ddaef988540de95021c8bd87c1afd3","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2019-07-25 13:10:53.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"565cac1bc9799ee675a2e6df6d5775c0348637da","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-25 15:54:02.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/84/670184/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/c8bfd44/ : FAILURE in 2h 41m 05s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/84/670184/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/e147eb9/ : FAILURE in 1h 03m 54s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/84/670184/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/0c6f059/ : FAILURE in 1h 18m 53s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"8e7c968c0b739f21f97587aec01ab54a362ea7e0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-25 16:05:37.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/84/670184/6/check/puppet-openstack-lint-ubuntu-bionic/20c7a22/ : SUCCESS in 6m 07s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/84/670184/6/check/puppet-openstack-syntax-5-ubuntu-bionic/3cea9c7/ : SUCCESS in 4m 26s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/84/670184/6/check/puppet-openstack-syntax-6-ubuntu-bionic/ad59111/ : SUCCESS in 6m 26s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/84/670184/6/check/puppet-openstack-unit-5.5-centos-7/7278238/ : SUCCESS in 34m 06s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/84/670184/6/check/puppet-openstack-unit-6.0-ubuntu-bionic/52a3370/ : SUCCESS in 17m 14s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/84/670184/6/check/puppet-openstack-unit-latest-ubuntu-bionic/e799570/ : SUCCESS in 17m 39s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/84/670184/6/check/tripleo-ci-centos-7-undercloud-containers/b6e6a7a/ : SUCCESS in 1h 47m 56s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/84/670184/6/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/898abba/ : FAILURE in 1h 51m 38s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/84/670184/6/check/tripleo-ci-centos-7-containers-multinode/d18a1eb/ : SUCCESS in 2h 28m 04s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/84/670184/6/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/f5556f8/ : SUCCESS in 2h 27m 17s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/84/670184/6/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/0b230c4/ : SUCCESS in 2h 27m 01s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/84/670184/6/check/tripleo-ci-centos-7-standalone/f39a082/ : SUCCESS in 2h 23m 34s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/84/670184/6/check/tripleo-ci-fedora-28-standalone/acccb5c/ : SUCCESS in 1h 59m 46s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/84/670184/6/check/tripleo-ci-centos-7-standalone-upgrade/5ba51a7/ : FAILURE in 1h 45m 13s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone http://logs.openstack.org/84/670184/6/check/tripleo-ci-centos-7-scenario001-standalone/1d78677/ : SUCCESS in 1h 52m 43s\n- tripleo-ci-centos-7-scenario002-standalone http://logs.openstack.org/84/670184/6/check/tripleo-ci-centos-7-scenario002-standalone/1816e79/ : SUCCESS in 1h 33m 44s\n- tripleo-ci-centos-7-scenario004-standalone http://logs.openstack.org/84/670184/6/check/tripleo-ci-centos-7-scenario004-standalone/175d494/ : SUCCESS in 1h 53m 31s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/84/670184/6/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/c0fa6f8/ : SUCCESS in 2h 21m 03s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"82a1b40fc12a23d5758b7ebd211ca9fca5603c18","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2019-07-27 07:59:45.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"d586a9b9ff824caceb80135e5b58ef9b797b0b2b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-27 11:01:36.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/84/670184/7/check/puppet-openstack-lint-ubuntu-bionic/9397a90/ : SUCCESS in 6m 27s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/84/670184/7/check/puppet-openstack-syntax-5-ubuntu-bionic/0254b3b/ : SUCCESS in 4m 57s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/84/670184/7/check/puppet-openstack-syntax-6-ubuntu-bionic/b0fd291/ : SUCCESS in 6m 08s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/84/670184/7/check/puppet-openstack-unit-5.5-centos-7/e96371b/ : SUCCESS in 38m 37s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/84/670184/7/check/puppet-openstack-unit-6.0-ubuntu-bionic/44cb779/ : SUCCESS in 15m 26s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/84/670184/7/check/puppet-openstack-unit-latest-ubuntu-bionic/ca7e6c3/ : SUCCESS in 20m 46s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/84/670184/7/check/tripleo-ci-centos-7-undercloud-containers/15b17df/ : SUCCESS in 1h 53m 13s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/84/670184/7/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/e0828d0/ : FAILURE in 2h 41m 49s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/84/670184/7/check/tripleo-ci-centos-7-containers-multinode/c145850/ : SUCCESS in 2h 42m 01s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/84/670184/7/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/ea0da62/ : SUCCESS in 2h 45m 25s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/84/670184/7/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/31a2b4a/ : SUCCESS in 2h 29m 03s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/84/670184/7/check/tripleo-ci-centos-7-standalone/8f7e15b/ : SUCCESS in 2h 12m 09s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/84/670184/7/check/tripleo-ci-fedora-28-standalone/6604800/ : SUCCESS in 2h 05m 08s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/84/670184/7/check/tripleo-ci-centos-7-standalone-upgrade/103dc42/ : SUCCESS in 2h 00m 15s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone http://logs.openstack.org/84/670184/7/check/tripleo-ci-centos-7-scenario001-standalone/32fac01/ : SUCCESS in 1h 38m 11s\n- tripleo-ci-centos-7-scenario002-standalone http://logs.openstack.org/84/670184/7/check/tripleo-ci-centos-7-scenario002-standalone/9c5bb0b/ : SUCCESS in 1h 28m 26s\n- tripleo-ci-centos-7-scenario004-standalone http://logs.openstack.org/84/670184/7/check/tripleo-ci-centos-7-scenario004-standalone/e39a1b2/ : SUCCESS in 1h 42m 45s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/84/670184/7/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/4f8791a/ : SUCCESS in 1h 56m 57s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"9fa399b6f7105cd9f1fc2bc9a58d9453a10e7445","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-27 11:39:22.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/84/670184/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/de2dd7f/ : TIMED_OUT in 3h 38m 46s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/84/670184/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/53670f1/ : SUCCESS in 2h 34m 28s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/84/670184/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/88e16b7/ : SUCCESS in 2h 48m 15s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"9fadc1d82815a80327c78f59c205b32a7895f6b0","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2019-07-27 12:02:52.000000000","message":"Patch Set 7:\n\ncheck-rdo","accounts_in_message":[],"_revision_number":7},{"id":"e5cd55370ac06ffdb33702db11cecaf4bafcc722","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-27 15:12:05.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/84/670184/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/07789d0/ : SUCCESS in 2h 33m 09s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/84/670184/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/b5b65ec/ : FAILURE in 3h 07m 44s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/84/670184/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/87aa32f/ : SUCCESS in 2h 58m 19s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"4bc7acaa32886f4df9f08f65f75887b3608f8e4e","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2019-07-28 09:49:20.000000000","message":"Patch Set 7:\n\ncheck-rdo","accounts_in_message":[],"_revision_number":7},{"id":"6d92ca3ba97435b24f948165dfca700377e8984b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-28 14:25:32.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/84/670184/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/d511fc6/ : FAILURE in 2h 04m 42s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/84/670184/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/1d9163e/ : SUCCESS in 2h 26m 44s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/84/670184/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/cc0fc52/ : TIMED_OUT in 4h 31m 50s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"1adfdc27e09b0ea5a32b0d861da9b19da2417de0","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-07-28 21:47:28.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"d489337a1d38e9af2e44b4f4ea999f8ca2f34a78","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-07-29 05:59:31.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"c511874b0f189f865f0c95451240647587493c00","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-29 05:59:43.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"2b13027a2da4f5e01421224683b273c1ab8b0bf1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-29 08:39:40.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/84/670184/7/gate/puppet-openstack-lint-ubuntu-bionic/a27593c/ : SUCCESS in 6m 58s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/84/670184/7/gate/puppet-openstack-syntax-5-ubuntu-bionic/be91bb3/ : SUCCESS in 6m 00s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/84/670184/7/gate/puppet-openstack-syntax-6-ubuntu-bionic/99509de/ : SUCCESS in 7m 25s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/84/670184/7/gate/puppet-openstack-unit-5.5-centos-7/fc1fd00/ : SUCCESS in 31m 53s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/84/670184/7/gate/puppet-openstack-unit-6.0-ubuntu-bionic/c4fec9d/ : SUCCESS in 17m 10s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/84/670184/7/gate/tripleo-ci-centos-7-undercloud-containers/ff29b24/ : SUCCESS in 1h 35m 59s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/84/670184/7/gate/tripleo-ci-centos-7-containers-multinode/4205307/ : SUCCESS in 2h 20m 19s\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/84/670184/7/gate/tripleo-ci-centos-7-standalone/aaa555e/ : SUCCESS in 1h 54m 42s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/84/670184/7/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/7ba8758/ : SUCCESS in 2h 21m 10s\n- tripleo-ci-centos-7-scenario001-standalone http://logs.openstack.org/84/670184/7/gate/tripleo-ci-centos-7-scenario001-standalone/2b73299/ : SUCCESS in 1h 51m 15s\n- tripleo-ci-centos-7-scenario002-standalone http://logs.openstack.org/84/670184/7/gate/tripleo-ci-centos-7-scenario002-standalone/ef5bc66/ : SUCCESS in 1h 46m 17s\n- tripleo-ci-centos-7-scenario004-standalone http://logs.openstack.org/84/670184/7/gate/tripleo-ci-centos-7-scenario004-standalone/3e42b47/ : SUCCESS in 1h 50m 50s","accounts_in_message":[],"_revision_number":7},{"id":"3cbea4655d46212220be0de03d8e31799090727e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-29 08:39:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"eab331cd17c2d685a2fcb04c6d7b9950ec230628","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2019-08-15 06:02:41.000000000","message":"Patch Set 7: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit dbae8506927bf77ed430db138461d846e81a2489","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"f1a593b6425f1e95ef789dd8ac9e32f095e347ff","revisions":{"126da2d719b5048e24df0bed6d8e9c49cacbd1fa":{"kind":"REWORK","_number":1,"created":"2019-07-10 20:26:06.000000000","uploader":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"ref":"refs/changes/84/670184/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/84/670184/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/84/670184/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/84/670184/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/84/670184/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/84/670184/1"}}},"commit":{"parents":[{"commit":"610691ca6561c7630d243d5aa23f7c25a60c9a5c","subject":"Merge \"Allow custom per-service bind_options for haproxy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/610691ca6561c7630d243d5aa23f7c25a60c9a5c"}]}],"author":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2019-07-10 20:25:07.000000000","tz":120},"committer":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2019-07-10 20:25:07.000000000","tz":120},"subject":"WIP Initial support for tls_priorities","message":"WIP Initial support for tls_priorities\n\nDepends-On: https://review.opendev.org/670155\nChange-Id: I703b5a429f48063474aace85bc45d948f5c91435\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/126da2d719b5048e24df0bed6d8e9c49cacbd1fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/126da2d719b5048e24df0bed6d8e9c49cacbd1fa"}]},"branch":"refs/heads/master"},"341257f6e2dac39f2b74600e75d5516b5ad378b2":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-07-11 06:48:50.000000000","uploader":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"ref":"refs/changes/84/670184/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/84/670184/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/84/670184/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/84/670184/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/84/670184/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/84/670184/2"}}},"commit":{"parents":[{"commit":"610691ca6561c7630d243d5aa23f7c25a60c9a5c","subject":"Merge \"Allow custom per-service bind_options for haproxy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/610691ca6561c7630d243d5aa23f7c25a60c9a5c"}]}],"author":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2019-07-10 20:25:07.000000000","tz":120},"committer":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2019-07-11 06:48:36.000000000","tz":120},"subject":"WIP Initial support for tls_priorities","message":"WIP Initial support for tls_priorities\n\nDepends-On: I1971810f6a90f244ed5ced972a5fe7fde29dde86\nChange-Id: I703b5a429f48063474aace85bc45d948f5c91435\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/341257f6e2dac39f2b74600e75d5516b5ad378b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/341257f6e2dac39f2b74600e75d5516b5ad378b2"}]},"branch":"refs/heads/master"},"5454d3191f802be0f797b13d60ec33f1cc7fe73a":{"kind":"REWORK","_number":3,"created":"2019-07-17 17:42:45.000000000","uploader":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"ref":"refs/changes/84/670184/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/84/670184/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/84/670184/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/84/670184/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/84/670184/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/84/670184/3"}}},"commit":{"parents":[{"commit":"610691ca6561c7630d243d5aa23f7c25a60c9a5c","subject":"Merge \"Allow custom per-service bind_options for haproxy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/610691ca6561c7630d243d5aa23f7c25a60c9a5c"}]}],"author":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2019-07-10 20:25:07.000000000","tz":120},"committer":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2019-07-17 17:41:18.000000000","tz":120},"subject":"WIP Initial support for tls_priorities","message":"WIP Initial support for tls_priorities\n\nWe add initial support for being able to specify tls priorities in\npacemaker. For bundles this will happen via an env variable because\npacemaker_remote is started normally as a process and there is no\nsourcing of /etc/sysconfig/pacemaker.\n\nWIP as I want to test this review fully on stein too (a partial version\nof this patch was already tested on queens)\n\nDepends-On: I1971810f6a90f244ed5ced972a5fe7fde29dde86\nChange-Id: I703b5a429f48063474aace85bc45d948f5c91435\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/5454d3191f802be0f797b13d60ec33f1cc7fe73a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/5454d3191f802be0f797b13d60ec33f1cc7fe73a"}]},"branch":"refs/heads/master"},"ef012d87e9b1621e505138fac8f6a3bd33cc74a9":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2019-07-22 15:46:52.000000000","uploader":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"ref":"refs/changes/84/670184/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/84/670184/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/84/670184/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/84/670184/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/84/670184/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/84/670184/4"}}},"commit":{"parents":[{"commit":"610691ca6561c7630d243d5aa23f7c25a60c9a5c","subject":"Merge \"Allow custom per-service bind_options for haproxy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/610691ca6561c7630d243d5aa23f7c25a60c9a5c"}]}],"author":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2019-07-10 20:25:07.000000000","tz":120},"committer":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2019-07-22 15:41:41.000000000","tz":120},"subject":"Initial support for tls_priorities","message":"Initial support for tls_priorities\n\nWe add initial support for being able to specify tls priorities in\npacemaker. For bundles this will happen via an env variable because\npacemaker_remote is started normally as a process and there is no\nsourcing of /etc/sysconfig/pacemaker.\n\nTested on both queens and stein. Via a deploy and a redeploy against\nexisting cloud. Observed that:\nA) We got PCMK_tls_priorities inside /etc/sysconfig/pacemaker with the\nvalue that was passed in THT\nB) Containers had the following env variable set:\n  \"PCMK_tls_priorities\u003dnormal\",\n\nDepends-On: I1971810f6a90f244ed5ced972a5fe7fde29dde86\nChange-Id: I703b5a429f48063474aace85bc45d948f5c91435\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/ef012d87e9b1621e505138fac8f6a3bd33cc74a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/ef012d87e9b1621e505138fac8f6a3bd33cc74a9"}]},"branch":"refs/heads/master"},"be822af9bfae7f6611ed636be22a09a7b34d36cf":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2019-07-22 15:48:47.000000000","uploader":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"ref":"refs/changes/84/670184/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/84/670184/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/84/670184/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/84/670184/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/84/670184/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/84/670184/5"}}},"commit":{"parents":[{"commit":"610691ca6561c7630d243d5aa23f7c25a60c9a5c","subject":"Merge \"Allow custom per-service bind_options for haproxy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/610691ca6561c7630d243d5aa23f7c25a60c9a5c"}]}],"author":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2019-07-10 20:25:07.000000000","tz":120},"committer":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2019-07-22 15:47:42.000000000","tz":120},"subject":"Initial support for tls_priorities","message":"Initial support for tls_priorities\n\nWe add initial support for being able to specify tls priorities in\npacemaker. For bundles this will happen via an env variable because\npacemaker_remote is started normally as a process and there is no\nsourcing of /etc/sysconfig/pacemaker.\n\nTested on both queens and stein. Via a deploy and a redeploy against\nexisting cloud. Observed that:\nA) We got PCMK_tls_priorities inside /etc/sysconfig/pacemaker with the\nvalue that was passed in THT\nB) Containers had the following env variable set:\n  \"PCMK_tls_priorities\u003dnormal\",\n\nThe \u0027-e\u0027 addition is a noop in case the PCMK_tls_priorities is unset\nso that we do not change the signature of the resources and hence do\nnot needlessly restart the HA resource.\n\nDepends-On: I1971810f6a90f244ed5ced972a5fe7fde29dde86\nChange-Id: I703b5a429f48063474aace85bc45d948f5c91435\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/be822af9bfae7f6611ed636be22a09a7b34d36cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/be822af9bfae7f6611ed636be22a09a7b34d36cf"}]},"branch":"refs/heads/master"},"e6f9ffc09ab467c09854ea1065c8997200d6eb70":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2019-07-24 05:29:19.000000000","uploader":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"ref":"refs/changes/84/670184/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/84/670184/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/84/670184/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/84/670184/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/84/670184/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/84/670184/6"}}},"commit":{"parents":[{"commit":"7605e4e18cf4c73dd6623b5467f8ccf7e29d4c28","subject":"Merge \"Add tcp-check connect port line to haproxy redis stanza\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/7605e4e18cf4c73dd6623b5467f8ccf7e29d4c28"}]}],"author":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2019-07-10 20:25:07.000000000","tz":120},"committer":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2019-07-24 05:29:19.000000000","tz":0},"subject":"Initial support for tls_priorities","message":"Initial support for tls_priorities\n\nWe add initial support for being able to specify tls priorities in\npacemaker. For bundles this will happen via an env variable because\npacemaker_remote is started normally as a process and there is no\nsourcing of /etc/sysconfig/pacemaker.\n\nTested on both queens and stein. Via a deploy and a redeploy against\nexisting cloud. Observed that:\nA) We got PCMK_tls_priorities inside /etc/sysconfig/pacemaker with the\nvalue that was passed in THT\nB) Containers had the following env variable set:\n  \"PCMK_tls_priorities\u003dnormal\",\n\nThe \u0027-e\u0027 addition is a noop in case the PCMK_tls_priorities is unset\nso that we do not change the signature of the resources and hence do\nnot needlessly restart the HA resource.\n\nDepends-On: I1971810f6a90f244ed5ced972a5fe7fde29dde86\nChange-Id: I703b5a429f48063474aace85bc45d948f5c91435\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/e6f9ffc09ab467c09854ea1065c8997200d6eb70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/e6f9ffc09ab467c09854ea1065c8997200d6eb70"}]},"branch":"refs/heads/master"},"f1a593b6425f1e95ef789dd8ac9e32f095e347ff":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2019-07-27 07:59:45.000000000","uploader":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"ref":"refs/changes/84/670184/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/84/670184/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/84/670184/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/84/670184/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/84/670184/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/84/670184/7"}}},"commit":{"parents":[{"commit":"f92c3c90c31c77a56e8bf7d81596fdd85f7f45bf","subject":"Only assign redis_vip and redis_bind_opts if enabled","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/f92c3c90c31c77a56e8bf7d81596fdd85f7f45bf"}]}],"author":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2019-07-10 20:25:07.000000000","tz":120},"committer":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2019-07-27 07:59:45.000000000","tz":0},"subject":"Initial support for tls_priorities","message":"Initial support for tls_priorities\n\nWe add initial support for being able to specify tls priorities in\npacemaker. For bundles this will happen via an env variable because\npacemaker_remote is started normally as a process and there is no\nsourcing of /etc/sysconfig/pacemaker.\n\nTested on both queens and stein. Via a deploy and a redeploy against\nexisting cloud. Observed that:\nA) We got PCMK_tls_priorities inside /etc/sysconfig/pacemaker with the\nvalue that was passed in THT\nB) Containers had the following env variable set:\n  \"PCMK_tls_priorities\u003dnormal\",\n\nThe \u0027-e\u0027 addition is a noop in case the PCMK_tls_priorities is unset\nso that we do not change the signature of the resources and hence do\nnot needlessly restart the HA resource.\n\nDepends-On: I1971810f6a90f244ed5ced972a5fe7fde29dde86\nChange-Id: I703b5a429f48063474aace85bc45d948f5c91435\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/f1a593b6425f1e95ef789dd8ac9e32f095e347ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/f1a593b6425f1e95ef789dd8ac9e32f095e347ff"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
