)]}'
{"id":"openstack%2Fpuppet-tripleo~671526","triplet_id":"openstack%2Fpuppet-tripleo~master~I261eb30b52a3baee3b0e6d47e8f32f3c579930bf","project":"openstack/puppet-tripleo","branch":"master","topic":"haproxy-redis-fix","hashtags":[],"change_id":"I261eb30b52a3baee3b0e6d47e8f32f3c579930bf","subject":"Add tcp-check connect port line to haproxy redis stanza","status":"MERGED","created":"2019-07-18 17:11:00.000000000","updated":"2019-07-24 01:12:25.000000000","submitted":"2019-07-24 01:12:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"671526-1563930745784-1c1abdd6","meta_rev_id":"8f93b4fdded864545f0f0e8ba28d30be87ab1a1d","_number":671526,"virtual_id_number":671526,"owner":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"date":"2019-07-23 07:05:34.000000000","_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"date":"2019-07-23 13:03:47.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":8042,"name":"Jiří Stránský","email":"jistr@redhat.com","username":"jistr"},{"value":0,"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":2,"date":"2019-07-24 01:12:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-23 22:08:49.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8042,"name":"Jiří Stránský","email":"jistr@redhat.com","username":"jistr"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-07-22 08:11:26.000000000","_account_id":8042,"name":"Jiří Stránský","email":"jistr@redhat.com","username":"jistr"},{"value":0,"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-19 15:19:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"},{"value":1,"date":"2019-07-22 08:23:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":8042,"name":"Jiří Stránský","email":"jistr@redhat.com","username":"jistr"},{"value":1,"date":"2019-07-22 08:27:40.000000000","_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":8042,"name":"Jiří Stránský","email":"jistr@redhat.com","username":"jistr"},{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},{"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-18 17:45:39.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-07-19 15:19:41.000000000","updated_by":{"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"},"reviewer":{"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"},"state":"REVIEWER"},{"updated":"2019-07-22 08:11:26.000000000","updated_by":{"_account_id":8042,"name":"Jiří Stránský","email":"jistr@redhat.com","username":"jistr"},"reviewer":{"_account_id":8042,"name":"Jiří Stránský","email":"jistr@redhat.com","username":"jistr"},"state":"REVIEWER"},{"updated":"2019-07-22 08:23:44.000000000","updated_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"REVIEWER"},{"updated":"2019-07-23 07:05:34.000000000","updated_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"REVIEWER"},{"updated":"2019-07-23 13:03:47.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-07-23 22:08:49.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-24 01:12:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d6f6d92bd6ff7674e8a3f7a8e6f5ae6a67ff3fdb","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2019-07-18 17:11:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"905e5c2dd19b920b34771e49bf65ef7f3ca49aec","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-07-18 17:45:39.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"108209dd348dc946ecc8f5fdbfdb709922b7da97","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-18 19:58:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/26/671526/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/08403c1/ : FAILURE in 25m 36s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/26/671526/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/2897678/ : FAILURE in 26m 51s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/26/671526/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/3ce7ffa/ : FAILURE in 2h 43m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"da6ab2eed6fb7201c7daa7f48d01249d8433e930","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-07-18 20:02:14.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"309345f1315258713a5ceb4c1dd91e7d22b4f893","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-18 21:11:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/26/671526/1/check/puppet-openstack-lint-ubuntu-bionic/0828ded/ : FAILURE in 5m 19s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/26/671526/1/check/puppet-openstack-syntax-5-ubuntu-bionic/84608e8/ : SUCCESS in 4m 31s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/26/671526/1/check/puppet-openstack-syntax-6-ubuntu-bionic/b11096f/ : SUCCESS in 5m 08s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/26/671526/1/check/puppet-openstack-unit-5.5-centos-7/193e6d0/ : SUCCESS in 28m 27s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/26/671526/1/check/puppet-openstack-unit-6.0-ubuntu-bionic/597eb3e/ : SUCCESS in 16m 57s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/26/671526/1/check/puppet-openstack-unit-latest-ubuntu-bionic/4ff328a/ : SUCCESS in 17m 47s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/26/671526/1/check/tripleo-ci-centos-7-undercloud-containers/e0abf9e/ : SUCCESS in 1h 41m 41s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/26/671526/1/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/2af3891/ : FAILURE in 2h 14m 08s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/26/671526/1/check/tripleo-ci-centos-7-containers-multinode/428b054/ : SUCCESS in 2h 25m 42s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/26/671526/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/d28375f/ : FAILURE in 1h 14m 09s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/26/671526/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/e483fd3/ : FAILURE in 2h 46m 11s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/26/671526/1/check/tripleo-ci-centos-7-standalone/afb6be0/ : SUCCESS in 1h 53m 20s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/26/671526/1/check/tripleo-ci-fedora-28-standalone/4323e02/ : FAILURE in 28m 08s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/26/671526/1/check/tripleo-ci-centos-7-standalone-upgrade/1880d41/ : FAILURE in 1h 37m 02s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone http://logs.openstack.org/26/671526/1/check/tripleo-ci-centos-7-scenario001-standalone/77ff0cd/ : FAILURE in 53m 22s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/26/671526/1/check/tripleo-ci-centos-7-scenario003-standalone/a4b531f/ : FAILURE in 48m 42s\n- tripleo-ci-centos-7-scenario004-standalone http://logs.openstack.org/26/671526/1/check/tripleo-ci-centos-7-scenario004-standalone/cc624f3/ : FAILURE in 55m 28s\n- tripleo-ci-centos-7-scenario012-standalone http://logs.openstack.org/26/671526/1/check/tripleo-ci-centos-7-scenario012-standalone/ea9544b/ : FAILURE in 1h 00m 51s (non-voting)\n- tripleo-ci-centos-7-scenario010-multinode-oooq-container http://logs.openstack.org/26/671526/1/check/tripleo-ci-centos-7-scenario010-multinode-oooq-container/8f24601/ : FAILURE in 1h 58m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d8b61311672143ba2f80f476aacbf084fb04b9ab","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-07-18 21:42:07.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"32d5f3dbfa9bd1bc6457ec41f81259e46b52ca84","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 00:18:34.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/26/671526/1/check/puppet-openstack-lint-ubuntu-bionic/50dd779/ : FAILURE in 4m 23s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/26/671526/1/check/puppet-openstack-syntax-5-ubuntu-bionic/04685ea/ : SUCCESS in 4m 40s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/26/671526/1/check/puppet-openstack-syntax-6-ubuntu-bionic/d4974d2/ : SUCCESS in 4m 17s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/26/671526/1/check/puppet-openstack-unit-5.5-centos-7/1180bd3/ : SUCCESS in 32m 09s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/26/671526/1/check/puppet-openstack-unit-6.0-ubuntu-bionic/8840c44/ : SUCCESS in 15m 02s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/26/671526/1/check/puppet-openstack-unit-latest-ubuntu-bionic/ee4a5c6/ : SUCCESS in 15m 22s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/26/671526/1/check/tripleo-ci-centos-7-undercloud-containers/0a0e59a/ : SUCCESS in 1h 39m 43s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/26/671526/1/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/fae7725/ : FAILURE in 1h 40m 30s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/26/671526/1/check/tripleo-ci-centos-7-containers-multinode/2bc788a/ : SUCCESS in 2h 02m 56s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/26/671526/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/68489e5/ : SUCCESS in 2h 17m 06s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/26/671526/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/ab3514c/ : FAILURE in 2h 31m 11s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/26/671526/1/check/tripleo-ci-centos-7-standalone/6fc9827/ : SUCCESS in 1h 44m 23s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/26/671526/1/check/tripleo-ci-fedora-28-standalone/d7ca8ca/ : FAILURE in 25m 40s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/26/671526/1/check/tripleo-ci-centos-7-standalone-upgrade/83b708f/ : FAILURE in 2h 00m 12s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone http://logs.openstack.org/26/671526/1/check/tripleo-ci-centos-7-scenario001-standalone/49d88ae/ : SUCCESS in 1h 47m 37s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/26/671526/1/check/tripleo-ci-centos-7-scenario003-standalone/1b977b6/ : SUCCESS in 1h 24m 03s\n- tripleo-ci-centos-7-scenario004-standalone http://logs.openstack.org/26/671526/1/check/tripleo-ci-centos-7-scenario004-standalone/b5d3ee9/ : SUCCESS in 1h 43m 13s\n- tripleo-ci-centos-7-scenario012-standalone http://logs.openstack.org/26/671526/1/check/tripleo-ci-centos-7-scenario012-standalone/8e81207/ : SUCCESS in 1h 57m 00s (non-voting)\n- tripleo-ci-centos-7-scenario010-multinode-oooq-container http://logs.openstack.org/26/671526/1/check/tripleo-ci-centos-7-scenario010-multinode-oooq-container/c6d0478/ : FAILURE in 1h 54m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3dea91487f359bc4d1b2e038bc6493831847561b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-19 00:30:14.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/26/671526/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/c722f28/ : FAILURE in 1h 38m 01s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/26/671526/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/5ee26e8/ : FAILURE in 2h 37m 16s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/26/671526/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/f244b6d/ : FAILURE in 2h 45m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6df1987b8f94bb430ed045adceb6db597542de32","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-07-19 02:16:41.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0f0128ca5edd608b4a434fb453254909ceab6027","author":{"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"},"date":"2019-07-19 04:48:56.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"d97effdf9743fb23d43497d20d29c6f0a4d07aac","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-19 04:50:36.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/26/671526/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/c623153/ : FAILURE in 2h 31m 24s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/26/671526/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/540a6f0/ : FAILURE in 2h 29m 27s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/26/671526/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/780d196/ : FAILURE in 1h 54m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ac22e0c1e7cc6f2e1f1d1259d3c4ae51f5022253","author":{"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"},"date":"2019-07-19 04:52:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"64152f67245c751e572cf11dbc9401e6fccdd84c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 05:18:39.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/26/671526/1/check/puppet-openstack-lint-ubuntu-bionic/24af308/ : FAILURE in 6m 51s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/26/671526/1/check/puppet-openstack-syntax-5-ubuntu-bionic/38c62c1/ : SUCCESS in 5m 55s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/26/671526/1/check/puppet-openstack-syntax-6-ubuntu-bionic/c62039f/ : SUCCESS in 5m 04s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/26/671526/1/check/puppet-openstack-unit-5.5-centos-7/c87c522/ : SUCCESS in 28m 24s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/26/671526/1/check/puppet-openstack-unit-6.0-ubuntu-bionic/bb18e1b/ : SUCCESS in 15m 00s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/26/671526/1/check/puppet-openstack-unit-latest-ubuntu-bionic/425ebb1/ : SUCCESS in 15m 39s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/26/671526/1/check/tripleo-ci-centos-7-undercloud-containers/9534590/ : SUCCESS in 1h 36m 55s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/26/671526/1/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/af73f60/ : FAILURE in 1h 58m 28s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/26/671526/1/check/tripleo-ci-centos-7-containers-multinode/d22b447/ : SUCCESS in 2h 14m 03s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/26/671526/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/5f9e249/ : SUCCESS in 2h 06m 30s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/26/671526/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/5154a77/ : FAILURE in 2h 29m 35s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/26/671526/1/check/tripleo-ci-centos-7-standalone/91bdd04/ : SUCCESS in 1h 47m 16s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/26/671526/1/check/tripleo-ci-fedora-28-standalone/06b3d04/ : FAILURE in 24m 49s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/26/671526/1/check/tripleo-ci-centos-7-standalone-upgrade/e989edc/ : FAILURE in 1h 43m 41s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone http://logs.openstack.org/26/671526/1/check/tripleo-ci-centos-7-scenario001-standalone/72d4814/ : SUCCESS in 1h 32m 43s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/26/671526/1/check/tripleo-ci-centos-7-scenario003-standalone/097012c/ : SUCCESS in 1h 16m 25s\n- tripleo-ci-centos-7-scenario004-standalone http://logs.openstack.org/26/671526/1/check/tripleo-ci-centos-7-scenario004-standalone/b49f251/ : FAILURE in 52m 20s\n- tripleo-ci-centos-7-scenario012-standalone http://logs.openstack.org/26/671526/1/check/tripleo-ci-centos-7-scenario012-standalone/57114f0/ : SUCCESS in 1h 44m 17s (non-voting)\n- tripleo-ci-centos-7-scenario010-multinode-oooq-container http://logs.openstack.org/26/671526/1/check/tripleo-ci-centos-7-scenario010-multinode-oooq-container/e555420/ : FAILURE in 2h 02m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3488751500108f36051562ecf9e6ee67ac8e0484","author":{"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"},"date":"2019-07-19 05:22:05.000000000","message":"Patch Set 1:\n\nstandalone failed because of:\n\n2019-07-19 03:04:03 | TASK [Run tripleo-container-image-prepare logged to /var/log/tripleo-container-image-prepare.log] ***\n2019-07-19 03:27:06 | fatal: [undercloud]: FAILED! \u003d\u003e {\"censored\": \"the output has been hidden due to the fact that \u0027no_log: true\u0027 was specified for this result\", \"changed\": true}\n2019-07-19 03:27:06 | \n\n2019-07-19 03:26:10.651 17123 ERROR root [  ] Image prepare failed: 401 Client Error: Unauthorized for url: http://mirror.bhs1.ovh.openstack.org:8082/v2/tripleomaster/centos-binary-nova-libvirt/blobs/sha256:4c3152bda61158839e096bc06c363fb74f80e1b4c8a568f4365207f4542a65c9","accounts_in_message":[],"_revision_number":1},{"id":"9fcdfd67071da5f57cc07882b362183af80f4ee8","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2019-07-19 06:33:54.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"665147b76ae428bc0d7d784dd87b7084849821f6","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2019-07-19 06:37:46.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"d8bf9e9fa743f508b8273f8b89feb03caa8529bf","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2019-07-19 06:40:12.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"705361410155a7979c21c7b2840ea185952860c0","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2019-07-19 07:13:39.000000000","message":"Patch Set 3: Workflow-1\n\nStill probably broken in the ssl case. Testing a fix","accounts_in_message":[],"_revision_number":3},{"id":"945dfb25d8eb520ed9b2f60181c26a1b89d0e239","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2019-07-19 07:53:34.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"afc57576692d2bfc9605dac7668650a26b545111","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2019-07-19 07:55:00.000000000","message":"Patch Set 3:\n\nI have tested the non-tls case and it works okay with haproxy from both 7.6 and 7.7:\nredis,controller-0.internalapi.localdomain,0,0,0,0,,0,0,0,,0,,0,0,0,0,UP,1,1,0,0,0,44,0,,1,19,1,,0,,2,0,,0,L7OK,0,0,,,,,,,0,,,,0,0,,,,,-1,(tcp-check),,0,0,0,0,\nredis,controller-1.internalapi.localdomain,0,0,0,0,,0,0,0,,0,,0,0,0,0,DOWN,1,1,0,1,1,40,40,,1,19,2,,0,,2,0,,0,L7TOUT,,2001,,,,,,,0,,,,0,0,,,,,-1,,,0,0,0,0,\nredis,controller-2.internalapi.localdomain,0,0,0,0,,0,0,0,,0,,0,0,0,0,DOWN,1,1,0,1,1,40,40,,1,19,3,,0,,2,0,,0,L7TOUT,,2001,,,,,,,0,,,,0,0,,,,,-1,,,0,0,0,0,\n\nI *think* it TLS should also work, but am trying to test that as well","accounts_in_message":[],"_revision_number":3},{"id":"44c324a46d9a913a1a82ceeaf37d349b665c85e0","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2019-07-19 08:29:00.000000000","message":"Patch Set 4:\n\nunlikely that I get to test TLS today, so I think we can merge this as is anyway","accounts_in_message":[],"_revision_number":4},{"id":"60a8f3a156274921f76f51e55671b2f6312e973d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 10:40:07.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/26/671526/4/check/puppet-openstack-lint-ubuntu-bionic/fe97cbd/ : SUCCESS in 4m 43s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/26/671526/4/check/puppet-openstack-syntax-5-ubuntu-bionic/40076f4/ : SUCCESS in 3m 55s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/26/671526/4/check/puppet-openstack-syntax-6-ubuntu-bionic/9a7a1c9/ : SUCCESS in 6m 04s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/26/671526/4/check/puppet-openstack-unit-5.5-centos-7/11a7dc7/ : SUCCESS in 29m 40s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/26/671526/4/check/puppet-openstack-unit-6.0-ubuntu-bionic/194144f/ : SUCCESS in 13m 58s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/26/671526/4/check/puppet-openstack-unit-latest-ubuntu-bionic/b22de21/ : SUCCESS in 18m 06s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/26/671526/4/check/tripleo-ci-centos-7-undercloud-containers/d7dac5e/ : SUCCESS in 1h 30m 57s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/26/671526/4/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/8f6bd99/ : FAILURE in 2h 15m 08s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/26/671526/4/check/tripleo-ci-centos-7-containers-multinode/eb13fb2/ : SUCCESS in 2h 19m 37s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/26/671526/4/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/8e941dd/ : SUCCESS in 2h 24m 01s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/26/671526/4/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/8e19863/ : FAILURE in 2h 42m 20s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/26/671526/4/check/tripleo-ci-centos-7-standalone/e9e8ee1/ : SUCCESS in 2h 14m 48s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/26/671526/4/check/tripleo-ci-fedora-28-standalone/5fb161d/ : FAILURE in 26m 52s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/26/671526/4/check/tripleo-ci-centos-7-standalone-upgrade/c32c1bc/ : FAILURE in 2h 02m 30s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone http://logs.openstack.org/26/671526/4/check/tripleo-ci-centos-7-scenario001-standalone/c53b143/ : SUCCESS in 1h 54m 02s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/26/671526/4/check/tripleo-ci-centos-7-scenario003-standalone/ccbb0cb/ : SUCCESS in 1h 18m 17s\n- tripleo-ci-centos-7-scenario004-standalone http://logs.openstack.org/26/671526/4/check/tripleo-ci-centos-7-scenario004-standalone/f909be4/ : SUCCESS in 1h 41m 30s\n- tripleo-ci-centos-7-scenario012-standalone http://logs.openstack.org/26/671526/4/check/tripleo-ci-centos-7-scenario012-standalone/6d35462/ : SUCCESS in 2h 18m 13s (non-voting)\n- tripleo-ci-centos-7-scenario010-multinode-oooq-container http://logs.openstack.org/26/671526/4/check/tripleo-ci-centos-7-scenario010-multinode-oooq-container/1724fc8/ : FAILURE in 2h 09m 14s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"00a79bad8c560bbbf98f59347e9640c48c299bc8","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-19 10:51:37.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/26/671526/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/f00044f/ : SUCCESS in 2h 32m 34s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/26/671526/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/442a457/ : FAILURE in 2h 28m 47s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/26/671526/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/a4074c0/ : SUCCESS in 2h 56m 26s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"be2cb5fa12cf417c78f9efeee4cd18f332650732","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2019-07-19 11:09:45.000000000","message":"Patch Set 4:\n\ncheck-rdo","accounts_in_message":[],"_revision_number":4},{"id":"642a02ece81f3b13373837eb983dba6af2518f5e","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-19 14:28:14.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/26/671526/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/d15b726/ : SUCCESS in 3h 13m 34s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 finger://ze06.softwarefactory-project.io:7979/386ef4d7eaa94378abeea90d97481cca : RETRY_LIMIT in 6s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/26/671526/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/373f8b6/ : FAILURE in 29m 46s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"da78ac09741e51d614cc733f84b5371b63eec40f","author":{"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"},"date":"2019-07-19 15:19:41.000000000","message":"Patch Set 4: Code-Review+1\n\ntls-everywhere seems ok:\n\nlisten redis\n  bind 172.17.1.102:6379 transparent\n  balance first\n  option tcp-check\n  option tcplog\n  tcp-check connect port 6379 ssl\n  tcp-check send AUTH\\ jdYxAd9KwsPng7NfgYe3ux3Er\\r\\n\n  tcp-check send PING\\r\\n\n  tcp-check expect string +PONG\n  tcp-check send info\\ replication\\r\\n\n  tcp-check expect string role:master\n  tcp-check send QUIT\\r\\n\n  tcp-check expect string +OK\n  server controller-0.internalapi.redhat.local 172.17.1.10:6379 ca-file /etc/ipa/ca.crt check check-ssl fall 5 inter 2000 on-marked-down shutdown-sessions rise 2\n  server controller-1.internalapi.redhat.local 172.17.1.30:6379 ca-file /etc/ipa/ca.crt check check-ssl fall 5 inter 2000 on-marked-down shutdown-sessions rise 2\n  server controller-2.internalapi.redhat.local 172.17.1.12:6379 ca-file /etc/ipa/ca.crt check check-ssl fall 5 inter 2000 on-marked-down shutdown-sessions rise 2\n\n\n\n[root@controller-0 haproxy]# echo \"show info;show stat\" | nc -U stats  |grep ^redis\nredis,FRONTEND,,,0,0,4096,0,0,0,0,0,0,,,,,OPEN,,,,,,,,,1,19,0,,,,0,0,0,0,,,,,,,,,,,0,0,0,,,0,0,0,0,,,,,,,,\nredis,controller-0.internalapi.redhat.local,0,0,0,0,,0,0,0,,0,,0,0,0,0,UP,1,1,0,0,0,39,0,,1,19,1,,0,,2,0,,0,L7OK,0,1,,,,,,,0,,,,0,0,,,,,-1,(tcp-check),,0,0,0,0,\nredis,controller-1.internalapi.redhat.local,0,0,0,0,,0,0,0,,0,,0,0,0,0,DOWN,1,1,0,1,1,35,35,,1,19,2,,0,,2,0,,0,L7TOUT,,2001,,,,,,,0,,,,0,0,,,,,-1,,,0,0,0,0,\nredis,controller-2.internalapi.redhat.local,0,0,0,0,,0,0,0,,0,,0,0,0,0,DOWN,1,1,0,1,1,35,35,,1,19,3,,0,,2,0,,0,L7TOUT,,2001,,,,,,,0,,,,0,0,,,,,-1,,,0,0,0,0,\nredis,BACKEND,0,0,0,0,410,0,0,0,0,0,,0,0,0,0,UP,1,1,0,,0,39,0,,1,19,0,,0,,1,0,,0,,,,,,,,,,,,,,0,0,0,0,0,0,-1,,,0,0,0,0,","accounts_in_message":[],"_revision_number":4},{"id":"90e4c3d55b34cf029a6fd51899b2374ca63620e0","author":{"_account_id":8042,"name":"Jiří Stránský","email":"jistr@redhat.com","username":"jistr"},"date":"2019-07-22 08:11:26.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"aeb304fafb0b7162bb69258b193feb2ff5326b5b","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2019-07-22 08:23:44.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"82eadfce73545d1073beaa8eeedc9b9696cbd6dd","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2019-07-22 08:27:40.000000000","message":"Patch Set 4: Workflow+1\n\ntested both tls and non-tls, merging this","accounts_in_message":[],"_revision_number":4},{"id":"7f0fb4d3d8d5849b0a12c2d792700a994e40d8e1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 08:27:52.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"97306a29ad6a89a3b98ca8c0671e0370a054948e","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2019-07-22 08:37:06.000000000","message":"Patch Set 4: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit c71f507b64b2197d76308409cf0d7f406ff32eff","accounts_in_message":[],"_revision_number":4},{"id":"846222dca75adc24a38ef9d6f99531517156c3ee","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 10:54:30.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/26/671526/4/gate/puppet-openstack-lint-ubuntu-bionic/fed5c85/ : SUCCESS in 7m 41s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/26/671526/4/gate/puppet-openstack-syntax-5-ubuntu-bionic/e1518f2/ : SUCCESS in 6m 27s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/26/671526/4/gate/puppet-openstack-syntax-6-ubuntu-bionic/50ecf60/ : SUCCESS in 6m 55s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/26/671526/4/gate/puppet-openstack-unit-5.5-centos-7/67e39ba/ : SUCCESS in 31m 20s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/26/671526/4/gate/puppet-openstack-unit-6.0-ubuntu-bionic/9c6cb73/ : SUCCESS in 19m 27s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/26/671526/4/gate/tripleo-ci-centos-7-undercloud-containers/5610169/ : FAILURE in 1h 08m 31s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/26/671526/4/gate/tripleo-ci-centos-7-containers-multinode/1887b66/ : SUCCESS in 2h 21m 18s\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/26/671526/4/gate/tripleo-ci-centos-7-standalone/11b184f/ : FAILURE in 57m 35s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/26/671526/4/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/a371bbe/ : FAILURE in 1h 07m 12s\n- tripleo-ci-centos-7-scenario001-standalone http://logs.openstack.org/26/671526/4/gate/tripleo-ci-centos-7-scenario001-standalone/6a84805/ : FAILURE in 53m 39s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/26/671526/4/gate/tripleo-ci-centos-7-scenario003-standalone/d585837/ : SUCCESS in 1h 18m 37s\n- tripleo-ci-centos-7-scenario004-standalone http://logs.openstack.org/26/671526/4/gate/tripleo-ci-centos-7-scenario004-standalone/2c59375/ : SUCCESS in 1h 44m 18s","accounts_in_message":[],"_revision_number":4},{"id":"933d5b4d62b0e88f1088805306dc40e3d15bd3c2","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2019-07-22 12:03:02.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"143f31f501a54eaa2dc93f71b9669131025d1c67","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2019-07-22 12:03:15.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"66b8ba677516da9391de3fc7b0a8b9a441562c2d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-22 14:54:11.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/26/671526/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/e8768b4/ : SUCCESS in 2h 31m 44s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/26/671526/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/1093e4a/ : FAILURE in 1h 06m 30s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/26/671526/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/e7b9bd9/ : SUCCESS in 2h 49m 11s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"ca8dc699a516033fec73874bfca3c1a830b8af8c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 14:55:09.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/26/671526/4/check/puppet-openstack-lint-ubuntu-bionic/5822434/ : SUCCESS in 5m 39s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/26/671526/4/check/puppet-openstack-syntax-5-ubuntu-bionic/95e85df/ : SUCCESS in 4m 53s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/26/671526/4/check/puppet-openstack-syntax-6-ubuntu-bionic/a3e9219/ : SUCCESS in 4m 28s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/26/671526/4/check/puppet-openstack-unit-5.5-centos-7/f67e7d6/ : SUCCESS in 31m 57s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/26/671526/4/check/puppet-openstack-unit-6.0-ubuntu-bionic/e0ce4a9/ : SUCCESS in 16m 51s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/26/671526/4/check/puppet-openstack-unit-latest-ubuntu-bionic/79e5a3e/ : SUCCESS in 19m 17s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/26/671526/4/check/tripleo-ci-centos-7-undercloud-containers/7415ffe/ : SUCCESS in 1h 35m 29s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/26/671526/4/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/a492a4d/ : FAILURE in 1h 27m 08s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/26/671526/4/check/tripleo-ci-centos-7-containers-multinode/9c0e00d/ : SUCCESS in 2h 19m 05s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/26/671526/4/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/d4cab5e/ : SUCCESS in 2h 17m 11s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/26/671526/4/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/b49b4c2/ : SUCCESS in 2h 43m 01s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/26/671526/4/check/tripleo-ci-centos-7-standalone/e741864/ : SUCCESS in 2h 13m 05s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/26/671526/4/check/tripleo-ci-fedora-28-standalone/c787528/ : FAILURE in 1h 35m 41s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/26/671526/4/check/tripleo-ci-centos-7-standalone-upgrade/bc8b3d2/ : FAILURE in 2h 01m 00s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone http://logs.openstack.org/26/671526/4/check/tripleo-ci-centos-7-scenario001-standalone/94816f0/ : SUCCESS in 1h 42m 38s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/26/671526/4/check/tripleo-ci-centos-7-scenario003-standalone/650a3df/ : SUCCESS in 1h 11m 21s\n- tripleo-ci-centos-7-scenario004-standalone http://logs.openstack.org/26/671526/4/check/tripleo-ci-centos-7-scenario004-standalone/157a463/ : SUCCESS in 1h 43m 10s\n- tripleo-ci-centos-7-scenario012-standalone http://logs.openstack.org/26/671526/4/check/tripleo-ci-centos-7-scenario012-standalone/d942ea6/ : SUCCESS in 1h 46m 51s (non-voting)\n- tripleo-ci-centos-7-scenario010-multinode-oooq-container http://logs.openstack.org/26/671526/4/check/tripleo-ci-centos-7-scenario010-multinode-oooq-container/a0ed437/ : FAILURE in 2h 01m 15s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"b97eb3f963d44e760ba6af00f0e1581177fe6e07","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 14:55:33.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"1b63e69804970f93c55189e49cee1625f86fcbdf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 17:29:38.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/26/671526/4/gate/puppet-openstack-lint-ubuntu-bionic/deefcc7/ : SUCCESS in 6m 08s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/26/671526/4/gate/puppet-openstack-syntax-5-ubuntu-bionic/fd9597c/ : SUCCESS in 7m 48s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/26/671526/4/gate/puppet-openstack-syntax-6-ubuntu-bionic/e980d03/ : SUCCESS in 5m 36s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/26/671526/4/gate/puppet-openstack-unit-5.5-centos-7/8ac7aa9/ : SUCCESS in 28m 53s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/26/671526/4/gate/puppet-openstack-unit-6.0-ubuntu-bionic/38d2ab1/ : SUCCESS in 19m 04s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/26/671526/4/gate/tripleo-ci-centos-7-undercloud-containers/9a32899/ : SUCCESS in 1h 47m 54s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/26/671526/4/gate/tripleo-ci-centos-7-containers-multinode/ea6415a/ : SUCCESS in 2h 26m 54s\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/26/671526/4/gate/tripleo-ci-centos-7-standalone/4951f2a/ : SUCCESS in 1h 44m 26s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/26/671526/4/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/abd4a55/ : SUCCESS in 2h 20m 02s\n- tripleo-ci-centos-7-scenario001-standalone http://logs.openstack.org/26/671526/4/gate/tripleo-ci-centos-7-scenario001-standalone/b0610aa/ : FAILURE in 1h 19m 03s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/26/671526/4/gate/tripleo-ci-centos-7-scenario003-standalone/cfb6824/ : SUCCESS in 1h 19m 49s\n- tripleo-ci-centos-7-scenario004-standalone http://logs.openstack.org/26/671526/4/gate/tripleo-ci-centos-7-scenario004-standalone/2051ecc/ : SUCCESS in 1h 35m 57s","accounts_in_message":[],"_revision_number":4},{"id":"f4f2b9f49a00105dc141dde7fae7d665936c65fa","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2019-07-22 20:46:08.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"b3209b96141492000d5ebb1a8c5f73e4edb950ca","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 23:28:40.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/26/671526/4/check/puppet-openstack-lint-ubuntu-bionic/eba894a/ : SUCCESS in 4m 43s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/26/671526/4/check/puppet-openstack-syntax-5-ubuntu-bionic/2cb33ea/ : SUCCESS in 4m 28s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/26/671526/4/check/puppet-openstack-syntax-6-ubuntu-bionic/870e1fd/ : SUCCESS in 5m 36s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/26/671526/4/check/puppet-openstack-unit-5.5-centos-7/0cf5c59/ : SUCCESS in 31m 43s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/26/671526/4/check/puppet-openstack-unit-6.0-ubuntu-bionic/bd8c1fc/ : SUCCESS in 16m 05s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/26/671526/4/check/puppet-openstack-unit-latest-ubuntu-bionic/7c6d194/ : SUCCESS in 15m 23s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/26/671526/4/check/tripleo-ci-centos-7-undercloud-containers/ec11ebb/ : SUCCESS in 1h 35m 01s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/26/671526/4/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/adaaf14/ : FAILURE in 2h 39m 11s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/26/671526/4/check/tripleo-ci-centos-7-containers-multinode/e229552/ : SUCCESS in 2h 39m 06s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/26/671526/4/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/95a4647/ : SUCCESS in 2h 22m 58s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/26/671526/4/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/bff0b4e/ : SUCCESS in 2h 23m 43s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/26/671526/4/check/tripleo-ci-centos-7-standalone/c12290d/ : SUCCESS in 1h 51m 22s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/26/671526/4/check/tripleo-ci-fedora-28-standalone/6eb9ace/ : SUCCESS in 1h 38m 40s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/26/671526/4/check/tripleo-ci-centos-7-standalone-upgrade/de99a0d/ : FAILURE in 1h 58m 33s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone http://logs.openstack.org/26/671526/4/check/tripleo-ci-centos-7-scenario001-standalone/0024439/ : FAILURE in 1h 12m 41s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/26/671526/4/check/tripleo-ci-centos-7-scenario003-standalone/53ec773/ : SUCCESS in 1h 19m 40s\n- tripleo-ci-centos-7-scenario004-standalone http://logs.openstack.org/26/671526/4/check/tripleo-ci-centos-7-scenario004-standalone/750f956/ : SUCCESS in 1h 41m 05s\n- tripleo-ci-centos-7-scenario012-standalone http://logs.openstack.org/26/671526/4/check/tripleo-ci-centos-7-scenario012-standalone/ded63a6/ : SUCCESS in 1h 46m 58s (non-voting)\n- tripleo-ci-centos-7-scenario010-multinode-oooq-container http://logs.openstack.org/26/671526/4/check/tripleo-ci-centos-7-scenario010-multinode-oooq-container/8af4cd0/ : FAILURE in 1h 53m 56s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"278dc0e092a3fa76d7ebe9107b10a7a870eaf9d3","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-22 23:30:13.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/26/671526/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/0274e2d/ : SUCCESS in 2h 31m 03s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/26/671526/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/072afb5/ : SUCCESS in 2h 28m 04s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/26/671526/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/5fd1d26/ : SUCCESS in 2h 42m 32s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"fe108d9d0065dd378eb36899a93f17abca16df8d","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2019-07-23 04:21:56.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"ad775267161df07fd057f1860bf24873d8b11772","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-23 07:03:18.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/26/671526/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/707e20a/ : SUCCESS in 2h 37m 14s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/26/671526/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/9af96e4/ : SUCCESS in 2h 28m 06s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/26/671526/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/71277a0/ : SUCCESS in 2h 38m 28s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e76ef597250abc57dd93e67f7406b92a5a44dafd","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2019-07-23 07:05:34.000000000","message":"Patch Set 4:\n\n\u003e recheck\n\nsorry that scenario1 is an actual bug https://bugs.launchpad.net/tripleo/+bug/1837451 but that should help https://review.opendev.org/#/c/672197/1","accounts_in_message":[],"_revision_number":4},{"id":"f9fbe5a6640d13eeb9d1ae0f1115b09c0aef1093","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 07:09:38.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/26/671526/4/check/puppet-openstack-lint-ubuntu-bionic/2e61aa6/ : SUCCESS in 5m 49s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/26/671526/4/check/puppet-openstack-syntax-5-ubuntu-bionic/ab24a35/ : FAILURE in 4m 23s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/26/671526/4/check/puppet-openstack-syntax-6-ubuntu-bionic/5fc314d/ : SUCCESS in 5m 22s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/26/671526/4/check/puppet-openstack-unit-5.5-centos-7/c6b74e0/ : SUCCESS in 30m 33s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/26/671526/4/check/puppet-openstack-unit-6.0-ubuntu-bionic/d80337a/ : SUCCESS in 18m 31s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/26/671526/4/check/puppet-openstack-unit-latest-ubuntu-bionic/7d82573/ : SUCCESS in 17m 46s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/26/671526/4/check/tripleo-ci-centos-7-undercloud-containers/7706985/ : SUCCESS in 1h 26m 24s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/26/671526/4/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/0dc8616/ : FAILURE in 1h 54m 59s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/26/671526/4/check/tripleo-ci-centos-7-containers-multinode/763d524/ : SUCCESS in 2h 30m 10s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/26/671526/4/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/8eb88b7/ : SUCCESS in 2h 20m 07s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/26/671526/4/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/7da615b/ : SUCCESS in 2h 29m 10s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/26/671526/4/check/tripleo-ci-centos-7-standalone/77bb10d/ : SUCCESS in 1h 40m 12s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/26/671526/4/check/tripleo-ci-fedora-28-standalone/1d85df0/ : SUCCESS in 1h 43m 59s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/26/671526/4/check/tripleo-ci-centos-7-standalone-upgrade/8e533b9/ : FAILURE in 1h 52m 27s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone http://logs.openstack.org/26/671526/4/check/tripleo-ci-centos-7-scenario001-standalone/cba92cf/ : FAILURE in 1h 13m 59s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/26/671526/4/check/tripleo-ci-centos-7-scenario003-standalone/4ec2eb4/ : SUCCESS in 1h 11m 28s\n- tripleo-ci-centos-7-scenario004-standalone http://logs.openstack.org/26/671526/4/check/tripleo-ci-centos-7-scenario004-standalone/2f82476/ : FAILURE in 1h 32m 41s\n- tripleo-ci-centos-7-scenario012-standalone http://logs.openstack.org/26/671526/4/check/tripleo-ci-centos-7-scenario012-standalone/83457e5/ : SUCCESS in 1h 47m 40s (non-voting)\n- tripleo-ci-centos-7-scenario010-multinode-oooq-container http://logs.openstack.org/26/671526/4/check/tripleo-ci-centos-7-scenario010-multinode-oooq-container/dd26320/ : FAILURE in 1h 43m 41s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"7dcc3507d81cc0f8c0a1a7c4ef833794231a475d","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-07-23 13:03:47.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"0b2c415aca7483931f9f9e965d976d11df2ee795","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 15:48:56.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/26/671526/4/check/puppet-openstack-lint-ubuntu-bionic/dd58d85/ : SUCCESS in 7m 46s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/26/671526/4/check/puppet-openstack-syntax-5-ubuntu-bionic/b4a91e8/ : FAILURE in 5m 08s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/26/671526/4/check/puppet-openstack-syntax-6-ubuntu-bionic/015cbef/ : FAILURE in 5m 09s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/26/671526/4/check/puppet-openstack-unit-5.5-centos-7/729bdc0/ : SUCCESS in 29m 48s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/26/671526/4/check/puppet-openstack-unit-6.0-ubuntu-bionic/72e3c08/ : SUCCESS in 21m 57s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/26/671526/4/check/puppet-openstack-unit-latest-ubuntu-bionic/8abdab2/ : SUCCESS in 18m 40s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/26/671526/4/check/tripleo-ci-centos-7-undercloud-containers/629828c/ : SUCCESS in 1h 31m 01s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/26/671526/4/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/7669da5/ : FAILURE in 2h 10m 59s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/26/671526/4/check/tripleo-ci-centos-7-containers-multinode/0caad07/ : SUCCESS in 2h 37m 51s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/26/671526/4/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/2f964e0/ : SUCCESS in 2h 23m 53s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/26/671526/4/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/0eebd12/ : SUCCESS in 2h 26m 38s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/26/671526/4/check/tripleo-ci-centos-7-standalone/f1e6395/ : SUCCESS in 1h 59m 24s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/26/671526/4/check/tripleo-ci-fedora-28-standalone/768d37c/ : FAILURE in 1h 46m 28s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/26/671526/4/check/tripleo-ci-centos-7-standalone-upgrade/ff9f78a/ : FAILURE in 1h 49m 35s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone http://logs.openstack.org/26/671526/4/check/tripleo-ci-centos-7-scenario001-standalone/a36ffe1/ : FAILURE in 1h 13m 49s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/26/671526/4/check/tripleo-ci-centos-7-scenario003-standalone/f09aa66/ : SUCCESS in 1h 13m 06s\n- tripleo-ci-centos-7-scenario004-standalone http://logs.openstack.org/26/671526/4/check/tripleo-ci-centos-7-scenario004-standalone/421e3b1/ : SUCCESS in 1h 59m 07s\n- tripleo-ci-centos-7-scenario012-standalone http://logs.openstack.org/26/671526/4/check/tripleo-ci-centos-7-scenario012-standalone/57712f1/ : SUCCESS in 1h 40m 18s (non-voting)\n- tripleo-ci-centos-7-scenario010-multinode-oooq-container http://logs.openstack.org/26/671526/4/check/tripleo-ci-centos-7-scenario010-multinode-oooq-container/12a251a/ : FAILURE in 1h 45m 20s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"574927cdd1e9175d4d2e3819aa6db9b740998d4d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-23 16:22:23.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/26/671526/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/48e864d/ : FAILURE in 1h 36m 03s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/26/671526/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/0b69a0d/ : SUCCESS in 2h 29m 42s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/26/671526/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/15fbc41/ : FAILURE in 3h 17m 12s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c34364f1cb4debd817f520280cbcebe5c0bfe439","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2019-07-23 19:07:07.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"7574f1dd429539d8ad4232b2a95ac112e7f31984","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 21:53:46.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/26/671526/4/check/puppet-openstack-lint-ubuntu-bionic/3330824/ : SUCCESS in 5m 09s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/26/671526/4/check/puppet-openstack-syntax-5-ubuntu-bionic/98d4f2f/ : SUCCESS in 6m 07s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/26/671526/4/check/puppet-openstack-syntax-6-ubuntu-bionic/6b9c359/ : SUCCESS in 5m 26s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/26/671526/4/check/puppet-openstack-unit-5.5-centos-7/ce5b28a/ : SUCCESS in 33m 52s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/26/671526/4/check/puppet-openstack-unit-6.0-ubuntu-bionic/b68c9a1/ : SUCCESS in 17m 29s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/26/671526/4/check/puppet-openstack-unit-latest-ubuntu-bionic/f11485f/ : SUCCESS in 18m 03s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/26/671526/4/check/tripleo-ci-centos-7-undercloud-containers/df6a172/ : SUCCESS in 1h 39m 12s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/26/671526/4/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/7b01d38/ : FAILURE in 2h 07m 59s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/26/671526/4/check/tripleo-ci-centos-7-containers-multinode/885b4a6/ : SUCCESS in 2h 29m 47s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/26/671526/4/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/785508d/ : SUCCESS in 2h 12m 36s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/26/671526/4/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/62effca/ : SUCCESS in 2h 37m 18s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/26/671526/4/check/tripleo-ci-centos-7-standalone/3488dc2/ : SUCCESS in 1h 54m 36s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/26/671526/4/check/tripleo-ci-fedora-28-standalone/b6d77e3/ : SUCCESS in 1h 51m 08s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/26/671526/4/check/tripleo-ci-centos-7-standalone-upgrade/d61265c/ : FAILURE in 1h 56m 02s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone http://logs.openstack.org/26/671526/4/check/tripleo-ci-centos-7-scenario001-standalone/e5be24f/ : SUCCESS in 1h 43m 14s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/26/671526/4/check/tripleo-ci-centos-7-scenario003-standalone/146702b/ : SUCCESS in 1h 18m 04s\n- tripleo-ci-centos-7-scenario004-standalone http://logs.openstack.org/26/671526/4/check/tripleo-ci-centos-7-scenario004-standalone/0632d57/ : SUCCESS in 1h 49m 20s\n- tripleo-ci-centos-7-scenario012-standalone http://logs.openstack.org/26/671526/4/check/tripleo-ci-centos-7-scenario012-standalone/c66af11/ : FAILURE in 1h 50m 41s (non-voting)\n- tripleo-ci-centos-7-scenario010-multinode-oooq-container http://logs.openstack.org/26/671526/4/check/tripleo-ci-centos-7-scenario010-multinode-oooq-container/0aeb2de/ : FAILURE in 1h 53m 21s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"6cac5397fdb4133c81adcb8c505230d32bdda5be","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 21:54:08.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"a7163f9bf7f10d02395aac4d12b6353a554f037d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-23 22:08:49.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/26/671526/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/1d73105/ : SUCCESS in 2h 30m 10s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/26/671526/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/9c5076d/ : SUCCESS in 2h 30m 24s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/26/671526/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/72ab25a/ : SUCCESS in 3h 00m 01s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"167d09be77ae4ec28c2956fe5254cd24ed127738","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 01:12:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"a9578beef4e0bf01019ddf8547381d86c8e4d540","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 01:12:25.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/26/671526/4/gate/puppet-openstack-lint-ubuntu-bionic/5888449/ : SUCCESS in 6m 18s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/26/671526/4/gate/puppet-openstack-syntax-5-ubuntu-bionic/06c3cbe/ : SUCCESS in 8m 14s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/26/671526/4/gate/puppet-openstack-syntax-6-ubuntu-bionic/33263b4/ : SUCCESS in 5m 11s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/26/671526/4/gate/puppet-openstack-unit-5.5-centos-7/2af84c4/ : SUCCESS in 31m 27s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/26/671526/4/gate/puppet-openstack-unit-6.0-ubuntu-bionic/dfa022f/ : SUCCESS in 17m 35s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/26/671526/4/gate/tripleo-ci-centos-7-undercloud-containers/fa060d6/ : SUCCESS in 1h 41m 45s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/26/671526/4/gate/tripleo-ci-centos-7-containers-multinode/0c93f29/ : SUCCESS in 2h 31m 27s\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/26/671526/4/gate/tripleo-ci-centos-7-standalone/6325540/ : SUCCESS in 1h 58m 55s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/26/671526/4/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/78b02dd/ : SUCCESS in 2h 20m 38s\n- tripleo-ci-centos-7-scenario001-standalone http://logs.openstack.org/26/671526/4/gate/tripleo-ci-centos-7-scenario001-standalone/88f7e8c/ : SUCCESS in 1h 49m 00s\n- tripleo-ci-centos-7-scenario003-standalone http://logs.openstack.org/26/671526/4/gate/tripleo-ci-centos-7-scenario003-standalone/76f068a/ : SUCCESS in 1h 45m 50s\n- tripleo-ci-centos-7-scenario004-standalone http://logs.openstack.org/26/671526/4/gate/tripleo-ci-centos-7-scenario004-standalone/2f7c577/ : SUCCESS in 2h 00m 07s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"e1cc8d0b0b1e1e55a80774421712eb4126dbe9b0","revisions":{"be01e68256ac919c2d8ccedcf01c3a7a9a920b0a":{"kind":"REWORK","_number":1,"created":"2019-07-18 17:11:00.000000000","uploader":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"ref":"refs/changes/26/671526/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/26/671526/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/26/671526/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/26/671526/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/26/671526/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/26/671526/1"}}},"commit":{"parents":[{"commit":"de91989ebc8c18e1e53df39a2caa1af07cf37bca","subject":"Merge \"Force to use markdown to prevent pypi issue\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/de91989ebc8c18e1e53df39a2caa1af07cf37bca"}]}],"author":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2019-07-18 17:09:17.000000000","tz":120},"committer":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2019-07-18 17:09:17.000000000","tz":120},"subject":"Add tcp-check connect port line to haproxy redis stanza","message":"Add tcp-check connect port line to haproxy redis stanza\n\nWhen https://bugzilla.redhat.com/show_bug.cgi?id\u003d1677420 will be merged\nin haproxy (via haproxy-1.5.18-9.el7.x86_64) our redis backend will stop\nworking because the fix around tcp-check is now more correct and\naccording to the haproxy doc a tcp-check sections *must* start with\ntcp-check connect first. From\nhttp://cbonte.github.io/haproxy-dconv/1.5/configuration.html#4-tcp-check%20connect\n:\n\"\"\"\nWhen there are no TCP port configured on the server line neither server port\ndirective, then the \u0027tcp-check connect port \u003cport\u003e\u0027 must be the first step\nof the sequence.\n\"\"\"\n\nChange-Id: I261eb30b52a3baee3b0e6d47e8f32f3c579930bf\nCo-Authored-By: Luca Miccini \u003clmiccini@redhat.com\u003e\nCloses-Bug: #1837086\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/be01e68256ac919c2d8ccedcf01c3a7a9a920b0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/be01e68256ac919c2d8ccedcf01c3a7a9a920b0a"}]},"branch":"refs/heads/master"},"bc2cc4a7b7aebf68f62e71402edb00820317aa44":{"kind":"REWORK","_number":2,"created":"2019-07-19 06:33:54.000000000","uploader":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"ref":"refs/changes/26/671526/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/26/671526/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/26/671526/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/26/671526/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/26/671526/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/26/671526/2"}}},"commit":{"parents":[{"commit":"de91989ebc8c18e1e53df39a2caa1af07cf37bca","subject":"Merge \"Force to use markdown to prevent pypi issue\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/de91989ebc8c18e1e53df39a2caa1af07cf37bca"}]}],"author":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2019-07-18 17:09:17.000000000","tz":120},"committer":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2019-07-19 06:33:40.000000000","tz":120},"subject":"Add tcp-check connect port line to haproxy redis stanza","message":"Add tcp-check connect port line to haproxy redis stanza\n\nWhen https://bugzilla.redhat.com/show_bug.cgi?id\u003d1677420 will be merged\nin haproxy (via haproxy-1.5.18-9.el7.x86_64) our redis backend will stop\nworking because the fix around tcp-check is now more correct and\naccording to the haproxy doc a tcp-check sections *must* start with\ntcp-check connect first. From\nhttp://cbonte.github.io/haproxy-dconv/1.5/configuration.html#4-tcp-check%20connect\n:\n\"\"\"\nWhen there are no TCP port configured on the server line neither server port\ndirective, then the \u0027tcp-check connect port \u003cport\u003e\u0027 must be the first step\nof the sequence.\n\"\"\"\n\nChange-Id: I261eb30b52a3baee3b0e6d47e8f32f3c579930bf\nCo-Authored-By: Luca Miccini \u003clmiccini@redhat.com\u003e\nCloses-Bug: #1837086\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/bc2cc4a7b7aebf68f62e71402edb00820317aa44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/bc2cc4a7b7aebf68f62e71402edb00820317aa44"}]},"branch":"refs/heads/master"},"b3a916c8ad60b6b48e91c0e262e3a0c2a7c8e240":{"kind":"REWORK","_number":3,"created":"2019-07-19 06:40:12.000000000","uploader":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"ref":"refs/changes/26/671526/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/26/671526/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/26/671526/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/26/671526/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/26/671526/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/26/671526/3"}}},"commit":{"parents":[{"commit":"de91989ebc8c18e1e53df39a2caa1af07cf37bca","subject":"Merge \"Force to use markdown to prevent pypi issue\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/de91989ebc8c18e1e53df39a2caa1af07cf37bca"}]}],"author":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2019-07-18 17:09:17.000000000","tz":120},"committer":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2019-07-19 06:40:00.000000000","tz":120},"subject":"Add tcp-check connect port line to haproxy redis stanza","message":"Add tcp-check connect port line to haproxy redis stanza\n\nWhen https://bugzilla.redhat.com/show_bug.cgi?id\u003d1677420 will be merged\nin haproxy (via haproxy-1.5.18-9.el7.x86_64) our redis backend will stop\nworking because the fix around tcp-check is now more correct and\naccording to the haproxy doc a tcp-check sections *must* start with\ntcp-check connect first. From\nhttp://cbonte.github.io/haproxy-dconv/1.5/configuration.html#4-tcp-check%20connect\n:\n\"\"\"\nWhen there are no TCP port configured on the server line neither server port\ndirective, then the \u0027tcp-check connect port \u003cport\u003e\u0027 must be the first step\nof the sequence.\n\"\"\"\n\nChange-Id: I261eb30b52a3baee3b0e6d47e8f32f3c579930bf\nCo-Authored-By: Luca Miccini \u003clmiccini@redhat.com\u003e\nCloses-Bug: #1837086\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/b3a916c8ad60b6b48e91c0e262e3a0c2a7c8e240"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/b3a916c8ad60b6b48e91c0e262e3a0c2a7c8e240"}]},"branch":"refs/heads/master"},"e1cc8d0b0b1e1e55a80774421712eb4126dbe9b0":{"kind":"REWORK","_number":4,"created":"2019-07-19 07:53:34.000000000","uploader":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"ref":"refs/changes/26/671526/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/26/671526/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/26/671526/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/26/671526/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/26/671526/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/26/671526/4"}}},"commit":{"parents":[{"commit":"de91989ebc8c18e1e53df39a2caa1af07cf37bca","subject":"Merge \"Force to use markdown to prevent pypi issue\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/de91989ebc8c18e1e53df39a2caa1af07cf37bca"}]}],"author":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2019-07-18 17:09:17.000000000","tz":120},"committer":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2019-07-19 07:53:13.000000000","tz":120},"subject":"Add tcp-check connect port line to haproxy redis stanza","message":"Add tcp-check connect port line to haproxy redis stanza\n\nWhen https://bugzilla.redhat.com/show_bug.cgi?id\u003d1677420 will be merged\nin haproxy (via haproxy-1.5.18-9.el7.x86_64) our redis backend will stop\nworking because the fix around tcp-check is now more correct and\naccording to the haproxy doc a tcp-check sections *must* start with\ntcp-check connect first. From\nhttp://cbonte.github.io/haproxy-dconv/1.5/configuration.html#4-tcp-check%20connect\n:\n\"\"\"\nWhen there are no TCP port configured on the server line neither server port\ndirective, then the \u0027tcp-check connect port \u003cport\u003e\u0027 must be the first step\nof the sequence.\n\"\"\"\n\nChange-Id: I261eb30b52a3baee3b0e6d47e8f32f3c579930bf\nCo-Authored-By: Luca Miccini \u003clmiccini@redhat.com\u003e\nCloses-Bug: #1837086\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/e1cc8d0b0b1e1e55a80774421712eb4126dbe9b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/e1cc8d0b0b1e1e55a80774421712eb4126dbe9b0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
