)]}'
{"id":"openstack%2Fpuppet-tripleo~671986","triplet_id":"openstack%2Fpuppet-tripleo~stable%2Frocky~I261eb30b52a3baee3b0e6d47e8f32f3c579930bf","project":"openstack/puppet-tripleo","branch":"stable/rocky","topic":"haproxy-redis-fix-stable/stein-stable/rocky","hashtags":[],"change_id":"I261eb30b52a3baee3b0e6d47e8f32f3c579930bf","subject":"Add tcp-check connect port line to haproxy redis stanza","status":"MERGED","created":"2019-07-22 08:37:35.000000000","updated":"2019-07-25 01:07:05.000000000","submitted":"2019-07-25 01:07:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"671986-1564016825083-31ef7618","meta_rev_id":"c678290c72f4dce6064358df60dfecfcb660911c","_number":671986,"virtual_id_number":671986,"owner":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8042,"name":"Jiří Stránský","email":"jistr@redhat.com","username":"jistr"},{"value":0,"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":2,"date":"2019-07-25 01:07:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2019-07-22 14:00:26.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8042,"name":"Jiří Stránský","email":"jistr@redhat.com","username":"jistr"},"all":[{"value":2,"date":"2019-07-22 09:10:50.000000000","_account_id":8042,"name":"Jiří Stránský","email":"jistr@redhat.com","username":"jistr"},{"value":0,"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"all":[{"value":0,"_account_id":8042,"name":"Jiří Stránský","email":"jistr@redhat.com","username":"jistr"},{"value":1,"date":"2019-07-24 22:04:04.000000000","_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8042,"name":"Jiří Stránský","email":"jistr@redhat.com","username":"jistr"},{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-22 09:10:50.000000000","updated_by":{"_account_id":8042,"name":"Jiří Stránský","email":"jistr@redhat.com","username":"jistr"},"reviewer":{"_account_id":8042,"name":"Jiří Stránský","email":"jistr@redhat.com","username":"jistr"},"state":"REVIEWER"},{"updated":"2019-07-22 14:00:26.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-25 01:07:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3eb19e650c5a962c4524343809e5cc10210433d1","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2019-07-22 08:37:35.000000000","message":"Patch Set 1: Cherry Picked from branch stable/stein.","accounts_in_message":[],"_revision_number":1},{"id":"c394a13cfa8db115a65364773d50fd0caa057a65","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2019-07-22 08:37:43.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"c9740040bac79180a8bfc31bfcb54f8e072fbd49","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2019-07-22 08:37:58.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/queens as commit ab1ae917df5dc13566f98e82c9a5b6c3ed2b54a5","accounts_in_message":[],"_revision_number":1},{"id":"e180a59df4403a2672227380b8cd908f4d83e177","author":{"_account_id":8042,"name":"Jiří Stránský","email":"jistr@redhat.com","username":"jistr"},"date":"2019-07-22 09:10:50.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"37c0c9ca35f2d89ffc136b6173f41674e8a58174","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 12:08:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-oooq http://logs.openstack.org/86/671986/1/check/tripleo-ci-centos-7-undercloud-oooq/9f19577/ : SUCCESS in 1h 08m 13s\n- tripleo-ci-centos-7-undercloud-upgrades http://logs.openstack.org/86/671986/1/check/tripleo-ci-centos-7-undercloud-upgrades/c289024/ : SUCCESS in 2h 04m 57s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/86/671986/1/check/tripleo-ci-centos-7-undercloud-containers/feac692/ : SUCCESS in 1h 40m 58s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/86/671986/1/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/ab88b76/ : SUCCESS in 2h 02m 31s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/86/671986/1/check/tripleo-ci-centos-7-containers-multinode/b7b6743/ : SUCCESS in 2h 39m 19s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/86/671986/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/5e2ff1b/ : SUCCESS in 2h 40m 03s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/86/671986/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/c8788ac/ : SUCCESS in 2h 50m 43s (non-voting)\n- tripleo-ci-centos-7-standalone-rocky http://logs.openstack.org/86/671986/1/check/tripleo-ci-centos-7-standalone-rocky/907e0dd/ : SUCCESS in 1h 54m 24s\n- puppet-openstack-lint http://logs.openstack.org/86/671986/1/check/puppet-openstack-lint/2d1bd65/ : SUCCESS in 5m 25s\n- puppet-openstack-syntax-4 http://logs.openstack.org/86/671986/1/check/puppet-openstack-syntax-4/6469ada/ : SUCCESS in 7m 12s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/86/671986/1/check/puppet-openstack-unit-4.8-centos-7/edcfec2/ : SUCCESS in 33m 15s\n- tripleo-ci-centos-7-scenario001-multinode-oooq-container http://logs.openstack.org/86/671986/1/check/tripleo-ci-centos-7-scenario001-multinode-oooq-container/1f5ddbc/ : SUCCESS in 3h 05m 48s (non-voting)\n- tripleo-ci-centos-7-scenario002-multinode-oooq-container http://logs.openstack.org/86/671986/1/check/tripleo-ci-centos-7-scenario002-multinode-oooq-container/6473292/ : SUCCESS in 3h 15m 19s (non-voting)\n- tripleo-ci-centos-7-scenario003-multinode-oooq-container http://logs.openstack.org/86/671986/1/check/tripleo-ci-centos-7-scenario003-multinode-oooq-container/d5e2f42/ : SUCCESS in 2h 44m 53s (non-voting)\n- tripleo-ci-centos-7-scenario004-multinode-oooq-container http://logs.openstack.org/86/671986/1/check/tripleo-ci-centos-7-scenario004-multinode-oooq-container/ba8d751/ : SUCCESS in 2h 52m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d1e9b96530aab6caf0cb84b277a542209e48c87b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-22 14:00:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/86/671986/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/4f5e6f6/ : RETRY_LIMIT in 12m 04s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/86/671986/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/768715b/ : RETRY_LIMIT in 11m 35s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/86/671986/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/010a858/ : SUCCESS in 3h 29m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"93391b6c3afe22552acae9999a34798971c7ecca","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2019-07-24 22:04:04.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8b2b4d1f44d7c3e2c6f64ab625cf670e6448ff0d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 22:04:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"633d2e6c676e4f43a7a959795bd3471187a156ac","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-25 01:07:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-undercloud-oooq http://logs.openstack.org/86/671986/1/gate/tripleo-ci-centos-7-undercloud-oooq/fd9bd20/ : SUCCESS in 1h 07m 05s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/86/671986/1/gate/tripleo-ci-centos-7-undercloud-containers/8c146c9/ : SUCCESS in 1h 39m 04s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/86/671986/1/gate/tripleo-ci-centos-7-containers-multinode/f0b7d8c/ : SUCCESS in 2h 59m 42s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/86/671986/1/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/d5963f8/ : SUCCESS in 2h 27m 50s\n- puppet-openstack-lint http://logs.openstack.org/86/671986/1/gate/puppet-openstack-lint/989ccf4/ : SUCCESS in 6m 08s\n- puppet-openstack-syntax-4 http://logs.openstack.org/86/671986/1/gate/puppet-openstack-syntax-4/583da36/ : SUCCESS in 4m 58s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/86/671986/1/gate/puppet-openstack-unit-4.8-centos-7/ce98ce2/ : SUCCESS in 51m 36s","accounts_in_message":[],"_revision_number":1},{"id":"b3e9116c7283103976891063da41008a08f989fa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-25 01:07:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9ca9f3706483f3802cd1498cd0879263269c059d","revisions":{"9ca9f3706483f3802cd1498cd0879263269c059d":{"kind":"REWORK","_number":1,"created":"2019-07-22 08:37:35.000000000","uploader":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"ref":"refs/changes/86/671986/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/86/671986/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/86/671986/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/86/671986/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/86/671986/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/86/671986/1"}}},"commit":{"parents":[{"commit":"9383bdf48fce7ab21b25d1bbcff6fc0fa5190fc9","subject":"Merge \"Add dateext and related parameters for logrotate\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/9383bdf48fce7ab21b25d1bbcff6fc0fa5190fc9"}]}],"author":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2019-07-18 17:09:17.000000000","tz":120},"committer":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2019-07-22 08:37:35.000000000","tz":0},"subject":"Add tcp-check connect port line to haproxy redis stanza","message":"Add tcp-check connect port line to haproxy redis stanza\n\nWhen https://bugzilla.redhat.com/show_bug.cgi?id\u003d1677420 will be merged\nin haproxy (via haproxy-1.5.18-9.el7.x86_64) our redis backend will stop\nworking because the fix around tcp-check is now more correct and\naccording to the haproxy doc a tcp-check sections *must* start with\ntcp-check connect first. From\nhttp://cbonte.github.io/haproxy-dconv/1.5/configuration.html#4-tcp-check%20connect\n:\n\"\"\"\nWhen there are no TCP port configured on the server line neither server port\ndirective, then the \u0027tcp-check connect port \u003cport\u003e\u0027 must be the first step\nof the sequence.\n\"\"\"\n\nChange-Id: I261eb30b52a3baee3b0e6d47e8f32f3c579930bf\nCo-Authored-By: Luca Miccini \u003clmiccini@redhat.com\u003e\nCloses-Bug: #1837086\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/9ca9f3706483f3802cd1498cd0879263269c059d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/9ca9f3706483f3802cd1498cd0879263269c059d"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
