)]}'
{"id":"openstack%2Fpuppet-tripleo~672056","triplet_id":"openstack%2Fpuppet-tripleo~master~If598da717653a383a2d3b3373c56517f8bca832f","project":"openstack/puppet-tripleo","branch":"master","topic":"sqla_collectd_master","hashtags":[],"change_id":"If598da717653a383a2d3b3373c56517f8bca832f","subject":"Add SQLAlchemy-collectd support","status":"MERGED","created":"2019-07-22 14:22:11.000000000","updated":"2019-08-22 05:22:52.000000000","submitted":"2019-08-22 05:22:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"672056-1566451372464-3b288848","meta_rev_id":"849dc06281743600e124db39676d9bacffca2047","_number":672056,"virtual_id_number":672056,"owner":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"date":"2019-08-13 01:32:57.000000000","_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-08-22 05:22:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-05 17:02:19.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2019-08-22 02:26:19.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":2,"date":"2019-08-13 13:33:39.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2019-08-22 02:26:19.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-01 11:17:40.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"},{"updated":"2019-08-05 17:02:19.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-13 13:33:39.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-08-22 02:26:19.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-08-22 05:22:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"aa20467b9861c15f2a85cadd35874d109a24b9b5","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2019-07-22 14:22:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6fa8f855a85c6fc52afde2769bb2c3de4693cc2b","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2019-07-22 14:33:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"74460e76ec9771426b424ac9bb8157c6770607a5","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-22 17:36:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/56/672056/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/55785c1/ : SUCCESS in 3h 12m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"91022d78b0adda4f8c32d4c78ffa8aeb3d2b78d5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 19:07:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/56/672056/1/check/puppet-openstack-lint-ubuntu-bionic/68fc2c9/ : FAILURE in 7m 20s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/56/672056/1/check/puppet-openstack-syntax-5-ubuntu-bionic/fe07299/ : FAILURE in 5m 30s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/56/672056/1/check/puppet-openstack-syntax-6-ubuntu-bionic/fd9e8eb/ : FAILURE in 4m 59s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/56/672056/1/check/puppet-openstack-unit-5.5-centos-7/372ca76/ : SUCCESS in 34m 31s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/56/672056/1/check/puppet-openstack-unit-6.0-ubuntu-bionic/c86da0c/ : SUCCESS in 17m 47s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/56/672056/1/check/puppet-openstack-unit-latest-ubuntu-bionic/7ab26da/ : SUCCESS in 16m 37s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/56/672056/1/check/tripleo-ci-centos-7-undercloud-containers/7e3989e/ : SUCCESS in 1h 46m 44s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/56/672056/1/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/cd76301/ : FAILURE in 2h 12m 14s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/56/672056/1/check/tripleo-ci-centos-7-containers-multinode/2cfca27/ : SUCCESS in 2h 17m 16s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/56/672056/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/e062bfc/ : SUCCESS in 2h 20m 01s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/56/672056/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/3355d8d/ : FAILURE in 2h 49m 55s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/56/672056/1/check/tripleo-ci-centos-7-standalone/bc1840e/ : SUCCESS in 1h 47m 19s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/56/672056/1/check/tripleo-ci-fedora-28-standalone/bc100e6/ : SUCCESS in 1h 57m 33s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/56/672056/1/check/tripleo-ci-centos-7-standalone-upgrade/cc44b05/ : FAILURE in 1h 49m 50s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone http://logs.openstack.org/56/672056/1/check/tripleo-ci-centos-7-scenario001-standalone/3df10fb/ : FAILURE in 1h 28m 31s","accounts_in_message":[],"_revision_number":1},{"id":"d708ebb690d95ef4e19e9191f8b531db887661df","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2019-07-22 22:16:02.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f7a65b5cdeb4dbee2ac9e1bc1fbebb0da453a998","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-23 00:11:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/56/672056/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/0259f17/ : FAILURE in 1h 53m 51s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a4ec7753712aa9489c637c282482ff7a59a6c027","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 00:53:20.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/56/672056/2/check/puppet-openstack-lint-ubuntu-bionic/eb608b7/ : FAILURE in 5m 11s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/56/672056/2/check/puppet-openstack-syntax-5-ubuntu-bionic/ca13467/ : SUCCESS in 4m 31s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/56/672056/2/check/puppet-openstack-syntax-6-ubuntu-bionic/10a6600/ : SUCCESS in 4m 38s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/56/672056/2/check/puppet-openstack-unit-5.5-centos-7/3b3e24c/ : SUCCESS in 29m 54s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/56/672056/2/check/puppet-openstack-unit-6.0-ubuntu-bionic/8544e00/ : SUCCESS in 14m 09s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/56/672056/2/check/puppet-openstack-unit-latest-ubuntu-bionic/a28fb67/ : SUCCESS in 16m 36s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/56/672056/2/check/tripleo-ci-centos-7-undercloud-containers/269c9e6/ : SUCCESS in 1h 52m 50s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/56/672056/2/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/05eeb28/ : FAILURE in 2h 15m 15s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/56/672056/2/check/tripleo-ci-centos-7-containers-multinode/8ae084d/ : SUCCESS in 2h 16m 22s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/56/672056/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/e401211/ : SUCCESS in 2h 21m 06s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/56/672056/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/b91460e/ : SUCCESS in 2h 30m 02s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/56/672056/2/check/tripleo-ci-centos-7-standalone/b57ce7e/ : SUCCESS in 1h 41m 45s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/56/672056/2/check/tripleo-ci-fedora-28-standalone/bfe5b1a/ : SUCCESS in 1h 43m 59s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/56/672056/2/check/tripleo-ci-centos-7-standalone-upgrade/91616e9/ : FAILURE in 2h 00m 25s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone http://logs.openstack.org/56/672056/2/check/tripleo-ci-centos-7-scenario001-standalone/305dac8/ : FAILURE in 1h 11m 07s","accounts_in_message":[],"_revision_number":2},{"id":"c48c0f64055e028cbdeb1e483f43300c64fae588","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2019-07-23 01:21:08.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"55393bf6ecc0f086da852290ca8a04bfaf6a964f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-23 04:04:11.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/56/672056/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/0b361da/ : SUCCESS in 2h 41m 18s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"cd1603c1873a42532040db4299c7dcc6fb1f8d68","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 04:12:06.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/56/672056/3/check/puppet-openstack-lint-ubuntu-bionic/8c20c63/ : SUCCESS in 4m 42s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/56/672056/3/check/puppet-openstack-syntax-5-ubuntu-bionic/caca98d/ : SUCCESS in 4m 36s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/56/672056/3/check/puppet-openstack-syntax-6-ubuntu-bionic/f5f99e1/ : SUCCESS in 4m 45s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/56/672056/3/check/puppet-openstack-unit-5.5-centos-7/00e01b1/ : SUCCESS in 31m 23s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/56/672056/3/check/puppet-openstack-unit-6.0-ubuntu-bionic/01f1199/ : SUCCESS in 15m 14s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/56/672056/3/check/puppet-openstack-unit-latest-ubuntu-bionic/480b7e8/ : SUCCESS in 16m 38s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/56/672056/3/check/tripleo-ci-centos-7-undercloud-containers/0ca5c67/ : SUCCESS in 1h 29m 04s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/56/672056/3/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/6c30977/ : FAILURE in 2h 04m 39s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/56/672056/3/check/tripleo-ci-centos-7-containers-multinode/6682a64/ : SUCCESS in 2h 20m 10s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/56/672056/3/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/75cef75/ : SUCCESS in 1h 57m 03s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/56/672056/3/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/304ad2c/ : SUCCESS in 2h 24m 43s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/56/672056/3/check/tripleo-ci-centos-7-standalone/17ae0be/ : SUCCESS in 1h 52m 51s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/56/672056/3/check/tripleo-ci-fedora-28-standalone/956a795/ : SUCCESS in 1h 43m 48s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/56/672056/3/check/tripleo-ci-centos-7-standalone-upgrade/403d757/ : FAILURE in 1h 57m 50s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone http://logs.openstack.org/56/672056/3/check/tripleo-ci-centos-7-scenario001-standalone/16463b9/ : FAILURE in 1h 14m 36s","accounts_in_message":[],"_revision_number":3},{"id":"d2e1741d8bfe52a4ae3d9863a8494a001c0ea863","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2019-07-23 07:56:47.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nI don\u0027t see a config for including sqlalchemy in \nhttp://logs.openstack.org/56/672056/3/check/tripleo-ci-centos-7-scenario001-standalone/16463b9/logs/undercloud/var/log/config-data/collectd/etc/collectd.d/","accounts_in_message":[],"_revision_number":3},{"id":"7665a6f2dcd76e0be1ee668ebddbe87db7a5d22c","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2019-07-24 14:38:35.000000000","message":"Patch Set 3:\n\n(1 comment)\n\n\u003e (1 comment)\n \u003e \n \u003e I don\u0027t see a config for including sqlalchemy in\n \u003e http://logs.openstack.org/56/672056/3/check/tripleo-ci-centos-7-scenario001-standalone/16463b9/logs/undercloud/var/log/config-data/collectd/etc/collectd.d/\n\nAre you referring to having it included in a tempest scenario? my understanding is that this would be on the tripleo-heat-templates side under ci/environments/, which I didn\u0027t yet implement because I would need to make the t-h-t patch depend on this one and I\u0027m not sure how to do that, is there something that needs to be on the puppet side for that to be in effect also?","accounts_in_message":[],"_revision_number":3},{"id":"28d901ec05a3f7ceaced29955ec601581489899b","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2019-08-01 11:17:40.000000000","message":"Patch Set 3:\n\nA, Mike, yes, you are right. Getting this included in ci runs requires more configuration.\n\nThen it\u0027s just the nit of 2010 instead of 2019 for the coyright.","accounts_in_message":[],"_revision_number":3},{"id":"bf49317f17165a1e9e8bcd6fa39b36c8b69e2143","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2019-08-05 14:56:21.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a4446471664ddbdb1cbc7e9cbee725e84daf6d64","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-05 17:02:19.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/56/672056/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/c06608f/ : FAILURE in 2h 05m 09s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"af0bdab6c34b9109dbaebb6eb7a51099fc4981ac","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 17:34:24.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://logs.opendev.org/56/672056/4/check/puppet-openstack-lint-ubuntu-bionic/89e1217/ : SUCCESS in 5m 13s\n- puppet-openstack-syntax-5-ubuntu-bionic https://logs.opendev.org/56/672056/4/check/puppet-openstack-syntax-5-ubuntu-bionic/256b2cd/ : SUCCESS in 5m 35s\n- puppet-openstack-syntax-6-ubuntu-bionic https://logs.opendev.org/56/672056/4/check/puppet-openstack-syntax-6-ubuntu-bionic/dc47c05/ : SUCCESS in 4m 16s\n- puppet-openstack-unit-5.5-centos-7 https://logs.opendev.org/56/672056/4/check/puppet-openstack-unit-5.5-centos-7/f387950/ : SUCCESS in 37m 52s\n- puppet-openstack-unit-6.0-ubuntu-bionic https://logs.opendev.org/56/672056/4/check/puppet-openstack-unit-6.0-ubuntu-bionic/45b58fd/ : SUCCESS in 24m 33s\n- puppet-openstack-unit-latest-ubuntu-bionic https://logs.opendev.org/56/672056/4/check/puppet-openstack-unit-latest-ubuntu-bionic/68cec7a/ : SUCCESS in 23m 05s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/56/672056/4/check/tripleo-ci-centos-7-undercloud-containers/04f04d5/ : SUCCESS in 1h 44m 43s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://logs.opendev.org/56/672056/4/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/efd914f/ : FAILURE in 1h 09m 45s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/56/672056/4/check/tripleo-ci-centos-7-containers-multinode/c2d3b4c/ : SUCCESS in 2h 26m 22s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/56/672056/4/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/84704ec/ : SUCCESS in 2h 25m 54s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/56/672056/4/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/8ac197d/ : SUCCESS in 2h 32m 48s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/56/672056/4/check/tripleo-ci-centos-7-standalone/a22cb96/ : SUCCESS in 1h 58m 26s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/56/672056/4/check/tripleo-ci-fedora-28-standalone/7899e21/ : SUCCESS in 1h 46m 56s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://logs.opendev.org/56/672056/4/check/tripleo-ci-centos-7-standalone-upgrade/d47b4f9/ : FAILURE in 41m 59s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://logs.opendev.org/56/672056/4/check/tripleo-ci-centos-7-scenario001-standalone/c78ff0d/ : SUCCESS in 2h 05m 46s","accounts_in_message":[],"_revision_number":4},{"id":"56a669c9c9274119da8617408d6bbe6d48dfe4e9","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-12 22:15:54.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"793758cb1863bb2cf3bbfd5e2cdcf36a62072238","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2019-08-13 01:32:57.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"0ca383eabdc9cb890b0eaeddd3355080d1ce0e15","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-13 13:33:39.000000000","message":"Patch Set 4: Code-Review+2\n\nk just making sure it\u0027s accounted for","accounts_in_message":[],"_revision_number":4},{"id":"05108bf5fe28b6ee32026cb9e0c2b719c4f4fbf7","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-08-22 02:26:19.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"6f3cef1d2ae3c1bc30d18202c4d713dc862de64e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 02:26:31.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"2f1cf59c1b9979d96033e997cabbf2581ebcddef","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 05:22:52.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/233c9d2b06d34a1ca95a030617f6d4ba : SUCCESS in 6m 03s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/00fdc94700104590a1d654c5dc184caf : SUCCESS in 5m 17s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fa15e67aed7048249a834c076e62742b : SUCCESS in 5m 15s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/71d1c02772a841eca184a2b595c5a679 : SUCCESS in 31m 53s\n- puppet-openstack-unit-6.0-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9f53e56d773c4ac1b73c2d7efe55a119 : SUCCESS in 17m 57s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/19fd48a9b5134290a611269f7eb275e7 : SUCCESS in 1h 45m 58s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/745fa979123948a6af9272e38c711c31 : SUCCESS in 2h 25m 28s\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/e38ba8093f6a42d995a96302cce0325e : SUCCESS in 1h 46m 56s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/790ad7af080d492ca9f8030328da1364 : SUCCESS in 2h 48m 53s\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/53b61779ebdc4e7291fe0499214c0ddc : SUCCESS in 1h 51m 57s","accounts_in_message":[],"_revision_number":4},{"id":"adce7cb12d801655953bf979adcecacf31c30d1f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 05:22:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"e2e602972f9c34d27b9c8facea4b200557bd9b53","revisions":{"b9fae505462160e6bed1373542ab873169c6838d":{"kind":"REWORK","_number":1,"created":"2019-07-22 14:22:11.000000000","uploader":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"ref":"refs/changes/56/672056/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/56/672056/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/56/672056/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/56/672056/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/56/672056/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/56/672056/1"}}},"commit":{"parents":[{"commit":"de91989ebc8c18e1e53df39a2caa1af07cf37bca","subject":"Merge \"Force to use markdown to prevent pypi issue\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/de91989ebc8c18e1e53df39a2caa1af07cf37bca"}]}],"author":{"name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","date":"2019-07-17 19:30:27.000000000","tz":-240},"committer":{"name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","date":"2019-07-22 14:17:04.000000000","tz":-240},"subject":"Add SQLAlchemy-collectd support","message":"Add SQLAlchemy-collectd support\n\nThe SQLAlchemy-collectd plugin is now shipped in podman\ncontainers under Kolla, this allows heat templates\nto pull the plugin into the collectd configuration when\nthe collectd templates are being used.\n\nA corresponding set of template changes in tripleo-heat-templates\nwill enable this plugin.\n\nChange-Id: If598da717653a383a2d3b3373c56517f8bca832f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/b9fae505462160e6bed1373542ab873169c6838d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/b9fae505462160e6bed1373542ab873169c6838d"}]},"branch":"refs/heads/master"},"edbcecf7d6b79b8bd4d860878e7507d39c2f3d5a":{"kind":"REWORK","_number":2,"created":"2019-07-22 22:16:02.000000000","uploader":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"ref":"refs/changes/56/672056/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/56/672056/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/56/672056/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/56/672056/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/56/672056/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/56/672056/2"}}},"commit":{"parents":[{"commit":"de91989ebc8c18e1e53df39a2caa1af07cf37bca","subject":"Merge \"Force to use markdown to prevent pypi issue\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/de91989ebc8c18e1e53df39a2caa1af07cf37bca"}]}],"author":{"name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","date":"2019-07-17 19:30:27.000000000","tz":-240},"committer":{"name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","date":"2019-07-22 22:15:55.000000000","tz":-240},"subject":"Add SQLAlchemy-collectd support","message":"Add SQLAlchemy-collectd support\n\nThe SQLAlchemy-collectd plugin is now shipped in podman\ncontainers under Kolla, this allows heat templates\nto pull the plugin into the collectd configuration when\nthe collectd templates are being used.\n\nA corresponding set of template changes in tripleo-heat-templates\nwill enable this plugin.\n\nChange-Id: If598da717653a383a2d3b3373c56517f8bca832f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/edbcecf7d6b79b8bd4d860878e7507d39c2f3d5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/edbcecf7d6b79b8bd4d860878e7507d39c2f3d5a"}]},"branch":"refs/heads/master"},"1717a281af2be398eeff41c92d9c68ad4f0c7c96":{"kind":"REWORK","_number":3,"created":"2019-07-23 01:21:08.000000000","uploader":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"ref":"refs/changes/56/672056/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/56/672056/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/56/672056/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/56/672056/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/56/672056/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/56/672056/3"}}},"commit":{"parents":[{"commit":"de91989ebc8c18e1e53df39a2caa1af07cf37bca","subject":"Merge \"Force to use markdown to prevent pypi issue\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/de91989ebc8c18e1e53df39a2caa1af07cf37bca"}]}],"author":{"name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","date":"2019-07-17 19:30:27.000000000","tz":-240},"committer":{"name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","date":"2019-07-23 01:21:03.000000000","tz":-240},"subject":"Add SQLAlchemy-collectd support","message":"Add SQLAlchemy-collectd support\n\nThe SQLAlchemy-collectd plugin is now shipped in podman\ncontainers under Kolla, this allows heat templates\nto pull the plugin into the collectd configuration when\nthe collectd templates are being used.\n\nA corresponding set of template changes in tripleo-heat-templates\nwill enable this plugin.\n\nChange-Id: If598da717653a383a2d3b3373c56517f8bca832f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/1717a281af2be398eeff41c92d9c68ad4f0c7c96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/1717a281af2be398eeff41c92d9c68ad4f0c7c96"}]},"branch":"refs/heads/master"},"e2e602972f9c34d27b9c8facea4b200557bd9b53":{"kind":"REWORK","_number":4,"created":"2019-08-05 14:56:21.000000000","uploader":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"ref":"refs/changes/56/672056/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/56/672056/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/56/672056/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/56/672056/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/56/672056/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/56/672056/4"}}},"commit":{"parents":[{"commit":"8e4de1b75ce67a253d5c872d3de18128cc17f2a9","subject":"Merge \"Enable service token in nova and cinder\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/8e4de1b75ce67a253d5c872d3de18128cc17f2a9"}]}],"author":{"name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","date":"2019-07-17 19:30:27.000000000","tz":-240},"committer":{"name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","date":"2019-08-05 14:56:16.000000000","tz":-240},"subject":"Add SQLAlchemy-collectd support","message":"Add SQLAlchemy-collectd support\n\nThe SQLAlchemy-collectd plugin is now shipped in podman\ncontainers under Kolla, this allows heat templates\nto pull the plugin into the collectd configuration when\nthe collectd templates are being used.\n\nA corresponding set of template changes in tripleo-heat-templates\nwill enable this plugin.\n\nChange-Id: If598da717653a383a2d3b3373c56517f8bca832f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/e2e602972f9c34d27b9c8facea4b200557bd9b53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/e2e602972f9c34d27b9c8facea4b200557bd9b53"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
