)]}'
{"id":"openstack%2Fpuppet-tripleo~674069","triplet_id":"openstack%2Fpuppet-tripleo~master~I8148a039ab9dcbc97baff141aae6ebab4c27e16d","project":"openstack/puppet-tripleo","branch":"master","topic":"bug/1835414","hashtags":[],"change_id":"I8148a039ab9dcbc97baff141aae6ebab4c27e16d","subject":"Add missing base apache configs when internal TLS","status":"MERGED","created":"2019-08-01 15:11:38.000000000","updated":"2019-08-09 14:13:59.000000000","submitted":"2019-08-09 14:13:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"674069-1565360039368-4300ec55","meta_rev_id":"fee07212a9784f3c3e44ae9bd9966835683254f2","_number":674069,"virtual_id_number":674069,"owner":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-08-09 14:13:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-09 10:40:05.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-09 07:30:01.000000000","_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2019-08-05 15:12:19.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2019-08-05 15:33:36.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-05 15:12:19.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-08-05 15:33:36.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-08-09 10:40:05.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-09 14:13:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"99b4a052b97d6204893904d3f7d42bda00c50bfa","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2019-08-01 15:11:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"761120a2be6646472677d0f59a8648a49f37f6f6","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-01 18:22:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/69/674069/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/528c0fb/ : FAILURE in 1h 13m 15s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/69/674069/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/b16c701/ : FAILURE in 1h 14m 33s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/69/674069/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/60d2aa8/ : SUCCESS in 3h 02m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e4371e18a86c280ae50808d40aee51fc748ad9c5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 01:29:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://logs.opendev.org/69/674069/1/check/puppet-openstack-lint-ubuntu-bionic/ff8102e/ : SUCCESS in 5m 46s\n- puppet-openstack-syntax-5-ubuntu-bionic https://logs.opendev.org/69/674069/1/check/puppet-openstack-syntax-5-ubuntu-bionic/ab45890/ : SUCCESS in 5m 06s\n- puppet-openstack-syntax-6-ubuntu-bionic https://logs.opendev.org/69/674069/1/check/puppet-openstack-syntax-6-ubuntu-bionic/ab725f1/ : SUCCESS in 5m 44s\n- puppet-openstack-unit-5.5-centos-7 https://logs.opendev.org/69/674069/1/check/puppet-openstack-unit-5.5-centos-7/3cb2972/ : SUCCESS in 27m 36s\n- puppet-openstack-unit-6.0-ubuntu-bionic https://logs.opendev.org/69/674069/1/check/puppet-openstack-unit-6.0-ubuntu-bionic/d6d2b02/ : SUCCESS in 17m 07s\n- puppet-openstack-unit-latest-ubuntu-bionic https://logs.opendev.org/69/674069/1/check/puppet-openstack-unit-latest-ubuntu-bionic/cdeec4c/ : SUCCESS in 17m 37s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/69/674069/1/check/tripleo-ci-centos-7-undercloud-containers/a596ba3/ : SUCCESS in 1h 35m 32s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://logs.opendev.org/69/674069/1/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/2355199/ : TIMED_OUT in 3h 12m 08s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/69/674069/1/check/tripleo-ci-centos-7-containers-multinode/c8d25d4/ : SUCCESS in 2h 20m 57s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/69/674069/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/83dff5e/ : SUCCESS in 2h 08m 39s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/69/674069/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/8fe300f/ : SUCCESS in 2h 53m 11s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/69/674069/1/check/tripleo-ci-centos-7-standalone/d72361e/ : SUCCESS in 1h 43m 53s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/69/674069/1/check/tripleo-ci-fedora-28-standalone/8cf0084/ : FAILURE in 58m 15s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://logs.opendev.org/69/674069/1/check/tripleo-ci-centos-7-standalone-upgrade/1a09d50/ : FAILURE in 1h 38m 33s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://logs.opendev.org/69/674069/1/check/tripleo-ci-centos-7-scenario001-standalone/c17d6b4/ : SUCCESS in 1h 36m 13s\n- tripleo-ci-centos-7-scenario002-standalone https://logs.opendev.org/69/674069/1/check/tripleo-ci-centos-7-scenario002-standalone/6b13f00/ : SUCCESS in 1h 31m 45s\n- tripleo-ci-centos-7-scenario003-standalone https://logs.opendev.org/69/674069/1/check/tripleo-ci-centos-7-scenario003-standalone/6b8ba58/ : SUCCESS in 1h 07m 55s\n- tripleo-ci-centos-7-scenario004-standalone https://logs.opendev.org/69/674069/1/check/tripleo-ci-centos-7-scenario004-standalone/3babb6d/ : SUCCESS in 1h 38m 44s\n- tripleo-ci-centos-7-scenario012-standalone https://logs.opendev.org/69/674069/1/check/tripleo-ci-centos-7-scenario012-standalone/a3db5f9/ : FAILURE in 1h 49m 05s (non-voting)\n- tripleo-ci-centos-7-scenario010-multinode-oooq-container https://logs.opendev.org/69/674069/1/check/tripleo-ci-centos-7-scenario010-multinode-oooq-container/2b443d3/ : FAILURE in 1h 51m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7577896eb55e4c3b1d94bbc0edb0022eb7852649","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2019-08-02 08:02:32.000000000","message":"Patch Set 1:\n\ncheck-rdo","accounts_in_message":[],"_revision_number":1},{"id":"e767bcd7faad26b0a167e4dcdb2364633cf489b6","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-02 10:59:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/69/674069/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/263d71f/ : SUCCESS in 2h 36m 39s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/69/674069/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/182218b/ : SUCCESS in 2h 44m 57s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/69/674069/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/28de0cf/ : SUCCESS in 2h 55m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ec10b480a2e2ce81e22f01b47858396daa322a11","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2019-08-05 11:40:59.000000000","message":"Patch Set 1:\n\nso we have a plenty of places that include the apache base config, when internal TLS is enabled. Those in the patch are missing that step.","accounts_in_message":[],"_revision_number":1},{"id":"35a1dd1fdffc0734652dcabae5a54ae23e8e4e39","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-08-05 12:37:31.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7135166cc94b57fb251502477d13530cc5945717","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-05 13:52:51.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"33ec5a510bff1bd0aae91ee3dd05d72beea2a4f2","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-08-05 14:21:56.000000000","message":"Patch Set 1: -Code-Review\n\noh yeah, good remark for swift","accounts_in_message":[],"_revision_number":1},{"id":"aef7fe29bbfb323486e1e34f862d3db5503c4c8d","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2019-08-05 14:54:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c57e4258aea70bd4b07c77036e0c9f9b27759fb2","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-05 15:12:19.000000000","message":"Patch Set 1: Code-Review+2\n\nevidently we wrap swift in apache when running under tls","accounts_in_message":[],"_revision_number":1},{"id":"683033f082e547209d5c19e81634d07b1c94494e","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-08-05 15:33:36.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7cadd2257347804f2f8b9f3bfd8b94c403bb196b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 15:33:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"611e914612f4dee30086978a448d9329a6d75702","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 20:43:20.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://logs.opendev.org/69/674069/1/gate/puppet-openstack-lint-ubuntu-bionic/1bcc8ca/ : SUCCESS in 7m 11s\n- puppet-openstack-syntax-5-ubuntu-bionic https://logs.opendev.org/69/674069/1/gate/puppet-openstack-syntax-5-ubuntu-bionic/7669f0e/ : SUCCESS in 5m 12s\n- puppet-openstack-syntax-6-ubuntu-bionic https://logs.opendev.org/69/674069/1/gate/puppet-openstack-syntax-6-ubuntu-bionic/850bd28/ : SUCCESS in 6m 21s\n- puppet-openstack-unit-5.5-centos-7 https://logs.opendev.org/69/674069/1/gate/puppet-openstack-unit-5.5-centos-7/1f4855a/ : SUCCESS in 30m 33s\n- puppet-openstack-unit-6.0-ubuntu-bionic https://logs.opendev.org/69/674069/1/gate/puppet-openstack-unit-6.0-ubuntu-bionic/b4c005d/ : SUCCESS in 18m 30s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/69/674069/1/gate/tripleo-ci-centos-7-undercloud-containers/5493512/ : SUCCESS in 2h 15m 37s\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/69/674069/1/gate/tripleo-ci-centos-7-containers-multinode/ec3721e/ : FAILURE in 1h 07m 32s\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/69/674069/1/gate/tripleo-ci-centos-7-standalone/37c35e5/ : SUCCESS in 2h 39m 52s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/69/674069/1/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/00d85f3/ : SUCCESS in 2h 56m 43s\n- tripleo-ci-centos-7-scenario001-standalone https://logs.opendev.org/69/674069/1/gate/tripleo-ci-centos-7-scenario001-standalone/4dcc50d/ : FAILURE in 1h 28m 27s\n- tripleo-ci-centos-7-scenario002-standalone https://logs.opendev.org/69/674069/1/gate/tripleo-ci-centos-7-scenario002-standalone/b4b0719/ : SUCCESS in 1h 55m 14s\n- tripleo-ci-centos-7-scenario003-standalone https://logs.opendev.org/69/674069/1/gate/tripleo-ci-centos-7-scenario003-standalone/9474a4e/ : SUCCESS in 1h 28m 46s\n- tripleo-ci-centos-7-scenario004-standalone https://logs.opendev.org/69/674069/1/gate/tripleo-ci-centos-7-scenario004-standalone/fd703ad/ : FAILURE in 1h 35m 52s","accounts_in_message":[],"_revision_number":1},{"id":"c226bc6f3339446ec5298ac549e28c1def08f7c7","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-05 23:38:58.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4872b4907d61fdcce991432d2536c7b31e829e0a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 02:17:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://logs.opendev.org/69/674069/1/check/puppet-openstack-lint-ubuntu-bionic/4e90f9d/ : SUCCESS in 4m 12s\n- puppet-openstack-syntax-5-ubuntu-bionic https://logs.opendev.org/69/674069/1/check/puppet-openstack-syntax-5-ubuntu-bionic/1c9c9b9/ : SUCCESS in 3m 55s\n- puppet-openstack-syntax-6-ubuntu-bionic https://logs.opendev.org/69/674069/1/check/puppet-openstack-syntax-6-ubuntu-bionic/3986af9/ : SUCCESS in 4m 17s\n- puppet-openstack-unit-5.5-centos-7 https://logs.opendev.org/69/674069/1/check/puppet-openstack-unit-5.5-centos-7/b7267b9/ : SUCCESS in 35m 48s\n- puppet-openstack-unit-6.0-ubuntu-bionic https://logs.opendev.org/69/674069/1/check/puppet-openstack-unit-6.0-ubuntu-bionic/0db732e/ : SUCCESS in 16m 56s\n- puppet-openstack-unit-latest-ubuntu-bionic https://logs.opendev.org/69/674069/1/check/puppet-openstack-unit-latest-ubuntu-bionic/4d9490f/ : SUCCESS in 18m 25s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/69/674069/1/check/tripleo-ci-centos-7-undercloud-containers/f3db787/ : SUCCESS in 1h 39m 40s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://logs.opendev.org/69/674069/1/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/3230d5b/ : FAILURE in 2h 09m 50s (non-voting)\n- tripleo-ci-centos-7-containers-multinode finger://ze02.openstack.org/995974acb87e4bb2a7f182b2cf751a18 : POST_FAILURE in 2h 28m 15s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates finger://ze05.openstack.org/01c082bf2be34e768ebd647b9ffbc19d : POST_FAILURE in 2h 20m 08s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades finger://ze03.openstack.org/1b4ef521a4934122b68e43e00b7cba0c : POST_FAILURE in 2h 32m 49s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/69/674069/1/check/tripleo-ci-centos-7-standalone/120c145/ : SUCCESS in 1h 49m 09s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/69/674069/1/check/tripleo-ci-fedora-28-standalone/a1918fd/ : SUCCESS in 1h 42m 19s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://logs.opendev.org/69/674069/1/check/tripleo-ci-centos-7-standalone-upgrade/4d1f8d2/ : SUCCESS in 2h 17m 39s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://logs.opendev.org/69/674069/1/check/tripleo-ci-centos-7-scenario001-standalone/bc43de9/ : FAILURE in 1h 04m 18s\n- tripleo-ci-centos-7-scenario002-standalone https://logs.opendev.org/69/674069/1/check/tripleo-ci-centos-7-scenario002-standalone/009e358/ : SUCCESS in 1h 39m 22s\n- tripleo-ci-centos-7-scenario003-standalone https://logs.opendev.org/69/674069/1/check/tripleo-ci-centos-7-scenario003-standalone/43950b5/ : SUCCESS in 1h 11m 47s\n- tripleo-ci-centos-7-scenario004-standalone https://logs.opendev.org/69/674069/1/check/tripleo-ci-centos-7-scenario004-standalone/cad9367/ : FAILURE in 1h 09m 31s\n- tripleo-ci-centos-7-scenario012-standalone https://logs.opendev.org/69/674069/1/check/tripleo-ci-centos-7-scenario012-standalone/dff8146/ : SUCCESS in 1h 44m 36s (non-voting)\n- tripleo-ci-centos-7-scenario010-multinode-oooq-container https://logs.opendev.org/69/674069/1/check/tripleo-ci-centos-7-scenario010-multinode-oooq-container/b85dcdd/ : FAILURE in 1h 52m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"42d2b9abf1ec5fc2d1d2ef982f6272ce3f847734","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-06 02:49:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/69/674069/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/bd0a3e7/ : SUCCESS in 2h 43m 17s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/69/674069/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/0640eee/ : SUCCESS in 2h 42m 20s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/69/674069/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/713e2c5/ : SUCCESS in 3h 08m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a5fda01b8a6701c17f02894668c28d8eea72b579","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2019-08-06 09:14:24.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9cdbbc4ecd9ea3a2c66c16ae978c65db43fa0bf9","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-06 13:25:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/69/674069/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/cbace73/ : SUCCESS in 3h 15m 15s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/69/674069/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/28de273/ : TIMED_OUT in 4h 09m 49s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/69/674069/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/9816252/ : SUCCESS in 3h 47m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"28a583b5bef6982c5e2be279c24a79ce1c01de99","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 14:40:22.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://logs.opendev.org/69/674069/1/check/puppet-openstack-lint-ubuntu-bionic/24c9b95/ : SUCCESS in 5m 20s\n- puppet-openstack-syntax-5-ubuntu-bionic https://logs.opendev.org/69/674069/1/check/puppet-openstack-syntax-5-ubuntu-bionic/14f4401/ : SUCCESS in 4m 23s\n- puppet-openstack-syntax-6-ubuntu-bionic https://logs.opendev.org/69/674069/1/check/puppet-openstack-syntax-6-ubuntu-bionic/4c0111e/ : SUCCESS in 5m 40s\n- puppet-openstack-unit-5.5-centos-7 https://logs.opendev.org/69/674069/1/check/puppet-openstack-unit-5.5-centos-7/e0e78ff/ : SUCCESS in 31m 59s\n- puppet-openstack-unit-6.0-ubuntu-bionic https://logs.opendev.org/69/674069/1/check/puppet-openstack-unit-6.0-ubuntu-bionic/2052b32/ : SUCCESS in 21m 24s\n- puppet-openstack-unit-latest-ubuntu-bionic https://logs.opendev.org/69/674069/1/check/puppet-openstack-unit-latest-ubuntu-bionic/c125abf/ : SUCCESS in 19m 46s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/69/674069/1/check/tripleo-ci-centos-7-undercloud-containers/a66a402/ : FAILURE in 1h 17m 10s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://logs.opendev.org/69/674069/1/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/577e4cc/ : FAILURE in 2h 12m 34s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/69/674069/1/check/tripleo-ci-centos-7-containers-multinode/e124b88/ : SUCCESS in 2h 29m 57s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/69/674069/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/c9798e1/ : SUCCESS in 3h 01m 08s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/69/674069/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/633f0e4/ : SUCCESS in 2h 37m 01s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/69/674069/1/check/tripleo-ci-centos-7-standalone/df86622/ : SUCCESS in 2h 07m 05s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/69/674069/1/check/tripleo-ci-fedora-28-standalone/12dd96e/ : SUCCESS in 1h 57m 10s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://logs.opendev.org/69/674069/1/check/tripleo-ci-centos-7-standalone-upgrade/a999d4b/ : FAILURE in 5m 25s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://logs.opendev.org/69/674069/1/check/tripleo-ci-centos-7-scenario001-standalone/84a94cf/ : FAILURE in 1h 13m 06s\n- tripleo-ci-centos-7-scenario002-standalone https://logs.opendev.org/69/674069/1/check/tripleo-ci-centos-7-scenario002-standalone/f0172a3/ : SUCCESS in 1h 28m 48s\n- tripleo-ci-centos-7-scenario003-standalone https://logs.opendev.org/69/674069/1/check/tripleo-ci-centos-7-scenario003-standalone/06111b7/ : SUCCESS in 1h 34m 07s\n- tripleo-ci-centos-7-scenario004-standalone https://logs.opendev.org/69/674069/1/check/tripleo-ci-centos-7-scenario004-standalone/0ff7af9/ : FAILURE in 1h 16m 40s\n- tripleo-ci-centos-7-scenario012-standalone https://logs.opendev.org/69/674069/1/check/tripleo-ci-centos-7-scenario012-standalone/208f121/ : SUCCESS in 2h 16m 37s (non-voting)\n- tripleo-ci-centos-7-scenario010-multinode-oooq-container https://logs.opendev.org/69/674069/1/check/tripleo-ci-centos-7-scenario010-multinode-oooq-container/58d1a5e/ : FAILURE in 2h 16m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1639bb5abf66e9d036825fff0600e7c0b68c7d8d","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2019-08-06 16:08:12.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d276d6f06938d1c85aee8899613b52b232ee78f5","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-06 19:10:48.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/69/674069/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/93ef02f/ : FAILURE in 1h 51m 44s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/69/674069/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/8dc2572/ : SUCCESS in 2h 54m 33s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/69/674069/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/0d3cdf6/ : SUCCESS in 2h 57m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"66b448ef830de637c491a8c07f5d5231437d3aa0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 23:05:09.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://logs.opendev.org/69/674069/1/check/puppet-openstack-lint-ubuntu-bionic/612c07a/ : SUCCESS in 4m 35s\n- puppet-openstack-syntax-5-ubuntu-bionic https://logs.opendev.org/69/674069/1/check/puppet-openstack-syntax-5-ubuntu-bionic/5e302da/ : SUCCESS in 4m 46s\n- puppet-openstack-syntax-6-ubuntu-bionic https://logs.opendev.org/69/674069/1/check/puppet-openstack-syntax-6-ubuntu-bionic/3f21c1f/ : SUCCESS in 4m 23s\n- puppet-openstack-unit-5.5-centos-7 https://logs.opendev.org/69/674069/1/check/puppet-openstack-unit-5.5-centos-7/65278c5/ : SUCCESS in 32m 05s\n- puppet-openstack-unit-6.0-ubuntu-bionic https://logs.opendev.org/69/674069/1/check/puppet-openstack-unit-6.0-ubuntu-bionic/39707c8/ : SUCCESS in 20m 52s\n- puppet-openstack-unit-latest-ubuntu-bionic https://logs.opendev.org/69/674069/1/check/puppet-openstack-unit-latest-ubuntu-bionic/86a509a/ : SUCCESS in 18m 46s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/69/674069/1/check/tripleo-ci-centos-7-undercloud-containers/b4490ab/ : FAILURE in 10m 01s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://logs.opendev.org/69/674069/1/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/406578d/ : FAILURE in 2h 43m 48s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/69/674069/1/check/tripleo-ci-centos-7-containers-multinode/82cc579/ : SUCCESS in 2h 36m 05s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/69/674069/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/910bf55/ : SUCCESS in 2h 33m 48s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/69/674069/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/ce02fcc/ : SUCCESS in 2h 23m 42s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/69/674069/1/check/tripleo-ci-centos-7-standalone/8cdcb36/ : SUCCESS in 2h 45m 18s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/69/674069/1/check/tripleo-ci-fedora-28-standalone/6dfe08f/ : SUCCESS in 1h 51m 34s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://logs.opendev.org/69/674069/1/check/tripleo-ci-centos-7-standalone-upgrade/80119c2/ : SUCCESS in 2h 04m 32s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://logs.opendev.org/69/674069/1/check/tripleo-ci-centos-7-scenario001-standalone/5457df0/ : SUCCESS in 1h 40m 09s\n- tripleo-ci-centos-7-scenario002-standalone https://logs.opendev.org/69/674069/1/check/tripleo-ci-centos-7-scenario002-standalone/c816b70/ : FAILURE in 1h 18m 52s\n- tripleo-ci-centos-7-scenario003-standalone https://logs.opendev.org/69/674069/1/check/tripleo-ci-centos-7-scenario003-standalone/a339601/ : SUCCESS in 1h 23m 00s\n- tripleo-ci-centos-7-scenario004-standalone https://logs.opendev.org/69/674069/1/check/tripleo-ci-centos-7-scenario004-standalone/8e71df5/ : SUCCESS in 1h 47m 47s\n- tripleo-ci-centos-7-scenario012-standalone https://logs.opendev.org/69/674069/1/check/tripleo-ci-centos-7-scenario012-standalone/679dd5e/ : SUCCESS in 1h 57m 42s (non-voting)\n- tripleo-ci-centos-7-scenario010-multinode-oooq-container https://logs.opendev.org/69/674069/1/check/tripleo-ci-centos-7-scenario010-multinode-oooq-container/3a18c2b/ : FAILURE in 1h 54m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"625e62d3c2466c6960493903fd86de7b6d886926","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2019-08-07 09:36:06.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c95bbd3dfa828761d6ca815f8a49a98caa318f93","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-07 12:49:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/69/674069/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/8e592ab/ : SUCCESS in 2h 55m 57s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/69/674069/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/5da8721/ : SUCCESS in 3h 03m 48s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/69/674069/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/633ea5f/ : SUCCESS in 3h 11m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c6c19bfeacdd387dd9d0e4d57c3c8476ab2211d8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 13:55:55.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://logs.opendev.org/69/674069/1/check/puppet-openstack-lint-ubuntu-bionic/433c141/ : SUCCESS in 4m 50s\n- puppet-openstack-syntax-5-ubuntu-bionic https://logs.opendev.org/69/674069/1/check/puppet-openstack-syntax-5-ubuntu-bionic/3e1c34e/ : SUCCESS in 5m 01s\n- puppet-openstack-syntax-6-ubuntu-bionic https://logs.opendev.org/69/674069/1/check/puppet-openstack-syntax-6-ubuntu-bionic/784333e/ : SUCCESS in 5m 45s\n- puppet-openstack-unit-5.5-centos-7 https://logs.opendev.org/69/674069/1/check/puppet-openstack-unit-5.5-centos-7/3a1412d/ : SUCCESS in 32m 55s\n- puppet-openstack-unit-6.0-ubuntu-bionic https://logs.opendev.org/69/674069/1/check/puppet-openstack-unit-6.0-ubuntu-bionic/6ec8bb5/ : SUCCESS in 18m 26s\n- puppet-openstack-unit-latest-ubuntu-bionic https://logs.opendev.org/69/674069/1/check/puppet-openstack-unit-latest-ubuntu-bionic/554bd4a/ : SUCCESS in 21m 04s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/69/674069/1/check/tripleo-ci-centos-7-undercloud-containers/6a336b4/ : SUCCESS in 1h 38m 24s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://logs.opendev.org/69/674069/1/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/4c5f405/ : FAILURE in 1h 53m 00s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/69/674069/1/check/tripleo-ci-centos-7-containers-multinode/4ccf555/ : SUCCESS in 2h 31m 56s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/69/674069/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/3514451/ : FAILURE in 15m 13s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/69/674069/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/50e7d58/ : SUCCESS in 2h 40m 16s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/69/674069/1/check/tripleo-ci-centos-7-standalone/2a865db/ : SUCCESS in 2h 04m 59s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/69/674069/1/check/tripleo-ci-fedora-28-standalone/0fb8219/ : SUCCESS in 2h 19m 53s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://logs.opendev.org/69/674069/1/check/tripleo-ci-centos-7-standalone-upgrade/2e5f7fc/ : SUCCESS in 2h 00m 48s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://logs.opendev.org/69/674069/1/check/tripleo-ci-centos-7-scenario001-standalone/419a936/ : SUCCESS in 2h 02m 22s\n- tripleo-ci-centos-7-scenario002-standalone https://logs.opendev.org/69/674069/1/check/tripleo-ci-centos-7-scenario002-standalone/ea8df48/ : SUCCESS in 1h 44m 53s\n- tripleo-ci-centos-7-scenario003-standalone https://logs.opendev.org/69/674069/1/check/tripleo-ci-centos-7-scenario003-standalone/44b0d57/ : SUCCESS in 1h 25m 30s\n- tripleo-ci-centos-7-scenario004-standalone https://logs.opendev.org/69/674069/1/check/tripleo-ci-centos-7-scenario004-standalone/dfbfebd/ : FAILURE in 53m 30s\n- tripleo-ci-centos-7-scenario012-standalone https://logs.opendev.org/69/674069/1/check/tripleo-ci-centos-7-scenario012-standalone/6df0906/ : SUCCESS in 2h 05m 03s (non-voting)\n- tripleo-ci-centos-7-scenario010-multinode-oooq-container https://logs.opendev.org/69/674069/1/check/tripleo-ci-centos-7-scenario010-multinode-oooq-container/e2bed70/ : FAILURE in 2h 04m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6a5fe1518f051d31ad52f46c43a23d3e8a66aa1c","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2019-08-08 09:16:06.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2a33541fe6dd987bd39a4590e0146eac88917e54","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-08 12:11:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/69/674069/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/0bfc363/ : FAILURE in 1h 59m 08s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/69/674069/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/dd8487a/ : SUCCESS in 2h 47m 10s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/69/674069/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/a876f16/ : SUCCESS in 2h 53m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a9593dd4df898e073be85656faa97ed57269a0ea","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 13:47:33.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://logs.opendev.org/69/674069/1/check/puppet-openstack-lint-ubuntu-bionic/174c0a7/ : SUCCESS in 5m 55s\n- puppet-openstack-syntax-5-ubuntu-bionic https://logs.opendev.org/69/674069/1/check/puppet-openstack-syntax-5-ubuntu-bionic/c34f415/ : SUCCESS in 6m 46s\n- puppet-openstack-syntax-6-ubuntu-bionic https://logs.opendev.org/69/674069/1/check/puppet-openstack-syntax-6-ubuntu-bionic/2c8a5ac/ : SUCCESS in 6m 05s\n- puppet-openstack-unit-5.5-centos-7 https://logs.opendev.org/69/674069/1/check/puppet-openstack-unit-5.5-centos-7/e6763e6/ : SUCCESS in 29m 49s\n- puppet-openstack-unit-6.0-ubuntu-bionic https://logs.opendev.org/69/674069/1/check/puppet-openstack-unit-6.0-ubuntu-bionic/d81ce56/ : SUCCESS in 19m 24s\n- puppet-openstack-unit-latest-ubuntu-bionic https://logs.opendev.org/69/674069/1/check/puppet-openstack-unit-latest-ubuntu-bionic/9e74dff/ : SUCCESS in 18m 57s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/69/674069/1/check/tripleo-ci-centos-7-undercloud-containers/ead7528/ : SUCCESS in 1h 37m 52s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://logs.opendev.org/69/674069/1/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/cd10874/ : FAILURE in 2h 37m 37s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/69/674069/1/check/tripleo-ci-centos-7-containers-multinode/7213db7/ : SUCCESS in 2h 30m 37s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/69/674069/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/25a9da6/ : SUCCESS in 2h 07m 35s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/69/674069/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/104ded9/ : SUCCESS in 2h 39m 29s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/69/674069/1/check/tripleo-ci-centos-7-standalone/0c4fa6b/ : SUCCESS in 1h 50m 00s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/69/674069/1/check/tripleo-ci-fedora-28-standalone/9c12a1e/ : SUCCESS in 1h 48m 51s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://logs.opendev.org/69/674069/1/check/tripleo-ci-centos-7-standalone-upgrade/ff05814/ : SUCCESS in 2h 07m 22s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://logs.opendev.org/69/674069/1/check/tripleo-ci-centos-7-scenario001-standalone/8426475/ : SUCCESS in 1h 51m 49s\n- tripleo-ci-centos-7-scenario002-standalone https://logs.opendev.org/69/674069/1/check/tripleo-ci-centos-7-scenario002-standalone/eea3a6a/ : SUCCESS in 1h 41m 33s\n- tripleo-ci-centos-7-scenario003-standalone https://logs.opendev.org/69/674069/1/check/tripleo-ci-centos-7-scenario003-standalone/eba2315/ : FAILURE in 48m 03s\n- tripleo-ci-centos-7-scenario004-standalone https://logs.opendev.org/69/674069/1/check/tripleo-ci-centos-7-scenario004-standalone/aaf5d5a/ : FAILURE in 48m 01s\n- tripleo-ci-centos-7-scenario012-standalone https://logs.opendev.org/69/674069/1/check/tripleo-ci-centos-7-scenario012-standalone/363a74c/ : SUCCESS in 1h 50m 30s (non-voting)\n- tripleo-ci-centos-7-scenario010-multinode-oooq-container https://logs.opendev.org/69/674069/1/check/tripleo-ci-centos-7-scenario010-multinode-oooq-container/1c182e4/ : FAILURE in 1h 59m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a3fffcc97d325e7c50d8e1373e48de8fa8aecc57","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2019-08-09 07:30:01.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"db31e856cc2d74cfc8a16df295e7e29a5589d2c2","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-09 10:40:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/69/674069/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/03a8a86/ : SUCCESS in 2h 40m 11s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/69/674069/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/17a1296/ : SUCCESS in 2h 50m 24s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/69/674069/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/be7f354/ : SUCCESS in 3h 08m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"16ba28a34565ef9e038e774164e589c646eb0809","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-09 11:47:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://logs.opendev.org/69/674069/1/check/puppet-openstack-lint-ubuntu-bionic/9eff138/ : SUCCESS in 4m 58s\n- puppet-openstack-syntax-5-ubuntu-bionic https://logs.opendev.org/69/674069/1/check/puppet-openstack-syntax-5-ubuntu-bionic/f73ece3/ : SUCCESS in 4m 26s\n- puppet-openstack-syntax-6-ubuntu-bionic https://logs.opendev.org/69/674069/1/check/puppet-openstack-syntax-6-ubuntu-bionic/f6ac88f/ : SUCCESS in 3m 56s\n- puppet-openstack-unit-5.5-centos-7 https://logs.opendev.org/69/674069/1/check/puppet-openstack-unit-5.5-centos-7/1472464/ : SUCCESS in 35m 11s\n- puppet-openstack-unit-6.0-ubuntu-bionic https://logs.opendev.org/69/674069/1/check/puppet-openstack-unit-6.0-ubuntu-bionic/d377ef1/ : SUCCESS in 19m 46s\n- puppet-openstack-unit-latest-ubuntu-bionic https://logs.opendev.org/69/674069/1/check/puppet-openstack-unit-latest-ubuntu-bionic/e8b6bae/ : SUCCESS in 21m 41s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/69/674069/1/check/tripleo-ci-centos-7-undercloud-containers/6a8b8be/ : SUCCESS in 1h 37m 13s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://logs.opendev.org/69/674069/1/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/0f9b6c0/ : FAILURE in 2h 07m 22s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/69/674069/1/check/tripleo-ci-centos-7-containers-multinode/370ac67/ : SUCCESS in 2h 29m 53s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/69/674069/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/f045fe4/ : SUCCESS in 2h 06m 45s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/69/674069/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/52a7724/ : SUCCESS in 2h 25m 59s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/69/674069/1/check/tripleo-ci-centos-7-standalone/0c6a044/ : SUCCESS in 1h 37m 47s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/69/674069/1/check/tripleo-ci-fedora-28-standalone/50875b2/ : SUCCESS in 2h 23m 03s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://logs.opendev.org/69/674069/1/check/tripleo-ci-centos-7-standalone-upgrade/786a8b1/ : SUCCESS in 2h 28m 10s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://logs.opendev.org/69/674069/1/check/tripleo-ci-centos-7-scenario001-standalone/1fe5143/ : SUCCESS in 1h 44m 39s\n- tripleo-ci-centos-7-scenario002-standalone https://logs.opendev.org/69/674069/1/check/tripleo-ci-centos-7-scenario002-standalone/61d7f8d/ : SUCCESS in 1h 39m 11s\n- tripleo-ci-centos-7-scenario003-standalone https://logs.opendev.org/69/674069/1/check/tripleo-ci-centos-7-scenario003-standalone/7d70887/ : SUCCESS in 1h 14m 41s\n- tripleo-ci-centos-7-scenario004-standalone https://logs.opendev.org/69/674069/1/check/tripleo-ci-centos-7-scenario004-standalone/4e4bc79/ : SUCCESS in 1h 43m 33s\n- tripleo-ci-centos-7-scenario012-standalone https://logs.opendev.org/69/674069/1/check/tripleo-ci-centos-7-scenario012-standalone/38aa2b0/ : SUCCESS in 1h 59m 26s (non-voting)\n- tripleo-ci-centos-7-scenario010-multinode-oooq-container https://logs.opendev.org/69/674069/1/check/tripleo-ci-centos-7-scenario010-multinode-oooq-container/4ae227a/ : FAILURE in 1h 52m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d8f1c3ec9ac983f611b0c9695029ef353d960b4b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-09 11:47:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"304b430afb72f98d77ae6e4a46de81264d625c31","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-09 14:13:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"766c0b24f83604855f248be956a9ee9317518c14","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-09 14:13:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://logs.opendev.org/69/674069/1/gate/puppet-openstack-lint-ubuntu-bionic/b28f630/ : SUCCESS in 5m 13s\n- puppet-openstack-syntax-5-ubuntu-bionic https://logs.opendev.org/69/674069/1/gate/puppet-openstack-syntax-5-ubuntu-bionic/f4ac0c7/ : SUCCESS in 6m 36s\n- puppet-openstack-syntax-6-ubuntu-bionic https://logs.opendev.org/69/674069/1/gate/puppet-openstack-syntax-6-ubuntu-bionic/2e439ab/ : SUCCESS in 4m 07s\n- puppet-openstack-unit-5.5-centos-7 https://logs.opendev.org/69/674069/1/gate/puppet-openstack-unit-5.5-centos-7/a48a747/ : SUCCESS in 29m 21s\n- puppet-openstack-unit-6.0-ubuntu-bionic https://logs.opendev.org/69/674069/1/gate/puppet-openstack-unit-6.0-ubuntu-bionic/5c10f15/ : SUCCESS in 19m 50s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/69/674069/1/gate/tripleo-ci-centos-7-undercloud-containers/b7e4204/ : SUCCESS in 1h 38m 04s\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/69/674069/1/gate/tripleo-ci-centos-7-containers-multinode/7a3d895/ : SUCCESS in 2h 20m 27s\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/69/674069/1/gate/tripleo-ci-centos-7-standalone/c0a8754/ : SUCCESS in 1h 45m 49s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/69/674069/1/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/7757af6/ : SUCCESS in 2h 19m 26s\n- tripleo-ci-centos-7-scenario001-standalone https://logs.opendev.org/69/674069/1/gate/tripleo-ci-centos-7-scenario001-standalone/89d1554/ : SUCCESS in 1h 30m 54s\n- tripleo-ci-centos-7-scenario002-standalone https://logs.opendev.org/69/674069/1/gate/tripleo-ci-centos-7-scenario002-standalone/07ffbc6/ : SUCCESS in 1h 37m 32s\n- tripleo-ci-centos-7-scenario003-standalone https://logs.opendev.org/69/674069/1/gate/tripleo-ci-centos-7-scenario003-standalone/f1e5c12/ : SUCCESS in 1h 16m 42s\n- tripleo-ci-centos-7-scenario004-standalone https://logs.opendev.org/69/674069/1/gate/tripleo-ci-centos-7-scenario004-standalone/e5f90dd/ : SUCCESS in 1h 40m 00s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3a5294ea9aa61d713632e031266dc9383c6bc8ec","revisions":{"3a5294ea9aa61d713632e031266dc9383c6bc8ec":{"kind":"REWORK","_number":1,"created":"2019-08-01 15:11:38.000000000","uploader":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"ref":"refs/changes/69/674069/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/69/674069/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/69/674069/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/69/674069/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/69/674069/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/69/674069/1"}}},"commit":{"parents":[{"commit":"5f0a05c052360d03cadcc634fb906a53494b4b6a","subject":"Merge \"Allow for customization of the Manila CephFS pool name\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/5f0a05c052360d03cadcc634fb906a53494b4b6a"}]}],"author":{"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","date":"2019-08-01 15:10:04.000000000","tz":120},"committer":{"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","date":"2019-08-01 15:10:04.000000000","tz":120},"subject":"Add missing base apache configs when internal TLS","message":"Add missing base apache configs when internal TLS\n\nSome services are missing the base apache configuration, when\nrunning with TLS and under WSGI. Address that in its base\nprofiles.\n\nRelated-Bug: #1835414\n\nChange-Id: I8148a039ab9dcbc97baff141aae6ebab4c27e16d\nSigned-off-by: Bogdan Dobrelya \u003cbdobreli@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/3a5294ea9aa61d713632e031266dc9383c6bc8ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/3a5294ea9aa61d713632e031266dc9383c6bc8ec"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
