)]}'
{"id":"openstack%2Fpuppet-tripleo~674553","triplet_id":"openstack%2Fpuppet-tripleo~master~Ib5f780349c12eebbf514a75293d8f4644dc5731f","project":"openstack/puppet-tripleo","branch":"master","topic":"ceph_dashboard","hashtags":[],"change_id":"Ib5f780349c12eebbf514a75293d8f4644dc5731f","subject":"Ceph grafana service certificate request","status":"MERGED","created":"2019-08-05 08:45:35.000000000","updated":"2019-08-13 15:16:02.000000000","submitted":"2019-08-13 15:16:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":22,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"674553-1565709362470-f7d3f53d","meta_rev_id":"22b83c1fcd3abaf5ee4a1da552adbd6b6d005ef4","_number":674553,"virtual_id_number":674553,"owner":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},{"value":0,"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"date":"2019-08-13 09:54:41.000000000","_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"value":0,"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"value":2,"date":"2019-08-13 15:16:02.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2019-08-13 09:05:38.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},{"value":0,"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},{"value":2,"date":"2019-08-12 22:14:57.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":1,"date":"2019-08-12 08:23:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"value":2,"date":"2019-08-09 10:02:20.000000000","_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-13 09:55:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"date":"2019-08-08 13:24:41.000000000","_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},{"value":0,"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},{"value":1,"date":"2019-08-12 22:14:57.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"value":0,"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-07 11:13:15.000000000","updated_by":{"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},"reviewer":{"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},"state":"REVIEWER"},{"updated":"2019-08-07 11:49:31.000000000","updated_by":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"reviewer":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"state":"REVIEWER"},{"updated":"2019-08-08 13:24:41.000000000","updated_by":{"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},"reviewer":{"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},"state":"REVIEWER"},{"updated":"2019-08-09 10:02:20.000000000","updated_by":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"reviewer":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"state":"REVIEWER"},{"updated":"2019-08-12 08:23:02.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"},{"updated":"2019-08-12 22:14:57.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-08-13 09:05:38.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-13 09:55:03.000000000","updated_by":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"reviewer":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"state":"REVIEWER"},{"updated":"2019-08-13 15:16:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7a68c6a939ef880e5d1d5c4622229e645eb43a60","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2019-08-05 08:45:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a08905e0a4211ebee00dae5aab3c93e20ea3d812","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-05 09:52:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/53/674553/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/8d6cffb/ : FAILURE in 49m 20s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/53/674553/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/dff2895/ : FAILURE in 49m 13s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/53/674553/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/bdce806/ : FAILURE in 1h 04m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e8d01f7785ad2c6dedf08e491a10be2742e0aba8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 10:17:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://logs.opendev.org/53/674553/1/check/puppet-openstack-lint-ubuntu-bionic/74ac55b/ : FAILURE in 5m 26s\n- puppet-openstack-syntax-5-ubuntu-bionic https://logs.opendev.org/53/674553/1/check/puppet-openstack-syntax-5-ubuntu-bionic/d13df30/ : FAILURE in 4m 50s\n- puppet-openstack-syntax-6-ubuntu-bionic https://logs.opendev.org/53/674553/1/check/puppet-openstack-syntax-6-ubuntu-bionic/b3cb806/ : FAILURE in 5m 21s\n- puppet-openstack-unit-5.5-centos-7 https://logs.opendev.org/53/674553/1/check/puppet-openstack-unit-5.5-centos-7/f62e54b/ : SUCCESS in 33m 03s\n- puppet-openstack-unit-6.0-ubuntu-bionic https://logs.opendev.org/53/674553/1/check/puppet-openstack-unit-6.0-ubuntu-bionic/d2f830b/ : SUCCESS in 18m 31s\n- puppet-openstack-unit-latest-ubuntu-bionic https://logs.opendev.org/53/674553/1/check/puppet-openstack-unit-latest-ubuntu-bionic/70ddbcb/ : SUCCESS in 18m 09s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/53/674553/1/check/tripleo-ci-centos-7-undercloud-containers/6565ac2/ : FAILURE in 57m 04s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://logs.opendev.org/53/674553/1/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/44274df/ : FAILURE in 1h 01m 16s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/53/674553/1/check/tripleo-ci-centos-7-containers-multinode/a4a5e13/ : FAILURE in 12m 27s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/53/674553/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/d14f5b3/ : FAILURE in 43m 25s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/53/674553/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/91fbca8/ : FAILURE in 45m 31s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/53/674553/1/check/tripleo-ci-centos-7-standalone/9e17e17/ : FAILURE in 48m 52s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/53/674553/1/check/tripleo-ci-fedora-28-standalone/de3e800/ : FAILURE in 44m 58s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://logs.opendev.org/53/674553/1/check/tripleo-ci-centos-7-standalone-upgrade/6b64c3e/ : FAILURE in 47m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4de3933d8daed925683d265764f16a9859828903","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2019-08-05 11:05:44.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d836de8a5bc720d8864aa5c04abbf0d018c6f771","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2019-08-05 11:53:01.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3eff44392a85acc205b46a6c77fd7834f4afea6d","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2019-08-05 12:12:21.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"56e214fd54f0b4410f57f32b9f3e6236ff4520a4","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2019-08-05 12:13:59.000000000","message":"Patch Set 4:\n\nI need to change the variable name because of [1] that presents the following error:\n\n```\n\nCould not parse for environment *root*: Unacceptable name. The name \u0027tripleo::certmonger::ceph_grafana-server\u0027 is unacceptable as the name of a Host Class Definition (file: /home/zuul/workspace/manifests/certmonger/ceph_grafana.pp, line: 48, column: 1)\n2019-08-05 11:59:23.182584 | ubuntu-bionic | Could not parse for environment *root*: Syntax error at \u0027-\u0027 (file: /home/zuul/workspace/manifests/profile/base/certmonger_user.pp, line: 172, column: 16)\n2019-08-05 11:59:23.402158 | ubuntu-bionic | ERROR\n2019-08-05 11:59:23.402572 | ubuntu-bionic | {\n2019-08-05 11:59:23.402688 | ubuntu-bionic |   \"delta\": \"0:01:56.420985\",\n2019-08-05 11:59:23.402796 | ubuntu-bionic |   \"end\": \"2019-08-05 11:59:23.186474\",\n2019-08-05 11:59:23.402902 | ubuntu-bionic |   \"msg\": \"non-zero return code\",\n2019-08-05 11:59:23.403005 | ubuntu-bionic |   \"rc\": 1,\n2019-08-05 11:59:23.403108 | ubuntu-bionic |   \"start\": \"2019-08-05 11:57:26.765489\"\n2019-08-05 11:59:23.403226 | ubuntu-bionic | }\n\n``` \n\n[1] https://logs.opendev.org/53/674553/3/check/puppet-openstack-syntax-5-ubuntu-bionic/5ab6929/job-output.txt.gz","accounts_in_message":[],"_revision_number":4},{"id":"ba80f1bc7d53c1da19c98d54a2bb89638711943c","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2019-08-05 12:31:08.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"91b458f0034bb32a85eb6b4a3dbdb97b7ac709c8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 15:16:40.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://logs.opendev.org/53/674553/5/check/puppet-openstack-lint-ubuntu-bionic/0ce0351/ : SUCCESS in 4m 38s\n- puppet-openstack-syntax-5-ubuntu-bionic https://logs.opendev.org/53/674553/5/check/puppet-openstack-syntax-5-ubuntu-bionic/f921de3/ : SUCCESS in 3m 40s\n- puppet-openstack-syntax-6-ubuntu-bionic https://logs.opendev.org/53/674553/5/check/puppet-openstack-syntax-6-ubuntu-bionic/fcbb759/ : SUCCESS in 4m 28s\n- puppet-openstack-unit-5.5-centos-7 https://logs.opendev.org/53/674553/5/check/puppet-openstack-unit-5.5-centos-7/f7013de/ : SUCCESS in 27m 41s\n- puppet-openstack-unit-6.0-ubuntu-bionic https://logs.opendev.org/53/674553/5/check/puppet-openstack-unit-6.0-ubuntu-bionic/7fec191/ : SUCCESS in 18m 21s\n- puppet-openstack-unit-latest-ubuntu-bionic https://logs.opendev.org/53/674553/5/check/puppet-openstack-unit-latest-ubuntu-bionic/25cf618/ : SUCCESS in 17m 15s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/53/674553/5/check/tripleo-ci-centos-7-undercloud-containers/556e728/ : SUCCESS in 1h 50m 30s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://logs.opendev.org/53/674553/5/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/9b62aed/ : FAILURE in 57m 31s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/53/674553/5/check/tripleo-ci-centos-7-containers-multinode/6c6a4a8/ : SUCCESS in 2h 36m 03s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/53/674553/5/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/a9a1f53/ : SUCCESS in 2h 29m 49s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/53/674553/5/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/6614e0e/ : SUCCESS in 2h 40m 40s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/53/674553/5/check/tripleo-ci-centos-7-standalone/9960ed4/ : SUCCESS in 1h 44m 48s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/53/674553/5/check/tripleo-ci-fedora-28-standalone/7123fce/ : SUCCESS in 1h 51m 28s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://logs.opendev.org/53/674553/5/check/tripleo-ci-centos-7-standalone-upgrade/f5bba50/ : FAILURE in 49m 04s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"4eb8a7acee559fc1f576db29653de263d83fab7e","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-05 16:12:53.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/53/674553/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/fcd74a3/ : TIMED_OUT in 3h 39m 14s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/53/674553/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/5813a58/ : SUCCESS in 3h 12m 22s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/53/674553/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/ded0a73/ : FAILURE in 2h 19m 19s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"3e234d03dbf333d052a6c80b2a3e9fe20a596859","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2019-08-05 17:05:44.000000000","message":"Patch Set 5:\n\ncheck-rdo","accounts_in_message":[],"_revision_number":5},{"id":"6a27f24e11f3104a8558427fa32aef1a1f89ff56","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-05 20:07:23.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/53/674553/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/30df079/ : SUCCESS in 2h 41m 34s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/53/674553/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/48c9aa1/ : FAILURE in 1h 14m 45s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/53/674553/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/1b4d359/ : SUCCESS in 2h 59m 30s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"26c0bf066f6c2fff69ca16d5ad91b59887806694","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2019-08-07 06:34:49.000000000","message":"Patch Set 5:\n\ncheck-rdo","accounts_in_message":[],"_revision_number":5},{"id":"c354bf6a93e50cffa6db8115d9327e2812815e95","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-07 09:45:00.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/53/674553/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/71e9414/ : SUCCESS in 2h 50m 03s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/53/674553/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/f8c2c49/ : SUCCESS in 2h 54m 07s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/53/674553/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/46e1641/ : SUCCESS in 3h 08m 35s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"77d485f11bb6ee63803e3f9013f363f10cd40800","author":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"date":"2019-08-07 10:56:28.000000000","message":"Patch Set 5:\n\n(9 comments)","accounts_in_message":[],"_revision_number":5},{"id":"aae89a613e5367b0f9e623a3540a266c3c98344e","author":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"date":"2019-08-07 10:59:59.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"276e7a6fa52aaf0b759cc1f2c647cced97f64927","author":{"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},"date":"2019-08-07 11:13:15.000000000","message":"Patch Set 5: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":5},{"id":"5aa4f160de8aa8079cb87df384049f5ebd2268b9","author":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"date":"2019-08-07 11:22:23.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"c32e6142318569ccb060e05d8a0fb5179c26406a","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2019-08-07 12:05:56.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"b9d3a11faf24f367e4eaec02cf9b9c74c511209a","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2019-08-07 12:21:23.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"6a465883f3b4472e2edff9857792209b0d5253f0","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2019-08-07 12:25:08.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"d08fcc853a95e717c3b9feb2945c1c6286fb800b","author":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"date":"2019-08-07 12:35:54.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"ab37f580bdeeaeb5ac81a6b5748c2b373b341a8c","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2019-08-07 12:37:49.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"2a1ef25adc79b892ab457cddc4783f8d5faf8723","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2019-08-07 12:39:06.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"9f234c3f57f84495efb599333d2f8db941b27ba7","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2019-08-07 12:40:49.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"dc6f3259d559e1d30fe0c5ef3b07a329e7fde54f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-07 15:41:25.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/53/674553/8/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/cf9e12b/ : SUCCESS in 2h 57m 27s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/53/674553/8/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/b50f27c/ : FAILURE in 2h 57m 54s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/53/674553/8/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/89bdcf2/ : FAILURE in 2h 29m 48s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"9afd38adf48c560efd4e69ca14f8816540e52204","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 19:27:20.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://logs.opendev.org/53/674553/8/check/puppet-openstack-lint-ubuntu-bionic/db7f5df/ : FAILURE in 5m 18s\n- puppet-openstack-syntax-5-ubuntu-bionic https://logs.opendev.org/53/674553/8/check/puppet-openstack-syntax-5-ubuntu-bionic/031f015/ : FAILURE in 4m 02s\n- puppet-openstack-syntax-6-ubuntu-bionic https://logs.opendev.org/53/674553/8/check/puppet-openstack-syntax-6-ubuntu-bionic/c404275/ : FAILURE in 4m 35s\n- puppet-openstack-unit-5.5-centos-7 https://logs.opendev.org/53/674553/8/check/puppet-openstack-unit-5.5-centos-7/46e0baf/ : SUCCESS in 39m 50s\n- puppet-openstack-unit-6.0-ubuntu-bionic https://logs.opendev.org/53/674553/8/check/puppet-openstack-unit-6.0-ubuntu-bionic/4c4e99c/ : SUCCESS in 18m 02s\n- puppet-openstack-unit-latest-ubuntu-bionic https://logs.opendev.org/53/674553/8/check/puppet-openstack-unit-latest-ubuntu-bionic/43c7096/ : SUCCESS in 20m 53s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/53/674553/8/check/tripleo-ci-centos-7-undercloud-containers/05ab0ef/ : FAILURE in 1h 10m 40s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://logs.opendev.org/53/674553/8/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/3f0d9e4/ : FAILURE in 1h 55m 50s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/53/674553/8/check/tripleo-ci-centos-7-containers-multinode/f0382c6/ : SUCCESS in 2h 45m 23s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/53/674553/8/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/93f631d/ : SUCCESS in 2h 34m 48s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/53/674553/8/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/135874c/ : SUCCESS in 2h 47m 20s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/53/674553/8/check/tripleo-ci-centos-7-standalone/2b821fb/ : SUCCESS in 1h 56m 32s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/53/674553/8/check/tripleo-ci-fedora-28-standalone/45b1040/ : SUCCESS in 1h 57m 36s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://logs.opendev.org/53/674553/8/check/tripleo-ci-centos-7-standalone-upgrade/f9324f4/ : SUCCESS in 2h 14m 52s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"fe0e2e46d7c24390c37ae1163ea1eee8082768ca","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2019-08-07 19:44:58.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"f5ebc7f179c90f7f0048816bfc6541e020bebd8d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-07 22:45:56.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/53/674553/9/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/381d5c8/ : SUCCESS in 2h 35m 49s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/53/674553/9/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/e678dc1/ : SUCCESS in 2h 56m 17s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/53/674553/9/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/d307f40/ : SUCCESS in 3h 00m 06s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"d2f60f7722aacb4660c28f6816613a1f3a51f0da","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 23:14:36.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://logs.opendev.org/53/674553/9/check/puppet-openstack-lint-ubuntu-bionic/637f321/ : SUCCESS in 5m 05s\n- puppet-openstack-syntax-5-ubuntu-bionic https://logs.opendev.org/53/674553/9/check/puppet-openstack-syntax-5-ubuntu-bionic/550a726/ : SUCCESS in 4m 02s\n- puppet-openstack-syntax-6-ubuntu-bionic https://logs.opendev.org/53/674553/9/check/puppet-openstack-syntax-6-ubuntu-bionic/8aecb76/ : SUCCESS in 4m 39s\n- puppet-openstack-unit-5.5-centos-7 https://logs.opendev.org/53/674553/9/check/puppet-openstack-unit-5.5-centos-7/58a8e2c/ : SUCCESS in 32m 58s\n- puppet-openstack-unit-6.0-ubuntu-bionic https://logs.opendev.org/53/674553/9/check/puppet-openstack-unit-6.0-ubuntu-bionic/5941a10/ : SUCCESS in 22m 16s\n- puppet-openstack-unit-latest-ubuntu-bionic https://logs.opendev.org/53/674553/9/check/puppet-openstack-unit-latest-ubuntu-bionic/c372c15/ : SUCCESS in 18m 11s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/53/674553/9/check/tripleo-ci-centos-7-undercloud-containers/bdf0231/ : SUCCESS in 1h 37m 39s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://logs.opendev.org/53/674553/9/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/b972250/ : FAILURE in 2h 04m 34s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/53/674553/9/check/tripleo-ci-centos-7-containers-multinode/328b73a/ : SUCCESS in 2h 33m 54s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/53/674553/9/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/bdecdcb/ : SUCCESS in 2h 19m 12s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/53/674553/9/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/61ab020/ : SUCCESS in 3h 08m 18s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/53/674553/9/check/tripleo-ci-centos-7-standalone/c6214c6/ : SUCCESS in 1h 52m 40s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/53/674553/9/check/tripleo-ci-fedora-28-standalone/c24029b/ : FAILURE in 2h 21m 03s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://logs.opendev.org/53/674553/9/check/tripleo-ci-centos-7-standalone-upgrade/426d3be/ : SUCCESS in 2h 52m 41s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"aa2e2e84865c899d73cff46974192ca468457498","author":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"date":"2019-08-09 10:02:20.000000000","message":"Patch Set 9: Code-Review+2\n\nwould like tls experts to review, but lgtm","accounts_in_message":[],"_revision_number":9},{"id":"adee283ecc357cecb844b8acf1d2bb32add7138f","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2019-08-12 08:23:02.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"b90eee160b70848d437239caf47a73565276d0c9","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-12 22:14:57.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"aa49ef1c73a6482ab0496d4fe400fd7fae58306c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 22:15:17.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"cc9c7f6525c5f152a3c33ce4feb0252a75155acd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-13 01:35:42.000000000","message":"Patch Set 9: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/44b60470785d4eba85af32f301c57785 : SUCCESS in 5m 55s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/81252cf5d78b475f994a0b9ebcd080fc : SUCCESS in 4m 46s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d4c0b4f1a3504248887f0acbc18ba4ef : SUCCESS in 6m 17s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/cae480066dcb42d9b0e2df9506601bf3 : FAILURE in 7m 07s\n- puppet-openstack-unit-6.0-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/55bedd2c62e04984845b41ca66f6abf4 : SUCCESS in 21m 33s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/bf48d936d1ce4dbe8b0d215c8e92a0ee : SUCCESS in 2h 15m 02s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/e350162104f94624b8d84738889af268 : TIMED_OUT in 3h 17m 42s\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/47635d8dc2ba4877b83bdaf882035a4f : SUCCESS in 2h 13m 47s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/3fe8e0035d99421c934902e8bbad612c : SUCCESS in 2h 17m 12s","accounts_in_message":[],"_revision_number":9},{"id":"e91765104646bc5feaa4e5a8356f6849c0bac3ec","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2019-08-13 05:20:30.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"a068e00da03a658ffcf81c9ffb632ac893dcc36a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-13 07:52:32.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2032e4c5a5be4758bd672fc9bb4659da : SUCCESS in 3m 39s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/322614642b884395bfe0aeb44ba3b03c : SUCCESS in 3m 37s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a2910242dd1a49e387358e19e434f5cc : SUCCESS in 3m 55s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/84e54afda8e5486a923e851e8c9b0ccc : SUCCESS in 33m 54s\n- puppet-openstack-unit-6.0-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fed36ca94d794916b70b4fb8113cbe0d : SUCCESS in 16m 37s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/03e039e1d1974a9ba017946b5a0aa947 : SUCCESS in 17m 12s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/7ee861092b9e4f5dbe64ec26ec6384c0 : SUCCESS in 1h 32m 35s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/a9e8be947bc84e8fbdea6a8b7991be94 : FAILURE in 1h 51m 45s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/d1fc80d5cf4c4bfaa75e0a0055564ac9 : SUCCESS in 1h 59m 17s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/4f6176ec619c40eabed26b146f2e9965 : SUCCESS in 2h 01m 37s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/5bf9c2f9b63b4ab690412cb96869260b : SUCCESS in 2h 26m 22s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/3da6ff131a584fc590c37abc832fc0b8 : FAILURE in 1h 50m 19s\n- tripleo-ci-fedora-28-standalone https://zuul.opendev.org/t/openstack/build/040db358a550411998b7adda89fb4235 : SUCCESS in 1h 42m 23s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/54945d535c5945079c35d124c859b61b : SUCCESS in 2h 15m 27s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"e3db781372109080f0f71b3cc9c7b523b616e459","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-13 09:05:38.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 finger://ze04.softwarefactory-project.io:7979/4547af2833e64f7ab9e9b45bc1e0a1a8 : POST_FAILURE in 3h 41m 33s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/53/674553/9/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/0c5608e/ : SUCCESS in 3h 07m 51s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 finger://ze04.softwarefactory-project.io:7979/5879ddd98e1641dd8a2598b953357de3 : POST_FAILURE in 3h 43m 49s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"8a1614a10a4572a31d27ab3958f02d5accc1080d","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2019-08-13 09:54:41.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"8e4818a4fb521bddbb637bd2f5199b97ec300e08","author":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"date":"2019-08-13 09:55:03.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"8e11b57af995af5f6be6ffa36a9e4c68b5bf5924","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-13 12:44:25.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f9849b9a7f0f454182ab41784232a2ce : SUCCESS in 5m 42s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/59a9baf719dc49a0b683704367ebdbfc : SUCCESS in 4m 48s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3ea281e2fd1b40a0989deccbc8810e3b : SUCCESS in 4m 58s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/b824acf0dd444e8186bc733e6bedbbed : SUCCESS in 28m 21s\n- puppet-openstack-unit-6.0-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7d298c85951f40ee858a03ad537b55cd : SUCCESS in 19m 22s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/21bd988209b54bc6a3097c4111ef2186 : SUCCESS in 20m 16s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/23be8c5aac0844c997b9488341ba71f4 : SUCCESS in 1h 37m 06s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/bce3f5f8652a4288baf34ea300d62090 : FAILURE in 2h 07m 29s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/4a4640416d0347b0aebdc9a442250844 : SUCCESS in 2h 37m 02s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/94a602b1966b46f2872c9abb6c5a14d5 : SUCCESS in 2h 05m 17s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/534affe3d52848d7afca47b02f8c96f3 : SUCCESS in 2h 39m 16s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/8925da523c6b457b99664445733c15d8 : SUCCESS in 1h 58m 00s\n- tripleo-ci-fedora-28-standalone https://zuul.opendev.org/t/openstack/build/8631c4e5a78540beb30480724939ba6a : SUCCESS in 1h 49m 55s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/f9837acd8af340fba7fd0b193e2f213c : SUCCESS in 2h 19m 23s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"b8d7b57e12feca4d8c013b92e0d0fdc620348e27","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-13 12:44:45.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"21f561a9655ed3d3dd95cb6c4cc19099a89d2180","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-13 14:03:22.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 finger://ze05.softwarefactory-project.io:7979/77f6b8318d7c4681a48c0e8cb327a04f : POST_FAILURE in 3h 35m 30s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 finger://ze01.softwarefactory-project.io:7979/0ce0c7a3dc1e4168a6e998121a5191c9 : POST_FAILURE in 3h 41m 46s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 finger://ze07.softwarefactory-project.io:7979/67b103c891a64af387f883b87b37f4f1 : POST_FAILURE in 4h 07m 45s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"c758eea66816db956932a2e3cec3e3ac524ea068","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-13 15:16:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":9},{"id":"26d51cee2aee9a13df8032fec598d35b00483044","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-13 15:16:02.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2bd69b0cc1374ddb9848fa78e29d5f9f : SUCCESS in 4m 17s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c95325b6de57498496f014f6fea25078 : SUCCESS in 3m 50s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/79f0bbec6a9b418486b30d9ab0f55ea1 : SUCCESS in 4m 55s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/36a85ad2a37e494fa64b71e86ffeee2c : SUCCESS in 32m 41s\n- puppet-openstack-unit-6.0-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f1a3c541310e4d93a255ddb6b589aaf6 : SUCCESS in 21m 01s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/2668ecf6a104451fba48d4ea5078e295 : SUCCESS in 1h 38m 01s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/d1910284610f4252b3663d1e2fcb4695 : SUCCESS in 2h 17m 06s\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/cf4435a4430841baaee7ff9813d29605 : SUCCESS in 1h 37m 47s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/eb554ca4ad28492eb99fe76d41739947 : SUCCESS in 2h 22m 52s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"dcf2263522440580508031c600ca1a15bb49ac2f","revisions":{"add6408b92b5c469e60153b23d6ad1783c866576":{"kind":"REWORK","_number":1,"created":"2019-08-05 08:45:35.000000000","uploader":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"ref":"refs/changes/53/674553/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/53/674553/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/53/674553/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/53/674553/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/53/674553/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/53/674553/1"}}},"commit":{"parents":[{"commit":"8e4de1b75ce67a253d5c872d3de18128cc17f2a9","subject":"Merge \"Enable service token in nova and cinder\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/8e4de1b75ce67a253d5c872d3de18128cc17f2a9"}]}],"author":{"name":"fpantano","email":"fpantano@redhat.com","date":"2019-08-05 08:45:49.000000000","tz":120},"committer":{"name":"fpantano","email":"fpantano@redhat.com","date":"2019-08-05 08:49:56.000000000","tz":120},"subject":"[WIP] Ceph grafana service certificate request","message":"[WIP] Ceph grafana service certificate request\n\nThis change adds the certmonger ceph_grafana class to\ndefine the necessary requirements to make the cert request\nfor the ceph dashboard service.\n\nChange-Id: Ib5f780349c12eebbf514a75293d8f4644dc5731f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/add6408b92b5c469e60153b23d6ad1783c866576"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/add6408b92b5c469e60153b23d6ad1783c866576"}]},"branch":"refs/heads/master"},"301d3fc1d7390375b40e6300d8d1c5cb3e692c31":{"kind":"REWORK","_number":2,"created":"2019-08-05 11:05:44.000000000","uploader":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"ref":"refs/changes/53/674553/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/53/674553/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/53/674553/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/53/674553/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/53/674553/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/53/674553/2"}}},"commit":{"parents":[{"commit":"8e4de1b75ce67a253d5c872d3de18128cc17f2a9","subject":"Merge \"Enable service token in nova and cinder\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/8e4de1b75ce67a253d5c872d3de18128cc17f2a9"}]}],"author":{"name":"fpantano","email":"fpantano@redhat.com","date":"2019-08-05 08:45:49.000000000","tz":120},"committer":{"name":"fpantano","email":"fpantano@redhat.com","date":"2019-08-05 11:09:57.000000000","tz":120},"subject":"[WIP] Ceph grafana service certificate request","message":"[WIP] Ceph grafana service certificate request\n\nThis change adds the certmonger ceph_grafana class to\ndefine the necessary requirements to make the cert request\nfor the ceph dashboard service.\n\nChange-Id: Ib5f780349c12eebbf514a75293d8f4644dc5731f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/301d3fc1d7390375b40e6300d8d1c5cb3e692c31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/301d3fc1d7390375b40e6300d8d1c5cb3e692c31"}]},"branch":"refs/heads/master"},"c9bd6ca0dc1f91215cc228094a36bc9f39943023":{"kind":"REWORK","_number":3,"created":"2019-08-05 11:53:01.000000000","uploader":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"ref":"refs/changes/53/674553/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/53/674553/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/53/674553/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/53/674553/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/53/674553/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/53/674553/3"}}},"commit":{"parents":[{"commit":"8e4de1b75ce67a253d5c872d3de18128cc17f2a9","subject":"Merge \"Enable service token in nova and cinder\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/8e4de1b75ce67a253d5c872d3de18128cc17f2a9"}]}],"author":{"name":"fpantano","email":"fpantano@redhat.com","date":"2019-08-05 08:45:49.000000000","tz":120},"committer":{"name":"fpantano","email":"fpantano@redhat.com","date":"2019-08-05 11:57:12.000000000","tz":120},"subject":"[WIP] Ceph grafana service certificate request","message":"[WIP] Ceph grafana service certificate request\n\nThis change adds the certmonger ceph_grafana class to\ndefine the necessary requirements to make the cert request\nfor the ceph dashboard service.\n\nChange-Id: Ib5f780349c12eebbf514a75293d8f4644dc5731f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/c9bd6ca0dc1f91215cc228094a36bc9f39943023"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/c9bd6ca0dc1f91215cc228094a36bc9f39943023"}]},"branch":"refs/heads/master"},"321d69572e3531fd121740a04f8bdf625a8a4174":{"kind":"REWORK","_number":4,"created":"2019-08-05 12:12:21.000000000","uploader":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"ref":"refs/changes/53/674553/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/53/674553/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/53/674553/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/53/674553/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/53/674553/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/53/674553/4"}}},"commit":{"parents":[{"commit":"8e4de1b75ce67a253d5c872d3de18128cc17f2a9","subject":"Merge \"Enable service token in nova and cinder\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/8e4de1b75ce67a253d5c872d3de18128cc17f2a9"}]}],"author":{"name":"fpantano","email":"fpantano@redhat.com","date":"2019-08-05 08:45:49.000000000","tz":120},"committer":{"name":"fpantano","email":"fpantano@redhat.com","date":"2019-08-05 12:16:29.000000000","tz":120},"subject":"[WIP] Ceph grafana service certificate request","message":"[WIP] Ceph grafana service certificate request\n\nThis change adds the certmonger ceph_grafana class to\ndefine the necessary requirements to make the cert request\nfor the ceph dashboard service.\n\nChange-Id: Ib5f780349c12eebbf514a75293d8f4644dc5731f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/321d69572e3531fd121740a04f8bdf625a8a4174"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/321d69572e3531fd121740a04f8bdf625a8a4174"}]},"branch":"refs/heads/master"},"9a3d41a573ed0fd76c1bfbf2bdc0259cf5685e8a":{"kind":"REWORK","_number":5,"created":"2019-08-05 12:31:08.000000000","uploader":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"ref":"refs/changes/53/674553/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/53/674553/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/53/674553/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/53/674553/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/53/674553/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/53/674553/5"}}},"commit":{"parents":[{"commit":"8e4de1b75ce67a253d5c872d3de18128cc17f2a9","subject":"Merge \"Enable service token in nova and cinder\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/8e4de1b75ce67a253d5c872d3de18128cc17f2a9"}]}],"author":{"name":"fpantano","email":"fpantano@redhat.com","date":"2019-08-05 08:45:49.000000000","tz":120},"committer":{"name":"fpantano","email":"fpantano@redhat.com","date":"2019-08-05 12:35:17.000000000","tz":120},"subject":"[WIP] Ceph grafana service certificate request","message":"[WIP] Ceph grafana service certificate request\n\nThis change adds the certmonger ceph_grafana class to\ndefine the necessary requirements to make the cert request\nfor the ceph dashboard service.\n\nChange-Id: Ib5f780349c12eebbf514a75293d8f4644dc5731f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/9a3d41a573ed0fd76c1bfbf2bdc0259cf5685e8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/9a3d41a573ed0fd76c1bfbf2bdc0259cf5685e8a"}]},"branch":"refs/heads/master"},"1383b221296d23767dc063849d6061800e181450":{"kind":"REWORK","_number":6,"created":"2019-08-07 12:21:23.000000000","uploader":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"ref":"refs/changes/53/674553/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/53/674553/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/53/674553/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/53/674553/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/53/674553/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/53/674553/6"}}},"commit":{"parents":[{"commit":"8e4de1b75ce67a253d5c872d3de18128cc17f2a9","subject":"Merge \"Enable service token in nova and cinder\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/8e4de1b75ce67a253d5c872d3de18128cc17f2a9"}]}],"author":{"name":"fpantano","email":"fpantano@redhat.com","date":"2019-08-05 08:45:49.000000000","tz":120},"committer":{"name":"fpantano","email":"fpantano@redhat.com","date":"2019-08-07 12:25:38.000000000","tz":120},"subject":"Ceph grafana service certificate request","message":"Ceph grafana service certificate request\n\nThis change adds the certmonger ceph_grafana class to\ndefine the necessary requirements to make the cert request\nfor the ceph dashboard service.\n\nChange-Id: Ib5f780349c12eebbf514a75293d8f4644dc5731f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/1383b221296d23767dc063849d6061800e181450"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/1383b221296d23767dc063849d6061800e181450"}]},"branch":"refs/heads/master"},"109ea77f3d8813f4a9dc9f4fdcbbf7c28bac47d7":{"kind":"REWORK","_number":7,"created":"2019-08-07 12:39:06.000000000","uploader":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"ref":"refs/changes/53/674553/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/53/674553/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/53/674553/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/53/674553/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/53/674553/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/53/674553/7"}}},"commit":{"parents":[{"commit":"8e4de1b75ce67a253d5c872d3de18128cc17f2a9","subject":"Merge \"Enable service token in nova and cinder\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/8e4de1b75ce67a253d5c872d3de18128cc17f2a9"}]}],"author":{"name":"fpantano","email":"fpantano@redhat.com","date":"2019-08-05 08:45:49.000000000","tz":120},"committer":{"name":"fpantano","email":"fpantano@redhat.com","date":"2019-08-07 12:43:00.000000000","tz":120},"subject":"Ceph grafana service certificate request","message":"Ceph grafana service certificate request\n\nThis change adds the certmonger ceph_grafana class to\ndefine the necessary requirements to make the cert request\nfor the ceph dashboard service.\n\nChange-Id: Ib5f780349c12eebbf514a75293d8f4644dc5731f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/109ea77f3d8813f4a9dc9f4fdcbbf7c28bac47d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/109ea77f3d8813f4a9dc9f4fdcbbf7c28bac47d7"}]},"branch":"refs/heads/master"},"d95a11002c88859e62e9dc8fcb008624e126cf93":{"kind":"REWORK","_number":8,"created":"2019-08-07 12:40:49.000000000","uploader":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"ref":"refs/changes/53/674553/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/53/674553/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/53/674553/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/53/674553/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/53/674553/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/53/674553/8"}}},"commit":{"parents":[{"commit":"8e4de1b75ce67a253d5c872d3de18128cc17f2a9","subject":"Merge \"Enable service token in nova and cinder\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/8e4de1b75ce67a253d5c872d3de18128cc17f2a9"}]}],"author":{"name":"fpantano","email":"fpantano@redhat.com","date":"2019-08-05 08:45:49.000000000","tz":120},"committer":{"name":"fpantano","email":"fpantano@redhat.com","date":"2019-08-07 12:45:05.000000000","tz":120},"subject":"Ceph grafana service certificate request","message":"Ceph grafana service certificate request\n\nThis change adds the certmonger ceph_grafana class to\ndefine the necessary requirements to make the cert request\nfor the ceph dashboard service.\n\nChange-Id: Ib5f780349c12eebbf514a75293d8f4644dc5731f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/d95a11002c88859e62e9dc8fcb008624e126cf93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/d95a11002c88859e62e9dc8fcb008624e126cf93"}]},"branch":"refs/heads/master"},"dcf2263522440580508031c600ca1a15bb49ac2f":{"kind":"REWORK","_number":9,"created":"2019-08-07 19:44:58.000000000","uploader":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"ref":"refs/changes/53/674553/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/53/674553/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/53/674553/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/53/674553/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/53/674553/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/53/674553/9"}}},"commit":{"parents":[{"commit":"8e4de1b75ce67a253d5c872d3de18128cc17f2a9","subject":"Merge \"Enable service token in nova and cinder\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/8e4de1b75ce67a253d5c872d3de18128cc17f2a9"}]}],"author":{"name":"fpantano","email":"fpantano@redhat.com","date":"2019-08-05 08:45:49.000000000","tz":120},"committer":{"name":"fpantano","email":"fpantano@redhat.com","date":"2019-08-07 19:49:09.000000000","tz":120},"subject":"Ceph grafana service certificate request","message":"Ceph grafana service certificate request\n\nThis change adds the certmonger ceph_grafana class to\ndefine the necessary requirements to make the cert request\nfor the ceph dashboard service.\n\nChange-Id: Ib5f780349c12eebbf514a75293d8f4644dc5731f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/dcf2263522440580508031c600ca1a15bb49ac2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/dcf2263522440580508031c600ca1a15bb49ac2f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
