)]}'
{"id":"openstack%2Fpuppet-tripleo~674604","triplet_id":"openstack%2Fpuppet-tripleo~master~I1296f3ee6b785acd00ab470311b02da498be6131","project":"openstack/puppet-tripleo","branch":"master","topic":"bug/1838972","hashtags":[],"change_id":"I1296f3ee6b785acd00ab470311b02da498be6131","subject":"Use MPM prefork for Centos 7/Fedora CI Train jobs","status":"ABANDONED","created":"2019-08-05 13:15:36.000000000","updated":"2019-10-21 09:20:00.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5aed12d65ca8c8c46132c1d6fb30da9f8ac64d56","_number":674604,"virtual_id_number":674604,"owner":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":-1,"date":"2019-08-26 10:15:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-26 10:19:47.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":-2,"date":"2019-09-02 07:27:10.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-05 13:24:24.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2019-08-05 13:32:46.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-08-26 10:15:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-26 10:19:47.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"87677cd7fbc7fce2ab455b47e18fbd5c011e1718","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2019-08-05 13:15:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"14835b80036bd025c2ab4336ec4b2dc5a6f326bf","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2019-08-05 13:22:46.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5c1eb2fa2a85aacae075f6d9e40ab91bfb5f7d39","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2019-08-05 13:25:31.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"3091eca1acdb5a44a06989382e5b51d77fa76cdb","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2019-08-05 13:32:48.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7dcd16c31d062e808238b692ce79393205c6d4d2","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-05 14:41:47.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/04/674604/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/5853276/ : FAILURE in 59m 01s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/04/674604/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/1427ac6/ : FAILURE in 57m 50s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/04/674604/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/1384990/ : FAILURE in 1h 10m 48s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2207f7479bb7e16b291ff822064fd721ce89e99c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 15:16:05.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://logs.opendev.org/04/674604/3/check/puppet-openstack-lint-ubuntu-bionic/5a2fdf8/ : FAILURE in 6m 52s\n- puppet-openstack-syntax-5-ubuntu-bionic https://logs.opendev.org/04/674604/3/check/puppet-openstack-syntax-5-ubuntu-bionic/a826eae/ : SUCCESS in 7m 00s\n- puppet-openstack-syntax-6-ubuntu-bionic https://logs.opendev.org/04/674604/3/check/puppet-openstack-syntax-6-ubuntu-bionic/46639b9/ : SUCCESS in 5m 57s\n- puppet-openstack-unit-5.5-centos-7 https://logs.opendev.org/04/674604/3/check/puppet-openstack-unit-5.5-centos-7/d61d753/ : FAILURE in 31m 59s\n- puppet-openstack-unit-6.0-ubuntu-bionic https://logs.opendev.org/04/674604/3/check/puppet-openstack-unit-6.0-ubuntu-bionic/902f6df/ : FAILURE in 20m 00s\n- puppet-openstack-unit-latest-ubuntu-bionic https://logs.opendev.org/04/674604/3/check/puppet-openstack-unit-latest-ubuntu-bionic/fec0d56/ : FAILURE in 18m 43s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/04/674604/3/check/tripleo-ci-centos-7-undercloud-containers/ffb51a5/ : FAILURE in 1h 22m 00s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://logs.opendev.org/04/674604/3/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/9d17b10/ : FAILURE in 55m 27s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/04/674604/3/check/tripleo-ci-centos-7-containers-multinode/02cfc6d/ : FAILURE in 1h 01m 26s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/04/674604/3/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/9984ed9/ : FAILURE in 1h 02m 05s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/04/674604/3/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/f800680/ : FAILURE in 1h 19m 25s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/04/674604/3/check/tripleo-ci-centos-7-standalone/b34bea8/ : FAILURE in 55m 16s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/04/674604/3/check/tripleo-ci-fedora-28-standalone/c54e458/ : SUCCESS in 1h 50m 02s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://logs.opendev.org/04/674604/3/check/tripleo-ci-centos-7-standalone-upgrade/410466b/ : FAILURE in 43m 51s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"022fec33a2ba6c0a153f153a591ea86e5d84384a","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2019-08-06 09:17:26.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"b6d6ff4bc3709d55ae8f3aef2960d3962e2e21a4","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2019-08-06 09:18:30.000000000","message":"Patch Set 3:\n\nneeds https://review.opendev.org/#/c/674580/","accounts_in_message":[],"_revision_number":3},{"id":"ee159ce1cd1a21d1cfa2c8cd226b2965d8d8dc5a","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2019-08-06 09:19:14.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"6d17f128e28d37ccecea685cc5d558cd25293eac","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 09:19:24.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n\nWarning:\n  Dependency cycle detected","accounts_in_message":[],"_revision_number":4},{"id":"c939e0284c4b35892849cc5e1a8c041c937efe4b","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2019-08-06 13:33:24.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"790b7c6120a694aba6200b7f66aee0f1a41f5b37","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 16:08:04.000000000","message":"Patch Set 5: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":5},{"id":"1df03503c12d1e7298d7d272f26ac6ebccfbf7d1","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-06 16:08:05.000000000","message":"Patch Set 5: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":5},{"id":"be77898bba4d1753ce6daff2465fe55d00a21e08","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2019-08-06 16:08:50.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"380a7491b67cfa8ecd8868151c0feecf304adc74","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-06 17:31:53.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/04/674604/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/4141c6f/ : FAILURE in 1h 03m 55s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/04/674604/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/82e5e6f/ : FAILURE in 59m 24s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/04/674604/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/5502af0/ : FAILURE in 1h 14m 21s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"72058e6eac3a5f66006142ba4d4cd01a07071003","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 23:45:52.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://logs.opendev.org/04/674604/5/check/puppet-openstack-lint-ubuntu-bionic/8e4ba0f/ : FAILURE in 5m 10s\n- puppet-openstack-syntax-5-ubuntu-bionic https://logs.opendev.org/04/674604/5/check/puppet-openstack-syntax-5-ubuntu-bionic/0166ac9/ : SUCCESS in 5m 19s\n- puppet-openstack-syntax-6-ubuntu-bionic https://logs.opendev.org/04/674604/5/check/puppet-openstack-syntax-6-ubuntu-bionic/ad47e74/ : SUCCESS in 5m 07s\n- puppet-openstack-unit-5.5-centos-7 https://logs.opendev.org/04/674604/5/check/puppet-openstack-unit-5.5-centos-7/a2559d9/ : FAILURE in 33m 39s\n- puppet-openstack-unit-6.0-ubuntu-bionic https://logs.opendev.org/04/674604/5/check/puppet-openstack-unit-6.0-ubuntu-bionic/8cb990c/ : FAILURE in 19m 12s\n- puppet-openstack-unit-latest-ubuntu-bionic https://logs.opendev.org/04/674604/5/check/puppet-openstack-unit-latest-ubuntu-bionic/feac025/ : FAILURE in 21m 00s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/04/674604/5/check/tripleo-ci-centos-7-undercloud-containers/6a4951b/ : FAILURE in 1h 14m 30s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://logs.opendev.org/04/674604/5/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/1613278/ : TIMED_OUT in 3h 21m 32s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/04/674604/5/check/tripleo-ci-centos-7-containers-multinode/a189cac/ : FAILURE in 1h 04m 17s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/04/674604/5/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/a8831a0/ : FAILURE in 1h 07m 42s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/04/674604/5/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/5d575d5/ : FAILURE in 1h 09m 16s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/04/674604/5/check/tripleo-ci-centos-7-standalone/5757989/ : FAILURE in 59m 03s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/04/674604/5/check/tripleo-ci-fedora-28-standalone/168820f/ : FAILURE in 1h 04m 21s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://logs.opendev.org/04/674604/5/check/tripleo-ci-centos-7-standalone-upgrade/13b5ef4/ : FAILURE in 2h 07m 04s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"89e8dcdd0f46742cca4e201f6494d7c7c156c08a","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2019-08-09 08:52:10.000000000","message":"Patch Set 5: Workflow-1","accounts_in_message":[],"_revision_number":5},{"id":"8556db53e8379507e5f9f58527d1852449e4d461","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2019-08-09 13:07:13.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"26b6b1feff045be982605901cfda571037609c27","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-09 14:16:03.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/04/674604/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/87f1770/ : FAILURE in 50m 56s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/04/674604/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/14212d3/ : FAILURE in 48m 35s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/04/674604/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/2e23c19/ : FAILURE in 1h 08m 03s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"b497751e7c062e1704951f6c8f3a53209164e43e","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2019-08-09 16:04:16.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":6},{"id":"9baa0dcee40c82370f34f72385e6239b871de4cc","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2019-08-12 07:42:20.000000000","message":"Restored","accounts_in_message":[],"_revision_number":6},{"id":"c05569e6df5b89356bc528062fb0074e64808a6c","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2019-08-12 07:42:27.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"baae13f37c1507fa71354673f838faaa74a9e2c3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 10:30:39.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://logs.opendev.org/04/674604/7/check/puppet-openstack-lint-ubuntu-bionic/caaae84/ : SUCCESS in 4m 26s\n- puppet-openstack-syntax-5-ubuntu-bionic https://logs.opendev.org/04/674604/7/check/puppet-openstack-syntax-5-ubuntu-bionic/363f110/ : SUCCESS in 4m 29s\n- puppet-openstack-syntax-6-ubuntu-bionic https://logs.opendev.org/04/674604/7/check/puppet-openstack-syntax-6-ubuntu-bionic/a8846fb/ : SUCCESS in 4m 01s\n- puppet-openstack-unit-5.5-centos-7 https://logs.opendev.org/04/674604/7/check/puppet-openstack-unit-5.5-centos-7/0cf0e1e/ : FAILURE in 28m 18s\n- puppet-openstack-unit-6.0-ubuntu-bionic https://logs.opendev.org/04/674604/7/check/puppet-openstack-unit-6.0-ubuntu-bionic/ea03677/ : FAILURE in 17m 26s\n- puppet-openstack-unit-latest-ubuntu-bionic https://logs.opendev.org/04/674604/7/check/puppet-openstack-unit-latest-ubuntu-bionic/66bee99/ : FAILURE in 18m 37s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/04/674604/7/check/tripleo-ci-centos-7-undercloud-containers/27ff301/ : SUCCESS in 1h 39m 38s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://logs.opendev.org/04/674604/7/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/13872ff/ : FAILURE in 1h 57m 05s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/04/674604/7/check/tripleo-ci-centos-7-containers-multinode/67c7d5e/ : SUCCESS in 2h 20m 58s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/04/674604/7/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/1d4a8fd/ : SUCCESS in 2h 12m 51s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/04/674604/7/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/d7c58d1/ : SUCCESS in 2h 27m 39s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/04/674604/7/check/tripleo-ci-centos-7-standalone/716dbd7/ : SUCCESS in 1h 49m 55s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/04/674604/7/check/tripleo-ci-fedora-28-standalone/9d316fe/ : SUCCESS in 1h 50m 51s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://logs.opendev.org/04/674604/7/check/tripleo-ci-centos-7-standalone-upgrade/ec7d6c4/ : SUCCESS in 2h 12m 19s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"d625c5c127add7a6a79fb740b71685ae8f24afae","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-12 11:36:24.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 finger://ze03.softwarefactory-project.io:7979/704683bf8aba43e8828d3a4c7d9a82d2 : POST_FAILURE in 3h 32m 10s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 finger://ze04.softwarefactory-project.io:7979/d56cd36871874df59d5dbf2bbb818fa6 : POST_FAILURE in 3h 30m 55s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 finger://ze04.softwarefactory-project.io:7979/0eb1528514264ce18ed382efc0b9d86b : POST_FAILURE in 3h 52m 17s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"9ea5438f05a9a12f83a83ccb6e10c6f5835dc838","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2019-08-12 15:08:14.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"2720a481a32818a510194eaf52013730fd23aa15","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-12 19:31:18.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 finger://ze04.softwarefactory-project.io:7979/5ba92f6c00ba44b884fc2c340f99670e : POST_FAILURE in 3h 31m 40s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 finger://ze04.softwarefactory-project.io:7979/b5c45827c39d47c9b25302b597b87f54 : POST_FAILURE in 3h 48m 10s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/04/674604/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/1af370d/ : SUCCESS in 3h 49m 45s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"556b0e0a7fb3c5242162d868c5118b40ae00a3cb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 22:04:26.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/30e21d4141074d00a5efde2eaf654115 : SUCCESS in 5m 52s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7e71309333244230b8862ed0daaf222c : SUCCESS in 4m 08s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6d51a930b65d43a5a454e3c4a61c8b1f : SUCCESS in 3m 39s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/c274461e9b5f4ffb81ad62fddddb0099 : FAILURE in 30m 31s\n- puppet-openstack-unit-6.0-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5e82b6d31f7e40e7bf723f5f0db3ad0f : FAILURE in 18m 18s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a563ca7c37db40429f10f5d81e60b42a : FAILURE in 18m 49s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/059045d4c58340b28944d9ff53e8d918 : SUCCESS in 1h 35m 11s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/4942d3e73f7d47fc8908bfe94afcb3f0 : FAILURE in 1h 48m 15s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/36f2086b286d4424a12b40f2928d04ec : SUCCESS in 2h 20m 09s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/4f63e79fdb904523a4179a895343d685 : SUCCESS in 2h 30m 38s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/1ae763ff4fcc47a0a4baea054af355a4 : SUCCESS in 2h 21m 03s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/dac0dda6f7114eb88ec2620d5dbb9c00 : SUCCESS in 1h 51m 33s\n- tripleo-ci-fedora-28-standalone https://zuul.opendev.org/t/openstack/build/024a9cbb42484ec98b99929efe818584 : SUCCESS in 1h 53m 07s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/7bfa8f0f33f5467587584461c5ee95d9 : SUCCESS in 2h 22m 18s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"88e2dc8049a983aea667e214cd8a5eceedbe4202","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2019-08-13 13:05:57.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"8cafab0ccd28051077082f74fe76ac09aa53d543","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2019-08-13 13:13:42.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"52d9f00bed6cff296b9b7f39750fee599e5c5707","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-13 16:23:23.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/04/674604/8/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/6dcc3ca/ : SUCCESS in 2h 45m 31s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/04/674604/8/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/9a26160/ : SUCCESS in 2h 53m 03s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/04/674604/8/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/379c34b/ : SUCCESS in 3h 06m 21s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"a7dce5824c0fdf058a06491f8112949178385db1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-13 16:41:35.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dd567f2bb17a4ec89277bf44e7480558 : SUCCESS in 4m 04s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b4955ae4757e4286be0ded03c5657fd5 : SUCCESS in 4m 00s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3e5762e6ecdb4e3db67b6eea5d786219 : SUCCESS in 3m 48s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/1e81f256406f4bdd886d73e9fd862aec : FAILURE in 31m 34s\n- puppet-openstack-unit-6.0-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/395913c062ea43dea811ad2b793a6e86 : FAILURE in 16m 46s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9dc0f6d339774279bdecc269db596f02 : FAILURE in 18m 18s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/fc8c3de1585c4a72b2f20d1ee345c77d : SUCCESS in 1h 35m 06s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/208e2e436f054a5e90710d9e05bd170b : FAILURE in 1h 57m 00s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/10b586263d8d4ddcabb02622fa3703dc : SUCCESS in 2h 39m 47s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/4f09567b7d344fc6b08e276711a83b55 : SUCCESS in 2h 38m 43s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/93267c0ff3114e75bb6823ceac813173 : FAILURE in 1h 03m 18s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/911678f0eaf247cd87f9ce2e9558094e : SUCCESS in 1h 46m 32s\n- tripleo-ci-fedora-28-standalone https://zuul.opendev.org/t/openstack/build/7e639af1666c494390849cfc28b276a3 : FAILURE in 1h 05m 26s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/09d964f7cd994ebe895d1c05f17c3db3 : SUCCESS in 2h 14m 13s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"6bdbba67e74a212e85edb34eb17f2922747f97ea","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2019-08-14 14:48:44.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"0350386d0b57210da1ed6976e84fd031598bfb02","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-14 17:54:04.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b55344ad28404dc784a03abcd0db0982 : SUCCESS in 3m 56s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b37dea88869f431085f9d62d6e9d30d3 : SUCCESS in 3m 49s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/29edaaa7612a4579be797a2f477e0685 : SUCCESS in 4m 00s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/c8d38b3253224a80a3cda6dc9ca8db08 : FAILURE in 28m 49s\n- puppet-openstack-unit-6.0-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/174c9d8af3a64841b6a6e77b17b889db : FAILURE in 17m 15s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/356fabd15d174cbe88b15a2a301ea614 : FAILURE in 21m 16s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/12dc387593ac4dcbac72b87003a7a85f : SUCCESS in 1h 38m 59s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/33a81d7c29ee46b1832109bfde958913 : FAILURE in 2h 12m 53s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/82db26c963534d4aa0a9222620bfdd6c : SUCCESS in 2h 29m 31s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/496e3a510e55430aa32acbf874188f27 : SUCCESS in 2h 22m 42s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/70f0a2dddb6b49049d8d709e2743e64a : SUCCESS in 3h 03m 35s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/669deefc84df4fa999260ccbc7b7c0fd : SUCCESS in 1h 58m 08s\n- tripleo-ci-fedora-28-standalone https://zuul.opendev.org/t/openstack/build/d45ba2ce273246418a273c01c4e8b492 : SUCCESS in 1h 47m 58s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/c85bd858d86a4a0a9ca9afd009a895e5 : SUCCESS in 2h 02m 06s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"f787cc84ce98d7d66157ee79932a3fca0916fbd8","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-14 17:54:55.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/04/674604/9/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/4f002ae/ : SUCCESS in 2h 35m 24s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/04/674604/9/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/b967176/ : SUCCESS in 2h 43m 02s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/04/674604/9/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/d2b2b66/ : SUCCESS in 3h 03m 57s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"0d99f450323661347a54bdc93551df675320f5fc","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2019-08-26 07:22:30.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"26501473208b620db0bcd5513c6461e4221fa244","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 10:15:03.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3f33d6f041674be98066f51cfd919008 : SUCCESS in 4m 46s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ed862a31581f419497ed8cd88d864808 : SUCCESS in 4m 16s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5eb94f588148476da27f1d5bd889153c : SUCCESS in 4m 17s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/a9e075e01aae4c4eacd7ca5bad51919a : FAILURE in 32m 09s\n- puppet-openstack-unit-6.0-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/36384895ad6344c6842571532ea7d2a6 : FAILURE in 18m 58s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cf700e95faf74e6e9cc156111d7136ee : FAILURE in 21m 32s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/c46fb40e7272406a95fad738392babbb : SUCCESS in 1h 31m 59s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/d1bb99a258904e97bea481a593b0db4e : FAILURE in 1h 47m 38s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/bf6a4c711f454e838776b8412f72c22b : SUCCESS in 2h 38m 49s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/f4111dd18e6049bbaafbbba6bc1d5742 : SUCCESS in 2h 24m 34s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/198e1c7c4f7947f1a86c3434e7384026 : FAILURE in 2h 22m 56s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/7a576400bf91485fb5766a189a1566af : SUCCESS in 2h 02m 25s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/996bd33df89a48c7803350de61117431 : SUCCESS in 2h 05m 56s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"4c85d7df8131370840990b7941fcb0df08d1d8f6","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-26 10:19:47.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/04/674604/10/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/89a3ea4/ : SUCCESS in 2h 41m 19s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/04/674604/10/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/861ede2/ : SUCCESS in 2h 48m 45s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/04/674604/10/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/0d29073/ : SUCCESS in 2h 56m 27s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"04c7ef30db43095a6680d99c80c5b56c3e1baac8","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2019-09-02 07:27:10.000000000","message":"Patch Set 10: Code-Review-2\n\nlet\u0027s wait for Ussury, where we\u0027d need no this change hopefully","accounts_in_message":[],"_revision_number":10},{"id":"d9371f4f58a3b5c8761a1f0e872845938c1ff0c2","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2019-10-21 09:20:00.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"da9048b2a967690b9bda88cf1a873777bfd11083","revisions":{"8fb1b24cdd2d0618478959dae366e1a81b75d4dc":{"kind":"REWORK","_number":1,"created":"2019-08-05 13:15:36.000000000","uploader":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"ref":"refs/changes/04/674604/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/04/674604/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/04/674604/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/04/674604/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/04/674604/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/04/674604/1"}}},"commit":{"parents":[{"commit":"8e4de1b75ce67a253d5c872d3de18128cc17f2a9","subject":"Merge \"Enable service token in nova and cinder\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/8e4de1b75ce67a253d5c872d3de18128cc17f2a9"}]}],"author":{"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","date":"2019-08-05 13:14:32.000000000","tz":120},"committer":{"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","date":"2019-08-05 13:14:32.000000000","tz":120},"subject":"Allow Centos 7 \u0026 Fedora CI passing for Apache base","message":"Allow Centos 7 \u0026 Fedora CI passing for Apache base\n\nThere is no /etc/httpd/conf.modules.d path for RHEL 7 OS family.\nAs of RHEL 8, apache module configs reside in /etc/httpd/conf.d.\n\nWhile the latter should be only used as of Train, we have to keep a w/a\nfor the missing former path, unless we get rid of Centos 7 CI/promo\njobs for Train.\n\nRelated-bug: #1838972\n\nChange-Id: I1296f3ee6b785acd00ab470311b02da498be6131\nSigned-off-by: Bogdan Dobrelya \u003cbdobreli@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/8fb1b24cdd2d0618478959dae366e1a81b75d4dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/8fb1b24cdd2d0618478959dae366e1a81b75d4dc"}]},"branch":"refs/heads/master"},"14bc846b1650b32029e2aadd76c1e08c9e77517d":{"kind":"REWORK","_number":2,"created":"2019-08-05 13:22:46.000000000","uploader":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"ref":"refs/changes/04/674604/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/04/674604/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/04/674604/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/04/674604/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/04/674604/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/04/674604/2"}}},"commit":{"parents":[{"commit":"8e4de1b75ce67a253d5c872d3de18128cc17f2a9","subject":"Merge \"Enable service token in nova and cinder\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/8e4de1b75ce67a253d5c872d3de18128cc17f2a9"}]}],"author":{"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","date":"2019-08-05 13:14:32.000000000","tz":120},"committer":{"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","date":"2019-08-05 13:22:23.000000000","tz":120},"subject":"Allow Centos 7 \u0026 Fedora CI passing for Apache base","message":"Allow Centos 7 \u0026 Fedora CI passing for Apache base\n\nThere is no /etc/httpd/conf.modules.d path for RHEL 7 OS family.\nAs of RHEL 8, apache module configs reside in /etc/httpd/conf.d.\n\nWhile the latter should be only used as of Train, we have to keep a w/a\nfor the missing former path, unless we get rid of Centos 7 CI/promo\njobs for Train.\n\nRevert this patch after no more Centos 7/Fedora CI jobs left for Train\n\nRelated-bug: #1838972\n\nChange-Id: I1296f3ee6b785acd00ab470311b02da498be6131\nSigned-off-by: Bogdan Dobrelya \u003cbdobreli@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/14bc846b1650b32029e2aadd76c1e08c9e77517d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/14bc846b1650b32029e2aadd76c1e08c9e77517d"}]},"branch":"refs/heads/master"},"5edce3226d0258680adc0b557f84f54568d69763":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-08-05 13:25:31.000000000","uploader":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"ref":"refs/changes/04/674604/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/04/674604/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/04/674604/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/04/674604/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/04/674604/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/04/674604/3"}}},"commit":{"parents":[{"commit":"8e4de1b75ce67a253d5c872d3de18128cc17f2a9","subject":"Merge \"Enable service token in nova and cinder\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/8e4de1b75ce67a253d5c872d3de18128cc17f2a9"}]}],"author":{"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","date":"2019-08-05 13:14:32.000000000","tz":120},"committer":{"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","date":"2019-08-05 13:25:10.000000000","tz":120},"subject":"Allow Centos 7 \u0026 Fedora CI passing for Apache base","message":"Allow Centos 7 \u0026 Fedora CI passing for Apache base\n\nThere is no /etc/httpd/conf.modules.d path for RHEL \u003e7 OS family.\nAs of RHEL 8, apache module configs reside in /etc/httpd/conf.d.\n\nWhile the latter should be only used as of Train, we have to keep a w/a\nfor the missing former path, unless we get rid of Centos 7 CI/promo\njobs for Train.\n\nRevert this patch after no more Centos 7/Fedora CI jobs left for Train\n\nRelated-bug: #1838972\n\nChange-Id: I1296f3ee6b785acd00ab470311b02da498be6131\nSigned-off-by: Bogdan Dobrelya \u003cbdobreli@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/5edce3226d0258680adc0b557f84f54568d69763"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/5edce3226d0258680adc0b557f84f54568d69763"}]},"branch":"refs/heads/master"},"f669261c3fda5a83a36bafd66df0e8babc87f56f":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2019-08-06 09:19:14.000000000","uploader":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"ref":"refs/changes/04/674604/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/04/674604/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/04/674604/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/04/674604/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/04/674604/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/04/674604/4"}}},"commit":{"parents":[{"commit":"8e4de1b75ce67a253d5c872d3de18128cc17f2a9","subject":"Merge \"Enable service token in nova and cinder\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/8e4de1b75ce67a253d5c872d3de18128cc17f2a9"}]}],"author":{"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","date":"2019-08-05 13:14:32.000000000","tz":120},"committer":{"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","date":"2019-08-06 09:18:48.000000000","tz":120},"subject":"Allow Centos 7 \u0026 Fedora CI passing for Apache base","message":"Allow Centos 7 \u0026 Fedora CI passing for Apache base\n\nThere is no /etc/httpd/conf.modules.d path for RHEL \u003e7 OS family.\nAs of RHEL 8, apache module configs reside in /etc/httpd/conf.d.\n\nWhile the latter should be only used as of Train, we have to keep a w/a\nfor the missing former path, unless we get rid of Centos 7 CI/promo\njobs for Train.\n\nRevert this patch after no more Centos 7/Fedora CI jobs left for Train\n\nRelated-bug: #1838972\nDepends-On: https://review.opendev.org/674604\n\nChange-Id: I1296f3ee6b785acd00ab470311b02da498be6131\nSigned-off-by: Bogdan Dobrelya \u003cbdobreli@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/f669261c3fda5a83a36bafd66df0e8babc87f56f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/f669261c3fda5a83a36bafd66df0e8babc87f56f"}]},"branch":"refs/heads/master"},"99d93fccd7cc472962ca1f70acaa32cc09fec07a":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2019-08-06 13:33:24.000000000","uploader":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"ref":"refs/changes/04/674604/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/04/674604/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/04/674604/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/04/674604/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/04/674604/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/04/674604/5"}}},"commit":{"parents":[{"commit":"8e4de1b75ce67a253d5c872d3de18128cc17f2a9","subject":"Merge \"Enable service token in nova and cinder\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/8e4de1b75ce67a253d5c872d3de18128cc17f2a9"}]}],"author":{"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","date":"2019-08-05 13:14:32.000000000","tz":120},"committer":{"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","date":"2019-08-06 13:33:05.000000000","tz":120},"subject":"Allow Centos 7 \u0026 Fedora CI passing for Apache base","message":"Allow Centos 7 \u0026 Fedora CI passing for Apache base\n\nThere is no /etc/httpd/conf.modules.d path for RHEL \u003e7 OS family.\nAs of RHEL 8, apache module configs reside in /etc/httpd/conf.d.\n\nWhile the latter should be only used as of Train, we have to keep a w/a\nfor the missing former path, unless we get rid of Centos 7 CI/promo\njobs for Train.\n\nRevert this patch after no more Centos 7/Fedora CI jobs left for Train\n\nRelated-bug: #1838972\nDepends-On: https://review.opendev.org/674580\n\nChange-Id: I1296f3ee6b785acd00ab470311b02da498be6131\nSigned-off-by: Bogdan Dobrelya \u003cbdobreli@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/99d93fccd7cc472962ca1f70acaa32cc09fec07a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/99d93fccd7cc472962ca1f70acaa32cc09fec07a"}]},"branch":"refs/heads/master"},"5076434107afbc62e4f5132c58b14dc1b13c5aad":{"kind":"REWORK","_number":6,"created":"2019-08-09 13:07:13.000000000","uploader":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"ref":"refs/changes/04/674604/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/04/674604/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/04/674604/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/04/674604/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/04/674604/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/04/674604/6"}}},"commit":{"parents":[{"commit":"8e4de1b75ce67a253d5c872d3de18128cc17f2a9","subject":"Merge \"Enable service token in nova and cinder\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/8e4de1b75ce67a253d5c872d3de18128cc17f2a9"}]}],"author":{"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","date":"2019-08-05 13:14:32.000000000","tz":120},"committer":{"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","date":"2019-08-09 13:06:59.000000000","tz":120},"subject":"Allow Centos 7 \u0026 Fedora CI passing for Apache base","message":"Allow Centos 7 \u0026 Fedora CI passing for Apache base\n\nThere is no /etc/httpd/conf.modules.d path for RHEL \u003e7 OS family.\nAs of RHEL 8, apache module configs reside in /etc/httpd/conf.d.\n\nWhile the latter should be only used as of Train, we have to keep a w/a\nfor the missing former path, unless we get rid of Centos 7 CI/promo\njobs for Train.\n\nRevert this patch after no more Centos 7/Fedora CI jobs left for Train\n\nRelated-bug: #1838972\n\nChange-Id: I1296f3ee6b785acd00ab470311b02da498be6131\nSigned-off-by: Bogdan Dobrelya \u003cbdobreli@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/5076434107afbc62e4f5132c58b14dc1b13c5aad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/5076434107afbc62e4f5132c58b14dc1b13c5aad"}]},"branch":"refs/heads/master"},"0ba1df2303d79ae38c089c1c57734383b5deffe8":{"kind":"REWORK","_number":7,"created":"2019-08-12 07:42:27.000000000","uploader":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"ref":"refs/changes/04/674604/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/04/674604/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/04/674604/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/04/674604/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/04/674604/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/04/674604/7"}}},"commit":{"parents":[{"commit":"8e4de1b75ce67a253d5c872d3de18128cc17f2a9","subject":"Merge \"Enable service token in nova and cinder\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/8e4de1b75ce67a253d5c872d3de18128cc17f2a9"}]}],"author":{"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","date":"2019-08-05 13:14:32.000000000","tz":120},"committer":{"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","date":"2019-08-12 07:32:11.000000000","tz":120},"subject":"Use MPM prefork for Centos 7/Fedora CI Train jobs","message":"Use MPM prefork for Centos 7/Fedora CI Train jobs\n\nFor Centos 7, apache RPM hardcodes prefork MPM in\n/etc/httpd/conf.modules.d/00-mpm.conf. That cannot be properly\nconfigured with container puppet tasks in t-h-t because as of RHEL 8 OS\nfamily, apache module configs reside in /etc/httpd/conf.d and there is\nno more /etc/httpd/conf.modules.d.\n\nTo w/a that for Centos 7 CI in Train, always use MPM prefork, so there\nis no more multiple MPM modules configuration conflicts.\n\nThere is no other/clean way for Kolla startup scripts to pick those OS\nversion dependent paths for kolla configs in t-h-t.\n\nRevert this patch after no more Centos 7/Fedora CI jobs left for Train.\n\nRelated-bug: #1838972\n\nChange-Id: I1296f3ee6b785acd00ab470311b02da498be6131\nSigned-off-by: Bogdan Dobrelya \u003cbdobreli@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/0ba1df2303d79ae38c089c1c57734383b5deffe8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/0ba1df2303d79ae38c089c1c57734383b5deffe8"}]},"branch":"refs/heads/master"},"684202c5288bdaaba23af645019c6f076d3e4f89":{"kind":"REWORK","_number":8,"created":"2019-08-13 13:13:42.000000000","uploader":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"ref":"refs/changes/04/674604/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/04/674604/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/04/674604/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/04/674604/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/04/674604/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/04/674604/8"}}},"commit":{"parents":[{"commit":"8e4de1b75ce67a253d5c872d3de18128cc17f2a9","subject":"Merge \"Enable service token in nova and cinder\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/8e4de1b75ce67a253d5c872d3de18128cc17f2a9"}]}],"author":{"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","date":"2019-08-05 13:14:32.000000000","tz":120},"committer":{"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","date":"2019-08-13 13:13:36.000000000","tz":120},"subject":"Use MPM prefork for Centos 7/Fedora CI Train jobs","message":"Use MPM prefork for Centos 7/Fedora CI Train jobs\n\nFor Centos 7, apache RPM hardcodes prefork MPM in\n/etc/httpd/conf.modules.d/00-mpm.conf. That cannot be properly\nconfigured with container puppet tasks in t-h-t because as of RHEL 8 OS\nfamily, apache module configs reside in /etc/httpd/conf.d and there is\nno more /etc/httpd/conf.modules.d.\n\nTo w/a that for Centos 7 CI in Train, always use MPM prefork, so there\nis no more multiple MPM modules configuration conflicts.\n\nThere is no other/clean way for Kolla startup scripts to pick those OS\nversion dependent paths for kolla configs in t-h-t.\n\nRevert this patch after no more Centos 7/Fedora CI jobs left for Train.\n\nRelated-bug: #1838972\n\nChange-Id: I1296f3ee6b785acd00ab470311b02da498be6131\nSigned-off-by: Bogdan Dobrelya \u003cbdobreli@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/684202c5288bdaaba23af645019c6f076d3e4f89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/684202c5288bdaaba23af645019c6f076d3e4f89"}]},"branch":"refs/heads/master"},"85599e6e1b039725c4e4ee6d23f3bafe09b1f11f":{"kind":"REWORK","_number":9,"created":"2019-08-14 14:48:44.000000000","uploader":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"ref":"refs/changes/04/674604/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/04/674604/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/04/674604/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/04/674604/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/04/674604/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/04/674604/9"}}},"commit":{"parents":[{"commit":"8e4de1b75ce67a253d5c872d3de18128cc17f2a9","subject":"Merge \"Enable service token in nova and cinder\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/8e4de1b75ce67a253d5c872d3de18128cc17f2a9"}]}],"author":{"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","date":"2019-08-05 13:14:32.000000000","tz":120},"committer":{"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","date":"2019-08-14 14:48:35.000000000","tz":120},"subject":"Use MPM prefork for Centos 7/Fedora CI Train jobs","message":"Use MPM prefork for Centos 7/Fedora CI Train jobs\n\nFor Centos 7, apache RPM hardcodes prefork MPM in\n/etc/httpd/conf.modules.d/00-mpm.conf. That cannot be properly\nconfigured with container puppet tasks in t-h-t because as of RHEL 8 OS\nfamily, apache module configs reside in /etc/httpd/conf.d and there is\nno more /etc/httpd/conf.modules.d.\n\nTo w/a that for Centos 7 CI in Train, always use MPM prefork, so there\nis no more multiple MPM modules configuration conflicts.\n\nThere is no other/clean way for Kolla startup scripts to pick those OS\nversion dependent paths for kolla configs in t-h-t.\n\nRevert this patch after no more Centos 7/Fedora CI jobs left for Train.\n\nRelated-bug: #1838972\n\nChange-Id: I1296f3ee6b785acd00ab470311b02da498be6131\nSigned-off-by: Bogdan Dobrelya \u003cbdobreli@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/85599e6e1b039725c4e4ee6d23f3bafe09b1f11f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/85599e6e1b039725c4e4ee6d23f3bafe09b1f11f"}]},"branch":"refs/heads/master"},"da9048b2a967690b9bda88cf1a873777bfd11083":{"kind":"REWORK","_number":10,"created":"2019-08-26 07:22:30.000000000","uploader":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"ref":"refs/changes/04/674604/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/04/674604/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/04/674604/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/04/674604/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/04/674604/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/04/674604/10"}}},"commit":{"parents":[{"commit":"8e4de1b75ce67a253d5c872d3de18128cc17f2a9","subject":"Merge \"Enable service token in nova and cinder\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/8e4de1b75ce67a253d5c872d3de18128cc17f2a9"}]}],"author":{"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","date":"2019-08-05 13:14:32.000000000","tz":120},"committer":{"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","date":"2019-08-26 07:22:21.000000000","tz":120},"subject":"Use MPM prefork for Centos 7/Fedora CI Train jobs","message":"Use MPM prefork for Centos 7/Fedora CI Train jobs\n\nFor Centos 7, apache RPM hardcodes prefork MPM in\n/etc/httpd/conf.modules.d/00-mpm.conf. That cannot be properly\nconfigured with container puppet tasks in t-h-t because as of RHEL 8 OS\nfamily, apache module configs reside in /etc/httpd/conf.d and there is\nno more /etc/httpd/conf.modules.d.\n\nTo w/a that for Centos 7 CI in Train, always use MPM prefork, so there\nis no more multiple MPM modules configuration conflicts.\n\nThere is no other/clean way for Kolla startup scripts to pick those OS\nversion dependent paths for kolla configs in t-h-t.\n\nRevert this patch after no more Centos 7/Fedora CI jobs left for Train.\n\nRelated-bug: #1838972\n\nChange-Id: I1296f3ee6b785acd00ab470311b02da498be6131\nSigned-off-by: Bogdan Dobrelya \u003cbdobreli@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/da9048b2a967690b9bda88cf1a873777bfd11083"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/da9048b2a967690b9bda88cf1a873777bfd11083"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
