)]}'
{"id":"openstack%2Fpuppet-tripleo~674757","triplet_id":"openstack%2Fpuppet-tripleo~master~Ibd817a4a34ee2944ee86b4a47e7621637d258da3","project":"openstack/puppet-tripleo","branch":"master","hashtags":[],"change_id":"Ibd817a4a34ee2944ee86b4a47e7621637d258da3","subject":"Add base class for neutron Mellanox agent","status":"MERGED","created":"2019-08-06 08:54:50.000000000","updated":"2019-08-27 07:14:13.000000000","submitted":"2019-08-27 02:47:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"674757-1566874051242-120218c1","meta_rev_id":"1739b0b9414835d50a7508b63f0a35e892d0d00b","_number":674757,"virtual_id_number":674757,"owner":{"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"date":"2019-08-22 06:30:28.000000000","_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":28714,"name":"Adrian Chiris","email":"adrianc@nvidia.com","username":"adrianc"},{"value":0,"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},{"value":2,"date":"2019-08-27 02:47:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2019-08-26 22:56:29.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},{"value":2,"date":"2019-08-20 13:43:51.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":28714,"name":"Adrian Chiris","email":"adrianc@nvidia.com","username":"adrianc"},{"value":0,"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2019-08-26 13:49:27.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"date":"2019-08-27 07:14:13.000000000","post_submit":true,"_account_id":28714,"name":"Adrian Chiris","email":"adrianc@nvidia.com","username":"adrianc"},{"value":0,"date":"2019-08-06 08:55:34.000000000","_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":28714,"name":"Adrian Chiris","email":"adrianc@nvidia.com","username":"adrianc"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-06 08:55:34.000000000","updated_by":{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},"reviewer":{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},"state":"REVIEWER"},{"updated":"2019-08-20 13:43:51.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-08-26 13:49:27.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-08-26 22:56:29.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-27 02:47:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-27 07:14:13.000000000","updated_by":{"_account_id":28714,"name":"Adrian Chiris","email":"adrianc@nvidia.com","username":"adrianc"},"reviewer":{"_account_id":28714,"name":"Adrian Chiris","email":"adrianc@nvidia.com","username":"adrianc"},"state":"REVIEWER"}],"messages":[{"id":"3b148dd6684a3eb775096c6af5333fa4a495f6e6","author":{"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},"date":"2019-08-06 08:54:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"65546eff2ead4b41e4c6fb47184f9ebfe23bf8b0","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-06 12:31:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/57/674757/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/7c76dbc/ : FAILURE in 2h 54m 48s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/57/674757/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/78ca10c/ : SUCCESS in 2h 53m 46s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/57/674757/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/aeb3e1f/ : SUCCESS in 3h 35m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"36a2787d2f1ab4a613377a1718f63951d908a3f9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 13:27:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://logs.opendev.org/57/674757/1/check/puppet-openstack-lint-ubuntu-bionic/a3dfa9b/ : SUCCESS in 5m 49s\n- puppet-openstack-syntax-5-ubuntu-bionic https://logs.opendev.org/57/674757/1/check/puppet-openstack-syntax-5-ubuntu-bionic/4d23b3e/ : SUCCESS in 6m 15s\n- puppet-openstack-syntax-6-ubuntu-bionic https://logs.opendev.org/57/674757/1/check/puppet-openstack-syntax-6-ubuntu-bionic/e768840/ : SUCCESS in 5m 38s\n- puppet-openstack-unit-5.5-centos-7 https://logs.opendev.org/57/674757/1/check/puppet-openstack-unit-5.5-centos-7/1646d63/ : SUCCESS in 32m 27s\n- puppet-openstack-unit-6.0-ubuntu-bionic https://logs.opendev.org/57/674757/1/check/puppet-openstack-unit-6.0-ubuntu-bionic/f84fe42/ : SUCCESS in 18m 25s\n- puppet-openstack-unit-latest-ubuntu-bionic https://logs.opendev.org/57/674757/1/check/puppet-openstack-unit-latest-ubuntu-bionic/a16eb57/ : SUCCESS in 22m 17s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/57/674757/1/check/tripleo-ci-centos-7-undercloud-containers/314e27d/ : SUCCESS in 1h 41m 25s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://logs.opendev.org/57/674757/1/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/ea35403/ : FAILURE in 2h 07m 38s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/57/674757/1/check/tripleo-ci-centos-7-containers-multinode/2bbf0b1/ : SUCCESS in 2h 19m 54s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/57/674757/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/74adcba/ : SUCCESS in 2h 25m 27s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/57/674757/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/49a5309/ : SUCCESS in 2h 28m 04s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/57/674757/1/check/tripleo-ci-centos-7-standalone/c417f0a/ : SUCCESS in 1h 50m 03s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/57/674757/1/check/tripleo-ci-fedora-28-standalone/a708cdf/ : SUCCESS in 1h 36m 25s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://logs.opendev.org/57/674757/1/check/tripleo-ci-centos-7-standalone-upgrade/4db928f/ : SUCCESS in 2h 18m 48s (non-voting)\n- tripleo-ci-centos-7-scenario010-multinode-oooq-container https://logs.opendev.org/57/674757/1/check/tripleo-ci-centos-7-scenario010-multinode-oooq-container/0672684/ : FAILURE in 54m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ae7441a026ff83a72db712004ff1c18bf8d50c53","author":{"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},"date":"2019-08-06 14:15:38.000000000","message":"Patch Set 1:\n\nrecheck-rdo","accounts_in_message":[],"_revision_number":1},{"id":"381bb4e40a2037b3ebf4eb6ca8252843a9c06e7a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-06 17:15:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/57/674757/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/48b1314/ : SUCCESS in 2h 57m 45s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/57/674757/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/b8309d4/ : SUCCESS in 2h 52m 22s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/57/674757/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/b94920d/ : SUCCESS in 2h 57m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"72d63024dcd1ba290d53d0221219915ea861986e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 20:26:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://logs.opendev.org/57/674757/1/check/puppet-openstack-lint-ubuntu-bionic/dae827f/ : SUCCESS in 5m 26s\n- puppet-openstack-syntax-5-ubuntu-bionic https://logs.opendev.org/57/674757/1/check/puppet-openstack-syntax-5-ubuntu-bionic/fc04337/ : SUCCESS in 4m 18s\n- puppet-openstack-syntax-6-ubuntu-bionic https://logs.opendev.org/57/674757/1/check/puppet-openstack-syntax-6-ubuntu-bionic/c5efc46/ : SUCCESS in 4m 22s\n- puppet-openstack-unit-5.5-centos-7 https://logs.opendev.org/57/674757/1/check/puppet-openstack-unit-5.5-centos-7/8322742/ : SUCCESS in 34m 04s\n- puppet-openstack-unit-6.0-ubuntu-bionic https://logs.opendev.org/57/674757/1/check/puppet-openstack-unit-6.0-ubuntu-bionic/5de1109/ : SUCCESS in 17m 19s\n- puppet-openstack-unit-latest-ubuntu-bionic https://logs.opendev.org/57/674757/1/check/puppet-openstack-unit-latest-ubuntu-bionic/9420218/ : SUCCESS in 19m 05s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/57/674757/1/check/tripleo-ci-centos-7-undercloud-containers/b269ed6/ : SUCCESS in 1h 39m 36s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://logs.opendev.org/57/674757/1/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/526e462/ : FAILURE in 1h 59m 29s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/57/674757/1/check/tripleo-ci-centos-7-containers-multinode/1217b2e/ : SUCCESS in 2h 16m 59s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/57/674757/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/e24a4c9/ : SUCCESS in 2h 19m 35s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/57/674757/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/092250f/ : SUCCESS in 2h 49m 33s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/57/674757/1/check/tripleo-ci-centos-7-standalone/cbab5e2/ : SUCCESS in 2h 04m 03s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/57/674757/1/check/tripleo-ci-fedora-28-standalone/3ec7ef8/ : FAILURE in 53m 35s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://logs.opendev.org/57/674757/1/check/tripleo-ci-centos-7-standalone-upgrade/ca04ca6/ : SUCCESS in 2h 03m 51s (non-voting)\n- tripleo-ci-centos-7-scenario010-multinode-oooq-container https://logs.opendev.org/57/674757/1/check/tripleo-ci-centos-7-scenario010-multinode-oooq-container/5bf2ec0/ : FAILURE in 2h 21m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"513d9bdc80ac5c9293babcbda48cdb003ec56d68","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-20 13:43:51.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"80a911c3962b3fac693770c0b55f6af6cf9f9153","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-08-22 02:26:56.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8eca883ea629cab9b68a217bbc5971c2022d8fb2","author":{"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},"date":"2019-08-22 06:30:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5223f870bdf78cba88c37801d920780229ad656a","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-22 13:40:11.000000000","message":"Patch Set 1:\n\nThis is a tripleo specific issue because containers. It doesn\u0027t belong in puppet-neutron because if doing it normally, we should have the package which will create these files.","accounts_in_message":[],"_revision_number":1},{"id":"5a5beb18570d3be887a6ba37ac69511ab55842eb","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-08-26 13:49:27.000000000","message":"Patch Set 1: -Code-Review Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d8dbcccc5a415e735e7caa96d2f90cd400038906","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 13:49:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"26639cb109915efa6b8fd07861856e8a0cd2e475","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 18:24:11.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/331ce08fe23f40a8a40a9c8425fa7b07 : SUCCESS in 5m 25s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1e098794b3d7469a8ba1148b01fcd72b : SUCCESS in 6m 10s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/069d72c1407346189db6c01136248091 : SUCCESS in 5m 24s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/539c9075760a46ea873f60d451b54e2e : SUCCESS in 28m 17s\n- puppet-openstack-unit-6.0-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d1b07d228c1142e883e792f8481efa91 : SUCCESS in 18m 22s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/44008f6c72cf4a11abb37ae67cdf61cd : FAILURE in 16m 08s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/03698eb9f0934cad8e722ddc0b56a637 : FAILURE in 22m 34s\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/4a63eb65ea014d12b4dfbf9676f68dd4 : FAILURE in 12m 44s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/dfe5b374341440119fd4facb4c20a751 : FAILURE in 27m 31s","accounts_in_message":[],"_revision_number":1},{"id":"bbeef85585ff0ede0f746daf2cf4aaffc7300652","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-26 19:57:08.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ab22abcf84cd898c3332740fa2ec3be4c0cc8b0c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-26 22:56:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/57/674757/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/5d24772/ : FAILURE in 1h 50m 43s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/57/674757/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/4d5349d/ : SUCCESS in 2h 57m 58s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/57/674757/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/521b7b1/ : FAILURE in 1h 25m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5378dbcc043860c17a4678729b0381750a554dea","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 23:04:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/44f7559689264117a95500ffcbccf4ec : SUCCESS in 5m 09s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b37ea34d8e14408fad54b4fc3028cc08 : SUCCESS in 5m 05s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4b26695e61c948cbacba9c3c85dda48d : SUCCESS in 5m 26s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/468e5cb0da554e11aefa25d196b2cec7 : SUCCESS in 32m 09s\n- puppet-openstack-unit-6.0-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/daff3e7a70f742c7b093f8783a875473 : SUCCESS in 19m 23s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6c3b652643b6426ab9436d5787392746 : SUCCESS in 24m 31s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/0f960dab443849abb750866398e6ca7e : SUCCESS in 1h 44m 56s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/036857e93994429ca85ff1b0eefe6932 : FAILURE in 2h 30m 34s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/86eeac26aa6a4dc8bcced0a7265816bd : SUCCESS in 3h 02m 31s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/7017da7753bf429ba6e5d2515283e466 : SUCCESS in 2h 49m 54s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/280d6862765c4bb8b57e1f220e41f1e9 : SUCCESS in 2h 29m 48s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/5d8b7edf2d884e84ab3647fade72bfeb : SUCCESS in 1h 47m 31s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/08f53af3990b49baa3080e80196b8617 : SUCCESS in 2h 00m 41s (non-voting)\n- tripleo-ci-centos-7-scenario010-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/eba884244f5f4e6ebb395ae41f1090bb : SUCCESS in 2h 31m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9c375ebfba26b7cae1bc305e9ad472c03461eddb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 23:05:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c27002caf6ed093d0d15db7cb502a6631409f040","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 02:47:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a119bc7e828247559634a27649c98469 : SUCCESS in 4m 44s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fd530a456d1b41eb9131a13b70b92619 : SUCCESS in 5m 20s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c154b0f0de8849d8a6fea99e9f84d50c : SUCCESS in 5m 05s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/f15ac01124e74260beda21051e846bc1 : SUCCESS in 34m 30s\n- puppet-openstack-unit-6.0-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3fd07bb00e91450998fa98c45dea8243 : SUCCESS in 17m 29s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/78d8ef855b1e4cf79fe2c48cb50a6891 : SUCCESS in 1h 59m 24s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/9fa27531caa94a50971c1b4ac3c257fa : SUCCESS in 2h 58m 24s\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/1db4ba5da1e64a58a85c71f79afb82ed : SUCCESS in 2h 06m 40s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/be2fc05a006a49a7821008a4f726960f : SUCCESS in 2h 46m 43s","accounts_in_message":[],"_revision_number":1},{"id":"e8ebbf50be3a5a475439638f17bcbe29b5facf1f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 02:47:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"969467960552400a4cd555be1db4d24f1e2013a2","revisions":{"969467960552400a4cd555be1db4d24f1e2013a2":{"kind":"REWORK","_number":1,"created":"2019-08-06 08:54:50.000000000","uploader":{"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},"ref":"refs/changes/57/674757/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/57/674757/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/57/674757/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/57/674757/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/57/674757/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/57/674757/1"}}},"commit":{"parents":[{"commit":"8e4de1b75ce67a253d5c872d3de18128cc17f2a9","subject":"Merge \"Enable service token in nova and cinder\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/8e4de1b75ce67a253d5c872d3de18128cc17f2a9"}]}],"author":{"name":"Hamdy Khader","email":"hamdyk@mellanox.com","date":"2019-08-06 08:53:24.000000000","tz":180},"committer":{"name":"Hamdy Khader","email":"hamdyk@mellanox.com","date":"2019-08-06 08:55:17.000000000","tz":180},"subject":"Add base class for neutron Mellanox agent","message":"Add base class for neutron Mellanox agent\n\nChange-Id: Ibd817a4a34ee2944ee86b4a47e7621637d258da3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/969467960552400a4cd555be1db4d24f1e2013a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/969467960552400a4cd555be1db4d24f1e2013a2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
