)]}'
{"id":"openstack%2Fpuppet-tripleo~674955","triplet_id":"openstack%2Fpuppet-tripleo~stable%2Fqueens~I8ca84798ba0e89cee9674dbe414de892d5d9e3f1","project":"openstack/puppet-tripleo","branch":"stable/queens","topic":"bug/1833584","hashtags":[],"change_id":"I8ca84798ba0e89cee9674dbe414de892d5d9e3f1","subject":"[queens only] Add registry login support","status":"MERGED","created":"2019-08-06 22:57:35.000000000","updated":"2019-08-20 05:17:06.000000000","submitted":"2019-08-20 05:17:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"674955-1566278226539-8c476a95","meta_rev_id":"227b324b77c18ab337541700ecbeed443ffc4569","_number":674955,"virtual_id_number":674955,"owner":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-08-20 05:17:06.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-07 05:32:01.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2019-08-07 16:12:03.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2019-08-20 00:53:41.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"date":"2019-08-19 19:46:55.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-07 05:32:01.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-20 00:53:41.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-08-20 05:17:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d578a738d093971f0e1b86ca61e6ee0b9b8ce9ab","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-06 22:57:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c3cd60e126da0ec600068295667ddf46af92f202","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-06 22:57:42.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"e4f96f7d6fd4a42d3d0312a8197c638e576d623b","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-06 23:03:48.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2a66409f3c97db0272e3da75ad64459d3209695d","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-06 23:33:16.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"72ebbfa39969c8d8eb55f6dbe4fd3510b3487878","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-06 23:36:56.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a348c3ee034047925dbef6fb50b7b60c99deadcd","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-06 23:37:56.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"1e0b31aa21d353083723ab90f664de5e81072640","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-06 23:38:09.000000000","message":"Topic changed from bug/1835657 to bug/1833584","accounts_in_message":[],"_revision_number":5},{"id":"0fb34f1df22666ee4f20d3ed0300856e26e40b5c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-07 00:14:44.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/55/674955/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/10e407c/ : FAILURE in 35m 03s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/55/674955/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/114d61e/ : FAILURE in 35m 04s","accounts_in_message":[],"_revision_number":5},{"id":"469e0146e9fd6e2a639919b537ee205aa69b8eb7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 00:43:37.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-oooq https://logs.opendev.org/55/674955/5/check/tripleo-ci-centos-7-undercloud-oooq/f4122a8/ : FAILURE in 26m 57s\n- tripleo-ci-centos-7-undercloud-upgrades https://logs.opendev.org/55/674955/5/check/tripleo-ci-centos-7-undercloud-upgrades/c661a48/ : FAILURE in 1h 04m 52s\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/55/674955/5/check/tripleo-ci-centos-7-containers-multinode/b368587/ : FAILURE in 31m 15s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/55/674955/5/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/227da24/ : FAILURE in 37m 23s (non-voting)\n- puppet-openstack-lint https://logs.opendev.org/55/674955/5/check/puppet-openstack-lint/66893dd/ : FAILURE in 7m 21s\n- puppet-openstack-syntax-4 https://logs.opendev.org/55/674955/5/check/puppet-openstack-syntax-4/e145eac/ : SUCCESS in 7m 37s\n- puppet-openstack-unit-4.8-centos-7 https://logs.opendev.org/55/674955/5/check/puppet-openstack-unit-4.8-centos-7/aeaceea/ : FAILURE in 39m 11s","accounts_in_message":[],"_revision_number":5},{"id":"db9eb9577741bf95548207e0fae15d2774e55c3e","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-07 03:13:00.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"4a061b2ef28bbdf4a357c4a12f4d7496cc8a149a","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-07 03:13:50.000000000","message":"Topic changed from bug/1835657 to bug/1833584","accounts_in_message":[],"_revision_number":6},{"id":"39a994d99da1746f19366f45ec4b12c5c213199d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 05:18:06.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-oooq https://logs.opendev.org/55/674955/6/check/tripleo-ci-centos-7-undercloud-oooq/2e5c199/ : SUCCESS in 1h 02m 38s\n- tripleo-ci-centos-7-undercloud-upgrades https://logs.opendev.org/55/674955/6/check/tripleo-ci-centos-7-undercloud-upgrades/ecf23c1/ : SUCCESS in 1h 14m 52s\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/55/674955/6/check/tripleo-ci-centos-7-containers-multinode/ffab8f4/ : SUCCESS in 2h 03m 50s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/55/674955/6/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/b579b66/ : FAILURE in 1h 49m 31s (non-voting)\n- puppet-openstack-lint https://logs.opendev.org/55/674955/6/check/puppet-openstack-lint/158b32b/ : SUCCESS in 4m 59s\n- puppet-openstack-syntax-4 https://logs.opendev.org/55/674955/6/check/puppet-openstack-syntax-4/688ac43/ : SUCCESS in 4m 38s\n- puppet-openstack-unit-4.8-centos-7 https://logs.opendev.org/55/674955/6/check/puppet-openstack-unit-4.8-centos-7/615092e/ : SUCCESS in 31m 55s","accounts_in_message":[],"_revision_number":6},{"id":"47563cef66d251c1e6e5602ab661b6fe250908c6","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-07 05:32:01.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/55/674955/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/0d8ea78/ : SUCCESS in 2h 18m 19s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/55/674955/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/b0532a5/ : SUCCESS in 2h 15m 47s","accounts_in_message":[],"_revision_number":6},{"id":"5bed163be3b4c15c30c332b844ff9dec0d54cf82","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-08-07 13:44:00.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"97a4ca73f3c9a2d20e758e537557909d377f785f","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-07 13:56:50.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"a08623061497250ce814020fa2ca8c8cdf44edb4","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-08-07 16:11:10.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"40e2e48c2610bec1db551784478a19d51f79ecca","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-08-07 16:12:03.000000000","message":"Patch Set 6: Code-Review+2\n\nyou\u0027re right, +2","accounts_in_message":[],"_revision_number":6},{"id":"74f74572d5f3385fd295a1ae5597b61ff356d039","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-07 17:06:07.000000000","message":"Patch Set 6: Workflow-1\n\nwant to test this actually works (a novel concept)","accounts_in_message":[],"_revision_number":6},{"id":"62a8507a037da6d8ee7063f92488fd4a1d813b14","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-19 19:46:55.000000000","message":"Patch Set 6: -Workflow\n\ntested and it worked","accounts_in_message":[],"_revision_number":6},{"id":"f3324ebc0cebf5f74689651fd160f2b0058a3c93","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-08-20 00:53:41.000000000","message":"Patch Set 6: Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"60c027f274845db3f1b3fecdd40f899767f1d3c6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 00:53:54.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"51027cb25cfd25b024c73f256bd49da6c1d289b7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 05:17:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"227b324b77c18ab337541700ecbeed443ffc4569","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 05:17:06.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-undercloud-oooq https://zuul.opendev.org/t/openstack/build/8814d33bc49949c7a86f90cf72535ea3 : SUCCESS in 1h 13m 33s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/f54f58e44d8c4b2e841fdbd8a5ee16bd : SUCCESS in 2h 20m 05s\n- puppet-openstack-lint https://zuul.opendev.org/t/openstack/build/e4febe4b78a6474ca4491942f6c0b9b4 : SUCCESS in 7m 54s\n- puppet-openstack-syntax-4 https://zuul.opendev.org/t/openstack/build/b4db4ea458844bfeb72f53c49bc8a21b : SUCCESS in 7m 20s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/a9270a3b3ebe4ca8ac9dc7d6cbbdaaad : SUCCESS in 40m 23s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"e3fae2b04a3e42cba88396e7abebb37f9978d81f","revisions":{"09904ba88158bdd173ea22130bc74eb2cac0d129":{"kind":"REWORK","_number":1,"created":"2019-08-06 22:57:35.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/55/674955/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/55/674955/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/55/674955/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/55/674955/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/55/674955/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/55/674955/1"}}},"commit":{"parents":[{"commit":"d64f7bfbb422da62306be9d7fa60d03db29754e1","subject":"IHA robustness improvements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/d64f7bfbb422da62306be9d7fa60d03db29754e1"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-08-06 22:55:12.000000000","tz":-360},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-08-06 22:57:32.000000000","tz":-360},"subject":"[queens only] Add registry login support","message":"[queens only] Add registry login support\n\nQueens uses puppet to configure docker so we need to add support for\nlogging into the registry in the puppet class itself.\n\nChange-Id: I8ca84798ba0e89cee9674dbe414de892d5d9e3f1\nRelated-Bug: #1835657\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/09904ba88158bdd173ea22130bc74eb2cac0d129"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/09904ba88158bdd173ea22130bc74eb2cac0d129"}]},"branch":"refs/heads/stable/queens"},"73d6e00c809eaaf81a56b2965b2a008d57fc7af6":{"kind":"REWORK","_number":2,"created":"2019-08-06 23:03:48.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/55/674955/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/55/674955/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/55/674955/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/55/674955/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/55/674955/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/55/674955/2"}}},"commit":{"parents":[{"commit":"d64f7bfbb422da62306be9d7fa60d03db29754e1","subject":"IHA robustness improvements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/d64f7bfbb422da62306be9d7fa60d03db29754e1"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-08-06 22:55:12.000000000","tz":-360},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-08-06 23:03:42.000000000","tz":-360},"subject":"[queens only] Add registry login support","message":"[queens only] Add registry login support\n\nQueens uses puppet to configure docker so we need to add support for\nlogging into the registry in the puppet class itself.\n\nChange-Id: I8ca84798ba0e89cee9674dbe414de892d5d9e3f1\nRelated-Bug: #1835657\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/73d6e00c809eaaf81a56b2965b2a008d57fc7af6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/73d6e00c809eaaf81a56b2965b2a008d57fc7af6"}]},"branch":"refs/heads/stable/queens"},"c6100cf4758e4bd4cfcb39ba87f67351fc0f1471":{"kind":"REWORK","_number":3,"created":"2019-08-06 23:33:16.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/55/674955/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/55/674955/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/55/674955/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/55/674955/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/55/674955/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/55/674955/3"}}},"commit":{"parents":[{"commit":"d64f7bfbb422da62306be9d7fa60d03db29754e1","subject":"IHA robustness improvements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/d64f7bfbb422da62306be9d7fa60d03db29754e1"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-08-06 22:55:12.000000000","tz":-360},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-08-06 23:33:13.000000000","tz":-360},"subject":"[queens only] Add registry login support","message":"[queens only] Add registry login support\n\nQueens uses puppet to configure docker so we need to add support for\nlogging into the registry in the puppet class itself.\n\nChange-Id: I8ca84798ba0e89cee9674dbe414de892d5d9e3f1\nRelated-Bug: #1835657\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/c6100cf4758e4bd4cfcb39ba87f67351fc0f1471"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/c6100cf4758e4bd4cfcb39ba87f67351fc0f1471"}]},"branch":"refs/heads/stable/queens"},"60bc78f1901da1f0e4e01d9b0facca2618259fa2":{"kind":"REWORK","_number":4,"created":"2019-08-06 23:36:56.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/55/674955/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/55/674955/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/55/674955/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/55/674955/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/55/674955/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/55/674955/4"}}},"commit":{"parents":[{"commit":"d64f7bfbb422da62306be9d7fa60d03db29754e1","subject":"IHA robustness improvements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/d64f7bfbb422da62306be9d7fa60d03db29754e1"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-08-06 22:55:12.000000000","tz":-360},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-08-06 23:36:52.000000000","tz":-360},"subject":"[queens only] Add registry login support","message":"[queens only] Add registry login support\n\nQueens uses puppet to configure docker so we need to add support for\nlogging into the registry in the puppet class itself.\n\nChange-Id: I8ca84798ba0e89cee9674dbe414de892d5d9e3f1\nRelated-Bug: #1835657\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/60bc78f1901da1f0e4e01d9b0facca2618259fa2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/60bc78f1901da1f0e4e01d9b0facca2618259fa2"}]},"branch":"refs/heads/stable/queens"},"ccc447c24f00bec4e85d0b3d6e49bbc1d95b8577":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2019-08-06 23:37:56.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/55/674955/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/55/674955/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/55/674955/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/55/674955/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/55/674955/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/55/674955/5"}}},"commit":{"parents":[{"commit":"d64f7bfbb422da62306be9d7fa60d03db29754e1","subject":"IHA robustness improvements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/d64f7bfbb422da62306be9d7fa60d03db29754e1"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-08-06 22:55:12.000000000","tz":-360},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-08-06 23:37:48.000000000","tz":-360},"subject":"[queens only] Add registry login support","message":"[queens only] Add registry login support\n\nQueens uses puppet to configure docker so we need to add support for\nlogging into the registry in the puppet class itself.\n\nChange-Id: I8ca84798ba0e89cee9674dbe414de892d5d9e3f1\nRelated-Bug: #1833584\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/ccc447c24f00bec4e85d0b3d6e49bbc1d95b8577"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/ccc447c24f00bec4e85d0b3d6e49bbc1d95b8577"}]},"branch":"refs/heads/stable/queens"},"e3fae2b04a3e42cba88396e7abebb37f9978d81f":{"kind":"REWORK","_number":6,"created":"2019-08-07 03:13:00.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/55/674955/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/55/674955/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/55/674955/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/55/674955/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/55/674955/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/55/674955/6"}}},"commit":{"parents":[{"commit":"d64f7bfbb422da62306be9d7fa60d03db29754e1","subject":"IHA robustness improvements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/d64f7bfbb422da62306be9d7fa60d03db29754e1"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-08-06 22:55:12.000000000","tz":-360},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-08-07 03:12:30.000000000","tz":-360},"subject":"[queens only] Add registry login support","message":"[queens only] Add registry login support\n\nQueens uses puppet to configure docker so we need to add support for\nlogging into the registry in the puppet class itself.\n\nChange-Id: I8ca84798ba0e89cee9674dbe414de892d5d9e3f1\nRelated-Bug: #1833584\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/e3fae2b04a3e42cba88396e7abebb37f9978d81f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/e3fae2b04a3e42cba88396e7abebb37f9978d81f"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
