)]}'
{"id":"openstack%2Fpuppet-tripleo~678092","triplet_id":"openstack%2Fpuppet-tripleo~stable%2Frocky~I6085bcd2f91f84577dcf59d3c93afb24ab718a49","project":"openstack/puppet-tripleo","branch":"stable/rocky","topic":"bug/1840180-stable/rocky","hashtags":[],"change_id":"I6085bcd2f91f84577dcf59d3c93afb24ab718a49","subject":"Add ability to install kernel related packages","status":"MERGED","created":"2019-08-22 19:40:46.000000000","updated":"2019-08-23 17:20:31.000000000","submitted":"2019-08-23 17:20:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"678092-1566580831773-51c6cb98","meta_rev_id":"4ce20b39cbf696c86455fc04dac488c680c4b47f","_number":678092,"virtual_id_number":678092,"owner":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2019-08-23 17:20:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-22 22:48:54.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2019-08-23 14:13:24.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2019-08-23 14:13:24.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-22 22:48:54.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-23 14:13:24.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-08-23 17:20:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"65faff74146d755f9687ed9b7fae2121b27c8ac2","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-22 19:40:46.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"79145da5687bb15c0655e834c59d95632ed5aec9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 22:39:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-oooq https://zuul.opendev.org/t/openstack/build/c586e39ee92a43eba607f55d1f231c74 : SUCCESS in 1h 03m 47s\n- tripleo-ci-centos-7-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/65d373a4bb7f41138f79abb9cbab7412 : SUCCESS in 2h 16m 52s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/4777197eec20420fb93abc993b5a4e99 : SUCCESS in 1h 47m 45s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/c89232d364c9474fa4c4ea200a3009e9 : SUCCESS in 2h 40m 05s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/d2fed86b7a294a4eabc1ac0819e40b0d : SUCCESS in 2h 23m 59s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/c2a3237ab09b401996b353f8221eaa94 : SUCCESS in 2h 29m 59s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/6d5d4f62dc9848a4a39b2269163f3845 : SUCCESS in 2h 50m 48s (non-voting)\n- tripleo-ci-centos-7-standalone-rocky https://zuul.opendev.org/t/openstack/build/dccdd3995acb45428ae77c1efd103e6f : SUCCESS in 1h 47m 00s\n- puppet-openstack-lint https://zuul.opendev.org/t/openstack/build/4c27e0ca284b412781a496abf83a07fd : SUCCESS in 5m 30s\n- puppet-openstack-syntax-4 https://zuul.opendev.org/t/openstack/build/01d698dc67e2418782b27606e2cefcc3 : SUCCESS in 4m 57s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/1a6a37a2f4304545b5d3ad69f1cbad90 : SUCCESS in 36m 06s","accounts_in_message":[],"_revision_number":1},{"id":"a7a787a65f6858daaff43a9b19b83293ad51e146","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-22 22:48:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/92/678092/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/a299da6/ : SUCCESS in 2h 47m 39s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/92/678092/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/c7076bd/ : SUCCESS in 2h 38m 33s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/92/678092/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/35b6c99/ : SUCCESS in 3h 06m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"64403a854ce9268aced9904ec00f195413ddd697","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-08-23 14:13:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1257050b968fb0330fb0656f08b4081ba8798241","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 14:13:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b6fe0176451fd2e653279e8e3bbb14a5c2a5cd95","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 17:20:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"215b0ce2b3df4e818b92cef2d2af47d25b60ab4d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 17:20:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-undercloud-oooq https://zuul.opendev.org/t/openstack/build/b37cdfac840840749cedfb3795ae6058 : SUCCESS in 1h 08m 52s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/f3db31fa1e5247a38faedfa9098e0772 : SUCCESS in 1h 33m 36s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/bc3e3b39fd9f4522b7364cd1c55a61ea : SUCCESS in 2h 25m 05s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/51f5123c6934424dbbdac2b1f47e0de0 : SUCCESS in 3h 00m 31s\n- puppet-openstack-lint https://zuul.opendev.org/t/openstack/build/8f6940a0314a4e8e8390a4636499f22a : SUCCESS in 5m 37s\n- puppet-openstack-syntax-4 https://zuul.opendev.org/t/openstack/build/c3788085d3ab4a919a2515c8a5d391f9 : SUCCESS in 6m 48s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/111167f9b7524a838f80180947c355a8 : SUCCESS in 32m 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8b12ef9d866ea1dcd8396552c6075acc9d0371d1","revisions":{"8b12ef9d866ea1dcd8396552c6075acc9d0371d1":{"kind":"REWORK","_number":1,"created":"2019-08-22 19:40:46.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/92/678092/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/92/678092/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/92/678092/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/92/678092/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/92/678092/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/92/678092/1"}}},"commit":{"parents":[{"commit":"cb5102003b811d1835b8f03eb2b83f612094a788","subject":"Make the rabbitmq-ready exec more stringent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/cb5102003b811d1835b8f03eb2b83f612094a788"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-08-14 19:27:37.000000000","tz":-360},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-08-22 19:40:46.000000000","tz":0},"subject":"Add ability to install kernel related packages","message":"Add ability to install kernel related packages\n\nAdded support to have the kernel class install packages prior to the\nkmod loading.\n\nChange-Id: I6085bcd2f91f84577dcf59d3c93afb24ab718a49\nRelated-Bug: #1840180\n(cherry picked from commit cf8dc4658a4dbbffc145c24898be280ca47d902e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/8b12ef9d866ea1dcd8396552c6075acc9d0371d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/8b12ef9d866ea1dcd8396552c6075acc9d0371d1"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
