)]}'
{"id":"openstack%2Fpuppet-tripleo~678093","triplet_id":"openstack%2Fpuppet-tripleo~stable%2Fqueens~I6085bcd2f91f84577dcf59d3c93afb24ab718a49","project":"openstack/puppet-tripleo","branch":"stable/queens","topic":"bug/1840180-stable/queens","hashtags":[],"change_id":"I6085bcd2f91f84577dcf59d3c93afb24ab718a49","subject":"Add ability to install kernel related packages","status":"MERGED","created":"2019-08-22 19:40:53.000000000","updated":"2019-08-23 16:34:06.000000000","submitted":"2019-08-23 16:34:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"678093-1566578046580-f069c2db","meta_rev_id":"3f39b904699c5ca51e8a4b95b1e6683dcaad49cb","_number":678093,"virtual_id_number":678093,"owner":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2019-08-23 16:34:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2019-08-22 23:00:36.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2019-08-23 14:13:17.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2019-08-23 14:13:17.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-22 23:00:36.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-23 14:13:17.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-08-23 16:34:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c912e1c3270e39f7047c4b8c2550a0e02c8759b3","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-22 19:40:53.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"7f7eca78588a38ca47da71e159b3d75e627c703a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 21:57:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-oooq https://zuul.opendev.org/t/openstack/build/962fc9a15de8470fae7ea40af46b746c : SUCCESS in 1h 07m 39s\n- tripleo-ci-centos-7-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/59a7e19cec6b4cda8f45327372c2b1d2 : SUCCESS in 1h 17m 09s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/868d628500504dfcaa5c1fa2a4f5225c : SUCCESS in 2h 09m 28s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/006d53198fa143bf97233ab028b8bcd2 : FAILURE in 1h 57m 26s (non-voting)\n- puppet-openstack-lint https://zuul.opendev.org/t/openstack/build/da60a1f106d948ada6745200652dd82e : SUCCESS in 7m 21s\n- puppet-openstack-syntax-4 https://zuul.opendev.org/t/openstack/build/4baee4fc6dd54563b6104f697b4b7d18 : SUCCESS in 5m 47s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/0aadbbe654384f6990a6a77e93163ab8 : SUCCESS in 33m 14s","accounts_in_message":[],"_revision_number":1},{"id":"28590d8bc9b854d078b0362e24c90d4eb216d6b5","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-22 23:00:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/93/678093/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/059b46e/ : FAILURE in 3h 18m 18s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/93/678093/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/5a8bf2b/ : SUCCESS in 2h 13m 58s","accounts_in_message":[],"_revision_number":1},{"id":"5320c991c1777c4401cdc1ff120858de62db5c68","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-08-23 14:13:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7604b181531beda9d935b07fd89f2c4a13302093","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 14:13:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2ce7c399db2d53810c61f343dfd5accf46119bd0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 16:34:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a83480e264e2fe653680b5b77df768c9db6c39f9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 16:34:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-undercloud-oooq https://zuul.opendev.org/t/openstack/build/5e65560b1e8249e68c41ed4037c062da : SUCCESS in 1h 02m 46s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/6de864a0339448549a74a6f5c2b8f576 : SUCCESS in 2h 13m 35s\n- puppet-openstack-lint https://zuul.opendev.org/t/openstack/build/e7c0d036e69e4a74b5a10dcb2c80c762 : SUCCESS in 6m 27s\n- puppet-openstack-syntax-4 https://zuul.opendev.org/t/openstack/build/9d4b3e5271ba4a83936ab572b5c7cd74 : SUCCESS in 7m 35s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/4f412b7a60b24836bda5cb8fcffeee54 : SUCCESS in 41m 14s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"81bc798ae00415826e121ed6c6419806eccbf16c","revisions":{"81bc798ae00415826e121ed6c6419806eccbf16c":{"kind":"REWORK","_number":1,"created":"2019-08-22 19:40:53.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/93/678093/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/93/678093/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/93/678093/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/93/678093/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/93/678093/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/93/678093/1"}}},"commit":{"parents":[{"commit":"051e05a91dc89021de45644a8a34ffb93c2ddce7","subject":"Make the rabbitmq-ready exec more stringent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/051e05a91dc89021de45644a8a34ffb93c2ddce7"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-08-14 19:27:37.000000000","tz":-360},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-08-22 19:40:53.000000000","tz":0},"subject":"Add ability to install kernel related packages","message":"Add ability to install kernel related packages\n\nAdded support to have the kernel class install packages prior to the\nkmod loading.\n\nChange-Id: I6085bcd2f91f84577dcf59d3c93afb24ab718a49\nRelated-Bug: #1840180\n(cherry picked from commit cf8dc4658a4dbbffc145c24898be280ca47d902e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/81bc798ae00415826e121ed6c6419806eccbf16c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/81bc798ae00415826e121ed6c6419806eccbf16c"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
