)]}'
{"id":"openstack%2Fpuppet-tripleo~684384","triplet_id":"openstack%2Fpuppet-tripleo~master~Id872c55cfc6b958fef3ccda2d923f821a1fe6a13","project":"openstack/puppet-tripleo","branch":"master","topic":"bug/1845153","hashtags":[],"change_id":"Id872c55cfc6b958fef3ccda2d923f821a1fe6a13","subject":"Add \u0027ipversion\u0027 to firewall/rule.pp","status":"MERGED","created":"2019-09-24 15:42:28.000000000","updated":"2019-10-01 11:22:31.000000000","submitted":"2019-10-01 11:21:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"684384-1569928884458-02882d20","meta_rev_id":"c872a20a927667af020d71049fdc9f1ede1227f9","_number":684384,"virtual_id_number":684384,"owner":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},{"value":0,"date":"2019-09-29 12:45:13.000000000","_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-01 11:21:24.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-09-26 09:16:51.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2019-09-30 19:07:34.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2019-09-27 19:17:04.000000000","_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},{"value":0,"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":2,"date":"2019-09-25 18:36:44.000000000","_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2019-09-30 19:07:34.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2019-09-27 19:17:04.000000000","_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},{"value":0,"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":0,"date":"2019-09-26 10:00:05.000000000","_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-24 16:17:53.000000000","updated_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"reviewer":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"state":"REVIEWER"},{"updated":"2019-09-26 09:16:51.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-26 10:00:05.000000000","updated_by":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"reviewer":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"state":"REVIEWER"},{"updated":"2019-09-27 19:17:04.000000000","updated_by":{"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},"reviewer":{"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},"state":"REVIEWER"},{"updated":"2019-09-30 19:07:34.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-10-01 11:21:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a77eda689fe3d4baef0b330aa89672b441f2ab6a","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2019-09-24 15:42:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"50f8799edaa367e0ce7493e9ee412b1b8b5d0850","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2019-09-24 16:17:53.000000000","message":"Patch Set 1: Code-Review+2\n\nthank you, that\u0027s well done","accounts_in_message":[],"_revision_number":1},{"id":"d9b56dac976618f3b5f717bf9ef0dd3439aca263","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-24 18:51:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/84/684384/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/26460af/ : FAILURE in 3h 08m 13s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/84/684384/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/ec5448d/ : FAILURE in 2h 18m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9e14bf8cf2bb7baf9ca11ae3768f0668b397d6b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 23:57:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7725c88a3dc441dc8fad4304659a7a05 : SUCCESS in 3m 51s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/15fd85657b30461d80596186d2d580d9 : SUCCESS in 3m 56s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3f51354aa8d04faaa5252be888ad9c00 : SUCCESS in 4m 25s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/b21b9e0d4359450b906384326e1c90e8 : SUCCESS in 37m 44s\n- puppet-openstack-unit-6.0-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cc876c04666e44c6adca90d448a19134 : SUCCESS in 18m 09s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e3af1329087e4fc88b36759f6dec0ebe : SUCCESS in 20m 26s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/cac8d408bafe494793e067f51913c785 : SUCCESS in 2h 42m 02s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/652d17278a004db1b8975855259bceb7 : FAILURE in 2h 51m 01s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/3406dac625d64ea5a6ed6a2afd54edfd : FAILURE in 1h 30m 29s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/de8a282b5c57419e8765db94aaba4f16 : SUCCESS in 2h 49m 34s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/3c98b12ea59043f98088a1856ea1a953 : SUCCESS in 3h 43m 11s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/e126378fb18c46b3ace2d841a1e7c123 : SUCCESS in 2h 35m 46s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/13e86fb2cc2d4c998fae3368ab0c66cb : FAILURE in 2h 43m 50s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/75922eb8591d44cc87c296127a306772 : SUCCESS in 3m 44s","accounts_in_message":[],"_revision_number":1},{"id":"4442fc760ea93c97f32e9218ca9ff7af98c4b1e6","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2019-09-25 14:00:58.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8c22ca095979288c239af7ea2a1075390d8b9838","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-25 17:55:04.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/84/684384/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/ac2dc4a/ : FAILURE in 3h 03m 27s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/84/684384/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/8f27123/ : FAILURE in 3h 45m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1c287feeec366dfd7783d42874831425f4b35e08","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2019-09-25 18:36:44.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"9acee4eb7b285110ebf5d06bc66fea62f880a856","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-25 21:54:00.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/84/684384/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/9734b94/ : FAILURE in 3h 16m 31s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/84/684384/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/adc6845/ : FAILURE in 3h 14m 35s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"05a8f61a89048a20b1e1c4cf497f3c826f43fb3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 01:21:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ff89af4fdf6b42dfa5d729d9f05644db : SUCCESS in 5m 07s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/51a29c2cb5a1480a907804d7d056ce2d : SUCCESS in 3m 49s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/813d119484e84f259ac82cacf8cb6ff6 : SUCCESS in 3m 53s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/66dd73cae8144582973c71d8300cda6c : SUCCESS in 37m 37s\n- puppet-openstack-unit-6.0-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fc84718035d4446d88fd45c17d30d09d : SUCCESS in 17m 04s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b9e5378e937c40e7905216fdb4c72466 : SUCCESS in 23m 03s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/b47b6995aea843189c443ccb14fd3993 : SUCCESS in 2h 07m 06s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/c48d77f874d6421792c1b54cbbba12f6 : FAILURE in 2h 39m 30s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/8a5669c534e64f70a2aa9b5284c9a934 : SUCCESS in 3h 19m 53s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/3890d0cfc41d456ab26ccd40626b4161 : SUCCESS in 2h 56m 52s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/f722accc1697458c9a6cdea7083f5bff : SUCCESS in 3h 14m 55s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/54f39493f58f49418c460ffe2cb4c941 : SUCCESS in 2h 15m 31s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/a7f3559194004398bc661646d148a080 : FAILURE in 2h 20m 47s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ff79991156844ed497bd46ed59119bac : SUCCESS in 3m 48s","accounts_in_message":[],"_revision_number":2},{"id":"7605a05cbd768971c0a91ff4c472feb7da59f450","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2019-09-26 05:37:55.000000000","message":"Patch Set 2:\n\ncheck-rdo","accounts_in_message":[],"_revision_number":2},{"id":"9101efa1dbd3ea6e45a7b7c1b1c3ff3077bf0876","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-26 09:16:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/84/684384/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/d720e25/ : SUCCESS in 3h 07m 07s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/84/684384/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/40adaf4/ : FAILURE in 3h 38m 08s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"fac071ce8a3e00e3515fd08d5e2bc2f203629ae9","author":{"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},"date":"2019-09-27 19:17:04.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"1226a3a013f089910d4421da210feee90324c4c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 19:17:23.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ab368663fdcac6e6d7e662482545f7fa04fd6d14","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 23:00:29.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0dad5c212fb24b02a364edb0cc6abb18 : SUCCESS in 4m 22s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/110e8e088d0e44aaad2c9e5fbe914d89 : SUCCESS in 6m 11s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/efdbcdad94df43689513c8f86706f5d0 : SUCCESS in 4m 14s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/2161dfb1406847319c5859d38c7ff132 : SUCCESS in 36m 12s\n- puppet-openstack-unit-6.0-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a76240f9f0e1453a9c55206cb71e45ca : SUCCESS in 21m 40s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/aa14c0a6e5584be4802b93748b7d7e3b : SUCCESS in 2h 40m 58s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/e46e68c95b93422aa0bf56b82e724729 : TIMED_OUT in 3h 36m 58s\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/e83fc7397ce14730aa342a8735b3feda : SUCCESS in 2h 43m 41s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/3d670eb053bc40dc80738515df1b7c02 : SUCCESS in 2h 38m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/66768f40130b41c1a123ca70c6d0b9a4 : SUCCESS in 5m 13s","accounts_in_message":[],"_revision_number":2},{"id":"ed3347337851ab4086bcca42f85276883e76f4aa","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2019-09-28 14:10:27.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"2843f59234948e52db97e8e817210e1192c4fccb","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-28 17:23:59.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/84/684384/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/5d2549f/ : SUCCESS in 3h 02m 49s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/84/684384/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/1645607/ : FAILURE in 3h 11m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4f420e0ca060c69e3c27162ba2d7d5f6619d9a59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-28 17:45:51.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/548773637ee54933a685ab32009657de : SUCCESS in 4m 12s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/aa65035547c2437aad26d174106bd629 : SUCCESS in 3m 58s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7a01f7d17570465ca340d6f15f2c68a3 : SUCCESS in 5m 49s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/34b7f6c2c6d747fb8bc06deb36d1dda3 : FAILURE in 10m 12s\n- puppet-openstack-unit-6.0-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0ba2a87ced774d9f8b172c5379c86577 : SUCCESS in 18m 09s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f51c2255ab984b54ad0e3a28e16b3474 : SUCCESS in 22m 08s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/262fb2c82e3d418795dbbc8630099a94 : SUCCESS in 2h 33m 11s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/f11e55d6f8814fafb59e6cb75e580afd : FAILURE in 1h 59m 06s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/ac4273d9d305488baf2888487da705a3 : SUCCESS in 3h 26m 13s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/19bef5e1080e4ea4b03a15a2cbada3cc : SUCCESS in 3h 10m 11s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/0b284e8ca22d4b5f90a7c5de824679af : SUCCESS in 3h 04m 24s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/20d60b678d734862bfffd5853c6d0840 : SUCCESS in 2h 42m 41s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/1c34375ea11e42c5a31d22367eba0a28 : FAILURE in 1h 57m 37s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f16edd83f3ca431b8d378a02ea536c15 : SUCCESS in 4m 44s","accounts_in_message":[],"_revision_number":2},{"id":"a09184ece8dd2a82abd0de8edbbf0e86f3436bfa","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2019-09-29 12:45:13.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"0fa56ae81708b4f5bbd343336bb21f2ac8220eb9","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-29 16:14:15.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/84/684384/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/58798ae/ : SUCCESS in 3h 05m 16s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/84/684384/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/e533f68/ : FAILURE in 3h 24m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5fdbf0b6a3327462edf5013f90df0a11974d78d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-29 16:19:46.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c668ad362e3c422096f07f8e28040532 : SUCCESS in 4m 18s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fbea36d1c2ae4d2ca159d461fe0e077c : SUCCESS in 3m 56s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/08ff7a61ce0048f6a42c01524366976b : SUCCESS in 3m 45s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/a39015526e6c4240850171f38248a964 : FAILURE in 7m 01s\n- puppet-openstack-unit-6.0-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/14b715e1ae284d10985dcfc41ebc9a59 : SUCCESS in 20m 39s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/efb2804776104a31adae4ae30d046c20 : SUCCESS in 20m 25s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/c24cc52bcf8e492d8227ef3e596a9415 : SUCCESS in 2h 00m 12s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/2be7a358e2094d7ca48ec0dcbf76e910 : FAILURE in 2h 13m 33s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/9c9990a95df4498189ef7cc35e767623 : SUCCESS in 3h 27m 39s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/72741928c9984352a287e01d790b9e80 : SUCCESS in 2h 54m 32s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/b9d0703f6b9749d0873596cdd5fb6407 : SUCCESS in 3h 02m 26s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/4380b0ce51b34658afcca51f8d802e9e : SUCCESS in 2h 21m 09s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/677c0e9a70844b51be360d8d1f3934e0 : FAILURE in 2h 00m 12s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3897fe5549f044658fd7cf2f441659ba : SUCCESS in 3m 23s","accounts_in_message":[],"_revision_number":2},{"id":"e136beded3fc18803607d27a0c6f226c006b3fd9","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-09-30 19:07:34.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"117efbbc1d32548fe50acaeddd55716ef6980942","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-30 22:17:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a0903e6907a1456ea4e9e45247e4c9dd : SUCCESS in 4m 13s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c3b3ac789beb460e8f2ae6530b8237dd : SUCCESS in 3m 39s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0e26d0d05cb74a048da3aaad7d55f178 : SUCCESS in 4m 22s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/53c2e0b7468c48a59c9b51fc6718b777 : SUCCESS in 36m 23s\n- puppet-openstack-unit-6.0-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7a5a6167f781483ba81c26b0f77dad12 : SUCCESS in 19m 35s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/940a838ada9b4375959f98f67b2a3d59 : SUCCESS in 20m 16s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/696da57d08ba4b618973cae9301fc1cf : SUCCESS in 1h 54m 14s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/8dc5a559e2924bd69dc14e798eb11125 : FAILURE in 1h 45m 46s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/6730404f70144a0c8efc8dde1510e3da : SUCCESS in 3h 00m 13s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/3aa3510a596c45e8bb2b704f01191eca : SUCCESS in 2h 31m 04s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/05fab030448842e5b7270c8baffaf6b1 : SUCCESS in 2h 47m 24s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/d634bc06d7ca415d965b7228050f9e03 : SUCCESS in 2h 11m 02s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/f59e09a392c146158f6552bc05e484f4 : FAILURE in 1h 42m 30s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/30b7ad986b9941efa1a64a0af8ad2ca6 : SUCCESS in 3m 50s","accounts_in_message":[],"_revision_number":2},{"id":"5dbe7444a683bc91231cfca19cf94b65ac4f5820","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-30 22:44:20.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4c8268be7b15d49baed944d476d5e5696d487f23","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 11:21:24.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2934ae4578d64a18ae4ed4f4d3edbb5a : SUCCESS in 4m 53s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dabdfbde87504327ac67d72e144e7cea : SUCCESS in 5m 31s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/175b7db361a942f8bb98f72a0865bd8c : SUCCESS in 4m 26s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/bc9cbad0623b429fbff90083857e99e8 : SUCCESS in 39m 02s\n- puppet-openstack-unit-6.0-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e7a890333e3b477e8f57007685db08c4 : SUCCESS in 25m 26s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/f0a057e916c7433ca83049e68c84307f : SUCCESS in 2h 07m 53s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/c8a5e2888be745ce9b2ff439bda0e71b : SUCCESS in 3h 19m 15s\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/f4870a20cbe848c28d3268cdf4282b24 : SUCCESS in 2h 34m 16s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/64de204783c34716859aee048ebf9853 : SUCCESS in 2h 45m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/87a9fc3afbda4d0d88c0bf9488d35292 : SUCCESS in 3m 42s","accounts_in_message":[],"_revision_number":2},{"id":"db91203e7beb13790d529a4cb241161e5a048364","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 11:21:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"134fa02605eb41d4e65b68d5eb858978079e17a1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 11:22:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/35b3b3e1180f4033a437f9a572e71326 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7264c75c37741f269fdd34c45d939019137b3977","revisions":{"6a37f2874b9bad3a24d56f1fc6f3e56099bdb817":{"kind":"REWORK","_number":1,"created":"2019-09-24 15:42:28.000000000","uploader":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"ref":"refs/changes/84/684384/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/84/684384/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/84/684384/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/84/684384/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/84/684384/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/84/684384/1"}}},"commit":{"parents":[{"commit":"f494131c2d4a72cf656434a031b8c33b47e1aa6a","subject":"Merge \"Use memcached for token caching in authtoken for telemetry services\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/f494131c2d4a72cf656434a031b8c33b47e1aa6a"}]}],"author":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2019-09-24 10:57:53.000000000","tz":120},"committer":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2019-09-24 15:36:37.000000000","tz":120},"subject":"Add \u0027ipversion\u0027 to firewall/rule.pp","message":"Add \u0027ipversion\u0027 to firewall/rule.pp\n\nAdd the posibility to add \u0027ipversion\u0027 to the firewall\nrule manifest.\n\nCloses-Bug: #1845153\nChange-Id: Id872c55cfc6b958fef3ccda2d923f821a1fe6a13\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/6a37f2874b9bad3a24d56f1fc6f3e56099bdb817"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/6a37f2874b9bad3a24d56f1fc6f3e56099bdb817"}]},"branch":"refs/heads/master"},"7264c75c37741f269fdd34c45d939019137b3977":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-09-25 18:36:44.000000000","uploader":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"ref":"refs/changes/84/684384/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/84/684384/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/84/684384/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/84/684384/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/84/684384/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/84/684384/2"}}},"commit":{"parents":[{"commit":"ebe6f1a3004bd56609ae6be922292504ac63704c","subject":"Merge \"Update log-driver value for podman\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/ebe6f1a3004bd56609ae6be922292504ac63704c"}]}],"author":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2019-09-24 10:57:53.000000000","tz":120},"committer":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2019-09-25 18:36:44.000000000","tz":0},"subject":"Add \u0027ipversion\u0027 to firewall/rule.pp","message":"Add \u0027ipversion\u0027 to firewall/rule.pp\n\nAdd the posibility to add \u0027ipversion\u0027 to the firewall\nrule manifest.\n\nCloses-Bug: #1845153\nChange-Id: Id872c55cfc6b958fef3ccda2d923f821a1fe6a13\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/7264c75c37741f269fdd34c45d939019137b3977"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/7264c75c37741f269fdd34c45d939019137b3977"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
