)]}'
{"id":"openstack%2Fpuppet-tripleo~688414","triplet_id":"openstack%2Fpuppet-tripleo~master~I8b9e341456d117358203095cf856d0206ec363d1","project":"openstack/puppet-tripleo","branch":"master","topic":"bug/1432387","hashtags":[],"change_id":"I8b9e341456d117358203095cf856d0206ec363d1","subject":"Ensure to remove cinder\u0027s outdated leftover lock files","status":"ABANDONED","created":"2019-10-14 14:40:02.000000000","updated":"2020-04-22 16:34:32.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"46e53b5fc25b31fb16302732c562f2843d615697","_number":688414,"virtual_id_number":688414,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-10-15 18:10:32.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-10-15 17:07:03.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},{"_account_id":17823,"name":"Michele Baldessari","email":"michele@redhat.com","username":"mbaldessari"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17823,"name":"Michele Baldessari","email":"michele@redhat.com","username":"mbaldessari"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":-1,"date":"2019-10-17 15:41:52.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},{"_account_id":17823,"name":"Michele Baldessari","email":"michele@redhat.com","username":"mbaldessari"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"_account_id":17823,"name":"Michele Baldessari","email":"michele@redhat.com","username":"mbaldessari"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-14 14:51:30.000000000","updated_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"reviewer":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"state":"REVIEWER"},{"updated":"2019-10-14 14:51:49.000000000","updated_by":{"_account_id":17823,"name":"Michele Baldessari","email":"michele@redhat.com","username":"mbaldessari"},"reviewer":{"_account_id":17823,"name":"Michele Baldessari","email":"michele@redhat.com","username":"mbaldessari"},"state":"REVIEWER"},{"updated":"2019-10-14 15:02:39.000000000","updated_by":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"reviewer":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"state":"REVIEWER"},{"updated":"2019-10-14 19:14:44.000000000","updated_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"REVIEWER"},{"updated":"2019-10-15 09:12:40.000000000","updated_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"REVIEWER"},{"updated":"2019-10-15 17:07:03.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-15 18:10:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1ec04282f179df097c17e19be486d34a6e986dcc","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-10-14 14:40:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d9a83ad6f4d6af4b481d043d1b5b9011691dba9e","author":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"date":"2019-10-14 15:02:39.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"76a58a7670906fab07a7a5761093861d0987564a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-14 17:23:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/14/688414/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/4817d45/ : FAILURE in 2h 18m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1a80e4c19805d9d8725c291e44a7c2999644da53","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-14 17:58:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e57e596e30b7481cad2f51ee92a9329c : SUCCESS in 4m 07s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4614b04fed534c6c83961dab936251e6 : SUCCESS in 4m 18s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2a469e8a800b4b2da317dec6dc9ecdf6 : SUCCESS in 4m 10s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/259ab47210c1419bb49edb23576962e5 : FAILURE in 7m 09s\n- puppet-openstack-unit-6.0-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ca736dc33d5c4f7eb39ff0813520d42f : FAILURE in 23m 20s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d75017affb194ecb9f613a6253ea0b29 : FAILURE in 22m 57s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/5255b740a1ae43769aabfc500d74c8d4 : SUCCESS in 2h 10m 28s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/a2980a8ededa4e5ead2d95a5de22b7d4 : FAILURE in 2h 09m 53s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/8e11ecb357e046d18ab9c5f52287325b : SUCCESS in 2h 56m 11s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/668c2081cf78461e82bda7a406703d81 : SUCCESS in 2h 28m 26s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/fc37d88e64d64ddaa828b1c663eadd02 : SUCCESS in 2h 40m 04s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/f71084e78c7849beb48cb7a083c6f0d6 : FAILURE in 1h 09m 44s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/71216570012b4607b647dc6aa9ff128c : FAILURE in 2h 13m 38s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/8635db2688834ffa92959a39eb594d17 : POST_FAILURE in 1h 39m 09s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/f050d4346ecd40f2bafcffdbf2678a1a : FAILURE in 1h 19m 35s","accounts_in_message":[],"_revision_number":1},{"id":"f236138cd80c4064633bb49764a6c204410a24e1","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2019-10-14 19:14:44.000000000","message":"Patch Set 1:\n\nThe puppet-openstack-* failures should go away when [1] merges.\n\nThe tripleo-ci-centos-7-* failures highlight a syntax error in the Depends-On for the puppet-cinder patch [2]. You need to specify the dependency by its Change-Id, or a full URL to its review.\n\n[1] https://review.opendev.org/688319\n[2] https://review.opendev.org/688413","accounts_in_message":[],"_revision_number":1},{"id":"b58db7436959d5fcc7779e4b34f313668024275e","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-10-15 07:32:40.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"bdf499d7f8e2d904b25936021b2c7d583880b753","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 08:20:26.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"a77cb7cf95a3fbb0f3e49f9b42d05731feef5412","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-15 08:20:27.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"3ba366d690ade5141a6567f6ac8b1d1b7e87165c","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2019-10-15 09:12:40.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"0a7a8ace9a344892ccb69061cd4edc08923199e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 12:03:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bd8b70adc41b41bba17c4abaff7b60f4 : SUCCESS in 4m 26s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/de0f788ac9524d548a9cf2b30011a5f2 : SUCCESS in 4m 09s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d3bf93fed4c143f3b9b3e94ecd6894fa : SUCCESS in 3m 36s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/18018d88ac6541fb87f49444bbbb3310 : SUCCESS in 41m 58s\n- puppet-openstack-unit-6.0-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/23bbf6b6483043868dfbbdd7ded78303 : SUCCESS in 21m 53s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/652089c2190c40f3a5f442cf17d6c685 : SUCCESS in 21m 52s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/8a9de464d5524c24b0aa37e8c26f4132 : SUCCESS in 2h 08m 56s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/b8b30c271b4e4e54802471ad90b75a01 : FAILURE in 2h 10m 45s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/6f8b98f946684647a3d2e9336657759b : SUCCESS in 2h 45m 05s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/23bae5e59eb94a12a0a3e20f4cd796d8 : SUCCESS in 2h 38m 22s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/5b2b627a638743b99bfcbc65e97426e7 : SUCCESS in 2h 30m 42s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/f5fc7aca461a404a933feb7e60f772ca : SUCCESS in 2h 12m 46s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/9877de3011734999b3b86088153d9d41 : FAILURE in 2h 12m 18s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/0339aca4d5194e55becfb2431fbe3a6b : SUCCESS in 2h 34m 28s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/be569982ae394d6d9bc2f77c8a3919e0 : SUCCESS in 1h 56m 55s","accounts_in_message":[],"_revision_number":2},{"id":"771e7be624cc3d4433de32603fe42b5ef59ad15e","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-15 12:31:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/14/688414/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/170647c/ : FAILURE in 3h 17m 47s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d0e31e8f02f5ca60eec49ef4a809b9f874ce9604","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-10-15 16:11:39.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3970b3db87c48bb95abfc0545558efc3ba34205d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-15 17:07:03.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/14/688414/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/c806ade/ : FAILURE in 43m 28s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"73d4c73df699e134c1d04034af0a59e983ea5c50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 18:10:32.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9170651742b346c4a7249710b324e626 : SUCCESS in 4m 39s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fb1ec88ace534f5db1f51ef0a06d8685 : FAILURE in 2m 59s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/86e88f3236e04cbe9cfdcac725f3fa48 : SUCCESS in 4m 53s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/0ef655dbb2404190880b4e33b5c16aee : FAILURE in 41m 33s\n- puppet-openstack-unit-6.0-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/534fef8c394047f0b83f2e4cc6fe40ed : FAILURE in 23m 25s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9d37494b92b94bd6b8e8b4819f6e100e : FAILURE in 21m 35s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/bc299e4ea127488289fac0ae4664f375 : FAILURE in 37m 02s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/ad8803e8ab83408199d99797b1f324c9 : FAILURE in 37m 26s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/e84cae5833244fc9ba32c06859104f0c : FAILURE in 50m 31s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/d48d49f42d524e96a820a2c16cfef31a : FAILURE in 44m 38s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/2a05f4ec75ae48b6ba88a552ff4433b1 : FAILURE in 52m 53s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/a1259231d79b4bbca0eb036684861fe4 : FAILURE in 39m 57s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/245882c71fb14b91bd94f9d91e98f7c2 : FAILURE in 32m 26s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/afafce8860d0422d85f1f40cb62219e6 : FAILURE in 34m 58s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/a0fa18a28c6f4a6680ad1f2a098de0f1 : FAILURE in 47m 34s","accounts_in_message":[],"_revision_number":3},{"id":"7f2ec34eaa96c20e3fd28c713457423ef193ac39","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-10-17 15:41:52.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"925a60f44c6469333de06fe69450e064450aa368","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-04-22 16:34:32.000000000","message":"Abandoned\n\nThese changes are no longer needed due to previous discussions on the ML.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"c9424046f202ca2f4fd747c2a28155e50c14ca8b","revisions":{"b541caba0bc0166c3d79715a1b8a7c8b2b371416":{"kind":"REWORK","_number":1,"created":"2019-10-14 14:40:02.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/14/688414/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/14/688414/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/14/688414/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/14/688414/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/14/688414/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/14/688414/1"}}},"commit":{"parents":[{"commit":"06e901c215cbe6b3f115167f67fb8bbad13eb725","subject":"Merge \"Remove Tacker service\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/06e901c215cbe6b3f115167f67fb8bbad13eb725"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2019-10-14 14:34:41.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2019-10-14 14:39:13.000000000","tz":120},"subject":"Ensure to remove cinder\u0027s outdated leftover lock files","message":"Ensure to remove cinder\u0027s outdated leftover lock files\n\nCinder to manage locks use fasteners, the fasteners module have a bug\nand it doesn\u0027t remove some lock files automatically after usage [1]\n\nCustomer file system growing slowly and so customer risk to facing some\nissues to file system usage after a long period.\n\nA fix was proposed to fasteners but it\u0027s not merged yet.This fix\npropose to switch from locks management based on files to locks\nmanaged through a unique file where locks will be managed by file offset.\nIt will introduce some design changes which can\u0027t be backported to\nstable branches.\n\nThis patch propose to introduce a new cron job to remove these lock files.\nHence, we could fix this issue on stable branches too.\n\nThe added cron job will remove all the lock files last accessed after 7\ndays to avoid to remove used lock files.\n\nThe lifetime is not configurable. We want to avoid lets user to\nconfigure a lifetime to short and so possibly introduce some race\nconditions with cinder.\n\nWhen the root cause will be fixed through the fasteners patch we can\nthen remove this cron job safely.\n\n[1] https://github.com/harlowja/fasteners/issues/26\n\nChange-Id: I8b9e341456d117358203095cf856d0206ec363d1\nCo-authored-by: Cédric Jeanneret \u003ccjeanner@redhat.com\u003e\nPartial-Bug: #1432387\nDepends-On: #688413\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/b541caba0bc0166c3d79715a1b8a7c8b2b371416"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/b541caba0bc0166c3d79715a1b8a7c8b2b371416"}]},"branch":"refs/heads/master"},"0e2012ed70820a40244c812c79ceb232113c1341":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-10-15 07:32:40.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/14/688414/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/14/688414/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/14/688414/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/14/688414/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/14/688414/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/14/688414/2"}}},"commit":{"parents":[{"commit":"06e901c215cbe6b3f115167f67fb8bbad13eb725","subject":"Merge \"Remove Tacker service\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/06e901c215cbe6b3f115167f67fb8bbad13eb725"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2019-10-14 14:34:41.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2019-10-15 07:31:49.000000000","tz":120},"subject":"Ensure to remove cinder\u0027s outdated leftover lock files","message":"Ensure to remove cinder\u0027s outdated leftover lock files\n\nCinder to manage locks use fasteners, the fasteners module have a bug\nand it doesn\u0027t remove some lock files automatically after usage [1]\n\nCustomer file system growing slowly and so customer risk to facing some\nissues to file system usage after a long period.\n\nA fix was proposed to fasteners but it\u0027s not merged yet.This fix\npropose to switch from locks management based on files to locks\nmanaged through a unique file where locks will be managed by file offset.\nIt will introduce some design changes which can\u0027t be backported to\nstable branches.\n\nThis patch propose to introduce a new cron job to remove these lock files.\nHence, we could fix this issue on stable branches too.\n\nThe added cron job will remove all the lock files last accessed after 7\ndays to avoid to remove used lock files.\n\nThe lifetime is not configurable. We want to avoid lets user to\nconfigure a lifetime to short and so possibly introduce some race\nconditions with cinder.\n\nWhen the root cause will be fixed through the fasteners patch we can\nthen remove this cron job safely.\n\n[1] https://github.com/harlowja/fasteners/issues/26\n\nChange-Id: I8b9e341456d117358203095cf856d0206ec363d1\nCo-authored-by: Cédric Jeanneret \u003ccjeanner@redhat.com\u003e\nPartial-Bug: #1432387\nDepends-On: If33825a3dae0dfd53e058d1d8277cee4a0423ea1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/0e2012ed70820a40244c812c79ceb232113c1341"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/0e2012ed70820a40244c812c79ceb232113c1341"}]},"branch":"refs/heads/master"},"c9424046f202ca2f4fd747c2a28155e50c14ca8b":{"kind":"REWORK","_number":3,"created":"2019-10-15 16:11:39.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/14/688414/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/14/688414/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/14/688414/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/14/688414/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/14/688414/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/14/688414/3"}}},"commit":{"parents":[{"commit":"06e901c215cbe6b3f115167f67fb8bbad13eb725","subject":"Merge \"Remove Tacker service\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/06e901c215cbe6b3f115167f67fb8bbad13eb725"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2019-10-14 14:34:41.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2019-10-15 16:11:24.000000000","tz":120},"subject":"Ensure to remove cinder\u0027s outdated leftover lock files","message":"Ensure to remove cinder\u0027s outdated leftover lock files\n\nCinder to manage locks use fasteners, the fasteners module have a bug\nand it doesn\u0027t remove some lock files automatically after usage [1]\n\nCustomer file system growing slowly and so customer risk to facing some\nissues to file system usage after a long period.\n\nA fix was proposed to fasteners but it\u0027s not merged yet.This fix\npropose to switch from locks management based on files to locks\nmanaged through a unique file where locks will be managed by file offset.\nIt will introduce some design changes which can\u0027t be backported to\nstable branches.\n\nThis patch propose to introduce a new cron job to remove these lock files.\nHence, we could fix this issue on stable branches too.\n\nThe added cron job will remove all the lock files last accessed after 7\ndays to avoid to remove used lock files.\n\nThe lifetime is not configurable. We want to avoid lets user to\nconfigure a lifetime to short and so possibly introduce some race\nconditions with cinder.\n\nWhen the root cause will be fixed through the fasteners patch we can\nthen remove this cron job safely.\n\n[1] https://github.com/harlowja/fasteners/issues/26\n\nChange-Id: I8b9e341456d117358203095cf856d0206ec363d1\nCo-authored-by: Cédric Jeanneret \u003ccjeanner@redhat.com\u003e\nPartial-Bug: #1432387\nDepends-On: If33825a3dae0dfd53e058d1d8277cee4a0423ea1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/c9424046f202ca2f4fd747c2a28155e50c14ca8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/c9424046f202ca2f4fd747c2a28155e50c14ca8b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
