)]}'
{"id":"openstack%2Fpuppet-tripleo~688992","triplet_id":"openstack%2Fpuppet-tripleo~stable%2Fqueens~I30f03bc8eb81db0243c137d4af08924adeebc951","project":"openstack/puppet-tripleo","branch":"stable/queens","topic":"fix-stable/stein-stable/rocky-stable/queens","hashtags":[],"change_id":"I30f03bc8eb81db0243c137d4af08924adeebc951","subject":"Deep merge hiera keys for mysqld_options","status":"MERGED","created":"2019-10-16 17:57:16.000000000","updated":"2019-10-17 11:45:02.000000000","submitted":"2019-10-17 11:45:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"688992-1571312702537-9e252c24","meta_rev_id":"4c373c705ad59a07aa8cfda65056fd8599ee50a3","_number":688992,"virtual_id_number":688992,"owner":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},{"value":0,"date":"2019-10-17 06:31:25.000000000","_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-17 11:45:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2019-10-16 19:59:33.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},"all":[{"value":2,"date":"2019-10-17 09:05:14.000000000","_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},{"value":0,"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},"all":[{"value":1,"date":"2019-10-17 09:05:14.000000000","_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},{"value":0,"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-16 19:59:33.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-17 09:05:14.000000000","updated_by":{"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},"reviewer":{"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},"state":"REVIEWER"},{"updated":"2019-10-17 11:45:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b153e95e746d762eb78205e5e4fe7da7d29f6def","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2019-10-16 17:57:16.000000000","message":"Patch Set 1: Cherry Picked from branch stable/rocky.","accounts_in_message":[],"_revision_number":1},{"id":"e882d2408497f8a210cf57fa06bc9eed28e07040","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-16 19:59:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/92/688992/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/bc45850/ : FAILURE in 2h 00m 17s","accounts_in_message":[],"_revision_number":1},{"id":"36b590bb11a8c351c4adabd3098070e69b0a18b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 22:46:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-oooq https://zuul.opendev.org/t/openstack/build/4e4a1ff86e0b401396e3b9a91893b494 : SUCCESS in 1h 41m 30s\n- tripleo-ci-centos-7-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/d91c84b505ff4b5781f96ce0ba91d414 : SUCCESS in 1h 41m 23s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/de715dc08568475892e2baa4277fb828 : SUCCESS in 2h 47m 55s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/86a9685d61e545adbc23946cc6f63a0c : SUCCESS in 2h 30m 39s (non-voting)\n- puppet-openstack-lint https://zuul.opendev.org/t/openstack/build/1cd0d849245b49bcab1b591a71183849 : SUCCESS in 5m 53s\n- puppet-openstack-syntax-4 https://zuul.opendev.org/t/openstack/build/824fcb54bfc040b0b63c73a2739bbd6a : SUCCESS in 6m 07s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/8524463935ce43f1aca09b81d4518526 : SUCCESS in 31m 53s","accounts_in_message":[],"_revision_number":1},{"id":"3d71f4c930b01c20d728a3a62c24ce86ce9091e8","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2019-10-17 06:31:25.000000000","message":"Patch Set 1:\n\ncheck-rdo","accounts_in_message":[],"_revision_number":1},{"id":"e2cc3d900038889759dc744669e36d6aaf4a8c60","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-17 08:19:35.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/92/688992/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/fd4adc0/ : FAILURE in 1h 46m 22s","accounts_in_message":[],"_revision_number":1},{"id":"47e4efc050474ee70ccb0f43d1a9851b41c54cde","author":{"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},"date":"2019-10-17 09:05:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1efc9fb6b448a62ea14bc0a460a30cf06d4148df","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-17 09:05:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8f9281390fbf9abcec25624fec4ce9c63326841b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-17 11:45:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-undercloud-oooq https://zuul.opendev.org/t/openstack/build/2d406b27e6d64ef18a5bb4ca7b5fe8ee : SUCCESS in 2h 09m 14s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/e14e8f82eb534845a41192737f99317a : SUCCESS in 2h 21m 02s\n- puppet-openstack-lint https://zuul.opendev.org/t/openstack/build/406fbcd1921d417c8a69ac2e07098d40 : SUCCESS in 4m 08s\n- puppet-openstack-syntax-4 https://zuul.opendev.org/t/openstack/build/77408149008f45ad9583e9bcd22673f6 : SUCCESS in 4m 13s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/23f0bbe02d8b4e93969add2a435a93ba : SUCCESS in 37m 58s","accounts_in_message":[],"_revision_number":1},{"id":"ecf6288a18300b090759274513df52fc0337df83","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-17 11:45:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"82fa648afad9835be368ba856e9ec95aa7b24165","revisions":{"82fa648afad9835be368ba856e9ec95aa7b24165":{"kind":"REWORK","_number":1,"created":"2019-10-16 17:57:16.000000000","uploader":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"ref":"refs/changes/92/688992/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/92/688992/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/92/688992/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/92/688992/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/92/688992/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/92/688992/1"}}},"commit":{"parents":[{"commit":"4b48d5a85628c9d20711cc7331ede8771f330879","subject":"Merge \"Include cors to gnocchi profile\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/4b48d5a85628c9d20711cc7331ede8771f330879"}]}],"author":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2019-10-14 17:19:51.000000000","tz":120},"committer":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2019-10-16 17:57:16.000000000","tz":0},"subject":"Deep merge hiera keys for mysqld_options","message":"Deep merge hiera keys for mysqld_options\n\nCurrently when adding some tuning options via hiera, galera won\u0027t start because\noverriding even a single mysql option will reset the whole key in the hash. So\nfor example, when adding:\n    tripleo::profile::base::database::mysql::mysql_server_options:\n      mysqld:\n        # MySQL InnoDB equally divided in 1GB instances\n        innodb_buffer_pool_instances: 2\n        # Query network write timeout raised to 120 seconds\n        net_write_timeout: 120\n        # Query network read timeout raised to 120 seconds\n        net_read_timeout: 120\n        # MySQL connection timeout set to 8 hours\n        connect_timeout: 28800\n\nThings will break because all the wsrep options that are set normally will be\noverridden and galera will refuse to start\n\nTested by passing the above hiera keys and observing the deploy complete\nsuccessfully and the settings correctly applied to galera/mysql on the overcloud.\n\nChange-Id: I30f03bc8eb81db0243c137d4af08924adeebc951\nCloses-Bug: #1848060\n(cherry picked from commit 7e78ebdc0f3678afddea3bdab2007c7b6ac92776)\n(cherry picked from commit d12fe87318264c9423e6bc0ac509fa2978261f1c)\n(cherry picked from commit 82e304a419698e1243765557b66ee2ebd83d03bc)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/82fa648afad9835be368ba856e9ec95aa7b24165"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/82fa648afad9835be368ba856e9ec95aa7b24165"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
