)]}'
{"id":"openstack%2Fpuppet-tripleo~695798","triplet_id":"openstack%2Fpuppet-tripleo~stable%2Ftrain~Ifae73ac2bf4481d0a11e89c0ea0916e85dd2db1d","project":"openstack/puppet-tripleo","branch":"stable/train","hashtags":[],"change_id":"Ifae73ac2bf4481d0a11e89c0ea0916e85dd2db1d","subject":"Improve stonith leves idempotency.","status":"MERGED","created":"2019-11-23 13:40:09.000000000","updated":"2019-11-25 21:18:30.000000000","submitted":"2019-11-25 21:18:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"695798-1574716710618-ca8dfbed","meta_rev_id":"34686cc703e1e745d0bf72d79d235ffed27b9b44","_number":695798,"virtual_id_number":695798,"owner":{"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-25 21:18:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-11-23 17:16:58.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2019-11-25 15:36:45.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-11-25 13:19:17.000000000","_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":1,"date":"2019-11-25 15:36:45.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-23 17:16:58.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-11-25 13:19:17.000000000","updated_by":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"reviewer":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"state":"REVIEWER"},{"updated":"2019-11-25 15:36:45.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-11-25 21:18:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6ce8582c4939587192c033c6e68ab227c6696622","author":{"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"},"date":"2019-11-23 13:40:09.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"166c112343ddcfde8e13466cb322c4c433770064","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-23 16:02:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/10c952d3ac984818909ef064d5608098 : SUCCESS in 4m 30s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/74096f97d282405f8ed758fab065fc8b : SUCCESS in 4m 29s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/edc6dfd5a215402c801c25662b9d2939 : SUCCESS in 40m 43s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/8eff2977d0334ffc87e5730bcaec9a71 : SUCCESS in 1h 30m 47s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/eb53cb8041bc4c4bbf6db128eb8edf0a : FAILURE in 1h 37m 07s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/41d9fc3c3c194d59bc1ceefc39d5bb6d : SUCCESS in 2h 21m 50s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/9205d58b16ef47b1901606a44ed99b26 : SUCCESS in 2h 07m 39s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/83088270ee324af5b91712e747e6e364 : SUCCESS in 1h 59m 24s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/fae046cf301d4603abc0113e07036d4a : SUCCESS in 1h 57m 12s","accounts_in_message":[],"_revision_number":1},{"id":"e7dec0418f086c2fe5d2f57dfb1cfcbd27c50eb1","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-23 17:16:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/98/695798/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/7a41c13/ : SUCCESS in 2h 32m 14s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/98/695798/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/f41348a/ : FAILURE in 3h 26m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4f4431762fc1176e3e52a8553b91ae34c553fabf","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2019-11-25 13:19:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"765a0d6c61b9172f7b908ed7e88fcbbfd90db861","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-11-25 15:36:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6deafa5aecf98f90e488e5bede82d84e00dbd136","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 15:37:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1a1546daf8fe7fb21fbbd3cdf28cce89cac6c5d5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 21:18:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bcc2f58a8a3f4c8ea37f78ee899a72c1 : SUCCESS in 5m 22s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1ba08bdd2bcc4e269d1db15f51084617 : SUCCESS in 5m 51s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/3ba5886df76340cd9aadd12bc7e30bd8 : SUCCESS in 40m 36s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/21ed4a056f7b4dde9c5ba83e0ba5d6e5 : SUCCESS in 1h 41m 50s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/3c3eeacb48124b68bd8322467032a5fc : SUCCESS in 2h 22m 03s\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/bd4124dbb08347d193ceb3aa2dee9d30 : SUCCESS in 2h 12m 39s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/64883e71b2f44370914421276d802dd4 : SUCCESS in 2h 02m 16s","accounts_in_message":[],"_revision_number":1},{"id":"34686cc703e1e745d0bf72d79d235ffed27b9b44","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 21:18:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"720cfa54bfbda9915134f70b70f2410473eba523","revisions":{"720cfa54bfbda9915134f70b70f2410473eba523":{"kind":"REWORK","_number":1,"created":"2019-11-23 13:40:09.000000000","uploader":{"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"},"ref":"refs/changes/98/695798/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-tripleo","ref":"refs/changes/98/695798/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/98/695798/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/98/695798/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-tripleo refs/changes/98/695798/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-tripleo refs/changes/98/695798/1"}}},"commit":{"parents":[{"commit":"bb277689eaa7d61d88be5c360a0428b466bd5a7a","subject":"Switch HA containers to k8s-file log-driver and make it a parameter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/bb277689eaa7d61d88be5c360a0428b466bd5a7a"}]}],"author":{"name":"Luca Miccini","email":"lmiccini@redhat.com","date":"2019-11-12 07:01:11.000000000","tz":60},"committer":{"name":"Luca Miccini","email":"lmiccini@redhat.com","date":"2019-11-23 13:40:09.000000000","tz":0},"subject":"Improve stonith leves idempotency.","message":"Improve stonith leves idempotency.\n\nThis commit improves the way stonith levels are set up and their\nresiliency against redeployments by introducing a stonith_levels\ncustom fact that collects the current stonith levels defined for\nthe specific server, so we can compare against the desired number\nof levels defined in hiera.\n\nIf these do not match (for example if there are additional levels\nthat are no longer necessary), the clean up step also introduced\nby this commit takes care of deleting the ones no longer necessary.\n\nChange-Id: Ifae73ac2bf4481d0a11e89c0ea0916e85dd2db1d\n(cherry picked from commit 7c30bd791fd5a9a5ee00003709752b057e3d86c8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/720cfa54bfbda9915134f70b70f2410473eba523"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-tripleo/commit/720cfa54bfbda9915134f70b70f2410473eba523"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
